)]}'
{"id":"openstack%2Fdesignate~932907","triplet_id":"openstack%2Fdesignate~master~I29b13b85c3f512d9be98fb21794cf30dbf8eddde","project":"openstack/designate","branch":"master","topic":"grenade-venv","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-10-22 17:22:25.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-10-23 18:02:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I29b13b85c3f512d9be98fb21794cf30dbf8eddde","subject":"Revert \"Disable GLOBAL_VENV in grenade\"","status":"MERGED","created":"2024-10-21 13:12:49.000000000","updated":"2024-10-23 18:05:42.000000000","submitted":"2024-10-23 18:02:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":932697,"submission_id":"932907-grenade-venv","meta_rev_id":"e8f96cef1bc866c1ffff5029075d7b8f48170821","_number":932907,"virtual_id_number":932907,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-10-23 18:02:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-10-22 17:22:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2024-10-23 16:57:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2024-10-21 14:34:18.000000000","_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2024-10-23 16:57:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-10-22 17:22:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-21 13:12:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-10-21 13:12:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-23 16:57:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f109c74031f669f1d0a076dfa4089814d0d803c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 13:12:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"75f40878790a0f8648b6005764080b9aa5dbf426","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 13:13:15.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"74d9472357ced38b3c022ca280bb2e485ed97dd2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 14:34:18.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"ba80cc31881d707c4513a0d53c08ea7877bbfc33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 14:38:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fde2bc3c07d7439ebb89bb42a3fd483c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4b38510412524799aeb7e2caa1131bb3 : SUCCESS in 46m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fa5c948aca0246429bded9e9ed9a09b1 : SUCCESS in 54m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7aba27c3f02e4721bbb98c824604ef55 : SUCCESS in 57m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6074cf2b2e704cd0ae280d1321d3021b : SUCCESS in 49m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/78aaadce220d43f2877d7b2100d3e3a5 : SUCCESS in 48m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d75a0580f0e34107acdc0b99bbcc6151 : SUCCESS in 51m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b48bfd5e68b546378f31604881bcf8cc : SUCCESS in 1h 08m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0c6565a39554c369a6132ad53b05313 : SUCCESS in 48m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5cdb3cb8119c4af59fb015a6a2b497c5 : SUCCESS in 58m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bd541ac2ec2c47dd9bd6067f37d3646f : SUCCESS in 5m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/36e66935884c4572870d4a72a79a4fe8 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/823868ed1b004dcab62352bf13c1c2c2 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/08280475e7f84c85b479922b4095b5d7 : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cc6456356f846aba6ea8470a0261b2b : SUCCESS in 6m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9b41a32fca0b48cbbc190c1bcc00c774 : SUCCESS in 34m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4fb2c631be224e99bad4a9fafa4e871d : SUCCESS in 4m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f3b9ece66c2645d9850ed3751caf2d5c : FAILURE in 48m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0048de74f401c9cd9938d5543f4a40dfe41333e0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-22 17:22:25.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"638219eab46ecaa18da901774d5d6a14478377ba","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-10-23 16:57:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be75fccbc80e6a26f2313d33653d055e3262fed2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 16:57:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"16c059081e969acfa02cde55f66c63f6eddb8d8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:02:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fd9d09fe8b54a3a867140b3c20d868e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/71a7158852f34a319e6d312a32e52275 : SUCCESS in 42m 15s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/673cafe85f374f8080b753fe838e8304 : SUCCESS in 49m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/60f0c16c3345477b8c3cf0770ad09761 : SUCCESS in 46m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3df9ffb956bf4179a6e9daf5db296ed5 : SUCCESS in 1h 03m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/752e4f60327648128aa63c8501f68a41 : SUCCESS in 48m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e2755106ef05434ca8ab5e36d5bf9853 : SUCCESS in 49m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c25f9b0dad64b9ea15a89a773eebd01 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f5946c90fd044fe387f9f73253a42be4 : SUCCESS in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f219cc004c684187b9cd1643d7408114 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d94fd6db4d842349af166c22b278f44 : SUCCESS in 6m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1e2bf93cac24294a8d1cea08bda8c3e : SUCCESS in 35m 30s","accounts_in_message":[],"_revision_number":1},{"id":"cc7b56bdfb706dda9b51c69e1e84fd98d0775cae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:02:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8f96cef1bc866c1ffff5029075d7b8f48170821","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 18:05:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04d5180b1bd54e4aad1b643083cff098\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c8d52f789254d898671a3715cf67efe : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769","revisions":{"c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769":{"kind":"REWORK","_number":1,"created":"2024-10-21 13:12:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/07/932907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/07/932907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/932907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/932907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/07/932907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/07/932907/1"}}},"commit":{"parents":[{"commit":"4b787aaba50ba59ecb9650d297d190ecbb951d49","subject":"Disable GLOBAL_VENV in grenade","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4b787aaba50ba59ecb9650d297d190ecbb951d49"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 13:12:49.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-10-21 13:12:49.000000000","tz":0},"subject":"Revert \"Disable GLOBAL_VENV in grenade\"","message":"Revert \"Disable GLOBAL_VENV in grenade\"\n\nThis reverts commit 4b787aaba50ba59ecb9650d297d190ecbb951d49.\n\nReason for revert:\nThe fix to support GLOBAL_VEN was backported to the source version.\n\nChange-Id: I29b13b85c3f512d9be98fb21794cf30dbf8eddde\nDepends-on: https://review.opendev.org/c/openstack/designate/+/932905\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
