)]}'
{"id":"openstack%2Fdesignate~932917","triplet_id":"openstack%2Fdesignate~master~I501ab049f4cd7543d1e88744e83f2a889c4d2be7","project":"openstack/designate","branch":"master","topic":"grenade-venv","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-10-23 11:14:50.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-14 08:27:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I501ab049f4cd7543d1e88744e83f2a889c4d2be7","subject":"Add SLURP job","status":"MERGED","created":"2024-10-21 15:10:43.000000000","updated":"2024-11-14 08:30:09.000000000","submitted":"2024-11-14 08:27:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"932917-grenade-venv","meta_rev_id":"73ebfd41e9007ab18ad80616015416ba7a569311","_number":932917,"virtual_id_number":932917,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 08:27:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-10-24 10:32:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2024-10-23 11:14:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-13 01:06:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-11-14 07:12:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-21 16:22:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-22 17:25:20.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2024-10-24 10:32:48.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-11-13 01:06:43.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"48ede0e22194f4cd8f8ecbbc2082bca05fa608bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-21 15:10:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0a3a16e262d793aeba8dd3356725f28ca903c5f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-21 16:22:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1b2693e18ab441c8bed12befcd24fb63\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/22b07db80e854e6ba7b7f3a3730b9c92 : SUCCESS in 41m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/006deb9f7d1e4068ad14f9a1be527f05 : SUCCESS in 1h 04m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/71f575e9957c48499e5c4804ae4f6699 : SUCCESS in 36m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/884411989af24487a7809ca59c557ae8 : SUCCESS in 38m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f71db7d9c17f4aab8d9fd3b2cab70f5e : SUCCESS in 37m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/315332d43d654bad892e28f9771563a3 : SUCCESS in 1h 10m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/192b3afff91842f497fdafec180514c5 : SUCCESS in 1h 05m 38s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/443e42a45eb9433da37cbcec67662b31 : SUCCESS in 1h 01m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4f984032520d446389c3c637e0a35947 : SUCCESS in 58m 01s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/21920e16d5304dfeb5584c861810558b : SUCCESS in 50m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b1f9f854989c4c1c9ab377302d502447 : SUCCESS in 50m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f0cb65cb17249d1866f3b84a71bc55f : SUCCESS in 5m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fed750ef74f5431e8b4f321a7e73a610 : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d67e23ffd614b0c81ce8654b4e2a882 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/064318abe5954502bd374e2c700eb78b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8d4e5be31c442d4951cd2a92a2694af : SUCCESS in 8m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6690e06abb034a71bc232c87f26b8959 : SUCCESS in 30m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/846918ced5a449518238c1bf50db20f1 : SUCCESS in 3m 52s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ccd084efe601472482b3e7d4c20a60d0 : SUCCESS in 52m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b6d80d973b7fa57591155d43c8b72abdab292505","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-22 17:25:20.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"6d7a183cf384b5fe96daa037fad13c676d9d92b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 06:15:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"3e9a97dc6bcef24df234daaa604ae69360db326d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-23 06:15:28.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"be8fcdf29deca337d3fe7d24474ece228e073667","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-23 07:36:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30d52ab530134588bd4e228bae6d00b8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dd3d0717721a44e987912d1a20b9ab43 : SUCCESS in 47m 16s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ca905ca0464b444b881d99b042c23578 : SUCCESS in 1h 04m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b207a4f13df94f36bf3fd4e568e5ffc5 : SUCCESS in 1h 02m 33s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8c86fa333fd047d9a25a8a94210fd222 : SUCCESS in 47m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1555eca27c4f470eba4589da81f76d2f : SUCCESS in 32m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/467ddfe3607845f0a75ff95b46451d3a : SUCCESS in 1h 06m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/420df39277df46c68bcb3bc6ff3add95 : SUCCESS in 1h 13m 41s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cd818eb1a4b24c6aaf49ca4cdfd65fce : SUCCESS in 1h 08m 40s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6e696788ad704b49ae2b23fe18b72b31 : SUCCESS in 1h 03m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7261d5fdcea74442943ee4e479fa36bf : SUCCESS in 31m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e4389ab63999429ba0e8d8bc45ea7f05 : SUCCESS in 38m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/51b3c42919ef4e158438a84b6168dffe : SUCCESS in 5m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de7236570e014cce80994fd00293daef : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e094df4b7b34ccfaa5e8c60806b56c9 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef94e77febfb48de9f7129f04b38916b : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8dde203a404ac3b2889754c706ced8 : SUCCESS in 5m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4082627f5b284fa896ede9021ad980d8 : SUCCESS in 38m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cae59dd1af054415878dd31b98d73b95 : SUCCESS in 4m 05s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/e00b523e4231412684d8c5460fa89550 : SUCCESS in 31m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"217c625acee6d10f40ff4503622068149ae7b1de","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-10-23 11:14:50.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"09b332ed56b23fcb50c690e1def868fcbc73aee9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-10-24 10:32:48.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"269848691141d88acbfead248ae93fd81ae89185","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-13 01:06:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"01727a457cb481a3569e5228ea6b7b5bb9b40d92","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-11-14 07:12:04.000000000","message":"Patch Set 2: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"0408536446e6db16bbbcf526ac5407cbf75ea9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 07:12:47.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"76cc8a50eabb8e68d34baa3be6d39bcaec82bf4e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 08:27:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c6e8b9f245143febe26267051b89da3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/eae2bef143b84010bd6ac9af14af645e : SUCCESS in 50m 54s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c452b7874938480288c5f66832ca0186 : SUCCESS in 47m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b12bd337644549a79caf07052dafcc18 : SUCCESS in 44m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/db83126dbe914f0b8fdb14de14f34ec9 : SUCCESS in 1h 08m 59s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c8d29c6ddf9b4488b981aa1449cbf245 : SUCCESS in 1h 05m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f6195d2468d14d17ace3da92735e7d2d : SUCCESS in 40m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/48c8573c0ed441a7a0af4bea870288d7 : SUCCESS in 42m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2893ed2434964414a9d2488f9a00c031 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd8cffae899547dda63408021abb7cf9 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/49432295bd654f0ebcd9174d54b495d6 : SUCCESS in 7m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3379cbec660248cea803e25cdc3f8add : SUCCESS in 6m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5743e490ed174a72a30b23af89f6d1a8 : SUCCESS in 41m 06s","accounts_in_message":[],"_revision_number":2},{"id":"997065a7dc86b6c43d76dcad0e3d805fa58925cb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 08:27:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"73ebfd41e9007ab18ad80616015416ba7a569311","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 08:30:09.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37cfca2085db4c319d3de2abd8d98d3f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c376be4c234c70ac13435acf271f36 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"2ec2d29e033bf96c7b74a6d9bffb90a8511711ef","revisions":{"b39eb0afd001721c87835e2262e4374354a81f7a":{"kind":"REWORK","_number":1,"created":"2024-10-21 15:10:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/932917/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/932917/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/932917/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/932917/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/932917/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/932917/1"}}},"commit":{"parents":[{"commit":"c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769","subject":"Revert \"Disable GLOBAL_VENV in grenade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 15:09:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 15:10:25.000000000","tz":540},"subject":"Add SLURP job","message":"Add SLURP job\n\n2025.1 is a SLURP release, and should support direct upgrade from\n2024.1. Add the job to verify this upgrade path.\n\nThis also switches the voting grenade job from pdns4 to bind, because\nbind job is both run in check and gate.\n\nDepends-on: https://review.opendev.org/c/openstack/designate/+/932906\nChange-Id: I501ab049f4cd7543d1e88744e83f2a889c4d2be7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b39eb0afd001721c87835e2262e4374354a81f7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b39eb0afd001721c87835e2262e4374354a81f7a"}]},"branch":"refs/heads/master"},"2ec2d29e033bf96c7b74a6d9bffb90a8511711ef":{"kind":"REWORK","_number":2,"created":"2024-10-23 06:15:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/17/932917/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/17/932917/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/932917/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/932917/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/17/932917/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/17/932917/2"}}},"commit":{"parents":[{"commit":"c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769","subject":"Revert \"Disable GLOBAL_VENV in grenade\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c51d5283e7e4cb62d3ea02d14a8e5bfeb3449769"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-21 15:09:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-23 06:14:46.000000000","tz":540},"subject":"Add SLURP job","message":"Add SLURP job\n\n2025.1 is a SLURP release, and should support direct upgrade from\n2024.1. Add the job to verify this upgrade path.\n\nDepends-on: https://review.opendev.org/c/openstack/designate/+/932906\nChange-Id: I501ab049f4cd7543d1e88744e83f2a889c4d2be7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ec2d29e033bf96c7b74a6d9bffb90a8511711ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2ec2d29e033bf96c7b74a6d9bffb90a8511711ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
