)]}'
{"id":"openstack%2Fdesignate~933027","triplet_id":"openstack%2Fdesignate~stable%2F2024.2~I4dccdfd171deba7d5dac713d59c47354b09ac367","project":"openstack/designate","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-07 01:32:57.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-11-07 00:32:38.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I4dccdfd171deba7d5dac713d59c47354b09ac367","subject":"Stable-only: Switch neutron job to stable version","status":"MERGED","created":"2024-10-22 16:12:25.000000000","updated":"2024-11-07 01:36:28.000000000","submitted":"2024-11-07 01:32:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"933027","meta_rev_id":"76bee2fbfd563410e37d25356a0dd146bee389b1","_number":933027,"virtual_id_number":933027,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-07 01:32:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-11-07 00:32:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-11-07 00:32:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-22 17:35:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-30 05:14:26.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"282f377f2edec417afd00d14492ab6d215217f4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-10-22 16:12:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"18b35816138a06d08d61bf6b853db360d579e1ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 17:35:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89ea464a1b9c4db0bfa011c02652d331\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a6f33a1e5d2445758aec01b88ab23ba8 : SUCCESS in 38m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3c7d545ed7ed4b45bb1d9903be519bea : SUCCESS in 59m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/be4b06a1966e42449cea0cdeb89c9b0d : SUCCESS in 52m 23s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/57e75fb1e06f4abdb3bedc846a41da13 : SUCCESS in 39m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e0befbbe96444e5cba5da23186625f04 : SUCCESS in 30m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fc08c6381aaf42d5a9d99c5617c1182d : SUCCESS in 1h 04m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e7ab0b1d17644cbf8cdd1f4d6fa50e9d : SUCCESS in 1h 16m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2d5de99e76744489bf5c8d3c150f03bc : SUCCESS in 44m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e93ad766a055436f8278e8758a630f82 : SUCCESS in 49m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3062d7e480bd49148b71fab1aec7860d : SUCCESS in 4m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143860ec08d5411081e3ba29d7fdec96 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46b551301dc644e68993d0bd0debd564 : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6372c1bff08d403a8203fb6440381838 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/129f3e7e2ecb409090f50feba4be32b3 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b9e83e674bfa4619b1d22bfa22651e38 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/1cab30386a3443d1889ed07432a51c2b : SUCCESS in 34m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/410a7825167f445da4523bb38d10e066 : SUCCESS in 4m 44s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f8b2e605328941f59445e40bae4cdd30 : FAILURE in 50m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"84f72789eefd110c5b77f2d28d723a582267ccd2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-07 00:32:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"279ca98791f99e3c34bb7bcff7d4931dbdacc9ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 00:33:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"717702ad583ca129a8ed3a94aaec07df138fbd96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 01:32:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e9032c7d0f84d3999ff9772965b1ffc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/82c65fc7f6964246b3ec2eb2a73f8dc0 : SUCCESS in 45m 56s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/23702470846243f3a993650ee22dabb1 : SUCCESS in 43m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8f3df0a2fa9d47348c7620ba4a729171 : SUCCESS in 41m 47s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e9d5c8c4132d491eaacb7e8756fe1ce5 : SUCCESS in 59m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3fed56e95914498f957a6a3173fa119e : SUCCESS in 36m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/426a21e3f5bd4cf6ab7ba015af177d53 : SUCCESS in 38m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/52b088dfb2b64edaa231d7c4f1637589 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8228ec64a0aa49d2b84531fb608f19bc : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/39ab9aee8072479d9267b442248db7d4 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/338d998e22684cb4bd7dbc67140f8258 : SUCCESS in 6m 40s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/28fdc066ea6e47ed8a6cdbae7f5cceb1 : SUCCESS in 27m 25s","accounts_in_message":[],"_revision_number":1},{"id":"1ba375ad3519b7ef2b44680427b485fcb5b519d7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 01:32:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"76bee2fbfd563410e37d25356a0dd146bee389b1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 01:36:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3df1d900ea9f4d098edcfb0db5d8b331\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e31866f8cd8e4714aba632df0e3e82d3 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e607108e0ff710c674ebada100f9cb5efeb777f3","revisions":{"e607108e0ff710c674ebada100f9cb5efeb777f3":{"kind":"REWORK","_number":1,"created":"2024-10-22 16:12:25.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/933027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/933027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/933027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/933027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/933027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/933027/1"}}},"commit":{"parents":[{"commit":"1bb5504e52ad94ecc714bf3ac614a7adc48f26cf","subject":"Update TOX_CONSTRAINTS_FILE for stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1bb5504e52ad94ecc714bf3ac614a7adc48f26cf"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-22 16:12:03.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-10-22 16:12:03.000000000","tz":540},"subject":"Stable-only: Switch neutron job to stable version","message":"Stable-only: Switch neutron job to stable version\n\nChange-Id: I4dccdfd171deba7d5dac713d59c47354b09ac367\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e607108e0ff710c674ebada100f9cb5efeb777f3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e607108e0ff710c674ebada100f9cb5efeb777f3"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
