)]}'
{"id":"openstack%2Fdesignate~934525","triplet_id":"openstack%2Fdesignate~master~I6d05e632714b6dede9d38bf6f9ed71af43a0ccea","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-13 22:07:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6d05e632714b6dede9d38bf6f9ed71af43a0ccea","subject":"Drop conflicting [DEFAULT] tcp_keepidle","status":"MERGED","created":"2024-11-09 08:17:16.000000000","updated":"2024-11-13 22:08:26.000000000","submitted":"2024-11-13 22:07:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934525","meta_rev_id":"5baab923642bc20a323e36a14b67930921ba6b5b","_number":934525,"virtual_id_number":934525,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-13 22:07:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-11-12 15:59:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-09 16:05:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-11-12 15:59:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-09 16:08:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-09 09:52:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-09 16:05:40.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-11-12 15:59:56.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"60e4d9023212298e9266363655ee1a71ad027e9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-09 08:17:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40d1f5d2f4730bddf0a15b2fd88b67f1fc717891","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-09 08:28:37.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"d6da2dbdcb10b35383d1e71d7806d38b3a70f162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-09 09:52:15.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0173deb7b90748acbcb7e47abf1d7892\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a28bb2328b894f26a9bd918ae88ad396 : SUCCESS in 41m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f0896c4f988e4f3b812cdc4cbb9f854c : SUCCESS in 53m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/46752265310141fbbb0591e382620db9 : SUCCESS in 40m 03s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/51cbab3b55d347ae8520b4f6357aab03 : SUCCESS in 51m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9f0cf56fcd64470786fcd47581d1c3a4 : SUCCESS in 47m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d5e217dea1e84908959bf17061125e22 : SUCCESS in 1h 17m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0fa1b20768a54fecb44e3f1244224c16 : SUCCESS in 1h 00m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/39c91e4c72d74dc2940d8b62c13f6d9c : SUCCESS in 37m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cf909fe928d04827b1ab4ef8cb629edc : SUCCESS in 49m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6538388844494e15951e02a08015eb41 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83689c0e0314413e8ad6f615e4c8169d : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/086a53720aee4f088ce5cd6aaa69d9f9 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5422da8d648843cf91928db2fa8a129c : SUCCESS in 7m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f14851ab90764ae8a83a028dc131e51d : SUCCESS in 8m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7a810ec6468c4aff8b685bec68f34de1 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7fab633dc1074d058ad9f508659fbe27 : SUCCESS in 27m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d24eca01d00b45c6b9159e6e07cc3e7a : SUCCESS in 5m 45s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/bc584b0636034aa993fc26e6f5c5e79b : SUCCESS in 48m 24s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c695ce6e3dae482b02896eda7d44256921ea7373","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-09 16:05:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"df64d951bc5044b0a1992f995242aa047ed0c28d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-09 16:08:03.000000000","message":"Patch Set 2: Review-Priority+2","accounts_in_message":[],"_revision_number":2},{"id":"880edd0799a35429041d26e84301de648fd1e178","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-12 15:59:56.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"391ed9cbbd505a62147c10e970841c18127e04c9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 16:00:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"137cdafc71607129fe5d3478ff6dd2aa7608c0fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 16:38:30.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/afe6e8de7b304798ae4813e687d9bd08\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6560749b8db24a60865a5b2f415ef1c8 : CANCELED\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ef47e69c62704057b3c0fd82df4b14da : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8019e3d5300e448cb17b3bad3958fa4a : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/80d76a5e8bb24c3c98114e132dcad070 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bb586393c34e465b88146a8e3cfa9957 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0433bf20591546e1bb25aa6d97b1d951 : FAILURE in 37m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07f17a7e957e4b5ab9d90f883c125fe9 : SUCCESS in 3m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4c733b6810146d88937aba2ece30629 : SUCCESS in 3m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9824f40151db490cb995971e29c72291 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d153ceb64c124c94b4e8c3fbd11005a2 : SUCCESS in 7m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/813c13c5698d461c88c94a22e61509df : SUCCESS in 4m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1755bbee9f324fa5be32616ce34f2370 : SUCCESS in 29m 34s","accounts_in_message":[],"_revision_number":2},{"id":"6218d2371c1012852ffc15c44c20b7fa97064b7a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-12 16:39:17.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"55a56c4dd0254e791ec0fe6e0cd1b2a54b3f9a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-12 17:58:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3bc6b9ed82f24846a2b7523a244be38c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/39848f1e71b04cfc8ee738002da919df : FAILURE in 33m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e6d1e01831294828bcdbf577ae067b1a : POST_FAILURE in 1h 02m 15s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/64fd975d291a4acfbf445e1f7cc08488 : FAILURE in 1h 06m 13s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/50dcac44910b46cc997ff48f82c72836 : FAILURE in 36m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e2d888a6d81741dd998217980b95606a : FAILURE in 41m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/27d6f93f67db423c9fc4faa59435a086 : FAILURE in 1h 12m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4427db265fdf436592fbbd373ba9f180 : FAILURE in 1h 13m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/07a9489baa8243e5b585edce85bca396 : FAILURE in 41m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/224f5a81ba3f42d69d54f314bec23252 : FAILURE in 40m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb27259db0ce4419896385929c7a7667 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/622680894f1542ca84b10edf2814e6cf : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4b9d983f323419ea7c2f3c57144ecd9 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ddf1e7257f224f528027e0d9ff836520 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5929d0ffdb2d49968109f22311943c6f : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/83fcf5fd38474e04849f26d7be8987bf : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1852cfeb288a4d45832c70166db9802c : SUCCESS in 36m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ee4960e10c304617a4b2470fed17090b : SUCCESS in 5m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/cce063a8f8cd453d84878ad5cd1dff7b : POST_FAILURE in 51m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2faf55f1be2683d3e58ebe203d3994fa02ce8931","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-13 00:38:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"3cf6e19f7dbe08e1ffc9548ebe55778aabac4dfa","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-13 00:40:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d34c1c8fba448a78706786de2932b2b96a168765","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 01:36:43.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92384d16962448feb0556d1ed6065774\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/018346b334ea4385a922253b2f948f39 : FAILURE in 33m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ef96d9cf70c04545887406de604bea5b : FAILURE in 39m 07s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/43860c8e577a4bbba7f8cfb641edad64 : FAILURE in 34m 15s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0fb3d959aedc434c9e329dcbfa3d2da7 : FAILURE in 30m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/78e76b2c4bcb402b9d4135e1fb7bc632 : FAILURE in 30m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dc471320d31b4e19a9f2dac6d68f58d4 : FAILURE in 53m 04s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/335aab2d96bd4eb9a0d89052020a3415 : FAILURE in 47m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e2a741185fc411287aa0b21409961c0 : FAILURE in 39m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/abae06ad59e947749f5f82a5c787746e : FAILURE in 41m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b421b9375dff494eb757fb048376c418 : SUCCESS in 4m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c09fa1e451724469aaf9302bb58ef0d7 : SUCCESS in 3m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cede9f68a56a49bfa6b837bd79ba4b50 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3aa432f3f84e4f29b4068e8fcf32be21 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2758255743664acfbe3b6d4a9d82a08c : SUCCESS in 4m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0770a9af99294823a4c1a8eaed43d2f1 : SUCCESS in 4m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/84b1d361609b43979314a2bd91431814 : SUCCESS in 28m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/24ead48b4d6a457bb913ed35a3538861 : SUCCESS in 4m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f4b803bc7b294d0ead8ddfee64403626 : FAILURE in 51m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d6791bd30063b9c79b05907ace6f94547dd4429d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-13 19:21:44.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"14a66a4d4ef6ad8de4e6a3fb7bb0de77e0387211","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 20:47:43.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ebdfcd1d3284b4ea5309639adc4ccf6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d3bea70371a347afafd259165fc0d5f3 : SUCCESS in 47m 02s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2514d15aa72c4c88b86f788963619cd5 : SUCCESS in 50m 40s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/70397ff7ccc94aad94bd4852335469f0 : SUCCESS in 53m 42s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e463f9a8814f4b2c878566284af86ad4 : SUCCESS in 47m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8cdf3039c98a4d11a0c83fc3a3c4a1c4 : SUCCESS in 48m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f32ec29f30c849619099d62a43bc8b1f : SUCCESS in 1h 10m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1de4ff08bbaa4ded843ea2a72bb29b9b : SUCCESS in 1h 18m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e0dd86686bd64cbcad7a3c86fe628d60 : SUCCESS in 49m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/054d4ed39d164b9ca545cdc32dfccb44 : SUCCESS in 47m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4a2832a4cec463daa0826ef32dcc262 : SUCCESS in 5m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01f67ae07cb84a49aac883fad5266839 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/19d3507ca8a845f292b08a4ab8158f82 : SUCCESS in 3m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a8e679bc5b54fdf967470be09a1415d : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00c3eb61072242a9bb65fcc5ffe9ebf8 : SUCCESS in 6m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b298b625ab54ce899f3a0e9e9b85886 : SUCCESS in 4m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c0fd6ff051e485f93667edb3fbbb20d : SUCCESS in 33m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/32e561ad13e84338907fd59b19e09fab : SUCCESS in 4m 13s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/92c22a6489864f7a836136661b1161bd : SUCCESS in 45m 53s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e550ced755f09bcd5e82a7701617e971f1e1d66b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 20:48:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"191a98a968b1e82581bb57c987b0ade740cdc0aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 22:07:19.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b6f036181e2497cbbfe8e8967591f6b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/db91a228d84946b891321956f8f27211 : SUCCESS in 40m 14s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/09085af83bb644f09fb60c23cd6067a0 : SUCCESS in 46m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7fdd94aef2614488a14013f64a3e4c46 : SUCCESS in 47m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e36dc918386c48918472d9ebdf2cd614 : SUCCESS in 1h 18m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bc819ab6101f42339bb8c866de142715 : SUCCESS in 45m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fce3bf35e6ce4f0880b32b258b2e5d33 : SUCCESS in 54m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e0e8fdcc64d4efeb3f630a526f94253 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/69b685d35a9b482fba9e2cef7c571e42 : SUCCESS in 6m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00b1ff16b6394efba43362a98728f291 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a35d7b1fe0349fbb7587b972a6025a9 : SUCCESS in 8m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8699c4e17b614aedbf5373bc4eede566 : SUCCESS in 6m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/469b49aab1144fa29c254a6b7f122664 : SUCCESS in 43m 24s","accounts_in_message":[],"_revision_number":2},{"id":"df1deaf345e2e73171736e7586db3ad202b74256","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 22:07:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5baab923642bc20a323e36a14b67930921ba6b5b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 22:08:26.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e36348b0134061973a7d6fd71cfbb9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8a82fdaaeea4586ac5177fbbefefddb : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b040727176b04d7cb4f96d43f0f5e14b : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"d25b7bc30135fef84a506d27f1c6adc69255a51b","revisions":{"4757ba6d434beb6926a62addb07b5d84beb1127f":{"kind":"REWORK","_number":1,"created":"2024-11-09 08:17:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/934525/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/934525/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/934525/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/934525/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/934525/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/934525/1"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:15:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:15:53.000000000","tz":540},"subject":"Drop conflicting [DEFAULT] tcp_keepidle","message":"Drop conflicting [DEFAULT] tcp_keepidle\n\nThis option is also declared in oslo.service and the conflict causes\ndoc build failure. Because this option is not used by designate, we\nshould drop it from designate side.\n\nAlso fix the [DEFAULT] backlog option which is defined but unused\n\nChange-Id: I6d05e632714b6dede9d38bf6f9ed71af43a0ccea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4757ba6d434beb6926a62addb07b5d84beb1127f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4757ba6d434beb6926a62addb07b5d84beb1127f"}]},"branch":"refs/heads/master"},"d25b7bc30135fef84a506d27f1c6adc69255a51b":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-11-09 08:28:37.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/934525/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/25/934525/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/934525/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/934525/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/25/934525/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/25/934525/2"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:15:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:28:37.000000000","tz":0},"subject":"Drop conflicting [DEFAULT] tcp_keepidle","message":"Drop conflicting [DEFAULT] tcp_keepidle\n\nThis option is also declared in oslo.service and the conflict causes\ndoc build to fail. Because this option is not used by designate, we\nshould drop it from designate side.\n\nAlso fix the [DEFAULT] backlog option which is defined but unused\n\nChange-Id: I6d05e632714b6dede9d38bf6f9ed71af43a0ccea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d25b7bc30135fef84a506d27f1c6adc69255a51b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d25b7bc30135fef84a506d27f1c6adc69255a51b"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
