)]}'
{"id":"openstack%2Fdesignate~934527","triplet_id":"openstack%2Fdesignate~master~I9d23f3c4e2c0fc40543cfe77568838d9dc417b16","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2024-11-22 16:24:21.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-01-06 04:34:01.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-01-06 03:28:00.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2024-11-22 14:39:30.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I9d23f3c4e2c0fc40543cfe77568838d9dc417b16","subject":"mdns: Expose option to customize tcp keepidle","status":"MERGED","created":"2024-11-09 08:34:02.000000000","updated":"2025-01-06 04:35:16.000000000","submitted":"2025-01-06 04:34:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934527","meta_rev_id":"17f2f74ce252922ad1114d95b33ab170e367fbe9","_number":934527,"virtual_id_number":934527,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-06 04:34:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-11-22 16:24:21.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-06 03:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-11-22 14:39:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-06 03:28:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-09 09:46:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-13 01:00:40.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-11-20 10:53:50.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2024-11-20 15:09:21.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-11-20 18:47:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"dcbefa268f8c9f720d21c88d63946015246b127c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-09 08:34:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38acd841bb32b0098e31b3c6cdd5583123b46ef5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-09 09:46:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1b2f4d6979f4c488b699392fa536774\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c36598a4e55940c096c8acb629684d80 : SUCCESS in 39m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6928f08559554831a1225514c000045c : SUCCESS in 42m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f39bdfa487df4732a2a35d30fc2dff33 : SUCCESS in 54m 06s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e8beb7aa14f647489b87c95065584ab6 : SUCCESS in 47m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/78de99e5cc3d4c768ad111c8b6f2ec9c : SUCCESS in 50m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04968f7aac0347c3b7c95c92246e62b7 : SUCCESS in 55m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/71891649fce54864a401c1da66dc79c4 : SUCCESS in 1h 06m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f8b39145afcb4de096cea218d99cea66 : SUCCESS in 40m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cef8b517a3b74e9fb67095fc13cc226a : SUCCESS in 39m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1a1dd3253114f0d9436e75c822dddc9 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68286d014166424eae17eab44b0c0672 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2d89a94b2454589b8fee1056578a3f2 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4d75feb59024b6db2d7c0c94b608e06 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51218a9cb2b487d92dc684f02953b1c : FAILURE in 4m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8eb3dda307e4ea2b2464e3a5a3ea9a1 : SUCCESS in 4m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b644c35e9c474fc0a06b3a94bad15044 : SUCCESS in 36m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b3f52bac6df54f6abef2c7f8157ea5c7 : SUCCESS in 4m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/4ce407e1b58f443d8ab069845b4d7940 : SUCCESS in 38m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"51b15c8cbf2d473c08dec0ed7f27f0ec2779d82d","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-13 01:00:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5b48944ad0e1a72193dbcc77ddc34fb72d9dabda","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-14 13:28:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"580f7e077f739747b8dafb39293aa25c7544c004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 14:41:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/676387bb1e294975aadc9e235c0516db\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/03bf5bd478874b60b52a22aabe4fdb0c : SUCCESS in 32m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bd91532ee7d748a9a169ed74838b1cae : SUCCESS in 46m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bcb5e56a7f14ada920b9984c0a53df1 : SUCCESS in 59m 45s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/483671f3fe86441b88869eeadb36e1d8 : SUCCESS in 51m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6794c7e385a34cc19c5a45bc4bf3d956 : SUCCESS in 49m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c4d16e7e60e947b0b1986145e0f5e9fe : SUCCESS in 1h 01m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7197996fd13c40268da91e55fea25061 : SUCCESS in 53m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/65d192385b784e78995e96bef3a373b8 : SUCCESS in 1h 07m 03s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/48d47c943a204fc39d3c2f9bab84ffc9 : SUCCESS in 54m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9b33fd3e807d402da0b40cd8c3b4f0a5 : SUCCESS in 52m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/17a78944db5e4e73aeb3c2279d1957d5 : SUCCESS in 36m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b8e09274bca4563a795eebab44df41b : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17d8af7946714354848007f8499a506a : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ec26e0bf04e4aedb1c5e4e984979f4c : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f81cbbad27a4bb4801673efa61ce535 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7e1679e569e346ab9b1d4e918de33f56 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b4a17a38bbf54167a30b94ae7e70923f : SUCCESS in 4m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2a5f7bc1074b4486aeae169f654adaad : SUCCESS in 28m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/39559c73467341ca87ed7162fc176a7e : SUCCESS in 4m 26s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6a8be16ebe3f48929c59637b314e1a38 : SUCCESS in 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa585a80231decbd76b898155850cab7eb981519","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-11-20 10:53:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a0f089858c6dd08e9db5ea289160d756acb8117","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 13:43:14.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ec1e85a1b7ad6131a777a0b2a0e04789759915f7","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 13:44:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e26e4dc84c87d2aca742cb0d4e4dd756f7fb676e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:54:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780de9d094344e2c8b90ebfb14bab5bb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/46e831e52eda4c9ca260e03917403dbd : SUCCESS in 45m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d4b8196818284a3aac42a43f1d019f7e : FAILURE in 48m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/11376d754412474585f1c75430c53157 : SUCCESS in 53m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/197d1877849b4af486b9b49da8ce7b95 : SUCCESS in 48m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2042bfda8ec74aaca890e3d673fe8100 : SUCCESS in 41m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e4834074901246b197b29b551ccc3b7f : SUCCESS in 1h 04m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/33129bba32164f4481993349e870c445 : SUCCESS in 1h 08m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/25d5aebd6ccf45a585ca7261aecbec3d : SUCCESS in 1h 05m 48s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1d1bcb49b8b44d3abbe7f399f0aee7b5 : SUCCESS in 1h 01m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/91b1398ded9644589d4932ff47ec67c8 : SUCCESS in 39m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4ad6d6124e6840ee97807aa27f70d172 : SUCCESS in 44m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/53ff4c58bfac432c87e2031a0073b41c : SUCCESS in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc600225bd3e40d785af63aa97bcd0cc : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68dba94f74354b7a8cb739eb6c51ffbf : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/387c80aa5d1b474f8e2422ffcf95821a : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99fdd8fc036c49e8922cd7c5407bb238 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ae7f6eef76944519a979798542b1e9c : SUCCESS in 4m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0424822047cb4af0982637898ae2d731 : SUCCESS in 32m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a47126f5ee8e4c0490353fd7512e845b : SUCCESS in 4m 04s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d48391d6bdc44705ad6d2088587da11a : SUCCESS in 50m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e7aa9e59e65d8e226ab3e21fd2c0d16001a8b851","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-11-20 15:09:21.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"284b27ed6da7cdab1203b4d3300f86929267f6dd","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 18:47:43.000000000","message":"Patch Set 2: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a3efaaadbfe18855e05d0a27dc1a7f3eb1f1e70b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 05:43:16.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"899f8f2ed06d0a19725fc5b748517fdc4ff024a0","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 05:43:45.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9ac86a0d9649e7c6d2dead7f761b7c2264ce66ad","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-22 05:44:00.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d459451295c9d1ea29b690b7eed58095454692e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-22 07:03:34.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/288c894875d941babf4b4c7fad0273a1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d876a25d45a14f199f7c7f73b3e88387 : SUCCESS in 52m 41s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6fd289110df141bf8190e8da892adf4d : SUCCESS in 59m 35s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6e22e1fee5540789aaf43f1ed8e7a33 : SUCCESS in 56m 27s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/373c29a5e47c49a6a02d90ab2c1c85c8 : SUCCESS in 49m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f5714c2bec949eeb14a5c1b8e7f04c7 : SUCCESS in 47m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8bf237255d474164bdeedf57b3a7376e : SUCCESS in 1h 06m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9f84db5e0ec1462088a6776b5c84d360 : SUCCESS in 1h 08m 02s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0f31b68b632b4adca58cced20c8c33b0 : SUCCESS in 1h 14m 06s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b57ce063621543ba89ace6b2d6281b5c : SUCCESS in 1h 12m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f7c8b92e24f642cea9833ba4d420c114 : SUCCESS in 43m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d490f11c651d41bd8118b4f79991262c : SUCCESS in 54m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c5a7941f2cd24615a92ba2c5701ebe14 : SUCCESS in 6m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/daa43ec9ad4b4bffae3778cd3e83c1e7 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1ca933ec9af4ab89a1539bfc40d858d : SUCCESS in 4m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/372905e6f17245b1b380e5322082ae3f : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ee2748201ba41afa4eebb06c67bb6f0 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0377b17535304098bb47c205cf680f57 : SUCCESS in 4m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/afbe7fb665c440988d2a9d999fffeb41 : SUCCESS in 45m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2dbbf5c96a3c475d8a38f3a29fa483ac : SUCCESS in 4m 02s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/aed33a225eef45f394c993862df5d93a : SUCCESS in 42m 19s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"4d8edc7ba0167e559ecca5dc59831f653711fac0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-22 14:39:30.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"57a8a33662e3d1b3a517efd69d82fef8d0ab83f9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-11-22 16:24:21.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"17226ecd8097cadb592e37c3b82a97287249f87f","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-06 03:28:00.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"c10851b93977c6de26418d0aae23dac66b39df9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 03:28:43.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"e59f468d0883553a7cd7ac028e83b58e374db6bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 04:34:01.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63ccde18ef91454f977e4108893add26\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c03978b2278849e190cc21ad87a81c49 : SUCCESS in 47m 44s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d45cec46cb7c44b1a03fc3301b34fb10 : SUCCESS in 51m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aa79d8e6740943d4bcdeb901f197547f : SUCCESS in 44m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/42064841c57c40d5bb9fe398b5b01d69 : SUCCESS in 59m 23s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/813af6159d3546758db1d46c21fbca5f : SUCCESS in 1h 03m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ae55fb378b244ff8a0eaeb9bb77a704d : SUCCESS in 38m 32s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ca0dd9f0eac4e6196268da5f060c0c1 : SUCCESS in 45m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97448ad34b1446e4979013383860da0c : SUCCESS in 3m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/daca7fd237954c7f8168e31ab686e347 : SUCCESS in 7m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8840b00182e4c4eb8fafd73c3556f16 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9f4089206f449fba9e52296c3fc1739 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e41646a1a5064b7a96ca70f59f3ae32c : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cadf50e5edb746e4b12e8037212b35c2 : SUCCESS in 45m 12s","accounts_in_message":[],"_revision_number":3},{"id":"ad32a60879cf2f8850e635114342bb4842a21e6f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 04:34:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"17f2f74ce252922ad1114d95b33ab170e367fbe9","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 04:35:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0038a0f74be049be9b799e7cf82a4a8a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/120cafb74a18422eaa9e462c640059f2 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2a9a5ff293c5498eaf1e90c9f52b1030 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"4c7d5d950f729b2074aaa087dd1d7b900ba8a32f","revisions":{"f9d310fbf0d1e405c0f08a5c80f796f209a7ca45":{"kind":"REWORK","_number":1,"created":"2024-11-09 08:34:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/934527/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/934527/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/934527/1"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:27:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:27:44.000000000","tz":540},"subject":"mdns: Expose option to customize tcp keepidle","message":"mdns: Expose option to customize tcp keepidle\n\n... to actually use the existing implementation to support it.\n\nChange-Id: I9d23f3c4e2c0fc40543cfe77568838d9dc417b16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9d310fbf0d1e405c0f08a5c80f796f209a7ca45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9d310fbf0d1e405c0f08a5c80f796f209a7ca45"}]},"branch":"refs/heads/master"},"60c6c8b1ec005d895f932eea21a21fff9aaf8038":{"kind":"REWORK","_number":2,"created":"2024-11-20 13:43:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/934527/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/934527/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/934527/2"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:27:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-20 13:43:04.000000000","tz":540},"subject":"mdns: Expose option to customize tcp keepidle","message":"mdns: Expose option to customize tcp keepidle\n\n... to actually use the existing implementation to support it.\n\nChange-Id: I9d23f3c4e2c0fc40543cfe77568838d9dc417b16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60c6c8b1ec005d895f932eea21a21fff9aaf8038"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/60c6c8b1ec005d895f932eea21a21fff9aaf8038"}]},"branch":"refs/heads/master"},"4c7d5d950f729b2074aaa087dd1d7b900ba8a32f":{"kind":"REWORK","_number":3,"created":"2024-11-22 05:43:16.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/27/934527/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/27/934527/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/27/934527/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/27/934527/3"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-09 08:27:07.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-22 05:43:05.000000000","tz":540},"subject":"mdns: Expose option to customize tcp keepidle","message":"mdns: Expose option to customize tcp keepidle\n\n... to actually use the existing implementation to support it.\n\nChange-Id: I9d23f3c4e2c0fc40543cfe77568838d9dc417b16\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c7d5d950f729b2074aaa087dd1d7b900ba8a32f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4c7d5d950f729b2074aaa087dd1d7b900ba8a32f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
