)]}'
{"id":"openstack%2Fdesignate~934561","triplet_id":"openstack%2Fdesignate~master~I7228f19f5aa2609a7cc244d3a9a3d850c1d50815","project":"openstack/designate","branch":"master","topic":"policy-yaml-file","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-11-14 20:09:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7228f19f5aa2609a7cc244d3a9a3d850c1d50815","subject":"Remove default override for config options policy_file","status":"MERGED","created":"2024-11-11 05:39:48.000000000","updated":"2024-11-14 20:10:40.000000000","submitted":"2024-11-14 20:09:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"934561-policy-yaml-file","meta_rev_id":"b7ff35409ab17b2c440f5fe18cc1b03c393cba1c","_number":934561,"virtual_id_number":934561,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-14 20:09:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-14 18:34:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-11-13 01:02:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-14 18:34:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-11 06:59:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-13 01:02:43.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2024-11-14 18:34:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"b44e80151faab5bb3577ffeea65036f80bc5a8c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:39:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42db71b20067cb5618fc4e61742544a211ae8072","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:55:39.000000000","message":"Topic policy-json-file removed","accounts_in_message":[],"_revision_number":1},{"id":"4ce1067c030054e5acc40fb66ada6de175ceb67d","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 05:55:43.000000000","message":"Topic set to policy-yaml-file","accounts_in_message":[],"_revision_number":1},{"id":"154bda00938fcbe245f123e7a943cb9552c573c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 06:59:41.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc53ce1f2e5b4939848e25ebd1812e1a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5b6f2487be544bc7a85808ddc28b7637 : FAILURE in 39m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f5420aca61464790856d4115148555e0 : FAILURE in 47m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a6b18de2314947b18575fdfe6f12b3db : FAILURE in 58m 07s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6199b3cda3c14a5ab4afeb7d090a482d : FAILURE in 31m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/264c33923b5647d493691d50aed6df14 : FAILURE in 40m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cd2ba8e8bc224dcba8f438e17bb396e5 : FAILURE in 56m 08s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0f782c16decd4ac68d91e1eaa495953c : FAILURE in 1h 12m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1493fcb420fc4de4882ced134ac93fd3 : FAILURE in 49m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2989962537c8482e874b0fc6747e8701 : FAILURE in 51m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90a6380da2e748dc94583c2dc3384ffa : SUCCESS in 5m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cccc68514d846d69ce231d1da836c9d : FAILURE in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2345af7d4db4597a8a90921cf98e3c3 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4db1ff1858b4ec48f74252e8879d79d : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04e5727aa75d4ea391966a9ddbcefe2a : FAILURE in 5m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/8d153f7ee42c4d018cd55c0f7b3352c9 : SUCCESS in 1m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4be00ac03b1d49d78cccd4fc0a916f36 : SUCCESS in 28m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c1c542797a774a86b7b5cf52568ab279 : SUCCESS in 4m 15s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b813befc1e444a39b77ba9dd14df2b14 : FAILURE in 50m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ea73a6884ce738c9189343c1bbf0390e59b2c73d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-11-11 20:20:38.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"306d53424c9bf898ad48e96044cf64cdc46085ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 21:42:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5606a247f8334370857c01b05e56f3b4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2e0290f564f64ccd8427f5140d364124 : FAILURE in 51m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/37ad5bbf21a04a41a8caabfaf1121c23 : FAILURE in 1h 01m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e568d3f120994395bc49338db0acb582 : FAILURE in 45m 02s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cad14924e6624782b479615cf0aa5034 : FAILURE in 36m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fac1c02c8b23461b935060fbb1a206f6 : FAILURE in 40m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/221dcd9f5dca49478a5e98ade93fa8e8 : FAILURE in 1h 15m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/32cfa6f9c36f4c0eb4522d59cefb54e7 : FAILURE in 50m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d3c87e78a0d44bfc8fbf2cf7d662de33 : FAILURE in 43m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/996b0d5cc63a48b2a0cc54609dd2ff21 : FAILURE in 32m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0dd16d8e72804430be6a3ae543fa4b94 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/203284c2b2184d46ac2c4f2618046fe0 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/091a85a170f84ca0833a510f613b8c49 : SUCCESS in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2a5b46efe8f84fbbae496c6265a6f0c7 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b257542e9c74f258da6fc1ab86d3a8d : FAILURE in 4m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c843690f3ca74c8d9e90a80c3a5fb870 : SUCCESS in 1m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6eaeb5984780478cb01ce393835bbf63 : SUCCESS in 37m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/44cb978bcd9647d499d8f491c91b66d8 : SUCCESS in 4m 08s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8bcbc2ec4d8545df96427f320d19642e : FAILURE in 39m 37s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92209985a9aef62084531e671749b7408ca32c8b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-13 01:02:43.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"c08a45071cef05a3140149fa33e5e04ce2666edf","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-14 13:28:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"46d455eec721881d9f547f688ea77d5547759f0f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 14:47:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/820c675352584ef8adb52184c4938a6f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad5761adde8c42ab9249cb3a0c3e3833 : SUCCESS in 43m 52s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e9e24ab0778247b1bb7395d26d439ecf : SUCCESS in 47m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0596fee5a97d4888803051c44037bed0 : SUCCESS in 56m 14s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c4cb001de8de4ba09f3dafc3b1c9d02c : SUCCESS in 48m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3f3e447e445a420f826d792c892da94f : SUCCESS in 45m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4a03a4b6766442d198fdf703b125407a : SUCCESS in 58m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b641838af18e4ce6a832b90791f25959 : SUCCESS in 59m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/63e7554889294bf19c21edcb09dd0668 : SUCCESS in 57m 48s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/951466b1d9004256b1d5083f444d2ccb : SUCCESS in 1h 10m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c2d6130002a2468a8ffb1508dd032d7b : SUCCESS in 52m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/18ecb59b4cd844a59f2c71fddab15504 : SUCCESS in 44m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e7aa88fd890f4edf9e05bccf7d4259c4 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6709a070a26a408484510484f54b6449 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4fa5e249e21b47f0b2a4446ec03d97e3 : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0dd9e781d954ed4a7dd41c68825bc15 : SUCCESS in 3m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5da57bf71a8a42ceb1ceba08080296a2 : SUCCESS in 7m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1fd0e140a70e413f89a147aa438dba88 : SUCCESS in 1m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e7ba89f6aac648a5ac8150ea14ebf437 : SUCCESS in 29m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fe4154178f004589a860c9a4f87172ac : SUCCESS in 4m 35s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/cf86348c2c5b41708e6819a871ce13cb : FAILURE in 37m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"12fabc437a7cccb0eddc803c84276ebcb14b84b6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-14 18:34:25.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"bdffa67164478be60295d4e2f4cb6ecfd8f1a38d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 18:35:05.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d3fe1d8762c506bb8dca4de840df1a0f5e89e388","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 20:09:42.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d218a3d0d3734661bed1d1603aed3045\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e743282876a6414f8736c39556085940 : SUCCESS in 48m 39s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9e166a3caabb42d394773835a91146e9 : SUCCESS in 41m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/377fe47d312343c1a7cf17d7f398f496 : SUCCESS in 42m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68c4ce78ef3740fa8c2f1ae2644aeefa : SUCCESS in 58m 09s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ace21aea2cd14da5a3727f178e4f3e85 : SUCCESS in 1h 03m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b49d276a5d37458e86beceb55a0406ee : SUCCESS in 46m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1192e2b02a6d4b3b954cfc8e5c89a601 : SUCCESS in 36m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8f56a7a294a46bf8a6c757ab5485ce3 : SUCCESS in 3m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5795528548f64d19b3bbcc829c3e9337 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7aee51c0302846eab5fbbbf5d4a7e9c5 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e8e76ad410841e38f171a7478da1a6c : SUCCESS in 7m 11s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9f84a42a6d2f424eb68f09f70b919190 : SUCCESS in 2m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1f5a84735184488798a178929d34dda4 : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":2},{"id":"e642112fd22034d92b58d08b4f198350b147c2fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 20:09:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b7ff35409ab17b2c440f5fe18cc1b03c393cba1c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-14 20:10:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bfe00c3c7a7c4317b174e45dd08a805c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9584f67a8d749c6bdcafef52a69abd6 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8a9a79ce7ff7a21f25d7e77034f7959d501955ba","revisions":{"bb48c7037f525232448039793ad2494151c48efa":{"kind":"REWORK","_number":1,"created":"2024-11-11 05:39:48.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/61/934561/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/934561/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/934561/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/934561/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/934561/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/934561/1"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:38:21.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:38:21.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: I7228f19f5aa2609a7cc244d3a9a3d850c1d50815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb48c7037f525232448039793ad2494151c48efa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bb48c7037f525232448039793ad2494151c48efa"}]},"branch":"refs/heads/master"},"8a9a79ce7ff7a21f25d7e77034f7959d501955ba":{"kind":"REWORK","_number":2,"created":"2024-11-11 20:20:38.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/61/934561/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/61/934561/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/934561/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/934561/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/61/934561/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/61/934561/2"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 05:38:21.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-11-11 20:20:24.000000000","tz":-480},"subject":"Remove default override for config options policy_file","message":"Remove default override for config options policy_file\n\nolso.policy 4.5.0[1] changed the config options policy_file\ndefault value to \u0027policy.yaml\u0027, which means it is changed\nfor all the OpenStack services and they do not need to\noverride the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides\nthe same configuration that services have overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/934012\n[2] https://review.opendev.org/c/openstack/requirements/+/934295\n\nChange-Id: I7228f19f5aa2609a7cc244d3a9a3d850c1d50815\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a9a79ce7ff7a21f25d7e77034f7959d501955ba"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8a9a79ce7ff7a21f25d7e77034f7959d501955ba"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
