)]}'
{"id":"openstack%2Fdesignate~935795","triplet_id":"openstack%2Fdesignate~master~I6da55bd967648771755ea1ee117bcaecaae69008","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-11-20 23:06:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6da55bd967648771755ea1ee117bcaecaae69008","subject":"Drop usage of pkg_resources","status":"MERGED","created":"2024-11-20 14:03:58.000000000","updated":"2024-11-20 23:07:27.000000000","submitted":"2024-11-20 23:06:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"935795","meta_rev_id":"ae99e3fd17416cacfc4c9632f0a909c65a8a20d8","_number":935795,"virtual_id_number":935795,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-20 23:06:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-11-20 18:48:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-20 21:42:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-20 21:42:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-20 15:19:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-20 18:48:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-11-20 21:42:59.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"12c7f8bb3ef3c66ed9145200c5f221183c50ddc9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-11-20 14:03:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"61a5ff0f4cf05418e399fed836f109996558a9f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 15:19:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d54b49f4368e4ea78135a159e2d425d3\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/03836ba516bb44129a163cf397dbe6d8 : SUCCESS in 36m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b138078087bc4786a741fb4747ff297e : SUCCESS in 44m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d79358743daf4acdad84ddc7bb266383 : SUCCESS in 58m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2cd558d6787d47c9bc219a4f1417153b : SUCCESS in 42m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5d37fdbbc85f4fda8bfb40643dd78554 : SUCCESS in 37m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f2592c00f5dc4a5ea5fc6768f96bb738 : SUCCESS in 1h 00m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f351f5f0945466c9d943f8eadc546fa : SUCCESS in 1h 00m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b0ecbfef431847879d341466fb718dc3 : SUCCESS in 55m 40s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/397dc532600243dd95ed498c502e2867 : SUCCESS in 1h 09m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/51857e781de4492f948cc4d5345f8838 : SUCCESS in 50m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d27731012d3140e2a443ceb305842383 : SUCCESS in 45m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/93ed3f2158664ea3ace68b000ff129d6 : SUCCESS in 5m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a9fd8f3258447639e2c669bf824c27e : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0a8286bce1b42ed820d63b19401e76f : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a44287154974729909f731fb8632162 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8ce327e7b9542ecad01c4415feee7cc : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/73f54d800a694ddea61dc327b157a30b : SUCCESS in 45m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b29fca492b924a8e9baa86c3020aa104 : SUCCESS in 4m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f542fe9aa91b45379b861a2ce1e58d23 : SUCCESS in 36m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b623093ff46bcd4a71ed31e80564ab43855b5606","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-11-20 18:48:29.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"596a9934499caacbc35a0d5504df507173e459c1","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-11-20 21:42:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9fad7d072e0135bf20b517bb8acc11b99d9e588a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 21:43:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24e0777a5d785102f54996b5f0ad5b5a92c13b7d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 23:06:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6138d4b14ef246c58798095d5bcdfa03\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6e583f313b654bb58c553c1604921973 : SUCCESS in 48m 35s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0b3a9b0c3891417990ddc49cb1afa741 : SUCCESS in 52m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/55b333aec2cb4276af19b850857ec476 : SUCCESS in 52m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f5f063f8c475489887494dd842074907 : SUCCESS in 1h 10m 33s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7cde90f8fe1146ed8dabe7e6c8f9dcab : SUCCESS in 1h 14m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a6d81f2327ae42bf8b64ec360595aad9 : SUCCESS in 46m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/56cae1eba54b4120ad08b1165fe9a28b : SUCCESS in 39m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07b18739e64141d98fd63e450f66909a : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb635fd304294a29b6291634e1f201ee : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/85a18a93b016400388d7a0eec99d2ffe : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ea8bd9422bc4998b41db98d30eef421 : SUCCESS in 4m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ab438c18bfa4320bca8db6f9afd1fa6 : SUCCESS in 30m 28s","accounts_in_message":[],"_revision_number":1},{"id":"025b51d390663222e777c5654d3d5f989b2f4b19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 23:06:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ae99e3fd17416cacfc4c9632f0a909c65a8a20d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 23:07:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a373aaba6fa34501967bd43cedc6dce5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a113714393d4f06ab7cac8ee6382af7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5ad4ded6a65f63355fc81fb4228e7e4f7cbaf973","revisions":{"5ad4ded6a65f63355fc81fb4228e7e4f7cbaf973":{"kind":"REWORK","_number":1,"created":"2024-11-20 14:03:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/95/935795/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/95/935795/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/935795/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/935795/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/935795/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/95/935795/1"}}},"commit":{"parents":[{"commit":"643fa563647a5aa74ac1db84aab2d0bcffaebff0","subject":"Merge \"Revert \"Disable GLOBAL_VENV in grenade\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/643fa563647a5aa74ac1db84aab2d0bcffaebff0"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-20 14:03:30.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-11-20 14:03:48.000000000","tz":540},"subject":"Drop usage of pkg_resources","message":"Drop usage of pkg_resources\n\nThe pkg_resources module was deprecated. It was removed from the core\nPython and has been moved to setuptools in Python 3.12.\n\nChange-Id: I6da55bd967648771755ea1ee117bcaecaae69008\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ad4ded6a65f63355fc81fb4228e7e4f7cbaf973"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5ad4ded6a65f63355fc81fb4228e7e4f7cbaf973"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
