)]}'
{"id":"openstack%2Fdesignate~936264","triplet_id":"openstack%2Fdesignate~master~I0f843f526f9af3b09a976664d820e64e1ef5ae79","project":"openstack/designate","branch":"master","topic":"fix-bug-2089474","attention_set":{"28338":{"account":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"last_update":"2024-11-26 15:19:45.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I0f843f526f9af3b09a976664d820e64e1ef5ae79","subject":"Fix 2089474: Add support for generic record types","status":"NEW","created":"2024-11-26 14:10:29.000000000","updated":"2024-12-30 13:53:07.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":5,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"f6f8be67ff9af1f4e1e2e6ac1f1a24d23e7b7594","_number":936264,"virtual_id_number":936264,"owner":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-12-02 19:17:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":-1,"date":"2024-12-30 13:53:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-26 15:19:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-12-01 14:32:10.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2024-12-30 13:53:07.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"37e5f26c3666cbf4d8788231bfa9c4e1a37a3098","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"date":"2024-11-26 14:10:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3684a92bd759bcdb1d4febcfacaad132464f00fe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-26 15:19:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c74e79c6ea3c47c5ae57a6b3c04bf690\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a69a4842dd434351a87e43f7b56f0323 : FAILURE in 26m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4850be8c68c24920b88e10ff1bcdb78c : FAILURE in 38m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7cc9bac985734fd2b7716ba9c1d00425 : FAILURE in 33m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a4aa2291d2e34074b161bb481c88c5ee : FAILURE in 28m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/65a2db5a328b49fc86409b337bc6ca4d : FAILURE in 25m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0436b9c1d3704fd4a0d645c80c517963 : FAILURE in 49m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/24b0908a178c4d1eb932f312bac9713a : FAILURE in 54m 29s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8244838b784e4c86b49266a270f3ea6a : FAILURE in 45m 14s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d73348e163a94ea3bca41e5fb15face0 : FAILURE in 23m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4b450e0bc6034f5bab801238de2d56e6 : FAILURE in 16m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ae7d43f745594cc18e982d947aecb067 : FAILURE in 28m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec59ab485bf649f2be60678c2996ce77 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4550c474ce614a9b911a25cf293dfcf2 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/86dbe8dd729a4463af7146d1a7f9fc72 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b61ddb6454c4d8c9531ca3dd56f72e9 : SUCCESS in 3m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d66e18256f9472b88fe29ab01c183f0 : SUCCESS in 6m 47s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5bf610f24e5a4c2ca3aa88957a1344d1 : SUCCESS in 1m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc2d45890195455d8464235b8327ced3 : FAILURE in 27m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5d4bc9551c4049d7ac940a2aef3475e1 : SUCCESS in 4m 23s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/00eed913c91a4e5ab3732c4f1a42d3fc : FAILURE in 25m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3f77a12a0633e9b7ca76b88297a420a716a338e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"date":"2024-11-27 11:21:03.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"81a17722a8996b4b7b7511b23cc14f4752adb217","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-27 12:35:12.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6f5180770604904be21004825869062\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/18b7bd1134e144b19b26133f5bff2229 : SUCCESS in 34m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a456e32737d64bc48bc64855ab24ad01 : SUCCESS in 46m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d360533a331d4879b81d5c218e4c9f83 : FAILURE in 55m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/62baae7c2b964cd084342976c7769dfc : SUCCESS in 46m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/43948a56945e49e6804a878ca27abdba : SUCCESS in 41m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/898d9caaf58a4872aa69b4b7edf42317 : SUCCESS in 1h 04m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4d224a34f4b7485b873f309fbe2d5843 : SUCCESS in 54m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/58d19fa6bb3a477d8ef14fb33749cb4e : SUCCESS in 1h 06m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0d1857d8d0c34b53aad8d602a1f30d8a : SUCCESS in 1h 02m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c0356b231a9f4abf9acd22276cd5cc6b : SUCCESS in 48m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8bc4afcca7b34438b4197c053f2fc9ff : SUCCESS in 51m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b76f92a0cae942deac266233a04cba95 : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a1170baaba7b48228ae928c5f16a9d15 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5907b31997cc4470b9deb7f8a0f9825e : SUCCESS in 4m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/542b85b109a5479dac926151ecdb1d72 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0faec4306474ba19e9ad67bcf3e5b67 : SUCCESS in 8m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/39f6e28a0afe4035aa860917009b7355 : SUCCESS in 1m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/acb43f3792a042148301b70ea3eae8bb : SUCCESS in 27m 16s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ff3e4b3f95424fe9b20b63e90c39105a : SUCCESS in 4m 12s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/e40e1ab82cdd4f32ae492d5732c9f449 : SUCCESS in 47m 29s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"67f4f11cda449e51f103e6b136ebac29749a2730","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2024-12-01 14:32:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c072aba6c06dfce7cdfaf51044215e08ce64419","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"date":"2024-12-02 16:05:33.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"9b273249b7476c9a745fa0d95bc7e9fb3fe69471","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 17:21:46.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87113a4f0bf7416bbd6cb91786f08369\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/41c5eb8899de46938dd733dd1e99f96f : SUCCESS in 51m 46s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f723eae7d47f4c2cab7344b262ecea0d : SUCCESS in 54m 37s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b61f227c339c4ee09aa43e0e1185c2dd : SUCCESS in 51m 23s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/522749a6cba3436ca3484870bafc5986 : SUCCESS in 43m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cbe569b80d614a0688ffbbf791644ae9 : SUCCESS in 45m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9ecb9ceac46b4f63a6ada136810cd779 : SUCCESS in 1h 03m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d6509085dfcd44198199fe8d78d33054 : SUCCESS in 1h 03m 14s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4dd16ba58ee941649d1f86667b2ec7de : SUCCESS in 1h 10m 22s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0660713bf1af44e28efd2a9214617240 : SUCCESS in 1h 04m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/49c50aabe8d94a7a84b0d599c452cafc : SUCCESS in 41m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9b566c6bd3c242938ac3f759a9b5e4b9 : SUCCESS in 40m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/89143e7f952c4124a32e06ce6f3d99a7 : FAILURE in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ee44b9fe2744188b8a9dea6ff9d3de2 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2875177b6859477e91df3563fd8207ca : FAILURE in 5m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e984f1ae63124965ab6c628cc3ceaa35 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/541107793db94630b37748f6dfe2adf6 : SUCCESS in 6m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4580832cc54543cdafff3ba8b9786cbb : SUCCESS in 3m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0c05698ffdb74e5cbb215f9553bd1b68 : SUCCESS in 47m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/aefd8d187aef465095e5313a64cf0a77 : FAILURE in 6m 56s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b2854336053e4dc9a375111c78bc6201 : SUCCESS in 42m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aef107cbdcada9bc2a50d36afb5e3235ccb4d5b0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"date":"2024-12-02 17:55:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"bf2ec2297b39aa319e350dbaa05a4cc4c3342574","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-12-02 19:17:03.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4737d2ea1f8b4c82b3782dff307b4110\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ae51bbd8e6794e1d969ff7cffae53d94 : SUCCESS in 50m 51s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9455b158901243bdbc65dab6da791866 : SUCCESS in 1h 03m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0c821738b7d649feb75f4a4632d4e567 : SUCCESS in 52m 23s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/26f647710e48457f8a592253dcab7e73 : SUCCESS in 58m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5a4ec626d4db4c58a3cba79c40f9e497 : SUCCESS in 52m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cfe74a41f9c74127bcb06adf8a66d655 : SUCCESS in 1h 14m 34s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/947f5c0e87c841e9857854398d268b9f : SUCCESS in 1h 06m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/bd2646c3b2ae467c90480c9c07d6d1b2 : SUCCESS in 56m 38s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/3c25c7c96f6c4ade8ca7dd429c993990 : SUCCESS in 1h 09m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4f370470ff724e54accd0266d1e098a2 : SUCCESS in 45m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a8a2dae6e4814f5c8e7aa028af68869e : SUCCESS in 58m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83d4d95f08ac48bda84852c65231930e : SUCCESS in 6m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6e699fc5fdd47ec87dbbbfd47f6085f : SUCCESS in 3m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e0f74565b2f4a378b3d224f0df04590 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41ba247531c144b68ecc13d6cd3120c5 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/932a40ff216443ef89d76940171f24f2 : SUCCESS in 6m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/24ace270b65b4b069bca52b89b7128a4 : SUCCESS in 2m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8b54869dbd64ec1b1ba8f8e466a8071 : SUCCESS in 43m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/25a199ff33ae4bb0a861578c42269bbb : SUCCESS in 6m 15s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/58494b04cfb94c058f70c290cea73485 : SUCCESS in 46m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f6f8be67ff9af1f4e1e2e6ac1f1a24d23e7b7594","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-12-30 13:53:07.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5b8262363afabf8539824b3efa48d13c03eab975","revisions":{"56448e0a42d7b807170c976940af859c0aed6323":{"kind":"REWORK","_number":1,"created":"2024-11-26 14:10:29.000000000","uploader":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"ref":"refs/changes/64/936264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/936264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/936264/1"}}},"commit":{"parents":[{"commit":"a0935ea53ae20fa85c4af49c8cfb3b8b38125bef","subject":"Merge \"Remove unused host/port options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0935ea53ae20fa85c4af49c8cfb3b8b38125bef"}]}],"author":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-11-26 14:04:45.000000000","tz":60},"committer":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-11-26 14:10:27.000000000","tz":60},"subject":"Fix 2089474: Add support for generic record types","message":"Fix 2089474: Add support for generic record types\n\nThis commit aims to add support for recordsets which\ntypes are in the form of TYPE[N] and the records data\nstarts with \u0027\\#\u0027. This has several advantages, such as\nallowing the use of record types unsupported by designate\nbut potentially supported by the backends.\n\nsee RFC 3597\n\nChange-Id: I0f843f526f9af3b09a976664d820e64e1ef5ae79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/56448e0a42d7b807170c976940af859c0aed6323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/56448e0a42d7b807170c976940af859c0aed6323"}]},"branch":"refs/heads/master"},"a26bd59e34246bd102dbcffbdefca1bf9cfe0df5":{"kind":"REWORK","_number":2,"created":"2024-11-27 11:21:03.000000000","uploader":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"ref":"refs/changes/64/936264/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/936264/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/936264/2"}}},"commit":{"parents":[{"commit":"a0935ea53ae20fa85c4af49c8cfb3b8b38125bef","subject":"Merge \"Remove unused host/port options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0935ea53ae20fa85c4af49c8cfb3b8b38125bef"}]}],"author":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-11-26 14:04:45.000000000","tz":60},"committer":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-11-27 11:20:31.000000000","tz":60},"subject":"Fix 2089474: Add support for generic record types","message":"Fix 2089474: Add support for generic record types\n\nThis commit aims to add support for recordsets which\ntypes are in the form of TYPE[N] and the records data\nstarts with \u0027\\#\u0027. This has several advantages, such as\nallowing the use of record types unsupported by designate\nbut potentially supported by the backends.\n\nsee RFC 3597\n\nChange-Id: I0f843f526f9af3b09a976664d820e64e1ef5ae79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a26bd59e34246bd102dbcffbdefca1bf9cfe0df5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a26bd59e34246bd102dbcffbdefca1bf9cfe0df5"}]},"branch":"refs/heads/master"},"02c724cd2ffbfe248fed792dee7d686fb24a1241":{"kind":"REWORK","_number":3,"created":"2024-12-02 16:05:33.000000000","uploader":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"ref":"refs/changes/64/936264/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/936264/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/936264/3"}}},"commit":{"parents":[{"commit":"a0935ea53ae20fa85c4af49c8cfb3b8b38125bef","subject":"Merge \"Remove unused host/port options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0935ea53ae20fa85c4af49c8cfb3b8b38125bef"}]}],"author":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-11-26 14:04:45.000000000","tz":60},"committer":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-12-02 16:05:08.000000000","tz":60},"subject":"Fix 2089474: Add support for generic record types","message":"Fix 2089474: Add support for generic record types\n\nThis commit aims to add support for recordsets which\ntypes are in the form of TYPE[N] and the records data\nstarts with \u0027\\#\u0027. This has several advantages, such as\nallowing the use of record types unsupported by designate\nbut potentially supported by the backends.\n\nsee RFC 3597\n\nChange-Id: I0f843f526f9af3b09a976664d820e64e1ef5ae79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02c724cd2ffbfe248fed792dee7d686fb24a1241"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02c724cd2ffbfe248fed792dee7d686fb24a1241"}]},"branch":"refs/heads/master"},"5b8262363afabf8539824b3efa48d13c03eab975":{"kind":"REWORK","_number":4,"created":"2024-12-02 17:55:30.000000000","uploader":{"_account_id":28338,"name":"para sitid","email":"parasitid@yahoo.fr","username":"parasitid"},"ref":"refs/changes/64/936264/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/64/936264/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/64/936264/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/64/936264/4"}}},"commit":{"parents":[{"commit":"a0935ea53ae20fa85c4af49c8cfb3b8b38125bef","subject":"Merge \"Remove unused host/port options\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a0935ea53ae20fa85c4af49c8cfb3b8b38125bef"}]}],"author":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-11-26 14:04:45.000000000","tz":60},"committer":{"name":"yann degat","email":"parasitid@yahoo.fr","date":"2024-12-02 17:55:11.000000000","tz":60},"subject":"Fix 2089474: Add support for generic record types","message":"Fix 2089474: Add support for generic record types\n\nThis commit aims to add support for recordsets which\ntypes are in the form of TYPE[N] and the records data\nstarts with \u0027\\#\u0027. This has several advantages, such as\nallowing the use of record types unsupported by designate\nbut potentially supported by the backends.\n\nsee RFC 3597\n\nChange-Id: I0f843f526f9af3b09a976664d820e64e1ef5ae79\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b8262363afabf8539824b3efa48d13c03eab975"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b8262363afabf8539824b3efa48d13c03eab975"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
