)]}'
{"id":"openstack%2Fdesignate~938472","triplet_id":"openstack%2Fdesignate~stable%2F2024.2~I379201bc3f71870cc49d1d5b3b50cfc9b99e4764","project":"openstack/designate","branch":"stable/2024.2","topic":"bug/2092394","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-01-13 23:53:38.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I379201bc3f71870cc49d1d5b3b50cfc9b99e4764","subject":"api: Handle duplicated query string paramters","status":"MERGED","created":"2025-01-06 12:58:35.000000000","updated":"2025-01-13 23:54:45.000000000","submitted":"2025-01-13 23:53:38.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"938472-bug/2092394","meta_rev_id":"22f617588e56a0ceb4505d3a237016cf146e3900","_number":938472,"virtual_id_number":938472,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-13 23:53:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-01-13 22:50:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-01-06 21:40:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-01-13 22:50:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-06 14:16:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-06 21:40:52.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-01-13 22:50:12.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"dc8fc272fc9bd0002552ddbcbfecca70ff1c59c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-01-06 12:58:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68937bc189f9731f294d91a5ef8c428840743f4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-06 14:16:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/802a7eb8d5dc4b3cb3ef25798f4aad75\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84294ae1d1ae41fabcf395b3a402b67b : SUCCESS in 32m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ec503f87d7a84b97a5fff0567be4877e : SUCCESS in 52m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/7ca1fc0219b84c9d9c43f86f06c49ff4 : SUCCESS in 47m 22s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c04afe49d2574fd0afec65e333e3a821 : SUCCESS in 52m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d8494828ab1a4dd79d8aad9ef5cdcc55 : SUCCESS in 48m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/21c0ab2c2f34459e8fc6d4a9727e65ed : SUCCESS in 1h 11m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/560a804fd4c2427f9325a0ad0bc138e9 : SUCCESS in 1h 15m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/497f8b441522489295d61c8112397cff : SUCCESS in 48m 29s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f7437f157936467abe655fb943acd241 : SUCCESS in 45m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52a62b93eff74f299bcefe5fce2e3379 : SUCCESS in 6m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efceb99aed014d0c9f756b675fa5b93f : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7a789d6bcdf46f1b29e0cd9a30802bb : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a0792e2514c4d0e819ab79a580b8a16 : SUCCESS in 6m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7995fab8f21d41efa9b5252240f17bfb : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6cfbab5aa4084f67ac531199e1b8f747 : SUCCESS in 5m 36s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/559c8e86b67b45ceac73e3f2ac770794 : SUCCESS in 28m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/46d85c3b3e9f4cb2bb5461b2450cee61 : SUCCESS in 3m 59s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/a166f2d289a24a2da0c7361e1938d652 : SUCCESS in 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"db4d1fd264516d1f96218208637e1ae1567e9490","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-01-06 21:40:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c1fc3a6fcbe71833f3951b4a4f74171833c6d768","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-13 22:50:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"04b2d3c6b0f9c8b163dbc888f7feea8f7b1edcc1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-13 22:50:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"83ab52b8fa4c1515de3ba11046fbb74314654957","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 22:50:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d9004718f3cfb355b3a30d3206245a8534b2273a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 23:53:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a699e39f79b54044b4fc8195c440ef3d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fcb67457e4b74fc3bcb18eb36137f48a : SUCCESS in 53m 16s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/631244980c304b948057471265f64757 : SUCCESS in 47m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ff4ce385c2474501aae45f1d7e636830 : SUCCESS in 49m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/11cc3ad4db4746938e5b64f7e5067a59 : SUCCESS in 1h 02m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8185eb27f87b4275a4327e3338bc7bc4 : SUCCESS in 54m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6c2b0cae30fa4d9a8e1700e8c33da527 : SUCCESS in 31m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/127dc05e31154b37ab814aa64361fc1c : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4674129f85194c93b13c6d2c31d78968 : SUCCESS in 4m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba2d39f01f06481895f41fe89272c23b : SUCCESS in 4m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/247a050a8df446119a9146265626e039 : SUCCESS in 6m 33s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/842b864547a944abb2a2338fb5ad66a7 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"3862693529f9883b0468707990962a6763d9b53a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 23:53:38.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"22f617588e56a0ceb4505d3a237016cf146e3900","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-13 23:54:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c573ebe571904738a6ab2abcfb919aca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/224d6384d873484d82975738c5b90941 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"505ea9b1245e07b28e12c1be3ca5d5e86d77efaf","revisions":{"505ea9b1245e07b28e12c1be3ca5d5e86d77efaf":{"kind":"REWORK","_number":1,"created":"2025-01-06 12:58:35.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/72/938472/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/72/938472/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/938472/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/938472/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/72/938472/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/72/938472/1"}}},"commit":{"parents":[{"commit":"a2f6f07e2ca7d1b67b8b221bed0529e93aebc90a","subject":"Merge \"Stable-only: Switch neutron job to stable version\" into stable/2024.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2f6f07e2ca7d1b67b8b221bed0529e93aebc90a"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-12-23 10:51:34.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-01-06 12:58:29.000000000","tz":0},"subject":"api: Handle duplicated query string paramters","message":"api: Handle duplicated query string paramters\n\nWe don\u0027t expect these anywhere, so raise a HTTP 400 (Bad Request) if we\nsee them.\n\nWhile here, we convert the method used to a classmethod since it doesn\u0027t\nneed access to instance state and makes testing ever so slightly easier.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I379201bc3f71870cc49d1d5b3b50cfc9b99e4764\nCloses-bug: #2092394\n(cherry picked from commit 0d77707b1fb37a21d2232ac5ecaf82df65a3f420)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/505ea9b1245e07b28e12c1be3ca5d5e86d77efaf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/505ea9b1245e07b28e12c1be3ca5d5e86d77efaf"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
