)]}'
{"id":"openstack%2Fdesignate~940090","triplet_id":"openstack%2Fdesignate~stable%2F2023.2~Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581","project":"openstack/designate","branch":"stable/2023.2","topic":"change_neutron_job","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-04-21 09:24:22.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"33685":{"account":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"last_update":"2025-04-21 09:36:16.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581","subject":"Support \"-\" in region name for FIP\u0027s","status":"MERGED","created":"2025-01-24 02:40:32.000000000","updated":"2025-04-21 10:20:30.000000000","submitted":"2025-04-21 10:19:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940090-change_neutron_job","cherry_pick_of_change":940089,"cherry_pick_of_patch_set":5,"meta_rev_id":"1ed5b43de5d728f49f01bc128528bc452bc2338e","_number":940090,"virtual_id_number":940090,"owner":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-04-21 10:19:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-04-21 09:24:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-04-16 18:31:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-04-16 21:00:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-04-21 09:24:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-04-21 09:24:22.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-24 04:07:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-11 10:02:51.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2025-04-16 18:31:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-04-16 21:00:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-21 09:24:22.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"c899bd9195fb4ab043573254ae2e025660477ec6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-01-24 02:40:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* designate/tests/unit/api/v2/test_floatingips.py","accounts_in_message":[],"_revision_number":1},{"id":"a50592345e68fd40dbd764ee51a8ba594c0d6014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 04:07:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a15458573d5c40c9b1ffc50e0d3df489\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/64e75c98d0864ffc9800a0b78f224756 : SUCCESS in 43m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/81810604063b4323a7cbc319773a6728 : SUCCESS in 46m 01s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/521abaaee5fb4b05a5966232fe62de98 : SUCCESS in 48m 09s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/80ee6b335fc2413ea48d033414b5dfb9 : SUCCESS in 43m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/28751999a70e4c9e8be0532e421d67bb : SUCCESS in 47m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a171cb61a33b405eab09628fc7d3778d : SUCCESS in 1h 04m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3070ea5084e4472dae5a7872594ecd2d : SUCCESS in 1h 14m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af7db8856b5548dca9f1c3506441618f : SUCCESS in 42m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6125acac5df8461eb50c3c4605ef1cf5 : SUCCESS in 47m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0e689ce153b34bd3bb2331e823ea75f7 : SUCCESS in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3687c3b8535427098957a23d2caea25 : FAILURE in 3m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37481ab8c0c043a7ba34eb8871d585ea : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bb41e881d35e4ebda4ce5e704e4fbab4 : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a44aba662a64ef58699291a4a015d7e : SUCCESS in 3m 24s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6e377b355c99488fac44badd56762690 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9818effc38e6477e842520aededbab7b : SUCCESS in 7m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d3facc7bef5049f8ab40a82bae93dc27 : FAILURE in 17m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/52c939b61fb3400e9f062dc24025e3cd : SUCCESS in 6m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d79d3dceee64e19fb3c25783b98e4b8e27a58aa5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-01-24 10:48:14.000000000","message":"Patch Set 2: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* designate/tests/unit/api/v2/test_floatingips.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7650ba61bdee8ef4f3f04326f35a3affd4fed3fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 12:02:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f020d2de57884a789d08096b6c156560\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/58245ab8a8c34ff2a7ba22d74a63b696 : SUCCESS in 52m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6c57cc4a61be4085ba8d6e9eb2505007 : SUCCESS in 53m 20s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/be82914d66e842cca275d589f5f90527 : SUCCESS in 44m 10s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6f7b782fd7f3451798d81f7cb0534d9c : SUCCESS in 37m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/bd35bdc2c87141a9b422781d1780abcb : SUCCESS in 52m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d226e85d0de74cebb538cddf5c52fe86 : SUCCESS in 57m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/68f96737b1d746b19e48359dc35d8438 : SUCCESS in 52m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/01056c86c5984542abe17cbc95ef8199 : SUCCESS in 35m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a20a00d7cc104c1388d72180136330ee : SUCCESS in 37m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3be9ab0cfb26475891a0af5f2fe63a2f : SUCCESS in 6m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62811dd4898b40dca1242f34f26d1c31 : FAILURE in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/50d8dbae27374e878602263e28998d4e : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd80803e1143460ba6e3fd066c2f4a44 : SUCCESS in 4m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dae5466f8ea6457582b22a00bac63ecc : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29758a67cf604dc58629ac76dcafbd13 : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b45234189524ace94ec399ec8d4ae07 : SUCCESS in 7m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6d78d85c42244ff0b0c4beef2bdf42c2 : FAILURE in 13m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/213d1689d2c8451c896ef679fb7f3a73 : SUCCESS in 4m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"c26eeb37c552a7f10ecda5a0ba866306cc5f0079","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-01-24 13:53:59.000000000","message":"Patch Set 3: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* designate/tests/unit/api/v2/test_floatingips.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"aa228eaabfc7425505075f5010aa7b603578a65e","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-01-24 15:30:06.000000000","message":"Patch Set 4: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* designate/tests/unit/api/v2/test_floatingips.py","accounts_in_message":[],"_revision_number":4},{"id":"90b73878e5f8ad746809e009326582eafeec0040","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-24 16:51:21.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5a9b147a28c749a8a363fa42c216c6b6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c1ca44440c434103836591207ce67449 : SUCCESS in 35m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/937bc77c33454400a449e6c6d38b0453 : SUCCESS in 55m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/14ffc52cd44145cd899c5e0489025b4d : SUCCESS in 42m 36s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ffaaab6930d046e18f90e7817e247ba3 : SUCCESS in 40m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/45370de6fc06454f97de56a4b059c3f8 : SUCCESS in 45m 30s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1c4291485f534e43baad6a4512f04976 : SUCCESS in 1h 06m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/987dd5d8d78d44f6b365a00f478f5bb6 : SUCCESS in 1h 00m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c362e18b8f9e4d608ca14c4b85e9c400 : SUCCESS in 34m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/48e3383309894c589360eacda9f16bc1 : SUCCESS in 37m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d4561fe72b794a1f80b9f1d8b859d370 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4371896762534034ab651106f70d13a5 : FAILURE in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3956113e26534530a9fde0e98ff4ea88 : SUCCESS in 4m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d3b6aa52ec3440f8828c54dbdc31142 : SUCCESS in 5m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efb21645b53148d4861afbfbdf4aa618 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/507e102db0d7494dbf814d6e48d6cd77 : SUCCESS in 7m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15e31a78592d425584c24b2cf68cb850 : SUCCESS in 7m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7f44576ddece494eaf6e45f0aeb51b44 : FAILURE in 14m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d9f69b8e32b34af1890b36d76f255173 : SUCCESS in 4m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"631797747526cefc5e2b0d7d3e118f06bdf77a6b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-01-25 00:47:22.000000000","message":"Patch Set 5: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* designate/tests/unit/api/v2/test_floatingips.py\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"e1b7484fc5ce30ba8d5be8c0db1336991e4e24de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 01:55:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5b868afff8a492297019c5d1d8bbcc2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a91451c422a64f9c8f3b42e94dbbed20 : SUCCESS in 40m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/70d5c918cb4948ae939586ca15ee8da2 : SUCCESS in 47m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b01b3e7957f9448096e547f8ac1d71a5 : SUCCESS in 40m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c03c62b813b44a6ebdd7eb51db158eb6 : SUCCESS in 36m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4837d95768ed4ce793bc972640f17cb9 : SUCCESS in 36m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8d8bf630799a443cbecc96cae71b8216 : SUCCESS in 49m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c91563f8d2974e1fa61e87611ad9c47c : SUCCESS in 1h 01m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e793da7613074d12ae2d410fe43bd0b9 : SUCCESS in 31m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/64a78987cd744d4ea328732f933f5ce9 : SUCCESS in 38m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dae14a888e5448b1a18e6a890aae024c : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9ea847a3694ad0bf49f9097f50c399 : FAILURE in 3m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7ec5cb21b7b4cf6b4abc5ab31e219f5 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cdcd651e8dde4dd28b2b9dd6583cb654 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f034dca82f7b4c01b55ab2ad202ed513 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/954bce07346640cf9b40a33f61a89cd5 : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/590a107b410a4f1fa175fa8823b22968 : SUCCESS in 6m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/17bacd153206408cb83d0480a66f2d2a : FAILURE in 14m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/35a1859d7d34422c93defcb7ae979e48 : SUCCESS in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"47423cb8ad388a9ab9799cf389a864c0d336f177","tag":"autogenerated:gerrit:abandon","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-01-25 02:21:52.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":5},{"id":"c10f57b19a385b59d1d23525c686592c673a839a","tag":"autogenerated:gerrit:restore","author":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"date":"2025-01-25 02:25:24.000000000","message":"Restored","accounts_in_message":[],"_revision_number":5},{"id":"59cc8fb652b10fa37fcfbb81b425e8f43ea51d6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 03:33:38.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bef7cf108f6444b9afca2fc963f53f81\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/44ab5850d8e04f70b8b8ca9f9779fff2 : SUCCESS in 45m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/08d7585daffc44adbfab5927208a63f0 : SUCCESS in 54m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2f17ce66d97c44fe9b28c36430551bff : SUCCESS in 35m 25s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3111048b565b4bf9bec889b2bb07a975 : SUCCESS in 38m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3e7aa12366a74868ac182074e19315f7 : SUCCESS in 38m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3d985c8c846e4924af6ffad1539f0168 : SUCCESS in 59m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f7b38ef07257435d8dc2d4cdb8c5130c : SUCCESS in 48m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c3b9429b8b0545d693add139ea939915 : SUCCESS in 36m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/47f2f60974224e748fc0d06d8375bd22 : SUCCESS in 39m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b8dfc2542cb4d4e9cabe49d14cbc995 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e6b95d4b4f4460da80ddf29c4e2851b : FAILURE in 7m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b6cdf611719f43d19eb7ddfba955da4e : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8782bcd5d8094a57a4e959108419534c : SUCCESS in 6m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc708df109d44299822aef381e79a0a1 : FAILURE in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f7bf5eec6b94d03a43eeca299c3ce84 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea01fd91835f4bafa7a16746ccdf6045 : SUCCESS in 20m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eec505dfc5bf4df19e61caa8a1cc4d86 : FAILURE in 19m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/93beaa8b6c4b42878b98990bd09cdf35 : SUCCESS in 4m 47s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"521f2ea3f4ead9fedf486608d06c2548e0b90aef","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-10 17:35:04.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a97d62a51f3394dbd9b9455d6f1a1a3788b31c9d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-10 17:35:37.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":6},{"id":"4e43c167f4d89a17a2ec1f31e792e0aac87f556d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-10 18:35:22.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/872522a538eb48008c81b8d4457b8453\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bbded0e0e8e342eba3a609f4b4efe91f : SUCCESS in 40m 36s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/558c955604ea468e8651f1d291ba16dc : SUCCESS in 46m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8bdc5c16e5e64fc0beb9795d5dba2388 : SUCCESS in 43m 38s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e0ed5adb42734bfcbea16cac6bb172b0 : SUCCESS in 38m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/226b4a7a6a1a471db9bf51e1c468842d : SUCCESS in 43m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e873195f59b946758a6bed0185634cc6 : SUCCESS in 54m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/446deff29ad74a0bad40d83d313fd157 : SUCCESS in 55m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b83ef4654bee4f0f95617b27c1a6fa9f : SUCCESS in 38m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c6d023996ac046068d713c49a9d9503c : SUCCESS in 34m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/693b31d7290e49b88e2bd10a2d3c803b : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d15fa815dbb43c284d16f100e07a421 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a4e2243a5ba4eb4b3184bd802b77228 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3d076b392c04c8e9b53f0ae5c181706 : SUCCESS in 4m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb4f19f8712d4e04b0e22e4e562af716 : SUCCESS in 2m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0a1389adf84405783741a6981d16f35 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/00292eaab7264251930ff19dea3551d3 : SUCCESS in 7m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6dd179762f2f4586b03c3b3f947db6a2 : FAILURE in 14m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f8dbe64f42fe4d81a13520a5424f2a1b : SUCCESS in 5m 51s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"febd81d6fec1fea6d4765e72f5e3a8182e0aec85","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-11 10:02:51.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"11e308749f7bf082c6f6b270a4ddfebbb08b49a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-11 11:25:09.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3774f36987714c86b32020961bcd95dd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5b6f624d17e04f36bfe1a5f227f91822 : SUCCESS in 34m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4a5eb5f6f8634f089e45440a51abd1ca : SUCCESS in 44m 52s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/970bd0e107e94148bd056fa3a8823adb : SUCCESS in 36m 49s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7fefd01838554312a0756741bca1bb78 : SUCCESS in 34m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d3dbd1789a454625a2fa1ef1597aa02f : SUCCESS in 37m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c2c7bb6d3b5549e8b3b426f8c1cc9f5b : SUCCESS in 54m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/69fcb79b9df641288da744a152216781 : SUCCESS in 54m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/081febc238624a1991cd841c812ad8ce : SUCCESS in 38m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2f5926a270674edc97aa6828d62e5e4f : SUCCESS in 49m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/88ec41bbdeef47a8bba85092976e3b80 : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64aa3ff8dd84d99847554d49ab453c9 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/41ec926c56644b23a7e623efb5bace88 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/50c084ec305446f08afb4a1480ad9911 : SUCCESS in 5m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/054717530a454a878785c8135404e48e : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30378ddc18a54c758307b07a2bd76be3 : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eeab73144dd444cb07d41e5b3ab2da0 : SUCCESS in 8m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d0d39a4a3a34c7484805b2cc23defae : FAILURE in 9m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/443c7515fc53468d906be331376424a6 : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"ebed9f81fda2a9ebfc15630ef135a292a908ec20","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-15 12:33:36.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"28e9d16906603bc57e21ba2f590e14423881249d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 13:37:10.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2428365a99bc496aac3d1bb68b0a8646\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7bc2803f79cc4331a1adb6685bf35930 : SUCCESS in 34m 53s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d8ca04949edd43209c078e0e972fa2ab : SUCCESS in 52m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/759e2e00c00a48afb395b4d4499d3c2b : SUCCESS in 48m 44s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/89ec94842c88492781379ccf755dcc1b : SUCCESS in 38m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7d67dc5aa17b44d0a47b8602884a8de4 : SUCCESS in 35m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/eba86340243241e2bf6863ea471f81df : SUCCESS in 58m 29s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4b8717b5a36a4bcdb06daa5979a63e47 : SUCCESS in 49m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2fd972b31888483ba936d38eb0e1e01b : SUCCESS in 39m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cf978e6492754fcbb9cabbb08b8464af : SUCCESS in 38m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/733b35a88ef8454baee9479fe08670e5 : SUCCESS in 4m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af25e4a0427048068b7554fffe31b515 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/774b6665946045dc92a4f79280dc7e8c : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb1953eb35054707878461bf5553927a : SUCCESS in 4m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7e5ca77da2164a3eb57017c2a4105305 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba2b5198aece4c7c80582324cbf13ada : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266a65545b834953a0b2bbc9f5c4636e : SUCCESS in 7m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/be2f89d00f764ad581acca3aa7453bf1 : FAILURE in 18m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/216307a3a13546b5b6d7a005f12d3c12 : SUCCESS in 4m 26s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"64ac608a1560856b35ee68449ceb1be8ff066c8c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 15:57:09.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"4c81b742a5a9345ca35b45b426439627e4c93ffb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 16:57:02.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/615da2c45bea46da92efe34ed22d00f7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7ebd16c8338a45008495e6171f196bec : SUCCESS in 45m 09s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/caf3f8c69c1e45edba9f40d92fe14478 : SUCCESS in 45m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d54ef6c23b34be6a688cdb15827fa07 : SUCCESS in 48m 40s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d79c14eecdd5440e854521ae9e3f29af : SUCCESS in 33m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e36e13192cd6423ea108b97ef0829c58 : SUCCESS in 36m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4f89492155934f709c930fc580b21a81 : SUCCESS in 52m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/67790234c6a64ebf83adae6d1aad1f8a : SUCCESS in 54m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/be4e74c20ea943769ee491427b0a9170 : SUCCESS in 33m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c69b3d98460247418c8053569e69b50b : SUCCESS in 28m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae8fb7fe3e004785999cccb4ce709d4b : FAILURE in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb10f817c0b640e4aca6a051996d76fc : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/806ce95bff52405bb9152e4322f72bea : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/22445422c97044c2b26e6b74b6a4c37c : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad2319bdf3714fef9a1b283589831998 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/613eaa810e6940f88678f2a309a9c70d : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee67928753274f888fdd2c4770c89dbc : SUCCESS in 7m 56s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/55ecd23b218a4e90b278878d1fe04b60 : SUCCESS in 26m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ec3a19a6d33a47c482cc0cfb3ce58633 : SUCCESS in 5m 41s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"b6cfd3361af5b590503caaa030e0a1c52ae742c6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 17:12:31.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e2a82d7f8139fc948dc05949f6a98e4234201207","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 18:15:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f90af29b97cf424bb038e168711e3e4d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2e27c084ae0448c093fff2dbe825f9af : SUCCESS in 38m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/60e04850f91241868357cbb130512a33 : SUCCESS in 52m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b20a62b1ccd4325b98a540aaf66840e : SUCCESS in 48m 12s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9d8758524b164636b1ba852b57e7d458 : SUCCESS in 37m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/372fbd9c999443b794a38040dd16500c : SUCCESS in 36m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b7b873f11a9b42fd9aa91d0a548f0a41 : SUCCESS in 57m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3123ced30c2644ab9dc2b7709f77ba1a : SUCCESS in 55m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/af315182f395483d91795f732cee8f0f : SUCCESS in 37m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/03b0d9f9870f4b4f8e9c97eb6e072601 : SUCCESS in 34m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9cc8bfac20e42b797dcd6fc1dc62b0e : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c1f657b92924cb8b7ddc25907296d1a : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2f959bce9cd14b97af8245ffb41665b4 : SUCCESS in 4m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67f063685965488081291f9ed1ea1da9 : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c328f99951434c8aaf2a023a7a6b006f : SUCCESS in 3m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f9232d2609442e2803ff6c77a802691 : SUCCESS in 3m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5aa7dbfb1d34ab6a8daec467927671a : SUCCESS in 3m 42s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/fb37f84181e748f5bf96dcecb8b261d1 : SUCCESS in 16m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8745e8c8d2f944e2bf81074378c95244 : SUCCESS in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"f81d8864016a0c57870cffa8b5fd5d2bd9094d30","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-16 18:31:20.000000000","message":"Patch Set 7: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"29defc0101ffeb6238197c3cde5812e1d7cec309","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-16 21:00:06.000000000","message":"Patch Set 7: Code-Review+1","accounts_in_message":[],"_revision_number":7},{"id":"27e95f65af0cee3584c37da12b5fad58152d5291","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-21 09:24:22.000000000","message":"Patch Set 7: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b004163ee8e6540a4b9c62e2bf5839c265406f07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 09:24:46.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"ddbd04575815e41de7496a1fae61ff816381c8c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 10:19:21.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6f512471ef243189f3646402e2d0e98\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4966754181c04821933001d1006bbe7b : SUCCESS in 35m 13s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/c37f514d61e144c19516dd638ce33521 : SUCCESS in 38m 12s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a51ad31089714c95bf802929b65b2a2d : SUCCESS in 30m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a7a32710a2544ee1a793ffb1386e565e : SUCCESS in 53m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/eb2ef08555cd49d3b8f811c77177bbe0 : SUCCESS in 37m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9b34445140804e95a1912bfbcdcd1b74 : SUCCESS in 35m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1270b0ef40aa4fd09c932cfc97bf93a0 : SUCCESS in 3m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/384f9800b2bf40e4b33571173d63d65c : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dcf0a77dd7484bda859a7515c91bea20 : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1bd6ed1bc1c4f1dac7fb663c2d94e46 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23458e38e8c64e108358448633567d14 : SUCCESS in 6m 47s\n- neutron-tempest-plugin-designate-scenario-2023-2 https://zuul.opendev.org/t/openstack/build/d05a8f8d2e03457ca78253f16c3dccb8 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":7},{"id":"4b7809180e9cb97ba2bd3b9961b68f59d1c1231c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 10:19:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"1ed5b43de5d728f49f01bc128528bc452bc2338e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 10:20:30.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2db609446bad46bcaeb095a8d7ede0db\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6d78113736f440fa61bd64d1e33bfde : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"eaecbaa8173a9c2efe1dccd2157928d2d0574133","revisions":{"2cfa82be82c595241c968e62d8b2f4942ba49a50":{"kind":"REWORK","_number":1,"created":"2025-01-24 02:40:32.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/940090/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/940090/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/940090/1"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:34:35.000000000","tz":240},"committer":{"name":"Mikhail","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:40:32.000000000","tz":0},"subject":"Support \"-\" in region name for FIP\u0027s","message":"Support \"-\" in region name for FIP\u0027s\n\nAfter moving to raw strings for regular expressions region names with a\n\"-\" in name have been broken because raw string duplicates backslashes\nand it was two backslashes in inital regular experession.\nWith this changes extra back slash deleted what is correct to r\u0027\u0027\nstring.\n\nCloses-Bug: 2096623\nChange-Id: Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2cfa82be82c595241c968e62d8b2f4942ba49a50"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2cfa82be82c595241c968e62d8b2f4942ba49a50"}]},"branch":"refs/heads/stable/2023.2"},"875b1f97e081d96bae01caa983939cd27546e304":{"kind":"REWORK","_number":2,"created":"2025-01-24 10:48:14.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/940090/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/940090/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/940090/2"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:34:35.000000000","tz":240},"committer":{"name":"Mikhail","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 10:48:14.000000000","tz":0},"subject":"Support \"-\" in region name for FIP\u0027s","message":"Support \"-\" in region name for FIP\u0027s\n\nAfter moving to raw strings for regular expressions region names with a\n\"-\" in name have been broken because raw string duplicates backslashes\nand it was two backslashes in inital regular experession.\nWith this changes extra back slash deleted what is correct to r\u0027\u0027\nstring.\n\nCloses-Bug: 2096623\nChange-Id: Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/875b1f97e081d96bae01caa983939cd27546e304"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/875b1f97e081d96bae01caa983939cd27546e304"}]},"branch":"refs/heads/stable/2023.2"},"446115ef06aa771d90b89a0004b217e7a714c4dd":{"kind":"REWORK","_number":3,"created":"2025-01-24 13:53:59.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/940090/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/940090/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/940090/3"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:34:35.000000000","tz":240},"committer":{"name":"Mikhail","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 13:53:59.000000000","tz":0},"subject":"Support \"-\" in region name for FIP\u0027s","message":"Support \"-\" in region name for FIP\u0027s\n\nAfter moving to raw strings for regular expressions region names with a\n\"-\" in name have been broken because raw string duplicates backslashes\nand it was two backslashes in inital regular experession.\nWith this changes extra back slash deleted what is correct to r\u0027\u0027\nstring.\n\nCloses-Bug: 2096623\nChange-Id: Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/446115ef06aa771d90b89a0004b217e7a714c4dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/446115ef06aa771d90b89a0004b217e7a714c4dd"}]},"branch":"refs/heads/stable/2023.2"},"c5978d38bebb1cfc32027d92f5e262673533ef06":{"kind":"REWORK","_number":4,"created":"2025-01-24 15:30:06.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/940090/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/940090/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/940090/4"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:34:35.000000000","tz":240},"committer":{"name":"Mikhail","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 15:30:06.000000000","tz":0},"subject":"Support \"-\" in region name for FIP\u0027s","message":"Support \"-\" in region name for FIP\u0027s\n\nAfter moving to raw strings for regular expressions region names with a\n\"-\" in name have been broken because raw string duplicates backslashes\nand it was two backslashes in inital regular experession.\nWith this changes extra back slash deleted what is correct to r\u0027\u0027\nstring.\n\nCloses-Bug: 2096623\nChange-Id: Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c5978d38bebb1cfc32027d92f5e262673533ef06"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c5978d38bebb1cfc32027d92f5e262673533ef06"}]},"branch":"refs/heads/stable/2023.2"},"8f028e4fa97ab007f0306cc02d6feb1b29c6558f":{"kind":"REWORK","_number":5,"created":"2025-01-25 00:47:22.000000000","uploader":{"_account_id":33685,"name":"Mikhail","display_name":"Mikhail Samoylov","email":"mikhailsamoiloff@gmail.com","username":"mikhails"},"ref":"refs/changes/90/940090/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/940090/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/940090/5"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:34:35.000000000","tz":240},"committer":{"name":"Mikhail","email":"mikhailsamoiloff@gmail.com","date":"2025-01-25 00:47:22.000000000","tz":0},"subject":"Support \"-\" in region name for FIP\u0027s","message":"Support \"-\" in region name for FIP\u0027s\n\nAfter moving to raw strings for regular expressions region names with a\n\"-\" in name have been broken because raw string duplicates backslashes\nand it was two backslashes in inital regular experession.\nWith this changes extra back slash deleted what is correct to r\u0027\u0027\nstring.\n\nCloses-Bug: 2096623\nChange-Id: Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f028e4fa97ab007f0306cc02d6feb1b29c6558f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8f028e4fa97ab007f0306cc02d6feb1b29c6558f"}]},"branch":"refs/heads/stable/2023.2"},"15794035db65a552b9d15d25a4c0e9f7766daf51":{"kind":"REWORK","_number":6,"created":"2025-04-10 17:35:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/90/940090/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/940090/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/940090/6"}}},"commit":{"parents":[{"commit":"2c817b3d7f01de44023f195c6e8de8853683a54a","subject":"Merge \"Add basic storage test to test config overrides\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c817b3d7f01de44023f195c6e8de8853683a54a"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:34:35.000000000","tz":240},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-10 17:34:33.000000000","tz":120},"subject":"Support \"-\" in region name for FIP\u0027s","message":"Support \"-\" in region name for FIP\u0027s\n\nAfter moving to raw strings for regular expressions region names with a\n\"-\" in name have been broken because raw string duplicates backslashes\nand it was two backslashes in inital regular experession.\nWith this changes extra back slash deleted what is correct to r\u0027\u0027\nstring.\n\nCloses-Bug: 2096623\nChange-Id: Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581\n(cherry picked from commit fed6500d226b3b85a89a7bf9c1d984f132c10caf)\n(cherry picked from commit 5490df4e6b60f00e31e0ff57f863e406728e4eef)\n(cherry picked from commit a33f7376407c57d86963b7f763c4ded7b4e16d3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15794035db65a552b9d15d25a4c0e9f7766daf51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15794035db65a552b9d15d25a4c0e9f7766daf51"}]},"branch":"refs/heads/stable/2023.2"},"eaecbaa8173a9c2efe1dccd2157928d2d0574133":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-16 15:57:09.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/90/940090/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/90/940090/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/90/940090/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/90/940090/7"}}},"commit":{"parents":[{"commit":"f0f9ad6c4dea2e67766574a8d8af7a771fb71bb0","subject":"Fix neutron job for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f0f9ad6c4dea2e67766574a8d8af7a771fb71bb0"}]}],"author":{"name":"mikhails","email":"mikhailsamoiloff@gmail.com","date":"2025-01-24 02:34:35.000000000","tz":240},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 15:56:49.000000000","tz":120},"subject":"Support \"-\" in region name for FIP\u0027s","message":"Support \"-\" in region name for FIP\u0027s\n\nAfter moving to raw strings for regular expressions region names with a\n\"-\" in name have been broken because raw string duplicates backslashes\nand it was two backslashes in inital regular experession.\nWith this changes extra back slash deleted what is correct to r\u0027\u0027\nstring.\n\nCloses-Bug: 2096623\nChange-Id: Ibdf8d93bdb8926b22c3ac978ccb72d795ad4c581\n(cherry picked from commit fed6500d226b3b85a89a7bf9c1d984f132c10caf)\n(cherry picked from commit 5490df4e6b60f00e31e0ff57f863e406728e4eef)\n(cherry picked from commit a33f7376407c57d86963b7f763c4ded7b4e16d3a)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eaecbaa8173a9c2efe1dccd2157928d2d0574133"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eaecbaa8173a9c2efe1dccd2157928d2d0574133"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
