)]}'
{"id":"openstack%2Fdesignate~940180","triplet_id":"openstack%2Fdesignate~master~I24dac5faf535e26d21b0cf1000a8707333f1a244","project":"openstack/designate","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"I24dac5faf535e26d21b0cf1000a8707333f1a244","subject":"Imported Translations from Zanata","status":"MERGED","created":"2025-01-25 04:06:51.000000000","updated":"2025-01-26 23:35:33.000000000","submitted":"2025-01-26 23:34:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940180-zanata/translations","meta_rev_id":"aaad62f8de9f1ad80c8a76db3f3a4a2ba4d2ec1d","_number":940180,"virtual_id_number":940180,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-01-26 23:34:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-01-26 22:42:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-01-25 13:33:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-01-26 22:42:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-25 05:14:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-01-25 13:31:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-01-25 13:31:54.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"CC"},{"updated":"2025-01-25 13:33:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-01-26 22:42:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d83baef21f62705b38e53df156b16d8e0b356a34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2025-01-25 04:06:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"88260a6bf142c28f2f8fb3002c4ee532ca6affc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-25 05:14:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b2764e6e81449f192a6235b800dfe29\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/10c5dd6ec567499c93e73f23d88f5a4d : SUCCESS in 41m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/3696daa870294091bc0c281febd28f37 : SUCCESS in 40m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cad0c8b3e55a468aa17265a8463c0ce9 : SUCCESS in 41m 18s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1bbff372cd894f2c93948cdf73c54a13 : SUCCESS in 47m 26s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8c4bf98d644044d191e8dfea06a8ee35 : SUCCESS in 35m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3cb415771ccb4526be2d9614ca27b9be : SUCCESS in 57m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78a445720414433cb8fad56fab1db7fb : SUCCESS in 58m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4c7afba7da6449e0921b4199e67897ce : SUCCESS in 1h 00m 16s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f1e6e9485e65492c8e86a563e86f9281 : SUCCESS in 1h 06m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3ba76f0bb3634c0296552b2016159de8 : SUCCESS in 42m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e8bd938f49af4fdbb2c4e424a44307f4 : SUCCESS in 46m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f5ad247a2384314a0aab5445b8014cc : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d82311d5a11448db47613160736f4d2 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c6465307b16e416abe9edbc3d286fce3 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f42a625758244d36a7667690df87d206 : SUCCESS in 6m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e92020ca6ed48b0aa5b9af652b57024 : SUCCESS in 7m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1da1eb691d4c48d09b8cb378765750a4 : SUCCESS in 4m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ae8c25574f1465a80099767d50b3e30 : SUCCESS in 45m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/977fc085811949808623f184ba04e2f7 : SUCCESS in 4m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9f30639fc46c41b4aee34abc4a7b19d2 : SUCCESS in 53m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"462c3781926ba0f90e726be38134006314af3850","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-01-25 13:33:41.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"72bee6e6a6d4cd83f7c1e36a729f4378329b55a9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-01-26 22:42:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4c0beaf94fcab7e45a3038725574722b2881310","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 22:42:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a76c1dcc091fe0ed5b4db8b97bb268f5eeeff7ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 23:34:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02808335c5f04ac69597bad4f9400a4d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d46f6341eff94de2b6506457cf8145ec : SUCCESS in 35m 24s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6729a098db754a2895cae71dc54ed22a : SUCCESS in 34m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/50c10f3c8a2d4ddd8b01cb3e904a3649 : SUCCESS in 36m 44s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ee2b9423f2f94028aa0496206ef92345 : SUCCESS in 51m 15s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1a6710c9e93d4fe8bfedadd2ee5c0aaf : SUCCESS in 49m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b196631fcf304451920e197b9e95f877 : SUCCESS in 33m 42s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/16ae349103fb40a2a77e3c43aa743fd7 : SUCCESS in 46m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fe75d9c66a949bf9c6f4b4410a637da : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/966fd17d75ea42a6a6e7ac88c9bf7679 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9929227c5ff84005a5da94e3a75ba99e : SUCCESS in 3m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4786e80343434eba8a66a511c36b9f57 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/36e3b12f1fb74f949cacd066c17abc1e : SUCCESS in 3m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/43b30dc32db14f25a2e1209d6dadbcea : SUCCESS in 33m 30s","accounts_in_message":[],"_revision_number":1},{"id":"5df9b1a413abf1f8a9c99786eda54fabb00d3d60","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 23:34:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"aaad62f8de9f1ad80c8a76db3f3a4a2ba4d2ec1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-26 23:35:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ddf813b1bc746e6bea674c67a864a12\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d2226e17f8e4ad197182f9d42aeb0be : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/107056df38ac40c195b89f1268cec34b : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1b562ec9a067f6dfc4c5019c3d6556bdea28792c","revisions":{"1b562ec9a067f6dfc4c5019c3d6556bdea28792c":{"kind":"REWORK","_number":1,"created":"2025-01-25 04:06:51.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/80/940180/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/940180/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/940180/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/940180/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/940180/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/940180/1"}}},"commit":{"parents":[{"commit":"d6e3f96cd844208b665ea720837b7069bd25a4a6","subject":"Merge \"doc: Use dnf instead of yum\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6e3f96cd844208b665ea720837b7069bd25a4a6"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-25 04:06:50.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2025-01-25 04:06:50.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: I24dac5faf535e26d21b0cf1000a8707333f1a244\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b562ec9a067f6dfc4c5019c3d6556bdea28792c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b562ec9a067f6dfc4c5019c3d6556bdea28792c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
