)]}'
{"id":"openstack%2Fdesignate~940675","triplet_id":"openstack%2Fdesignate~master~Ie8bebeffabe8de80b9fa9658d4b45928696085ba","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-02-05 07:05:51.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"Ie8bebeffabe8de80b9fa9658d4b45928696085ba","subject":"DNM: Testing designate tempest multipool","status":"ABANDONED","created":"2025-02-04 05:51:08.000000000","updated":"2025-02-20 00:53:44.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8059f49325497e709346957533db28f65b1e4a4f","_number":940675,"virtual_id_number":940675,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-05 08:02:58.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-04 07:11:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"82997e1c0b4bad27912915ea51d6ffc8de403990","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-04 05:51:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2455b886582a3087c18f988a8235ea6fdd76dd3d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-04 05:55:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f86ddb7e7425ffeaa7957608ce165250f849e3bb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-04 07:11:57.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2660bbb7b841404c846b044cb856dd7a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1669368cc4f34cb5abeea586efddb992 : SUCCESS in 52m 35s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d03f93a1f98f453ca13103784cfebed8 : SUCCESS in 51m 39s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fbc407b3ebd2420fbb8fbd313790d7af : SUCCESS in 1h 00m 31s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/e0e9aae8dcd64488a7a20758e9d4ff1b : FAILURE in 43m 39s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/063ed0a619e2419ab70ad2ada3d19943 : SUCCESS in 46m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/63bd83766f1b44dfbb4563a583edd19f : SUCCESS in 1h 02m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3753038fcb7044a69e1b2bc49c2593dc : SUCCESS in 1h 06m 00s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/691c896574924a37b0e1d0b9e6452b3a : SUCCESS in 1h 10m 24s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/990b742bcd084a10a7f594e39076351e : SUCCESS in 1h 03m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ec74742f24144917a393f474fa81114d : SUCCESS in 37m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/40f0971580444a61adf2177735764a40 : SUCCESS in 38m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ab53ea24c814390908c5396bb3fa44f : SUCCESS in 5m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c51b96c9be794fd98f4034088251ded8 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68ece3b329fd4c1ea9676a3992d652e7 : SUCCESS in 3m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ccf6bc9fd9b414ebf1e95a487f194b9 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6f65d6206bc47ee81b8befae653edda : SUCCESS in 7m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e0f029859744471ba5f019f95728ca6 : SUCCESS in 25m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ad8f145f07db411fa545138f3986fb43 : SUCCESS in 4m 39s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ce1bde638c184156889978bac7e32667 : SUCCESS in 52m 33s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f59a0844e3042fd897a52ef87ecd9aa0c28bbfe3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-05 07:05:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ceeaa08a9b3924781fdfff0ff6b9564584daaf48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 08:02:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9a922fbfb7e14c4e977e0e355da2ee0d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f234303fc12444d3b076318d2ba0ee24 : SUCCESS in 47m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/f76448ab198a4216a0271e670df8962b : SUCCESS in 45m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/195a8cb6b6b1451aa9afee9f76a97c21 : SUCCESS in 54m 24s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0a5beacc37a048a29c643a9a03c2e9f8 : SUCCESS in 39m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/49791125460c494d80c91502bfb8fa78 : SUCCESS in 45m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f41a41d727d540e48203b1981b4b0341 : SUCCESS in 50m 32s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0a2eb8ec4edf48968c370cfcbaa9a103 : SUCCESS in 50m 25s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/576c8b4e8eed4859b83e7da0d8a99a52 : SUCCESS in 49m 07s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/149cd0d6a9fd46359c17332eb90ad43d : SUCCESS in 52m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/20da76e6211e492fa55685179384a7e2 : SUCCESS in 45m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/072cb81b8b7a4e9493a1cf37868a3358 : SUCCESS in 45m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/49fddd9c5fce41d7a72c3acacfb92537 : SUCCESS in 6m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45055bb0bb1a4192bedcb758e76d053a : SUCCESS in 3m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75afe56682c540c093e682b9832b30df : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/483c315d34bc4eea95309fb7bd7dcd28 : SUCCESS in 4m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c38c3953891942f49e5fdf558138f4d9 : SUCCESS in 6m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/15d2a6095b9d44cfa13eb8e20e93e7ae : SUCCESS in 35m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4dd6853565a048fa80f4980a04589f7f : SUCCESS in 3m 29s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/195fc9f1795a49c99f291cf467ac807f : SUCCESS in 52m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8059f49325497e709346957533db28f65b1e4a4f","tag":"autogenerated:gerrit:abandon","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-20 00:53:44.000000000","message":"Abandoned\n\nTesting is done","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8e3ba800dc243a892b0b2561270b67bfdd72158d","revisions":{"1111ff5a00cb5f2ff3a655c80db97cf4c815ea1b":{"kind":"REWORK","_number":1,"created":"2025-02-04 05:51:08.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/75/940675/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/940675/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/940675/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/940675/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/940675/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/940675/1"}}},"commit":{"parents":[{"commit":"7ac4889a04f3d068932f16c5bb637df0466eaf76","subject":"Add pools.yaml attributes docs section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ac4889a04f3d068932f16c5bb637df0466eaf76"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-04 05:47:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-04 05:51:06.000000000","tz":0},"subject":"DNM: Testing designate tempest multipool","message":"DNM: Testing designate tempest multipool\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/920653\nChange-Id: Ie8bebeffabe8de80b9fa9658d4b45928696085ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1111ff5a00cb5f2ff3a655c80db97cf4c815ea1b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1111ff5a00cb5f2ff3a655c80db97cf4c815ea1b"}]},"branch":"refs/heads/master"},"8e3ba800dc243a892b0b2561270b67bfdd72158d":{"kind":"REWORK","_number":2,"created":"2025-02-04 05:55:06.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/75/940675/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/75/940675/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/940675/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/940675/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/75/940675/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/75/940675/2"}}},"commit":{"parents":[{"commit":"7ac4889a04f3d068932f16c5bb637df0466eaf76","subject":"Add pools.yaml attributes docs section","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7ac4889a04f3d068932f16c5bb637df0466eaf76"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-04 05:47:28.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-02-04 05:54:59.000000000","tz":0},"subject":"DNM: Testing designate tempest multipool","message":"DNM: Testing designate tempest multipool\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/920653\nChange-Id: Ie8bebeffabe8de80b9fa9658d4b45928696085ba\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e3ba800dc243a892b0b2561270b67bfdd72158d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8e3ba800dc243a892b0b2561270b67bfdd72158d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
