)]}'
{"id":"openstack%2Fdesignate~941574","triplet_id":"openstack%2Fdesignate~master~Ib0c8523e47c61638b3f579091a1955c96aa81a6b","project":"openstack/designate","branch":"master","topic":"bug/2098388","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-08-06 10:58:42.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-08-06 12:11:11.000000000","reason":"Change was submitted"},"32927":{"account":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"last_update":"2025-08-06 12:11:11.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-03-18 12:16:16.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ib0c8523e47c61638b3f579091a1955c96aa81a6b","subject":"Do AXFR with latest data","status":"MERGED","created":"2025-02-13 16:19:10.000000000","updated":"2025-08-06 12:12:11.000000000","submitted":"2025-08-06 12:11:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941574-bug/2098388","meta_rev_id":"0edae5ba2eb8f0792a2fdd3e954d661c3b0ec7ea","_number":941574,"virtual_id_number":941574,"owner":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-06 12:11:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-08-06 10:58:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-02 17:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-08-06 10:58:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-13 17:46:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 07:50:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-02-19 07:50:06.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-02-19 07:50:29.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-02-24 15:15:26.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-03-16 21:52:59.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REMOVED"}],"messages":[{"id":"690fed3d08fc7d4daa0ff2482311961a76013a9f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-02-13 16:19:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"76391fb6daff7ccac689ca8134766f05de973414","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-13 17:46:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1d3fe80ead142a49a7d19598feb65d4\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/65b33611d89d4d289e264057a8edd160 : SUCCESS in 46m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/01eae52944b14c56aade7fd723fde332 : SUCCESS in 51m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/12cf195c12a44dd98ec4fa8b2dfd3e26 : SUCCESS in 43m 15s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/23909433d1984dc18b23ba46367a9ffe : FAILURE in 42m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/89b28cc1b4c04643ae98eb25b0b9fee4 : SUCCESS in 48m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c41b734aec384e499c843c1a796687a7 : SUCCESS in 1h 02m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4e847f6270fb4bc3a61c50f0b33f678d : SUCCESS in 56m 28s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b6229d117cf242b6b8b00d493846ad92 : SUCCESS in 1h 21m 17s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/3599cf7260264f548254fd5f3ce2cd34 : SUCCESS in 1h 04m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f844734b89a1447abc84ccaddeff8e92 : SUCCESS in 52m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fd189dfde7b94155b07c8713bc64f2e9 : SUCCESS in 39m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c916c8b278e2406595a524a8d6943352 : SUCCESS in 7m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/db697124aa0c4315af8e9779491a814b : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c158378c2b74289830865640ba750f1 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/922320b7f8da409c85a7249da6169aea : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/498b80eff51843debe5bdc09eca1bab0 : SUCCESS in 6m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/055909ff9c124db186849ea7e1ad9a39 : SUCCESS in 45m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/119421067bce4ad6bd0bbf3cc0c8535e : SUCCESS in 5m 23s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/92afcc3330ac4d66a3ba774714a65687 : SUCCESS in 38m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1b9934baeb0392d8d90d3d6e1f85fbf1c6c4bf6d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-02-19 06:28:12.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d3d35348f2687f7fb7083c935953335dd10d33a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 07:42:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bceb2fe667c4d7f85a1b864a7ac4c92\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6f3ad1b6481446f18120a5811c9e5271 : SUCCESS in 33m 44s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1960d88dc4a5496a8b5a612780056af3 : SUCCESS in 46m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9eec061a862147c0856aaceb32aa4854 : SUCCESS in 41m 36s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2dddeb6c9eee4e47a7a4661de764b9fa : SUCCESS in 33m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fda8bfd972274efa96b09bd65590159b : SUCCESS in 38m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/60dd476dc7fc490caa9d1009ea00bf0f : SUCCESS in 54m 38s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a51a59df48414f0ba851999a0dbd124e : SUCCESS in 55m 41s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3afe293e66384226b7862e9f71f0e6d1 : SUCCESS in 56m 48s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4e705799014d4cd28a61b8a77c79dc52 : SUCCESS in 1h 08m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/33723c540e92478ab40de74a6e52b66b : SUCCESS in 51m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4fc1a25e45534e258930ed4ea9d86e01 : SUCCESS in 35m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94fff36d731144208e46c75df0138219 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c76ff9efd122494d97fce29d79d90af3 : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bed3156008a64022af4d57238e6e085a : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8adafdcf2990472a80ec73954ae66006 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15eaee5649004dd287372b56115a7f87 : SUCCESS in 7m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/206cd91d7b014bba913d3108c429caf3 : SUCCESS in 33m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e6389b33d16e46f794c2f9c89ad7cb79 : SUCCESS in 5m 02s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1316f758dffe4ed58cf4fa351dba0b63 : SUCCESS in 51m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8b702805fd9b85e5cc92804801f58a2150c7dbc3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-24 15:17:37.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2ee427f6ef98b2ffadccfb78bfc7fe279b86b793","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-02-24 19:44:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"ba6247c41f94fab72e786f86dac3139b0735eae7","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-02-24 19:46:03.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"19e0ed735f11ae7e29cdcad68958272afeb9f295","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 20:56:53.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0abf384709a346fc81e9f03df9729d5a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/275b392e1b6c4867860c313903409cf0 : SUCCESS in 53m 40s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fdc0262370544540bd48d79359cf7349 : SUCCESS in 1h 01m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/13174a8df00643e4b1ab973b43848aef : SUCCESS in 1h 00m 29s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/43a91b2d55224468a025f5ba2a55364b : SUCCESS in 49m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0cf4c122f6a54c6f9fa2bf9d16ce5111 : SUCCESS in 46m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/29480a1bbc324cf283fa60bdf1c92776 : SUCCESS in 1h 06m 41s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/abd371d2c76f4c2bb9f7ea6ada022479 : SUCCESS in 1h 01m 40s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b0bd610d4beb46f59656b3e726749d88 : SUCCESS in 59m 02s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fd81d9838f1d4e5e8bc0725b37db5853 : SUCCESS in 1h 00m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7bc01bcf4cce48b38b8815c1a0ed7e43 : SUCCESS in 52m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c10fc715b2004478b27531cbcb269aed : SUCCESS in 52m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e21358c61fd46a6a47998bc21c42173 : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1459856546644620bdac7ef369bcfbc4 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6310eb4c59d4beda10266ba4a01e273 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/27abf8c036ae423cbdbfa0b821d59499 : SUCCESS in 6m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c151232b2cc742d5bb20cdcbd5cf5838 : SUCCESS in 7m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6840764858814160b6979ef2454a6f6b : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/acdb2c374c2046f18a0168ada7c5b7a6 : SUCCESS in 41m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/de1bc617df1c44d2a27e96b68c518a09 : SUCCESS in 6m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b471c92c171740fa9efeb431e59138a3 : SUCCESS in 55m 14s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f65cdd22daebc917697b2e78025da46b6edac04e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-03-15 18:06:09.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"bfc58236df2b9ad2899c46ecd5b465f19d34c5d3","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-03-16 18:06:07.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1251f11d46a4c60de127223b006237ad5b18da8f","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-16 21:52:59.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_37598\u003e.","accounts_in_message":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"_revision_number":3},{"id":"2b91172096de6f8d73b7db3548a44751f38b209c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-03-17 12:52:45.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"70e9a75766f3c378e4749eda8ca07f97cb114c36","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-03-17 12:53:04.000000000","message":"Patch Set 5: Patch Set 4 was rebased","accounts_in_message":[],"_revision_number":5},{"id":"8d5f3915b959783efcb182dd87284e5665c332f6","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-03-17 12:54:24.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"624a70099d7aec8d5bda675905205a9a149737ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-17 14:16:38.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b76e777b8f6441748ca9576c4a3738d0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/51d02bba471240edaf81a98b5356202b : SUCCESS in 42m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/24f6205079e6442c892478666abb39b4 : SUCCESS in 52m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/86d6508af0144447aca5f35694ee9e54 : SUCCESS in 57m 49s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/ae293bf70339490392b4638b5a5f38df : SUCCESS in 40m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9ad252ec387d4d629ef4564b3bbc32de : SUCCESS in 49m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2923874e24fb4054b6d63e08d1cb18db : SUCCESS in 1h 16m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bc0da856c55748659d7e79d44b369c0e : SUCCESS in 55m 25s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4e35e87caa594ddf908da4f7d6123566 : SUCCESS in 1h 01m 30s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/59e737371894412d81ccc9ec6b85ac45 : SUCCESS in 56m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b02be2a4d728457d80c46198f53bfdfc : SUCCESS in 52m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c2f99fb7f91f4fe59ad3d91c35d36ced : SUCCESS in 55m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc611a82fbe943f689c734682f67180a : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b32c5e315d4e4a9daddbc96ace554d19 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54b300ba7f2a405fa33bfee5698aace4 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7d8f3f9240c145059b7ddb715bbd742c : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46f19c7b6e6849ad8892adc4e9125d86 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/49187ca80b7a40dc8c5843f12bd0c22a : SUCCESS in 4m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9eb345f7eac3450f8e464f1564008ee8 : SUCCESS in 37m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0002b044c63242b3bd9dbd27ae45f88b : SUCCESS in 2m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0f5c1486138743848320fbfe4f70ac52 : FAILURE in 42m 10s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1d10df477ab93fe1022ae96266ccb1a440698e15","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-03-18 12:16:16.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"31bdf862bd90df7233963798fc7eadf0ebbb70f0","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-24 09:53:29.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"53fe1f75018d6c2f9ab2803ffaa6c267dc6b90b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-05-22 19:26:20.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"a07930dae1e6abdde43baa30e9c9cd15c56bf3b9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 20:20:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e04deb22619345c39fb5b784566750ee\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e8c15af45ea546ab9adc5b644bf11078 : SUCCESS in 49m 20s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/978151083a0a41cc9f9aee264a27068b : SUCCESS in 39m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dd338e72192a4bf986a1ec8392f514cf : SUCCESS in 47m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/82f11de0b0b246f2ad8a93550c22a848 : FAILURE in 40m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/21b08ce392b146fdb8ace8305979e6eb : FAILURE in 42m 30s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9fa2adcce9bf4d8ab533fe56febe347a : FAILURE in 22m 02s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/51a2d7fa781c4f8e9ad6d827687e31d5 : FAILURE in 22m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3acfd5689ec444e7b768f1a94d5e7b4b : SUCCESS in 46m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/731db616aee24823ab676aa9bf99c480 : SUCCESS in 50m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9e9bbb569724c31a04d820784d5ccd4 : SUCCESS in 8m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3700f288e63d4c5d803f5e3e3b9d9256 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19aa601eb72a49258a579347a8014d2f : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/324c842bbb194597a4cf652cc1d1f148 : SUCCESS in 7m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7b68303e32c84d3ab2838ac270a6cbd5 : SUCCESS in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a23665d64f134c6c995fe1336f8eba3e : SUCCESS in 9m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3bfabcb763934b29a6554938535e80f4 : SUCCESS in 3m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ef6efdaba8844d79bde3dfefaab8a162 : SUCCESS in 50m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d6e43a12c5ea4df096ab38a4c1376101 : SUCCESS in 4m 39s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1225b1b8da874902bbbdd7896ce08eca : FAILURE in 43m 23s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8a2b481949a561df0bd4be84f4100d0a54de22b5","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-05-23 07:52:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e832e54afc96e8503af9b9818e8db85613f9a634","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 08:45:23.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/82e6789421ee4d83b2e2632ddc0ae36f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0d8328cc4e824a09a10c14b6211c1958 : SUCCESS in 43m 02s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/153e0eb87ee3401c87a94d6790bf773a : SUCCESS in 41m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ef77d4bf2fb5434f991120e1f32ff585 : SUCCESS in 45m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5b017bdb018845faa6aaee8261b33880 : FAILURE in 42m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6450bdd08e0d4a74b1b64c280bcb867b : FAILURE in 35m 35s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d74214cd6a534069841f1423c78be8ca : FAILURE in 17m 39s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/18016394679147daa91670e5facd60bf : FAILURE in 18m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2506c1bc9e794ebbbee4c75b2fca1084 : SUCCESS in 42m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a123298a92fb4f45af96ffe229012aa6 : SUCCESS in 42m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/572ac8f78504431e8f8b02ffb869df96 : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c33fa3497d84108acbb429e90f5f15e : SUCCESS in 4m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ca743c66f5e04ca685974c60fafb4afb : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cec49a99cb3842a19a1f303d8858c1d4 : SUCCESS in 4m 39s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/dd5853696e404380b457c5c515c7eeed : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b51cfea48b54296883598e2c8dcf874 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/157d6808f76941bba2efe1f828e83fe4 : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9cf0d86d86524049b5e14d04779af7e8 : SUCCESS in 32m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/45c30dacab9b4d69aafb94a537666f7f : SUCCESS in 4m 49s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/4292d76f4049475ea371707af58bce93 : SUCCESS in 48m 33s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"f300a5386f64097fae956641da2c809b6523c92e","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-05-26 12:54:00.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"66665d3a394f21346731ade8097176704a4dbcce","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-26 13:33:42.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"5b3642fdd038289a96599ceeb65900f85ef7303e","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-05-26 13:38:43.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"baafc038c168eb3bc2784a7b49ff477eba8ac93c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 13:49:22.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a1f6d95f367949579a1a47f018f9c3fb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4e82c8f1b51840adbe6d598e68296fa9 : SUCCESS in 47m 55s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/fafe95d198d34809918a565f27b98d21 : SUCCESS in 44m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a44631851696415eae4d8175c1cb0f2c : SUCCESS in 45m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/23f3c0e5931a4cdab9373c395c921d48 : FAILURE in 46m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/facab93cbdfa4c9cb165caa146a6b24d : FAILURE in 42m 10s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/fceec7171a0c4951b040d618ac74281e : FAILURE in 22m 17s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/12b0c13bd3c742b4936d3ee3a52f9709 : FAILURE in 21m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ea3e2fc1f1ae45f3a795cee733ff5941 : SUCCESS in 47m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/890f746ff22c4f68adcd8ee6579de4e2 : SUCCESS in 50m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/816308dcc3f345db9806fb2e698af1d9 : SUCCESS in 8m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/008fe464998a4c348729ba4e46ae8520 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dddef1643f5a476da3c6590f416ae2c6 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0dc867efc74343589cf6d86fc59f42a9 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4931ff9a078442b5afb1d15165a968b7 : SUCCESS in 7m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2628b6799a2e4c3fbd5f6c71bb440b2f : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7f333dacdda14b84aec845c69e5a369b : SUCCESS in 3m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7a4ac7e4e9140eaafadda907dfadcd1 : SUCCESS in 48m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c92f557a70ec4139849600c7aa0467ed : SUCCESS in 4m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/10a03061e76e4ea4bcd732699927c4f3 : FAILURE in 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"967d19d3735ed643aab5e6b3319e249060020816","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-06-02 17:29:03.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"5bc04eb3972a9541a7f8276b724f73e301b7a806","author":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"date":"2025-06-02 17:29:38.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"923c1c3d871569fbaff95c8af084e0800eb00bf5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 18:52:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/229a0f37909b46e6ae2152e435d81ccb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c7a935aca8eb4432b3b521f30ba3466f : SUCCESS in 38m 49s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/203ba276b63b47a08e9d3f3c1d120ba3 : SUCCESS in 44m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3a1d4ae1789c45de8b88d053c19bcd99 : SUCCESS in 41m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6d8c6899550049e6916c4a2204a662e4 : SUCCESS in 1h 02m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e659aea5ab104080b936a7e2a0a0412a : SUCCESS in 1h 08m 20s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/23934384c5bc45fbb75994ee186342c6 : SUCCESS in 1h 01m 15s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ba8f62f9d2bd403b94ff6ab8547c3a03 : SUCCESS in 1h 01m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c1f26b6d7abc4b2cbae3ce8d7e5bc6ee : SUCCESS in 42m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9df940d12e854a9386b68aadb123ad76 : SUCCESS in 46m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35c2f1d824364328820506bd8424515b : SUCCESS in 7m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b16ac432746495b8f57d68fe423b525 : SUCCESS in 3m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7b9311b9a35e4dd689e73cd24475ee7e : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b4a35e084c874e5db2aaf38f2ba5e38c : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a4319ba51f77401096bc2664db78dee2 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0e8529c8b8b42f3b1a168b5cc01f7db : SUCCESS in 8m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8e1ac482c794e31868dbfe53f637753 : SUCCESS in 3m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/db0f47504078494d93d64e1a6591fa55 : SUCCESS in 32m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bda4988d87dd4f3c85097e9d7f5ce19f : SUCCESS in 4m 45s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ad1c77635455494fb9cdc613c9609beb : FAILURE in 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"29ad5cf9e204ddba65557fceb97f99d192aa7def","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-06 10:58:42.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"049480e3b6aa8e7f46ab9bf57f3beab2e0ac4325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 10:59:09.000000000","message":"Patch Set 7: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":7},{"id":"3d5fa20f9ad7977b24e5d6199b64911b04977c37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 12:11:10.000000000","message":"Patch Set 7: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e5d42feb63b4719a64ede101ed04025\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/683e7626fbd246429c83a1a2cbf47beb : SUCCESS in 49m 30s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/ae7e62beb70048d4834afcd34c020b35 : SUCCESS in 40m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3cfff503f7834f6ca2ba36f7b212c9e5 : SUCCESS in 56m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f86607d283a14038a6de95272bb9f7c8 : SUCCESS in 1h 10m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ff9aba028f224311bacd5bc0e8b398a5 : SUCCESS in 45m 01s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/95ce88364fd8409aa831da4740408ac1 : SUCCESS in 47m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb2f0d69b49140ddac847d632237cb51 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dcb58281c19425b87881488f81dcbf6 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4553c3d88b044cc5852d25f8b1e1e947 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aca5ec8ac95648239634e74e1c110fb6 : SUCCESS in 10m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e98d3640fb548da849535c299db43bc : SUCCESS in 4m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/480c7384aad341d7af7a6b93ad8a051d : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":7},{"id":"acdd3370e9fd3559aafd8075324fbb4c312b27d8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 12:11:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":7},{"id":"0edae5ba2eb8f0792a2fdd3e954d661c3b0ec7ea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 12:12:11.000000000","message":"Patch Set 7:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1eb2f7a79da462e963888243596eb2c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dae32b2b0d454722adb4295d4b97ccc1 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/672a588627444ecf966a29d579c706f7 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"ac7cdcd95011eb044e0298ac7894a44b008e259d","revisions":{"25d8a48b039d85e1b5aaa230d2090c70c4916357":{"kind":"REWORK","_number":1,"created":"2025-02-13 16:19:10.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/74/941574/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/941574/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/941574/1"}}},"commit":{"parents":[{"commit":"fba08a7912718dc2971152132bc494eb1ae81cda","subject":"Merge \"More cleanup for the RBAC new default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fba08a7912718dc2971152132bc494eb1ae81cda"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:16:23.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:17:00.000000000","tz":180},"subject":"Do AXFR with latest data","message":"Do AXFR with latest data\n\nThe patch fixes 2 problems:\n\n1) Primary Designate update zone and notifies\nSecondary Designate simultaneously.\nThat\u0027s why Secondary Designate doesn\u0027t do AXFR\nor gets the previous version.\nNow Primary Designate updates zone and backend.\nIf backend is updated successfuly, then Designate notifies\nSecondary Designate.\n2) Secondary Designate does AXFR,\nbut zone is \"updated\" and \"pending\" and\nbackend is not updated.\nTask \"worker_periodic_recovery\" fixes it.\nNow Designate updates backend and sets right zone status\nafter AXFR.\n\nCloses-Bug: #2098388\nChange-Id: Ib0c8523e47c61638b3f579091a1955c96aa81a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/25d8a48b039d85e1b5aaa230d2090c70c4916357"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/25d8a48b039d85e1b5aaa230d2090c70c4916357"}]},"branch":"refs/heads/master"},"29b1165469418514b344aa73c827da04e0a7b91a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-19 06:28:12.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/74/941574/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/941574/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/941574/2"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:16:23.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2025-02-19 06:28:12.000000000","tz":0},"subject":"Do AXFR with latest data","message":"Do AXFR with latest data\n\nThe patch fixes 2 problems:\n\n1) Primary Designate update zone and notifies\nSecondary Designate simultaneously.\nThat\u0027s why Secondary Designate doesn\u0027t do AXFR\nor gets the previous version.\nNow Primary Designate updates zone and backend.\nIf backend is updated successfuly, then Designate notifies\nSecondary Designate.\n2) Secondary Designate does AXFR,\nbut zone is \"updated\" and \"pending\" and\nbackend is not updated.\nTask \"worker_periodic_recovery\" fixes it.\nNow Designate updates backend and sets right zone status\nafter AXFR.\n\nCloses-Bug: #2098388\nChange-Id: Ib0c8523e47c61638b3f579091a1955c96aa81a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29b1165469418514b344aa73c827da04e0a7b91a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/29b1165469418514b344aa73c827da04e0a7b91a"}]},"branch":"refs/heads/master","description":"Rebase"},"5a62497f3ce806a04fb34ba57c5c10266bc62683":{"kind":"REWORK","_number":3,"created":"2025-02-24 19:44:19.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/74/941574/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/941574/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/941574/3"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:16:23.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-24 19:41:18.000000000","tz":180},"subject":"Do AXFR with latest data","message":"Do AXFR with latest data\n\nThe patch fixes 2 problems:\n\n1) Primary Designate update zone and notifies\nSecondary Designate simultaneously.\nThat\u0027s why Secondary Designate doesn\u0027t do AXFR\nor gets the previous version.\nNow Primary Designate updates zone and backend.\nIf backend is updated successfuly, then Designate notifies\nSecondary Designate.\n2) Secondary Designate does AXFR,\nbut zone is \"updated\" and \"pending\" and\nbackend is not updated.\nTask \"worker_periodic_recovery\" fixes it.\nNow Designate updates backend and sets right zone status\nafter AXFR.\n\nCloses-Bug: #2098388\nChange-Id: Ib0c8523e47c61638b3f579091a1955c96aa81a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a62497f3ce806a04fb34ba57c5c10266bc62683"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5a62497f3ce806a04fb34ba57c5c10266bc62683"}]},"branch":"refs/heads/master"},"57c31dcf45026c829346f78e402040d5f6d49b85":{"kind":"REWORK","_number":4,"created":"2025-03-17 12:52:45.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/74/941574/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/941574/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/941574/4"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:16:23.000000000","tz":180},"committer":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-03-17 12:49:32.000000000","tz":180},"subject":"Do AXFR with latest data","message":"Do AXFR with latest data\n\nPrimary Designate updates zone and notifies\nSecondary Designate simultaneously.\nThat\u0027s why Secondary Designate doesn\u0027t do AXFR\nor gets the previous version.\nNow Primary Designate updates zone and backend.\nIf backend is updated successfuly, then Designate notifies\nSecondary Designate.\n\nCloses-Bug: #2098388\nChange-Id: Ib0c8523e47c61638b3f579091a1955c96aa81a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57c31dcf45026c829346f78e402040d5f6d49b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/57c31dcf45026c829346f78e402040d5f6d49b85"}]},"branch":"refs/heads/master"},"e7641aa20d3d2bfa0e569a3aa123e069e7567b12":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-03-17 12:53:04.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/74/941574/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/941574/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/941574/5"}}},"commit":{"parents":[{"commit":"a265c5070bd0f83d9352218f9de9475416a132b1","subject":"Fix periodic secondary refresh task","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a265c5070bd0f83d9352218f9de9475416a132b1"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:16:23.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2025-03-17 12:53:04.000000000","tz":0},"subject":"Do AXFR with latest data","message":"Do AXFR with latest data\n\nPrimary Designate updates zone and notifies\nSecondary Designate simultaneously.\nThat\u0027s why Secondary Designate doesn\u0027t do AXFR\nor gets the previous version.\nNow Primary Designate updates zone and backend.\nIf backend is updated successfuly, then Designate notifies\nSecondary Designate.\n\nCloses-Bug: #2098388\nChange-Id: Ib0c8523e47c61638b3f579091a1955c96aa81a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7641aa20d3d2bfa0e569a3aa123e069e7567b12"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e7641aa20d3d2bfa0e569a3aa123e069e7567b12"}]},"branch":"refs/heads/master","description":"Rebase"},"990938a58750f8beabbce3e3d8ea0484afbbb36e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-05-22 19:26:20.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/74/941574/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/941574/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/941574/6"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:16:23.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2025-05-22 19:26:20.000000000","tz":0},"subject":"Do AXFR with latest data","message":"Do AXFR with latest data\n\nPrimary Designate updates zone and notifies\nSecondary Designate simultaneously.\nThat\u0027s why Secondary Designate doesn\u0027t do AXFR\nor gets the previous version.\nNow Primary Designate updates zone and backend.\nIf backend is updated successfuly, then Designate notifies\nSecondary Designate.\n\nCloses-Bug: #2098388\nChange-Id: Ib0c8523e47c61638b3f579091a1955c96aa81a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/990938a58750f8beabbce3e3d8ea0484afbbb36e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/990938a58750f8beabbce3e3d8ea0484afbbb36e"}]},"branch":"refs/heads/master","description":"Rebase"},"ac7cdcd95011eb044e0298ac7894a44b008e259d":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-06-02 17:29:03.000000000","uploader":{"_account_id":32927,"name":"mitya-eremeev-2","display_name":"Mitya Eremeev","email":"mitossvyaz@mail.ru","username":"mitya-eremeev-2"},"ref":"refs/changes/74/941574/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/941574/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/941574/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/941574/7"}}},"commit":{"parents":[{"commit":"3484e178aa8b2c067b484d7e7c6be8f26cf57428","subject":"Merge \"Skip installation to speed up pep8\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3484e178aa8b2c067b484d7e7c6be8f26cf57428"}]}],"author":{"name":"Mitya_Eremeev","email":"mitossvyaz@mail.ru","date":"2025-02-13 16:16:23.000000000","tz":180},"committer":{"name":"mitya-eremeev-2","email":"mitossvyaz@mail.ru","date":"2025-06-02 17:29:03.000000000","tz":0},"subject":"Do AXFR with latest data","message":"Do AXFR with latest data\n\nPrimary Designate updates zone and notifies\nSecondary Designate simultaneously.\nThat\u0027s why Secondary Designate doesn\u0027t do AXFR\nor gets the previous version.\nNow Primary Designate updates zone and backend.\nIf backend is updated successfuly, then Designate notifies\nSecondary Designate.\n\nCloses-Bug: #2098388\nChange-Id: Ib0c8523e47c61638b3f579091a1955c96aa81a6b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac7cdcd95011eb044e0298ac7894a44b008e259d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ac7cdcd95011eb044e0298ac7894a44b008e259d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
