)]}'
{"id":"openstack%2Fdesignate~941791","triplet_id":"openstack%2Fdesignate~master~I8cfc2a3898ea2ca81d2ad84ce91a661ad44ee6f7","project":"openstack/designate","branch":"master","topic":"pep8-skip-install","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-06-01 15:34:53.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"37632":{"account":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"last_update":"2025-02-24 21:58:08.000000000","reason":"removed on reply"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-02-27 22:40:41.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-06-02 14:05:18.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-06-01 15:26:27.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I8cfc2a3898ea2ca81d2ad84ce91a661ad44ee6f7","subject":"Skip installation to speed up pep8","status":"MERGED","created":"2025-02-15 12:02:57.000000000","updated":"2025-06-02 16:56:17.000000000","submitted":"2025-06-02 16:55:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"941791-pep8-skip-install","meta_rev_id":"7103143e48eb941414abf642d929bd7de8930433","_number":941791,"virtual_id_number":941791,"owner":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 16:55:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"all":[{"value":1,"date":"2025-02-24 15:21:16.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2025-02-24 21:58:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":2,"date":"2025-02-27 22:40:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-01 15:26:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-01 15:26:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-15 13:25:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-24 15:21:16.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-02-24 15:23:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"state":"REVIEWER"},{"updated":"2025-02-24 15:23:55.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-02-27 22:56:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"9f0bf2fc8bd4e3a42f3f4826c1897f373be17608","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 12:02:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cdf0335f48520e2b21a3859e6d73dc506086011b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 12:04:51.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"dfeff052f4c5ecc41e07a0686f7b7daad4ff5941","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-15 12:04:59.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"6b63955ff766ec6691380909126587f1a67402a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-15 13:25:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2a62daf71294ab3be05024700173568\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1bd6e0881101409fb10a49c25a81e26c : SUCCESS in 2m 25s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9d349f99d96243fc8846069c0c97fc12 : SUCCESS in 35m 23s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/914b1dcf162448089d8a21df0454f2d2 : SUCCESS in 50m 33s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0dee36005c14d999a59c2aadf1620b0 : SUCCESS in 55m 00s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/ee535fb191ea40e5979b633252c646ff : SUCCESS in 37m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cb14c15facce4e6b99d63c867d33c599 : SUCCESS in 47m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/302f8b86a85549f887bf3f1668de68cc : SUCCESS in 1h 09m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1452c45996154da4ab6a832254f4625a : SUCCESS in 58m 40s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9b0d40d8d12040e58df57ba2af173fc7 : SUCCESS in 1h 11m 11s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/dac03988ccbe4c059070a865f3124b4e : SUCCESS in 1h 07m 38s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6d89256da4724db9b84ea01bec332342 : SUCCESS in 41m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e07da4d1ebf048eab9dc2bb6a55660bb : SUCCESS in 40m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf98580a80de4a3985676d16e1bf783d : SUCCESS in 7m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2733e043014442bd1aa6c539d8f2e9 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9c1f9c8b056048d4934ef42e7f02659a : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5130569b461746f89eff5962dc97fccd : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21e1c2408a95424ba8576b3dca47b6ec : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a51d6a1fe1f04efe99989f427546bd45 : SUCCESS in 3m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6047c1ea7ac245bab1490c25aff5271e : SUCCESS in 4m 08s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/38e07f250f5445b7b5002a48f5032243 : FAILURE in 50m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bb8fb34505b4ef211feacd017a9c4309d0981763","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-24 15:21:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b31e0a93d149b62d029676237c7ca4061520f829","author":{"_account_id":37632,"name":"Dmitriy Chubinidze","email":"dcu995@gmail.com","username":"chubinidzedr"},"date":"2025-02-24 21:58:08.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"82589d17de7e640bdc16082a0288f11156bc0c28","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-02-27 22:40:41.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6655276e8344c23664af79c3cfc6f7e6c61817f0","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-31 15:19:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"07b980150e9aef4dd218f2549d2f81cc58dcf093","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-01 15:26:27.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5cfee2769fd9e2ad709cee7fca87a2a3e3c5983c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-01 15:26:52.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ba0d08a47c0f868a4cc014951bd150a4cdcb17e9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-01 15:43:39.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/eccd0b5ade544e4cab565ff530b53260\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/92484838953d44bba96e64b25ed66371 : SUCCESS in 3m 44s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bd3f55f5fe8b4aa3b462a1ace0b57037 : CANCELED\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/3f10d5dd3eec4b7b81ae56064eaecde7 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4919cb2c2283499f9c2e87780572673e : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/65eca2030b9543228f7abde82b87570a : FAILURE in 15m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/79e6c8010e6a4ece93f3e9d629e5764c : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c91f65914a86406b8bc8e5457c2afed9 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47cb0a3dfd954a21bf5bf26715b48c52 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10ca4228b9c4424aa89c03fdf2aee29f : CANCELED\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0ce60f36cbb348339a3458d5194e4a6f : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce19c273a36b459890ac90b68293199b : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7595579fdb74cb882ccca76bd7b381a : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":2},{"id":"ea8db824db16bd46553a7bab7b4d2033a68446d3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-01 15:44:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"810ed311df2cf613e77161555ccadb2fb45553d4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-01 16:53:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b4b820fce3a43dc9f445ca0d5a108ab\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/33d3037f152049648696875c2fa0c44c : SUCCESS in 4m 17s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0c517cca202d4e64a875b7a2ed8dd9c7 : SUCCESS in 37m 24s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/054740814c9b428ea8d95a7f5e93d54c : SUCCESS in 33m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3cc2322311ed4a02b74f93f8c5f45bbc : SUCCESS in 35m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2b385d924da84b96bb5fb76686a3d54c : FAILURE in 40m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c0b603a2161347bca66646bc9ebc03bb : FAILURE in 40m 47s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/287ef0503ed54fec858f18cfd2eb63c1 : FAILURE in 20m 59s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4b4af840c2564a578d6165645a92681d : FAILURE in 24m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7e346dec739b4bcaac37229017163f59 : SUCCESS in 45m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8bc15645dd7b4b9c9bd4fd78d40cbed1 : SUCCESS in 46m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0cf43a74b7f64409bb7179e58e9b51f9 : SUCCESS in 7m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d4a9b57df904781aefbc34ef3f69d4f : SUCCESS in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8901077ca0604878aa76253d3af90cd1 : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dcc9448c08274bf288da808dca3475e2 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/73900180ddc1449082d69c576fbc7b73 : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13d5fac85a21482081aee9770da1e9e3 : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4597603635154ff68afa6d1e797aaf5d : SUCCESS in 4m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/730906c7bf7d48d3a3d8913161726b41 : SUCCESS in 4m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5fd0beee35da4fc399dc0c60f785f878 : SUCCESS in 46m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5ca5203057dd292d00d36fec176ef47b058ea48b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-01 17:07:12.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8064355cc389e9155ddc8892a74406030f687b49","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-02 14:05:18.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"afc86404d59d0c83e5b18b02f253176cb7114af2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 15:41:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c70d56a612d9488980e1ada94f82dc95\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/481eb3ed1d564390895760a34dfce443 : SUCCESS in 3m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ae6a0d6759414a479df3f263cee1a143 : SUCCESS in 44m 10s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/13d13ecc2b2c4ef580a28c6348c4c45c : SUCCESS in 51m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e46c9235bb4c4c2787d16f9fce3c653d : SUCCESS in 39m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9a84e41025be43c19a5c6259084c8185 : FAILURE in 45m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/45cebc33f2f04458b33e95156c863212 : FAILURE in 39m 45s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/70c17c712bcc45deba5a7d199d8d86fb : SUCCESS in 1h 04m 20s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e0db58e0fa514806b3dd89c0a24662a0 : SUCCESS in 1h 06m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9f01a0911ea0483cbf9fac0f55ef2695 : SUCCESS in 49m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c9faf65e71274a0d854639dbf2037fd9 : SUCCESS in 48m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86b5f0f658244655848eb53d3d487b47 : SUCCESS in 6m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e6a27ae19ddc4465b4d901829854549f : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1a5a0e131a044c9380939bf9801ce69d : SUCCESS in 4m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5eeee58535c4482ea3aabbfb80ded89e : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32ee659b4255447895c17fbb17432d55 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/208812b5b0dd4ff0b4d0b84cc5d8b3f9 : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a739f867a00a4935892c0b42e0851428 : SUCCESS in 3m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f394beac51f2430ea2f256245752799c : SUCCESS in 5m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3b78589aeafc4e1881e9b79843d1b27a : FAILURE in 45m 49s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8ff8935e648acb6bd0baba8f796525447179cabc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 15:42:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e9021480c4463aaeebb185587d949c03e5491fb7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 16:55:10.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f446d602493415fbfa9f11db58bd800\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/344e42f8de774046bf8263d2e8ccc000 : SUCCESS in 3m 57s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/df248d219cd744b481b4b901e988b422 : SUCCESS in 49m 13s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/909a1042e9dc462c8b564558b3b7a205 : SUCCESS in 42m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/598bb68473c64c82822cd9f6f59845f0 : SUCCESS in 41m 53s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/33e4888654384b6eb515b05514746248 : SUCCESS in 1h 05m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/16b44ae94c8b46559bc4aa43f57ef9d0 : SUCCESS in 44m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/20e5d4271eae4c11b257d6b11c684b22 : SUCCESS in 42m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/78e0ae52a9ab400eab09b49a2500109a : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d35ed5344c334cf4be6c61cefa370857 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e25e87ba2dd49c9aea509f4d6d41634 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/77d70d96e3854c27a0a2d9f20b911891 : SUCCESS in 7m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0eefaf7fdb4d460e81e3528c2c076da7 : SUCCESS in 4m 20s","accounts_in_message":[],"_revision_number":2},{"id":"cda038c9c15bf5499f2ba6b682ef7b7fdd692df8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 16:55:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7103143e48eb941414abf642d929bd7de8930433","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 16:56:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08d88ffe033c4914b5f3247a98be0a76\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/c94db540b42f4a368bee3c0d5c520f7a : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73640f6e5aa942c8bbf9cbf3cfb64683 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c360e5f28724ec9a54c57d3ef2d67e5 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"43784d2ac62c66244668744c5ba2d24ae5ba471d","revisions":{"8900c5020a0e2f6ff9fd9d244b32fe0495d127e7":{"kind":"REWORK","_number":1,"created":"2025-02-15 12:02:57.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/941791/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/941791/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/941791/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/941791/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/941791/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/941791/1"}}},"commit":{"parents":[{"commit":"fba08a7912718dc2971152132bc494eb1ae81cda","subject":"Merge \"More cleanup for the RBAC new default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fba08a7912718dc2971152132bc494eb1ae81cda"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-15 12:02:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-15 12:02:57.000000000","tz":0},"subject":"Skip installation to speed up pep8","message":"Skip installation to speed up pep8\n\nChange-Id: I8cfc2a3898ea2ca81d2ad84ce91a661ad44ee6f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8900c5020a0e2f6ff9fd9d244b32fe0495d127e7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8900c5020a0e2f6ff9fd9d244b32fe0495d127e7"}]},"branch":"refs/heads/master"},"43784d2ac62c66244668744c5ba2d24ae5ba471d":{"kind":"REWORK","_number":2,"created":"2025-02-15 12:04:51.000000000","uploader":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"ref":"refs/changes/91/941791/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/941791/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/941791/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/941791/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/941791/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/941791/2"}}},"commit":{"parents":[{"commit":"fba08a7912718dc2971152132bc494eb1ae81cda","subject":"Merge \"More cleanup for the RBAC new default\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fba08a7912718dc2971152132bc494eb1ae81cda"}]}],"author":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-15 12:02:57.000000000","tz":0},"committer":{"name":"Ivan Anfimov","email":"lazekteam@gmail.com","date":"2025-02-15 12:04:51.000000000","tz":0},"subject":"Skip installation to speed up pep8","message":"Skip installation to speed up pep8\n\nChange-Id: I8cfc2a3898ea2ca81d2ad84ce91a661ad44ee6f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43784d2ac62c66244668744c5ba2d24ae5ba471d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/43784d2ac62c66244668744c5ba2d24ae5ba471d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
