)]}'
{"id":"openstack%2Fdesignate~942576","triplet_id":"openstack%2Fdesignate~master~I467435fc07be00757269911b467478c2dd188350","project":"openstack/designate","branch":"master","topic":"fix_zone_move","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-03-18 16:13:46.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-02-27 10:13:53.000000000","reason":"\u003cGERRIT_ACCOUNT_13252\u003e replied on the change","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-03-18 17:39:21.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-03-16 08:52:52.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I467435fc07be00757269911b467478c2dd188350","subject":"Fix pool move zone target pool id","status":"MERGED","created":"2025-02-24 11:51:14.000000000","updated":"2026-03-18 17:40:31.000000000","submitted":"2026-03-18 17:39:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942576-fix_zone_move","meta_rev_id":"71ceff5b7a4d2ddd6d69d6221cb38dd70ea9b353","_number":942576,"virtual_id_number":942576,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-18 17:39:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-03-18 16:13:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-03-16 08:52:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-03-18 16:13:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-03-18 16:13:46.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-24 13:19:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-24 13:44:19.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-02-24 13:44:19.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-02-24 13:44:19.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-02-24 13:45:25.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-02-27 10:13:53.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REMOVED"},{"updated":"2025-02-27 10:17:51.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-04 15:03:11.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":28609,"name":"Arkady Shtempler","email":"ashtempl@redhat.com","username":"ashtempl"},"state":"CC"},{"updated":"2026-03-04 15:03:11.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":6681,"name":"Brent Eagles","email":"beagles@redhat.com","username":"beagles"},"state":"CC"}],"messages":[{"id":"61374bc582e070d47abd0cba0a77cbf7a42d4a13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-24 11:51:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"922012954054b9c34f3a1cb1c4e5218553825eaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-24 11:51:54.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"2b7c77f399867c174d9f267784a4a7c9d7cbfe1e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-24 13:19:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31b515602011414cb72d413187a32717\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32f766eb68a24dd4b50199c0cdb9090e : SUCCESS in 53m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ab00e1a3c822405898a8119f3cdb23f2 : SUCCESS in 1h 03m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/19bc84725c204dd3b557e5ed0c0d1007 : SUCCESS in 43m 56s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/840078bc22714940afc322efcc6e7923 : SUCCESS in 45m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fa0989ce07a14374808d116302b07fff : SUCCESS in 48m 49s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/cd364bfe43774abebe643582c8c22d5e : SUCCESS in 1h 18m 43s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/34cf5ba66845437c92d24b5ae3e8244f : SUCCESS in 56m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9b304118eb6d42b88ef36999bf1c91c1 : FAILURE in 17m 20s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/046967c5d0494497bdd4ef535a442748 : SUCCESS in 1h 18m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e20e6bd6e66445fdab0f6b26086900fe : SUCCESS in 47m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a8bb77f85c2415683ca9b4227b6a87c : SUCCESS in 52m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae6fbc0bd6624f309ea28994b23b8293 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f275ec47735d419189e3cd9b8467443e : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f27d83ca2dc242ed93a281d9f1f8d662 : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25bcc96c4a484b5e8b0d4c343c556366 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0dacf4be16e4d589126ae9bb441de30 : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73e4730555514af496a4399dfa579325 : SUCCESS in 3m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d614527254504b90b58415b0ec55b686 : SUCCESS in 44m 52s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/298a9c9dc2a04f4db2ac07504b3293f0 : SUCCESS in 4m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/7e9de74991df47d998506377d6fabcdb : SUCCESS in 48m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ed67cb6c10e48c27a35ef02dfb51f05fa2e57424","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-24 13:45:25.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"3681f571d133b65c9489f10113fce93f6599260e","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-02-27 10:13:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f9c75f6cfdad46ab3041eada2b67f6be5f5001bc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-27 10:17:51.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"46a80b62b52c1a2453419fd65af372e71ea090b8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-12 11:43:10.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"355171ad7bb95047e161f6469c6eeea9b6c8e912","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-12 11:43:28.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"960860ac432788585f33644eeb0b82d083b063da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-12 12:54:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77d2a9f1cf754eb7bb5825244ea9e907\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f3873f672bed4b57ae5b13cc0760f733 : SUCCESS in 1h 06m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/46fe0a32765140b9ab4f60f521b99763 : SUCCESS in 1h 02m 08s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2edc8e08ab5d4129aefec510b7a5222b : SUCCESS in 48m 24s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fa5d5028fac24e6caee341f6f6e77b56 : SUCCESS in 53m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee2f0ee8273f406286a734e01bba2505 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bedd0870a13e4c80af90f0f44ba62d7e : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dbe6f35fbaa4485c8ea2b11f49a85c35 : SUCCESS in 8m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fffc176265074552866b95dc88b4322e : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/836f9b976572484ab75181206e933858 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8be13fc81ab2401db7d19e1178a782f5 : SUCCESS in 5m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b002681128f149478fd6cbe32a3d13d0 : SUCCESS in 4m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e49076695824c3180bef1d3ef4ee19e : SUCCESS in 32m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0a89c0ab889643f0ba879ebcaddb331f : SUCCESS in 4m 21s","accounts_in_message":[],"_revision_number":3},{"id":"cb625288f7c89d33fb55f61d42df91a33f8e1dd9","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-03-16 08:52:52.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"2daeedf43f687f83676b2a96dc4c89eae16da270","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-03-18 16:13:46.000000000","message":"Patch Set 3: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"598dfdf3b2647c36769eacaa5e2ac0988b22dc33","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:14:15.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"7bdabcb586403345af656b83e2edf3352113eca2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:39:21.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f5c8f0169ffa4b3cb48ef611bb6d7e40\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/ce0645869c7e470bbb543bf4ae4d93b9 : SUCCESS in 55m 55s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f1e1ecbcf9f644b4896622e047e6673a : SUCCESS in 50m 41s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bb0ca6137e40452594cb5430fdecdb88 : SUCCESS in 45m 41s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/fffb57e95505469d83372c3ac50cb2df : SUCCESS in 27m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/75ed0f7c72834cea821fa176d7e2fc78 : SUCCESS in 1h 07m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2450ec1cccf4430995da753863869fa4 : SUCCESS in 1h 15m 49s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f90e4db14dc4437d89cd680faf274c4f : SUCCESS in 1h 13m 35s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9a98aa9ea0d04c46babf610e89a1387b : SUCCESS in 1h 02m 51s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/21487a6fd3fd436e83fb47cc369cfa35 : SUCCESS in 46m 52s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/7fe3fed859404a14acb5b4cc6dda5f41 : SUCCESS in 51m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a0379300c5a54d609f5bc64a6a3b0949 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3514d05fa95349c797fdb16cf1aea187 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0f8f020d18bf45c9a419374d7ebb9ee7 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dac8dff3385c4abba46aaf92d137adba : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4b8d060013684e53a978dec93f3454a1 : SUCCESS in 6m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6279ebd034f245bfa7161c3600cfc0fc : SUCCESS in 35m 51s","accounts_in_message":[],"_revision_number":3},{"id":"bc2eada355a9358e60a6eed8524767d7844e118a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:39:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"71ceff5b7a4d2ddd6d69d6221cb38dd70ea9b353","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 17:40:31.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2362ac3d9db542279ad04af13f32f087\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7eef7497e85a4cd29384c901c0452430 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/65c8e534852d418fae99debe9a134118 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7","revisions":{"e5b6e3974a0d96e9966eb093cbc7c858b2ccfcb0":{"kind":"REWORK","_number":1,"created":"2025-02-24 11:51:14.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/76/942576/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/942576/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/942576/1"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nChange-Id: I467435fc07be00757269911b467478c2dd188350\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5b6e3974a0d96e9966eb093cbc7c858b2ccfcb0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e5b6e3974a0d96e9966eb093cbc7c858b2ccfcb0"}]},"branch":"refs/heads/master"},"d33a1a7d2cd376d4e91a807701939fdac9b37071":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-02-24 11:51:54.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/76/942576/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/942576/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/942576/2"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:51:39.000000000","tz":60},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d33a1a7d2cd376d4e91a807701939fdac9b37071"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d33a1a7d2cd376d4e91a807701939fdac9b37071"}]},"branch":"refs/heads/master"},"7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":3,"created":"2025-09-12 11:43:28.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/76/942576/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/76/942576/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/76/942576/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/76/942576/3"}}},"commit":{"parents":[{"commit":"035453ffd1ba0f574a1d7f23b458222c0fd8fc68","subject":"Restructure Designate CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/035453ffd1ba0f574a1d7f23b458222c0fd8fc68"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-24 11:48:53.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-12 11:43:26.000000000","tz":0},"subject":"Fix pool move zone target pool id","message":"Fix pool move zone target pool id\n\nSo far, pool move zone was assigning target_pool_id after updating zone\nin storage, i.e. zone was not moved.\n\nThis patch fixes it.\n\nCloses-Bug: #2096866\nChange-Id: I467435fc07be00757269911b467478c2dd188350\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7fe4fba9a15805faf60abf8386e0ab9d93c8f0d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
