)]}'
{"id":"openstack%2Fdesignate~942773","triplet_id":"openstack%2Fdesignate~master~I082b3f5df4ce6733416b7edfacc61e90e8d926df","project":"openstack/designate","branch":"master","topic":"eventlet-removal-designate-executor","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-09-08 12:59:59.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-09-08 12:59:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I082b3f5df4ce6733416b7edfacc61e90e8d926df","subject":"DNM: changing designate eventlet executor to threading","status":"ABANDONED","created":"2025-02-26 10:11:54.000000000","updated":"2025-09-08 12:59:59.000000000","total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"be02cc52eaa378b3cfeee3cce91ed6eee41f7314","_number":942773,"virtual_id_number":942773,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-16 13:34:31.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 11:20:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-02 16:14:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-18 04:01:44.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-06-19 14:40:35.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"CC"}],"messages":[{"id":"00ad1f616aef0c5030401567eb618b4d8405b970","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:11:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f9092ed19c6801c74c684db0598d251cbebd89cc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:16:47.000000000","message":"Topic eventlet-removal removed","accounts_in_message":[],"_revision_number":1},{"id":"5496ea7b56e37c9165604dcc247f1077e0a8c5cc","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:17:05.000000000","message":"Topic set to eventlet-removal-designate-executor","accounts_in_message":[],"_revision_number":1},{"id":"fad089d64c32a3d2df69476a88b7560afc2a9a43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 11:20:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3325cf7fda54fae80af23794a1b540e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f9fec40d42ba4a8dbc0d434bda0c3da0 : SUCCESS in 44m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b9756d3d15d041ac8196cd0d18e4fc30 : SUCCESS in 57m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1aca6b42d2674967920069ce14feca2c : FAILURE in 52m 13s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d32fdd22c63e42b7b159d79564009260 : SUCCESS in 46m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be8a84ca549d4aaabebe0811685fa7e6 : SUCCESS in 39m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/613102c402774262a99c6d117670098e : SUCCESS in 57m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b12b51d222ed45f2b4331d12cd532edc : SUCCESS in 58m 11s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8a108b305f1b44d38ec437f513d3d7df : SUCCESS in 1h 03m 03s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/63bb7e0d398944b9a5b9256856407804 : SUCCESS in 52m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9996af4ad2594706a5991c54734deea9 : SUCCESS in 43m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a816ad39b864c8c850f5d879e9af718 : SUCCESS in 49m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fd967a12c07e4456b70811d1ba9034e5 : SUCCESS in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5244b9569fcb4d2a9c253986199f4b44 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b63b06e10654b429ea92b9494c69dfe : SUCCESS in 6m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/52ca407f79d649ffb23a1aa2640f8f09 : SUCCESS in 6m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/309fe7982fc04281a06b7bb2e4343bfc : SUCCESS in 6m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/03bf82f0752f44d78f966cf6e0ccefb7 : SUCCESS in 50m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8c024c673f4340d4b5ee9a4e3719a691 : SUCCESS in 4m 52s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d69487ec8ad94a829a10f69df6c34110 : SUCCESS in 42m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08ea4639b74334bf715b4ae7b8c9527f6cd47258","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-27 10:50:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"69ccc21269835472e2b87793df5ba49e2a7ecdfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-27 13:10:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef5ccfe20b114cfd90db27f7f01963ed\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8c75c372281f4127a7f98a06f355e9a1 : SUCCESS in 43m 45s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7a900ecca4ba4fbf96db818e2c3d44f9 : SUCCESS in 1h 00m 04s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ddf5882ec31540c78a5b09b77350353d : FAILURE in 51m 46s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/a3a6b2073e70492aa63a04b0285670cb : SUCCESS in 54m 41s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2bb91d3dfc6443e698ea8747826364d1 : SUCCESS in 48m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2b787a1014884dcc95562a108ace9988 : SUCCESS in 1h 01m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0a112195ce1d4fe0ad5a5f95b85c857c : FAILURE in 1h 06m 08s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/37becc97083a47f3bf0d47edd86e47f5 : SUCCESS in 1h 06m 49s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1525d81011e4463bb8d99d0371214594 : SUCCESS in 1h 12m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9e9d0b2760a14d00975cf306366b79d3 : FAILURE in 47m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d643e7d8db984b77b78bfefd7fc6385f : SUCCESS in 1h 42m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c146a20611b421487c9d75ffc9f83b5 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e02494fd303405b9afc778efc245a9f : SUCCESS in 8m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d0638d70bc81453e8b555a6621c786d6 : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4eb68158f96435894fbd717875946b5 : SUCCESS in 5m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/990ac682cc4442f2b03dd87ab6784092 : SUCCESS in 7m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7716a6f98cd4e6b86c6970b4d9ed000 : SUCCESS in 33m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6b09d7290199433c84c1540a73302f05 : SUCCESS in 6m 11s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3900305cfe394af2a186d8233cb88ae9 : SUCCESS in 46m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"86337cb65fcc4dc3368643c735eee7d5de50b80b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-03 11:35:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5c97c88593fa3ed5b0c9b0707197dcc2cf217c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 13:24:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/256ddda6194f4cb3a7bdb4e5150e3001\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ff72de5ad2e34eff99da3cca7f74cb5b : SUCCESS in 52m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/27e46f3be14742b0b965d8daba9ce3a6 : SUCCESS in 1h 03m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/31353783d2c9427c888b4e449b889db2 : SUCCESS in 1h 00m 40s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d11a8402bc7246a19f1b96bfe06a1bd5 : SUCCESS in 45m 00s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fe8135726c9141baabc6b4e2c4bc068a : SUCCESS in 38m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2e5715a1db2f4501b97d98cc9f210a98 : SUCCESS in 1h 14m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d23f7a47f9bd467180a7b4a39d9da3df : SUCCESS in 1h 01m 44s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1d82476bef544e9ba10a18673aed421b : SUCCESS in 1h 03m 45s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0dfb991963e14ef7b397cc55ced0a492 : SUCCESS in 1h 03m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dca6167dfc4547a99b64b025ddbb3624 : SUCCESS in 46m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/679d54ef6fac4cec85a8e3fb87ec5a0d : SUCCESS in 42m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb97a1104a242f99b98af515fb0b7ea : SUCCESS in 7m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcc76a485283490db8d4e4c7e28ef9d9 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d301f5970c9149459d4727eb6dedc3dc : SUCCESS in 5m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4df5e157fb24421ca00cde8e44bd0645 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ad6851a46f34b7ab7bae4b32a3a1947 : SUCCESS in 6m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/82547749067b4adfb2c7e662cea48d03 : SUCCESS in 46m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d9f6ce9c7ada4843aa8f4c2015c4f052 : SUCCESS in 5m 12s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c0375c11a8224d30b8204407ccfc156d : FAILURE in 57m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4eb1757b3c69c39fbf0e90f1e2199354f380f857","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-03 14:04:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f3ac374477dcb6d0d13579ee96631486fe81c001","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 15:57:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2db6c10b972148f79121693b9a668ea1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/117ad8ff954f49889b899e0961ade5c9 : SUCCESS in 1h 00m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/36fdb563e1fd4799b0f3f6dcfca7381a : TIMED_OUT in 1h 13m 32s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8e18e0011c63415abc3bd347bf51be59 : SUCCESS in 48m 47s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/b63851505e3b4753bbede193f89f6591 : SUCCESS in 36m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5156caed45534c23b421f9407fe46563 : SUCCESS in 44m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fda6ba2196bb4ead9bf2af66e5ad5ad1 : SUCCESS in 56m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eda72e0353df4771a873302925d2fc90 : SUCCESS in 58m 27s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/84de33b268274b5b9786e27530383fd9 : SUCCESS in 58m 32s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/61d83610e511443ba6cd50ff5db3824d : SUCCESS in 57m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0ac02088c85c46779bfeec0380c70474 : SUCCESS in 42m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/416d096da0494de9842192ea1864843c : SUCCESS in 38m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f157ac42fdd4eb3ae9049e811d7a041 : SUCCESS in 8m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/785963a457f5416291fd611704508628 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a73a7c610b14eeba7ff43838e03bd2b : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9677f1c5ac041feac079ca076850c9d : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c16fa7dd989648c6984746df2508ff91 : TIMED_OUT in 30m 37s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c95ce0f7cab4448b74958ccaf5b6b3e : SUCCESS in 41m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/77bdbed83ddd4d5886073fbc329cc437 : SUCCESS in 4m 50s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8f189a2848274b20a2dd38c10ccf8e2b : TIMED_OUT in 1h 12m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"324ec1f9d666c7cad021294e767e2086f04586c8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-01 11:42:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7a17ee92e164ba0b91c9066d0225054eec46b67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 12:53:44.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4bd9e204c9304ce3823ea9c945bf0b33\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/73977cecb58948d48a97d414fe4225be : SUCCESS in 48m 13s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fd18fd106bf84082b2925b6da995baf8 : SUCCESS in 51m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c3cd45a6e95f44c8a05bf1c1c38c7304 : SUCCESS in 41m 55s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/9a1a2468c400440482c814f0819104ea : SUCCESS in 36m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/63497084dc4a45708ed12cc40f103cc6 : SUCCESS in 47m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a5cfb738bcc249459d0bc1c1f72ba8f2 : FAILURE in 26m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/666abafcf5b243deb1d41bcfe1620cd3 : FAILURE in 24m 10s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8a8f0d67c4c1490fa22ba5933b4d0821 : SUCCESS in 51m 27s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8af23cb55b0c4efface59f30fb9cdd72 : SUCCESS in 1h 04m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2b6b00dd24cf469ab3df720d55f82d8a : SUCCESS in 42m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5dea2f1ff557453c9f4383a2df248e02 : SUCCESS in 43m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0db662639f6a4157a9f52a119d25b13c : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7cf37dbde35411fa945472b76d4d2d3 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3195038ee5104172b4c368e1acf58a6f : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8d5d33fa74a74663a32475de0bec3a24 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ec5dcb26f434fed898abf1d977e8382 : SUCCESS in 9m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e1543e04ea94b63a432ff9a8ce689c7 : SUCCESS in 31m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9741e450fa5b45928c46a173782b3eca : SUCCESS in 5m 15s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3265f88fe889471e8bd3bcd18110e889 : FAILURE in 39m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ca0df76c3c862eb59093c184a7e7147e3fa92c3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 08:24:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0bd77194b184664d971ca47a8ac1e5724b16e8e6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 10:17:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb75b4e774b43e0856e51846ea937cd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f33f64269b9f4015a5c6bd4f3202ea5c : SUCCESS in 46m 36s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/870af392b6c44c01b6749a2133f5d98c : SUCCESS in 56m 44s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c9b293d0b6dc42dd99d664c08e6f5a1f : SUCCESS in 48m 53s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/8db4a608849c400ca0b323df23fbc190 : SUCCESS in 46m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8e513531815c458cb6e2d352db3d9aa6 : SUCCESS in 41m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f83d6369370b48c0b81957a2349b1f99 : SUCCESS in 45m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7148b1efabb748bf8bc085fa60cb1813 : SUCCESS in 49m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3de1eeda8f9d454e943fa194cd2e1179 : SUCCESS in 1h 06m 16s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/894af1f3b2384ab8b2c96b2fbf6f1957 : SUCCESS in 53m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d00e39f15fcb42cb96e7a798afa20634 : SUCCESS in 41m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7b86395dfa354852a8ef7936cc5d6f10 : SUCCESS in 1h 41m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/746255672aed47c5aa176620feae713f : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa1841c19e7a4cebb5b80fac240c5b5b : SUCCESS in 4m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/83a5950e48e34b9a898ace8ab299fe90 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f7bb67cdd553497e9f435799de60fea3 : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a3d7899d7be400bbf0be5d18ae2f298 : SUCCESS in 8m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9fbbbf67b1da4add8f3b2368874579a6 : SUCCESS in 19m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9e6a0cf196d049278a50b51c22a53e11 : SUCCESS in 5m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/cbcbb42f58ba4045be6b48ce45513307 : SUCCESS in 46m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2042a12f9b293ef74ab64858171ebc9c93afbab7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 11:46:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4dd4f9e9dd625a4736769ef7e0a04fc577706cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 13:54:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f30e3300d24025953093491283bd5f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ff0239cf74ec41bf98b89603618b6d86 : SUCCESS in 38m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/03e6e98fb14649a198bd2381dc4ecaad : SUCCESS in 53m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0dd7bd7807a47adbc977326bf3b305a : SUCCESS in 50m 41s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e2b6a1b6c44d4cfe870926434fdadbf1 : SUCCESS in 40m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/20cacb77502d4674b2325fa7987bba89 : SUCCESS in 39m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/93682bf4f0dc42b492d0daa47ec6387b : SUCCESS in 1h 03m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e1430bc70b604124bba0c967a302df3e : SUCCESS in 59m 17s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2d9b15eb897c4773ba572769f41d6e36 : SUCCESS in 1h 02m 43s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e20aa78362ee473d8f771f1180047d0c : SUCCESS in 1h 01m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/580d6d131a8e479cb68e3d9e19e31b97 : SUCCESS in 43m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/91377164907443e9806bd09b31963644 : SUCCESS in 38m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff3e256e17024f9086faeb6008e34007 : SUCCESS in 3m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26e184b1f93c4af981b8943ddd3c467d : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf6a2c717aa44eeebde3104180085fcd : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0b0471e611b4438fb7d8de19fcd5ac06 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/45e4ff316fec407992fb24ba8b610e93 : SUCCESS in 11m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/638d124f8a76408cbf76b4c26b641e2e : SUCCESS in 32m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a49d7e21f1b54084a070dae9e21ccf26 : SUCCESS in 5m 27s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5ddf0e0daa7c47879e5674e582f8b41a : FAILURE in 44m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1bcc28780853e0d2ca6738411f50df25922a138b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 14:17:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"736904fd82fed3a75f144b384436ebf9667c8e82","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 16:23:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6b5006fe00034000a8846bc79d8b5c34\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ecde9378ea7648df8848dde7a45aee10 : SUCCESS in 44m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4a1b663cbfea4bc1910c2791cc11c4b7 : SUCCESS in 59m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d00dfc8ec2b24c4fad6eb0f1794f134c : SUCCESS in 44m 49s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/54888d9a66514dadac431eb3e032b1c5 : SUCCESS in 36m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/93dc187524d24e828d6ad02405a539c1 : SUCCESS in 45m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a23ea29187dc4981a396662b6ca30e40 : SUCCESS in 58m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f6fd68bc887342d290903d9a961293af : SUCCESS in 59m 54s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6e702931dbfd47659dfb0cc0ccfef86c : SUCCESS in 57m 36s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1b07ac4ed201417292fd3bbb4dac7b1b : SUCCESS in 1h 07m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/613b9939d8f34782bb9f8aec5aea83c4 : SUCCESS in 39m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d806ff54046b4501a301fad50ac5aab2 : SUCCESS in 41m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e18868875754f80aa419aa3f05ac505 : SUCCESS in 6m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa1ba950b8644e7bb3ea0513cfef7d8f : SUCCESS in 1m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/32001e7be66a4c9b8c9c787e03c08802 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44979614fbea426fabc31e352a8197cf : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/897540cc53c24dba9ca3ad79ec92f1e2 : SUCCESS in 9m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9fd6f24ff8de4b0f8aede47a0e574393 : SUCCESS in 36m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5ebe4376ebad4e718db3f1e582d4830e : SUCCESS in 5m 36s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8079350e65e247a583a7d63df6e3d289 : FAILURE in 40m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05d51c7c581dc66710f59ce6ccef36e205fd4f45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-26 16:12:56.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"5645242f717b0efc8cbb9b959f809f080809f02b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 17:00:03.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ed4e3d2bafb41f18aaf9457904f258b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9c0fedd8773f402189269e3249110596 : FAILURE in 19m 16s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c32d7d2da1d04e7cb83cd04405e3cfbf : FAILURE in 21m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c57d6bddb6344706af6673bc0ff51e5f : FAILURE in 22m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/24657bf82b9d43be98aafcef03d11067 : FAILURE in 41m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a10e12af6f44393a02b330a4b2c5d22 : FAILURE in 40m 50s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/be5d46b56d25486ea08972cd06e59e93 : FAILURE in 20m 42s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9940f4df02d346e9b8c4a7f7cae10549 : FAILURE in 21m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/94046201a546461e95ceb12b8af62b33 : FAILURE in 22m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/31aad49e5fa8456daeefd206a6447f29 : FAILURE in 20m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee65f62a039646b6aebc32533749ce6a : FAILURE in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ecc2e3a90bf642ffb225becf9090b51a : FAILURE in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5269031d03774b72a09778005fb206da : FAILURE in 3m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4ae8dcc743294dd6a6eaed77b2c81cdd : FAILURE in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/96c067510c474fdc9569561fdbccb7ef : FAILURE in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44814af59a3d4362a4124edebc3129fe : SUCCESS in 8m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f63d0d868cb741c4952640a38f063072 : SUCCESS in 2m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/032c47df4f5a46cea7d12fd6dac5eac4 : FAILURE in 28m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d0c5c8083de2450297e5a7826e63596a : FAILURE in 4m 06s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d9422467a1f84eb6b5043bba64005c28 : FAILURE in 21m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"751ee8d9db4b0497715e44dac9dfaf7f7f9ec50a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-26 17:13:56.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"dfbdb79bdfa1b94b868506fc2b559a5d0ff5b10d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-26 17:14:46.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"f482dd8c72df80803d37af1a5d3db106f73df091","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-26 17:19:57.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"b01f6e8c771d54502f984f348ba2599bb1099f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-26 18:03:40.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/86f66ca0a10242b0b453fd8ebe32d8dc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0d94617bce1a4da68998bf6c157890af : FAILURE in 16m 13s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/721b869e1c0a4cca80eb79a7144cdc24 : FAILURE in 14m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/26221fdeaffd451284680b34b12cd241 : FAILURE in 14m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0458e072d36c4ab5884d612b06ddc987 : FAILURE in 41m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/94f9a29ff6994215b2ac8352b0e00a0c : FAILURE in 29m 47s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/af98f2d86c1a41be85f15a8700558704 : FAILURE in 17m 55s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/91d10335341646f08b8527feff57e757 : FAILURE in 22m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/991fb9b63e3c48bb91f952668aaa52be : FAILURE in 22m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/61da170f52ce4169a3bd4c671eafe0d3 : FAILURE in 23m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2797297da09547ac90c2f42e7264ccda : FAILURE in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d920e38e16b454f89a9f251d0a599ae : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d26b7c01aca74f99b78c4e42d5c6b63a : FAILURE in 3m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7a01f361b89342059a40e695231dfd58 : FAILURE in 3m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/664b3d80d1b246a9b52ae55fcae144de : FAILURE in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f26f6bdef95a4bb1946cb36465d2c86b : SUCCESS in 9m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ef9efa8c5bc84595be3825826a3a1fd8 : SUCCESS in 1m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5c993fb254e7466180e138aa0134c216 : FAILURE in 25m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4b141fcb4ddc4b02b607daa36e0608f0 : FAILURE in 4m 52s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0cedc16effd840f0ab5b672da3b11050 : FAILURE in 17m 16s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"5d8927ae6d8af30a0ea1628a60ab28732a684aa8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-26 18:46:35.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"cdbc8e5110e425aca48314084040235127452a4b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 14:26:41.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"c5895e109a3dca5523be68c6bd3deae3c2ef2b11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 14:30:37.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2020aa786ae5abb6cc1d2ba803d11325fed46942","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 15:23:14.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"7bae35cefcd6dd701c74d659a21d3d3650474daa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 15:58:07.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5a8944c43f04bf7f2b7c8cc65038baae705ed466","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 16:05:00.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ac8d378183861cca520c8d5c159d9aba228c2817","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 17:07:04.000000000","message":"Patch Set 10: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/343400a3e89a4c59950f5abdada31974\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e49b31b055684ab1863790a3729ed927 : FAILURE in 17m 14s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/04ab1b4445c54e7b88d89f2d68f416bf : FAILURE in 23m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/acb14c0d177140eda6b5ef0ef439e91b : FAILURE in 19m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/46f02f1b83544f409cf5eec8386ea2cd : FAILURE in 41m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/54aaf295f1c642ed982b5dfce2f371e9 : FAILURE in 40m 25s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/50d4aa64bc1a4561890b8f7993e473c8 : FAILURE in 20m 18s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/02e59a10b01e4d3aa6c076d2e6231aa9 : FAILURE in 20m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/5bb210e62efc421da1fa267d4d78f423 : FAILURE in 27m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3aed393d37264dd6a83599a8377e6f00 : FAILURE in 22m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90af1631c97749bcb58778b780b18e01 : FAILURE in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594871dc5c5e469ab030294c6cc3fbba : FAILURE in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c072919b4984167a3658d5f9a93446b : FAILURE in 3m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b04e57d637e9452990462b791be437ed : FAILURE in 3m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6f9fa54b1d44824a631844d11b000a6 : FAILURE in 6m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0f3b95976b74b1c988d10fd813f914b : FAILURE in 6m 27s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a6f32a369ac34de9a8d6d1f6b0464ec5 : SUCCESS in 1m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/247421a29f8148bb89bc8dfb2233ce3d : FAILURE in 33m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b35a2cfaa256458d96ce97b68d247f31 : FAILURE in 3m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/063d7049063b4fb1a63ff27f9a701c6f : FAILURE in 24m 43s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"31ea1c5aa082f244abf02641313bfda14a3122c3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 10:09:06.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"1e6d002bf53130c6719209ba7ab413cdf80afdbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:33:18.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e6945916e6014ca7a8327e0bb2ebda8a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/914910cfec934ac6b128f7e36bf48af9 : FAILURE in 17m 20s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/8f3ad5a5cf8c43239d29be3fc7e8a0d5 : FAILURE in 16m 16s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4230bef6cb364fa1a604e7819bc6bdcc : FAILURE in 22m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e81c2d6b81264caa932773bcb26484a0 : FAILURE in 35m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e35214b755334739861205ffdc472511 : FAILURE in 36m 11s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/82cbbfdabd194c6d835b960827a78b9d : FAILURE in 17m 56s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a1fa555413ed4c3bb10f0c0c783e1ee6 : FAILURE in 17m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/03359c83275d47cd911a94f0cafb5334 : FAILURE in 19m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/071b1d2bcf874a38ab58765197455671 : FAILURE in 22m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9e62e97ce0ea4f66ab12fe834dfb6a28 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e083a189103047b98efbf71d31399e0a : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/96348bfd73d243dea77a82afa3fe3339 : FAILURE in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8cf1b7b4dd684d739f347ef6171ce747 : FAILURE in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2925b4de62604eefb88764984e74ce02 : FAILURE in 7m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/769e26a47e9342cba499ca779ce0d8a3 : FAILURE in 6m 22s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3300b22f5cbc458b82834b16e0919c51 : SUCCESS in 3m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd8fda6b76f64374bd6bd2e980d2a9a0 : FAILURE in 37m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/32b7718ea0524e38917ed50e8c609ee2 : FAILURE in 3m 28s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/eb2e9200de6f4902aac042c3febfc198 : FAILURE in 23m 02s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"05a4be5f4d2c0ac04fa52737a5934109f1626f41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 12:24:21.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"b3635783d53bf4ad3b0c717d131445a23aff0f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 13:14:48.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9047e555b05d43daa83f30ec07128c89\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2effc7b89d8e4af89e60fe6390965ced : FAILURE in 23m 16s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c3965b08a3834c68a467693bef8096d0 : FAILURE in 22m 04s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4f8c423e98a2490e9bc1985854ba92a3 : FAILURE in 23m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/67221b32664c45e2839b2985b0550f44 : FAILURE in 44m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5dcbc44f483e4fa693b8c5fdf4c00b6c : FAILURE in 41m 24s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1f99b28c44324ef68676b52083b3e9ae : FAILURE in 23m 44s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b378f36024e74b9b91d22dc3ccf83443 : FAILURE in 21m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f3ec02c5c5ac45029cafed9f7736768e : FAILURE in 19m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/88f3a66b7a5b4b2984d0a464a9a6efc3 : FAILURE in 24m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0fe0ec5ddb7f4a14a9d66a600a229416 : FAILURE in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a85fa9b4fdb4685ab91f9a9b8133a5f : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b609e937a2e441b0882a42930c3005b5 : FAILURE in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ba62e336706143ee8e417883e67d48df : FAILURE in 3m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aa46444bfe1748069f34e5d0bb938012 : FAILURE in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61f5394b6dcd4b88b19db18b48e44d80 : SUCCESS in 8m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2d560947500b4836a416950695e64853 : SUCCESS in 2m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e6626b5e103d477da1ce3aaf7067a4c0 : FAILURE in 30m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/84f49dc23b0148a9b49343c72d3feaa7 : FAILURE in 4m 12s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/18002e365ec242809277264b280ecf7d : FAILURE in 23m 05s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"75de5b50e01fa86ea7e360fcbcbe28d085f2b8d6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-02 16:14:01.000000000","message":"Patch Set 12: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ab6f02938df2bca9901c63ec797d955aeb85b2e2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 16:18:03.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"e153f1286fa1044e41920606ca4d37002e27a14f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 16:18:37.000000000","message":"Patch Set 13:\n\n(1 comment)","accounts_in_message":[],"_revision_number":13},{"id":"55c27a5f2240e69faabe227aa49e35d0c6b0b8b5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 16:20:00.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"c23be021d8c40854b304a4476bb098e63a57de84","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 17:41:30.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/92648dc48a0342f9962b960b0bba34d1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bae23ebde75f44efadace2ab692fa704 : FAILURE in 24m 57s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/02ed0ac773f942b8b1bd78011977a353 : FAILURE in 20m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/15a19925b5054afeb7bb06e17ba44fb9 : FAILURE in 21m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d7157105a2e64a26852003a40de8ad87 : FAILURE in 40m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/73941451ce0741988db3cd730ae552f2 : FAILURE in 43m 11s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0170601008d94d7d9c31e78cf64df946 : FAILURE in 42m 14s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6431cbab93764cce9c66c50aeb2a243e : FAILURE in 46m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/757a05e1f3a849648e80b9d65e3a8af8 : FAILURE in 21m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7048560c09084620989aea157b9e69a7 : FAILURE in 24m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be4a68b01068401ab3704db686db8f80 : FAILURE in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a4d78334591433ebfe14cda21492f57 : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79f09f101b9b4bbf8461b914a996103d : FAILURE in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/63b1090ebe0947b69726f94fa69f6ee9 : FAILURE in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a3206c5bbf634fdcb66fc9990766f32b : FAILURE in 6m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff2ead1430ea4c20957b36d2dff387ed : SUCCESS in 10m 18s\n- requirements-check https://zuul.opendev.org/t/openstack/build/626decc591c549c7aa1809757279ea8a : SUCCESS in 3m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/941539e0f22a424c93efc972441c686d : FAILURE in 31m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fbe7bd473f0342b886f12e20fc558124 : FAILURE in 5m 41s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f23d561a999c402b9f4830d1c1932a3a : FAILURE in 23m 39s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"22e98265183ddc50b7e28f1e049cd6bb85d47ff7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 14:52:11.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"71821c531af67b2c9999a7b047ecb3c2c3cbcea7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 16:33:01.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1dc4aeb157834e41a5a21fa769b2e3cd\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/01f3786b2c0e446581f698aae2b758ab : FAILURE in 24m 33s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/4a1573293fda45e395614dbc1adb04fa : FAILURE in 19m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3c48d4d772a248f68bacbde26512b4e0 : FAILURE in 21m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/36c6b53869584e5f97463389ae3fb054 : FAILURE in 43m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/228ebb0950084bfa9ba8c1793b11dbeb : FAILURE in 45m 00s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a63d25cecd8e4e38ba89c5f16bb2ddae : FAILURE in 40m 58s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c378142058734f33b42d2f2710972c99 : FAILURE in 44m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ae230246890c446a8e8c60e54efb4985 : FAILURE in 22m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f908602888884b72af60a103e76ae96a : FAILURE in 20m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d752f32788394cb18e244c1806a82b90 : FAILURE in 7m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d65e734fa6564c32852bc3035272c697 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b42310e5f7f843219b1b30cde5e3e5e7 : FAILURE in 6m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8ebbf6af9514fd8afe17c61360dc842 : FAILURE in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7af17ffb7ac24286b6d2456601cd9dea : FAILURE in 9m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271c2fb3731b423688f7f915bc3366d2 : SUCCESS in 9m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3baa90586b594c4c9ea9ca5949a615ba : SUCCESS in 2m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e68a7dc6101f49df8d9cb7783f807bfb : FAILURE in 35m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/885cea4a9440421c925cfc123fbdeeb8 : FAILURE in 5m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/7d9e793a4e0543038cdd4289b5eb8e0a : FAILURE in 22m 08s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"6e7e33896dd4e8355317a28595307016cf7dd52f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 16:55:20.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":16},{"id":"aaaf54e06f38211885a0c697015719904c1b1a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 18:11:02.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd4b5473316f48999a8d844c8b85434b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3275a072cd754146a34ea9f7fc4d6e09 : FAILURE in 22m 13s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/080b95c0712f489d8a84c572e7f9d094 : FAILURE in 20m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/6f3fcb7fc8a94964858ce127b95f3a97 : FAILURE in 21m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c82a70a2c26241c8a776a1ea297f1a78 : FAILURE in 47m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0e4d4727134c4d198c1ae4592c1a314d : FAILURE in 40m 14s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/dbee59fc0111416d8f0a31f864852c71 : FAILURE in 46m 25s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8ec7f800aa9942ca92d3f6ca756ec26c : FAILURE in 39m 28s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/92cf79fe00b244d1893ce3cbbc4a2f44 : FAILURE in 24m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bf5f52d165bd43b4b87fafc42f9cd537 : FAILURE in 21m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4603af1e87c34cea90210dcb43aa2891 : FAILURE in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4525e13d3acd4294a4c77796f2890617 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86858d8f56304d94ae889292fb27365a : FAILURE in 5m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/90b1cc9cc4cb44a28ae351fec97d052d : FAILURE in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5ed5b78a8c334e9e9b9a6f52cd5a3117 : FAILURE in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff3af315f4fa413496f84ad99decfeb9 : SUCCESS in 7m 33s\n- requirements-check https://zuul.opendev.org/t/openstack/build/51e68525fa2e4a6b9e7f31a7820b9286 : SUCCESS in 2m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1bf55e108ec74c70bad29275b2c2c17a : FAILURE in 28m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/80b7339ba2f6464292465729b349bc77 : FAILURE in 5m 05s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/efa2928dc4f5431eb27a971c5a33f272 : FAILURE in 24m 12s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"a53ebc42493132b2fe55f9c08d27a436cbfbf16d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-16 12:43:28.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"c6f87f92ef487b1736b145f32fe560a2a4ae3ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 13:34:31.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1e3c85b256094cedb3f73c803c1384f8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/771471dddc6d4d4993161b735ba4ebc9 : FAILURE in 18m 58s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/385e9860d3cf4a0ebb4ad7aa8b86cb13 : FAILURE in 23m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/10426f8df1524076bf4b1b92a8ac4cac : FAILURE in 23m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fc8d4f8a1ed24f2ca5b8ba23af2fdafb : FAILURE in 41m 50s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/43e4120bca9a4f72ae683b0d83267e78 : FAILURE in 43m 29s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0c7181ec10104843bb17cbecb3822ee9 : FAILURE in 40m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8dc9b9b789004e3596d44dfbf480a1c5 : FAILURE in 46m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cb387e416a104d08a52d4a931e263282 : FAILURE in 22m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/242a91b2749c4aec9aa70f9061b39530 : FAILURE in 10m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dcee31047d0a47dcacbc9d8ae501391f : FAILURE in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf59a02679864f0bab86a8239b1d668f : FAILURE in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d544cc88f4b24ba3b1cd24c8ab30b86b : FAILURE in 8m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e78270057df4ba1af48d17887340471 : FAILURE in 7m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/65fed1e6009741b79ee176de33f8d14e : FAILURE in 12m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/651a65c72d2140b096ca7b5e60a9fac4 : SUCCESS in 8m 59s\n- requirements-check https://zuul.opendev.org/t/openstack/build/96d49049b2ca49b686a0c5c19abba601 : SUCCESS in 3m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1b749d450e540e7ac3e5783b304ebb7 : FAILURE in 33m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7d0a9c23f9e94e51a2e80668bfcdd70d : FAILURE in 8m 32s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/77f076fe5f6e4bd381506ba62a75f783 : FAILURE in 24m 20s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"e7f5e53ea2f712c7ce4664a0706d585004ea1722","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-06-18 04:01:44.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"be02cc52eaa378b3cfeee3cce91ed6eee41f7314","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-08 12:59:59.000000000","message":"Abandoned\n\nThis change is obsolete and not needed anymore","accounts_in_message":[],"_revision_number":17}],"current_revision_number":17,"current_revision":"fdff50db53ebe4d87a860776bd3db9b450f07e59","revisions":{"ae9d8aee0790e0252a30ae1f2f7a455f7a2741f7":{"kind":"REWORK","_number":1,"created":"2025-02-26 10:11:54.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/1"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae9d8aee0790e0252a30ae1f2f7a455f7a2741f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ae9d8aee0790e0252a30ae1f2f7a455f7a2741f7"}]},"branch":"refs/heads/master"},"1be9122bf9ae0095e215262ecb1623ef022f3410":{"kind":"REWORK","_number":2,"created":"2025-05-26 16:12:56.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/2"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-26 16:12:43.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1be9122bf9ae0095e215262ecb1623ef022f3410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1be9122bf9ae0095e215262ecb1623ef022f3410"}]},"branch":"refs/heads/master"},"d650434bc97ce2f95e4d25c0ec5a66fba6793680":{"kind":"REWORK","_number":3,"created":"2025-05-26 17:13:56.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/3"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-26 17:13:49.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d650434bc97ce2f95e4d25c0ec5a66fba6793680"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d650434bc97ce2f95e4d25c0ec5a66fba6793680"}]},"branch":"refs/heads/master"},"ff424213b2da3fb273d3b57808b28b1704541a76":{"kind":"REWORK","_number":4,"created":"2025-05-26 17:14:46.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/4"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-26 17:14:39.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff424213b2da3fb273d3b57808b28b1704541a76"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff424213b2da3fb273d3b57808b28b1704541a76"}]},"branch":"refs/heads/master"},"163e9290232fb4541c235ad4d5379ffc7e443c84":{"kind":"REWORK","_number":5,"created":"2025-05-26 17:19:57.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/5"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-26 17:19:49.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/163e9290232fb4541c235ad4d5379ffc7e443c84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/163e9290232fb4541c235ad4d5379ffc7e443c84"}]},"branch":"refs/heads/master"},"1fcb0f419bdcf865c23fcbe365330c87757bc153":{"kind":"REWORK","_number":6,"created":"2025-05-30 14:26:41.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/6"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 14:26:34.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fcb0f419bdcf865c23fcbe365330c87757bc153"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1fcb0f419bdcf865c23fcbe365330c87757bc153"}]},"branch":"refs/heads/master"},"fcf3c874d205d1542483e9ef879c3401014ecb70":{"kind":"REWORK","_number":7,"created":"2025-05-30 14:30:37.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/7"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 14:30:30.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcf3c874d205d1542483e9ef879c3401014ecb70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fcf3c874d205d1542483e9ef879c3401014ecb70"}]},"branch":"refs/heads/master"},"2b007526795776e169669ba5b83f2f980a4c54eb":{"kind":"REWORK","_number":8,"created":"2025-05-30 15:23:14.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/8"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 15:23:07.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b007526795776e169669ba5b83f2f980a4c54eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b007526795776e169669ba5b83f2f980a4c54eb"}]},"branch":"refs/heads/master"},"e543cbe4ea3d68c9372bf40e032fa8b8810251ab":{"kind":"REWORK","_number":9,"created":"2025-05-30 15:58:07.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/9"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 15:58:00.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e543cbe4ea3d68c9372bf40e032fa8b8810251ab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e543cbe4ea3d68c9372bf40e032fa8b8810251ab"}]},"branch":"refs/heads/master"},"a3d9ca3b985ee76c7b4f0f39a3d4d9a543816536":{"kind":"REWORK","_number":10,"created":"2025-05-30 16:05:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/10"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 16:04:54.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3d9ca3b985ee76c7b4f0f39a3d4d9a543816536"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a3d9ca3b985ee76c7b4f0f39a3d4d9a543816536"}]},"branch":"refs/heads/master"},"dda73c9e9596935c61375392c48d692025fa449f":{"kind":"REWORK","_number":11,"created":"2025-06-02 10:09:06.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/11"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-02 10:08:59.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dda73c9e9596935c61375392c48d692025fa449f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dda73c9e9596935c61375392c48d692025fa449f"}]},"branch":"refs/heads/master"},"ecf226669a6f3fe916c2d9f7252eb2a051c9dfb1":{"kind":"REWORK","_number":12,"created":"2025-06-02 12:24:21.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/12"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-12 01:26:52.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ecf226669a6f3fe916c2d9f7252eb2a051c9dfb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ecf226669a6f3fe916c2d9f7252eb2a051c9dfb1"}]},"branch":"refs/heads/master"},"4413ef183838f8b03dbcab88d9a11e501daec095":{"kind":"REWORK","_number":13,"created":"2025-06-02 16:18:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/13"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-02 16:17:55.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4413ef183838f8b03dbcab88d9a11e501daec095"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4413ef183838f8b03dbcab88d9a11e501daec095"}]},"branch":"refs/heads/master"},"65096cb83b0f4951dd050f54124ea333024985e8":{"kind":"REWORK","_number":14,"created":"2025-06-02 16:20:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/14"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-02 16:19:53.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65096cb83b0f4951dd050f54124ea333024985e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/65096cb83b0f4951dd050f54124ea333024985e8"}]},"branch":"refs/heads/master"},"1250f96fb05f29eede8ba700a05ff99640df95ed":{"kind":"REWORK","_number":15,"created":"2025-06-03 14:52:11.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/15"}}},"commit":{"parents":[{"commit":"a395fd3ede778f139c3588958473b4c07c98593f","subject":"Merge \"docs: add OpenStack-Ansible as an installation method \"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a395fd3ede778f139c3588958473b4c07c98593f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-03 14:51:58.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1250f96fb05f29eede8ba700a05ff99640df95ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1250f96fb05f29eede8ba700a05ff99640df95ed"}]},"branch":"refs/heads/master"},"ff37f5daf315c793eafbbb50132fd60da6b29694":{"kind":"REWORK","_number":16,"created":"2025-06-03 16:55:20.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/16"}}},"commit":{"parents":[{"commit":"82652559ea8641b11c4227f0e9cf273b4c567b96","subject":"Merge \"Drop explicit executor argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82652559ea8641b11c4227f0e9cf273b4c567b96"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-03 16:55:10.000000000","tz":120},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff37f5daf315c793eafbbb50132fd60da6b29694"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff37f5daf315c793eafbbb50132fd60da6b29694"}]},"branch":"refs/heads/master"},"fdff50db53ebe4d87a860776bd3db9b450f07e59":{"kind":"REWORK","_number":17,"created":"2025-06-16 12:43:28.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/73/942773/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/73/942773/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/73/942773/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/73/942773/17"}}},"commit":{"parents":[{"commit":"82652559ea8641b11c4227f0e9cf273b4c567b96","subject":"Merge \"Drop explicit executor argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82652559ea8641b11c4227f0e9cf273b4c567b96"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:11:21.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-16 12:43:26.000000000","tz":0},"subject":"DNM: changing designate eventlet executor to threading","message":"DNM: changing designate eventlet executor to threading\n\nChange-Id: I082b3f5df4ce6733416b7edfacc61e90e8d926df\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fdff50db53ebe4d87a860776bd3db9b450f07e59"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fdff50db53ebe4d87a860776bd3db9b450f07e59"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
