)]}'
{"id":"openstack%2Fdesignate~942774","triplet_id":"openstack%2Fdesignate~master~I33dad118e506c79fa659f0aa9d5761cbe7eefb5f","project":"openstack/designate","branch":"master","topic":"eventlet-removal-designate-worker","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-02-26 13:14:46.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I33dad118e506c79fa659f0aa9d5761cbe7eefb5f","subject":"Remove eventlet from designate worker","status":"ABANDONED","created":"2025-02-26 10:21:08.000000000","updated":"2025-02-26 16:28:41.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"75b87a660b829660d7d1da26a6b45a94d48f93dd","_number":942774,"virtual_id_number":942774,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-26 11:40:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 10:36:18.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2025-02-26 11:40:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bb71892be9a6b4a7935b2f8594da7a0dbf759b41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:21:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6f4bba7d3be5f2d0fddac62fed943fbd349b17f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:22:48.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dccadc3d8fde1b25ba89811a7678fc5db6f51efb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:25:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4c3b2e27aca698f01afeb088f8e01724c553857","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-02-26 10:36:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6ec61ee8c81ef8a2c405aa93eda9c0ad438efaea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 11:40:24.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24101c7d8cdc4a3ba15cef9f26c98a1b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/db9391baa124478ba81575f5c981f1a0 : SUCCESS in 52m 27s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5c8c83da97c946cab16e192fd6e8b08f : SUCCESS in 1h 09m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d931b26254104e51bd306683f5f5b929 : SUCCESS in 58m 59s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/9de3a405b2194d1e95093939d1627728 : SUCCESS in 46m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ddbe8e17d59f4fb8923bbdfec14b02dd : SUCCESS in 49m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04aff59f0e47457bbfd949a1b7332309 : SUCCESS in 1h 01m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7ae6c185c1de4d8f802aa6ee3e885329 : SUCCESS in 1h 08m 22s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/21bdb7fe6f8b40f5954c67a1bc635e17 : SUCCESS in 1h 04m 29s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/67c89c4eed5649a2a72da2f5edd949cb : SUCCESS in 59m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3ca0eb9d805c40068043d9b854ea076a : SUCCESS in 52m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/abd69665047948ff8ff0983efcb2a630 : SUCCESS in 50m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd1df958e3f34c83899e689d7fbd23fe : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b43315bbf3a4cccb5c06d4914446b51 : SUCCESS in 3m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b8178dea52bd46f4b29a96406ac7bc71 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/160d47ae89054b5da8e838e7ba3ac8bf : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef75016f9c448999211defcc2adfe34 : SUCCESS in 4m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2944fa893b444bc185d0f1169fe9363c : SUCCESS in 34m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/600bdc81bc3049bfb5f9561081554871 : SUCCESS in 6m 51s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2d7f914db9ac44c099d02cc380f39a11 : SUCCESS in 48m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"98842ab580cf1424ef722844a12ccee01437fcda","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-02-26 12:49:58.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e6d6b64336e57b13af3806352b0fa3a2d64eba14","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 13:14:46.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"75b87a660b829660d7d1da26a6b45a94d48f93dd","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 16:28:41.000000000","message":"Abandoned\n\nI squashed both patches into https://review.opendev.org/c/openstack/designate/+/942777","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"c3d38dda8aae058fc46da2c809aed654a06be0b0","revisions":{"a2a4df2237de7b5186c7896ca121128d71e415e3":{"kind":"REWORK","_number":1,"created":"2025-02-26 10:21:08.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/942774/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/942774/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/942774/1"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:52.000000000","tz":60},"subject":"Remove eventlet from designate worker","message":"Remove eventlet from designate worker\n\nReplace usage of futurist.GreenThreadPoolExecutor due to its eventlet\nusage. Designate used it because of eventlet + py37 issues:\nhttps://bugs.launchpad.net/bugs/1782647\n\nChange-Id: I33dad118e506c79fa659f0aa9d5761cbe7eefb5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2a4df2237de7b5186c7896ca121128d71e415e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a2a4df2237de7b5186c7896ca121128d71e415e3"}]},"branch":"refs/heads/master"},"7eadae59a81bc005145ada0271d1f9f97c7bd9bb":{"kind":"REWORK","_number":2,"created":"2025-02-26 10:22:48.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/942774/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/942774/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/942774/2"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:22:42.000000000","tz":60},"subject":"Remove eventlet from designate worker","message":"Remove eventlet from designate worker\n\nReplace usage of futurist.GreenThreadPoolExecutor due to its eventlet\nusage. Designate used it because of eventlet + py37 issues:\nhttps://bugs.launchpad.net/bugs/1782647\n\nChange-Id: I33dad118e506c79fa659f0aa9d5761cbe7eefb5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7eadae59a81bc005145ada0271d1f9f97c7bd9bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7eadae59a81bc005145ada0271d1f9f97c7bd9bb"}]},"branch":"refs/heads/master"},"c3d38dda8aae058fc46da2c809aed654a06be0b0":{"kind":"REWORK","_number":3,"created":"2025-02-26 10:25:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/74/942774/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/74/942774/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/74/942774/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/74/942774/3"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:25:44.000000000","tz":60},"subject":"Remove eventlet from designate worker","message":"Remove eventlet from designate worker\n\nReplace usage of futurist.GreenThreadPoolExecutor due to its eventlet\nusage. Designate used it because of eventlet + py37 issues:\nhttps://bugs.launchpad.net/bugs/1782647\n\nChange-Id: I33dad118e506c79fa659f0aa9d5761cbe7eefb5f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3d38dda8aae058fc46da2c809aed654a06be0b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3d38dda8aae058fc46da2c809aed654a06be0b0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
