)]}'
{"id":"openstack%2Fdesignate~942777","triplet_id":"openstack%2Fdesignate~master~Ic5d334960fa1f7585e2a8490131d9c11a44f100f","project":"openstack/designate","branch":"master","topic":"eventlet-removal-designate-worker","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-04-04 16:07:59.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-10-22 23:07:59.000000000","reason":"Change was abandoned"},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2025-04-03 09:26:21.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_37598\u003e by clicking the attention icon","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-10-22 23:07:59.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ic5d334960fa1f7585e2a8490131d9c11a44f100f","subject":"Remove eventlet futurist","status":"ABANDONED","created":"2025-02-26 10:33:05.000000000","updated":"2025-10-22 23:07:59.000000000","total_comment_count":21,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"9547302c0c6366bac2eedf97c02b348c475e66ca","_number":942777,"virtual_id_number":942777,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-04 17:17:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 11:59:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-03-04 16:03:44.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2025-03-10 12:21:36.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-03-15 17:57:47.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-03-15 18:09:46.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"37ab47242741c01f29973f0ea88c7659c2487f6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:33:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9a96eb6078da067123054f5c080453363066af81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 11:59:14.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4af5ca0fa944d048b99aa0140defc47\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/66aaf3efddc74a1db6d6b0e5fde8a7c2 : FAILURE in 42m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/04827a8a71854d81b696715d666c50b3 : FAILURE in 50m 36s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9257ee9b758844baac9a7bd98b3fa28b : FAILURE in 54m 46s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/645c9b9417c44bcf9b9d3e68fa351a72 : FAILURE in 35m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eb4ab06ad2f143bdba4a4ec9571f4727 : FAILURE in 45m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/53b1c3191cf84654af51011e91d1438d : FAILURE in 1h 09m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/55a4e57569164ed6a98569d618919177 : FAILURE in 1h 17m 17s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a5a24eab858a4da294e81eb4541f90a4 : FAILURE in 1h 11m 29s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bf16d9a80a894105b7469506e3904777 : FAILURE in 1h 13m 11s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9f355222a5424ec4a8ba54dbe1ceae03 : FAILURE in 48m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/536cdac6c8ff4bc0a7c2518534d69aac : FAILURE in 43m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c847860fdd9845a4b0c5c3215b8f6864 : FAILURE in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd99d88079a64c49beec2bc1f2e7123e : FAILURE in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73f342a2282f47de8035bdeb72ccc385 : FAILURE in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b2457e629404fde98b21ddada7827b9 : FAILURE in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2567c09ea6004fa0bba0dd95b4587573 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b923a7c1eac54548aaba3ed8b5e6bd04 : SUCCESS in 49m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/77a0a40dd9624d8ab44260eeee71ea6e : FAILURE in 5m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/65e27f43336e41ee90d6c997abc77dab : FAILURE in 46m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a9dcc93793ab197e899e2f007cc8b3732c47f606","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 12:40:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"52003087a962b62e6d0d38d7df297dfc3048dee2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 14:05:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bc100e9b4b084ec2abebfd849dd2b86e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4dee7d6f514847f69616c242198815f7 : SUCCESS in 41m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/be8190a83a04461c9b8cf47573a68f89 : SUCCESS in 57m 41s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/abf7411bf534418a890a1287e570681f : SUCCESS in 1h 02m 04s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d2bc569f8ce34c0cb4d322de90c72a2e : SUCCESS in 39m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a3c5d7143da041729a4255ac4e2f8586 : FAILURE in 51m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ae94be3d95094c7ba0d3a93d5f2a9bf9 : SUCCESS in 1h 14m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ff763dff49e54bdca7366ba83b3c5676 : FAILURE in 1h 08m 40s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2cf4bf9c4f4a4db9867a67f4ad98a134 : SUCCESS in 1h 15m 04s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/35e68d87f29d44908b5da04e6ca98ef6 : FAILURE in 1h 21m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fd91dc3fca3b4b2e8fc23e82523f9522 : SUCCESS in 1h 00m 04s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/30d77ce18c054248839f8c4bfbb24654 : SUCCESS in 48m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/086a30cb60d34d99b2aa9ceee5f74e49 : SUCCESS in 8m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e11f960079224433877df6d8ff4c4842 : SUCCESS in 3m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/92eb6a472ea24aa5b2dea2ea859224b7 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f469dcd406ec478bb520115d6aabcb78 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc72b19ea50b449092d9ba2e78b19023 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2ecfaf0483524c62a93366b9364cbece : SUCCESS in 33m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5baf6b365c254d3eaa0473c02899a35b : SUCCESS in 7m 05s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/86f8c2905da549a1883c98bf7983431b : TIMED_OUT in 1h 12m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"928f5280431fe55ff6ea563857e4169551d3004a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 14:35:19.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"5823513b47a8ba84f7acd7cba211207e50273d28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 16:18:15.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a339a051267840c3b728cd2e600d0fd1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f50210aae10e4b8abf6a84169bb74804 : SUCCESS in 42m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/51998d4193cc46ff8f988ee4fc6dd67a : SUCCESS in 57m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/24f90d4a6a6c4ead9fb747f58de65cfa : TIMED_OUT in 1h 13m 24s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d551fd93453b4dd68c0a064671787974 : SUCCESS in 49m 10s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/28c1f0d963454479a02f48dee85349b5 : SUCCESS in 48m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a62fa873afc3460e9303251714d00a8d : SUCCESS in 1h 26m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/30b55ce982b94a098a8361a7803dd87a : SUCCESS in 1h 32m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4c154455edee48e4afab6c23f22958c1 : SUCCESS in 1h 10m 51s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6c1cb5cf49de4761adad751da325431b : SUCCESS in 1h 28m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1b858e2161974687a4b268654a4dabdf : SUCCESS in 48m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2babbeaef3f5481ea68a96504dc4fce4 : SUCCESS in 54m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5def8b02d5de495cbcd535d069d7c18c : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d73b25b4c10e4e689b847d9eaf87270f : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dee584191d14b0b9411c370427de16b : SUCCESS in 8m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d38cc31615014d199abcea7def16b5b5 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e9d19ca18c544b1a94a60afecc0bbc3 : SUCCESS in 6m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/31e2c50e5566405098457a70c06bf76c : SUCCESS in 31m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dff27493ae9849a5b96828126fa2c7b6 : SUCCESS in 4m 36s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/a72a9c8467d34174a06a8d8a1458a705 : FAILURE in 45m 02s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"ed8b88f9344808c03e25a547762cc37f21675448","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 16:27:56.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"449082db56fb78a4aef78b3efbe0bf03060fecc4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 16:28:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"77a35a6cab052d4dd011a06a4d8e3f3d00db7910","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 18:16:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a89049125cbb4e688f3d84a0afddc574\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f0963eeaca9492d89b335f59c2529d9 : TIMED_OUT in 1h 12m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bdef733eb36744079c435d3ec7d5df8a : TIMED_OUT in 1h 13m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/130792b52ee04a05abb89a280c1e3940 : TIMED_OUT in 1h 15m 56s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/5927abd613674168abce9e6bfeacccb0 : TIMED_OUT in 1h 12m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3b85148092c64126aa48f5e05966ecec : TIMED_OUT in 1h 12m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d9eec304dfb64e2f96d3c94fa09d8be0 : SUCCESS in 57m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b9a5d46a8d714429b63053e4d5b3e2c0 : SUCCESS in 1h 16m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9e45261a3bdf49ba82a12fb0c3c34513 : SUCCESS in 1h 39m 35s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/713eac61c5b84ddc93ec1851f16f398e : SUCCESS in 56m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fe956a19c5d243b084249f5d5f7d4502 : SUCCESS in 59m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b5b0b75dae7a4b3f8a0166f5176f83cf : SUCCESS in 45m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b42cf307678e45bca813fe03614da058 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d199c2535d6a41deaded3dfb28631c65 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c828f7b5c38948548069002b56a803d0 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f081a5a2cd3a4c51b286b44b5e6506d0 : SUCCESS in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1005ae5845f47149d13a09e523048fb : SUCCESS in 5m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/909618bf04f443bf912ff375b5b60415 : SUCCESS in 33m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d33099135a934f3c952755e76cb322f5 : SUCCESS in 4m 39s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8fd609d91b254bbfa703f97fe537b4b6 : SUCCESS in 40m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"102083d400fe6cd75710f0e82f6e5cbf37e2bd97","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-27 11:48:51.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a8c0713dc728bdc2f246debd938263c730f01e8c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-03 11:35:25.000000000","message":"Patch Set 3: -Workflow\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"167b7966cf6a95e6f5e2bd8874c84debce7900bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 12:55:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/182088ebf3024da9ade60a50842ef5ea\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fd51f2665d8a4657a2507808fdc58334 : SUCCESS in 48m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d599e7e4dd1148809849f11f7b2153be : SUCCESS in 48m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ed74e8e670d458f9117c26d9f7aa7ff : SUCCESS in 47m 55s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/90439cc77945483f9955bb8cbc726442 : SUCCESS in 39m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/73db2ba5f31346ddb0e11724f528c972 : SUCCESS in 44m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/69b3dd07bdab4a92aac099e7b2832629 : SUCCESS in 1h 04m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/90dc577ef2b64f63b71372b0bc0a7d0c : SUCCESS in 1h 01m 39s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/42fbad62570f4d91a8b4166029f4e811 : SUCCESS in 1h 09m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d59480adea1b412c92fb844fa71909b7 : SUCCESS in 1h 00m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/21535a35e77946eebfd15d0b75066a8b : SUCCESS in 52m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6badde878a844dc684b867a1f3144f47 : SUCCESS in 42m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c2cf5951cb644457b5154212f463bc0a : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e010d3cef1bb4ecbb963e2cd06ae7c6f : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/466d05a3b09e4c8f98ae187934f3dc63 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/60a982e9c87a4f1b8840d0fa28a9a198 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0eb4ee07397b432b95281b2540f82508 : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65fceabbc0584e289bfd6d0a1e52aae9 : SUCCESS in 45m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8229a570f822483e89cab523c3684ff9 : SUCCESS in 5m 19s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/572e91eca5674cfa9f2b19eb8753f848 : FAILURE in 52m 09s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"10002c6a2624b5e52500a8ebdfbba21cdf822e36","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-03 13:00:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1de84eb1cf12080aa7ab8c6cdf3df84706bf7be3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 14:33:06.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/00930e2f62b943bda2006708b8e55f8b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/30d0a10fdc0c41c59f8aedcdfd498d1a : SUCCESS in 41m 24s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/13b8b18142854011ac4fd704212c31a8 : SUCCESS in 1h 13m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d083367ab64b465580671a8a4cac53ff : SUCCESS in 50m 13s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/26451e08a83b4787a32c751c86965a01 : TIMED_OUT in 1h 12m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/227fa25cd81349c383d4e3b40801c155 : SUCCESS in 49m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3b2df98039914df380f5615bfc9e215f : SUCCESS in 1h 08m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0fd1be83f5494e0493ab31020b0c668f : SUCCESS in 57m 58s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3c44854c4b214bfe96f61a42d3ebb101 : SUCCESS in 1h 04m 44s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a5eac03403ea42b9b31d919c6b08953f : SUCCESS in 51m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bcee282001ad40c79022e33066bcb3ab : SUCCESS in 42m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5ec222c49bef4b3b8bc69cac6306a88e : SUCCESS in 50m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fd8c3777f2943ac82086236ab14fd51 : SUCCESS in 6m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4c76e0cb84a242f4be4de97892386124 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87b1240107e24715956ec541a4efca0d : SUCCESS in 5m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0638fb960214c79b9a26d622d74f475 : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75d699c4f75f4e3d968fb965b57adf94 : SUCCESS in 7m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2caae1af9d314115b26cd8465ec6eb80 : SUCCESS in 34m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/60573ec939ad4137a1eaa712dd406e6b : SUCCESS in 4m 59s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ac0767eb895c4c10bb225d6af97bd5be : SUCCESS in 46m 20s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"2aa4dcae3530234e8da56ca37d5572b3539fcb96","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-03-04 16:03:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"b3e59bb6fb6fb636966b0985fd3d24cd5e24bfe5","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-04 16:07:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"83b8380237d91e654b6573448ff89b0cdea3c195","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-04 16:08:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"7a5afe207310a5546c6e282fb1d41ecbb600c964","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-04 18:02:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/643e07996e764fad99f9bb4a335c3a0b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/98a177a03e7b48cc9e8ef02cea440c22 : SUCCESS in 40m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6923db0c6dc64ce7896b9325c69ee098 : SUCCESS in 1h 12m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5ee157bbac1477a95e6041a7e2c346d : SUCCESS in 1h 01m 13s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/0b99d830152f44b18c2815c0b8ce3c86 : SUCCESS in 38m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b761ebc4fd37499791a533ddcd851452 : SUCCESS in 40m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/726c6a617cb1453c8a8f619860daf8cf : SUCCESS in 1h 09m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/62c225555e0748dfbf4477465b6735f6 : SUCCESS in 1h 01m 47s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/40c2eb910168465eaace33cc92311dc6 : SUCCESS in 1h 10m 59s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e21030c0dd7d4c41b0f24fe01c5437a1 : SUCCESS in 1h 11m 48s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/905622840df84998ba8846bb39a3a2a9 : SUCCESS in 48m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/96ddfe53dc764c8ab40dddf3e050a298 : SUCCESS in 56m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e539a93249c345ee85c9d9d689888052 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/860f2453a6ac4e259f73ab3c65cc5269 : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/443589bfc7a249319ceb6f89929afc03 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b0f05fa35f6646d98dd5dc0621e9abc6 : SUCCESS in 6m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70117f9b5a5647dfb015e4be0cfb095d : SUCCESS in 7m 21s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3e2df3c0a3694037af27cc6bf29d0d5e : SUCCESS in 2m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d0c55fd15fb74363a47dc9059a66680f : SUCCESS in 31m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3d7ebcc87b38484cadbbed43924ef44b : SUCCESS in 9m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c6497c125be24ad9b9bdd008ec3fad8a : TIMED_OUT in 1h 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"10af68b680f3ffbd150d9b7b0b20062ffb4cd959","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-10 09:28:41.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"ed22e46e3ef19e3316a146334c50d87eeac97ce2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 10:49:57.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a67ff2d1c87c4673b0efacb7f45473ba\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bea96d58e4644c6bb3e1adfb7f1632cd : SUCCESS in 46m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/fb26a60c9fcd4c10849f71fb9a8b0133 : SUCCESS in 1h 04m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd76bf61317a42969231995f0faed18f : SUCCESS in 51m 05s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/a4f2fb34f9f547fbb32e8f7687941ee7 : SUCCESS in 48m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f581750742874d3485556b2262d37808 : SUCCESS in 50m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9f8560319b01498d9e490a05876b658e : SUCCESS in 1h 15m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a8fb897dd962410dbe800a07982f29d5 : SUCCESS in 1h 08m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/df0325b2d3e746468bd4d8b1339e2b36 : SUCCESS in 1h 10m 57s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4f67a96a473042fbaf1abcb7a64ab913 : SUCCESS in 1h 04m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/afcba16f6d6343f0b15cd3870983be5b : SUCCESS in 48m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9dc7dc3f4e414bdcaff2c07bc3397ba1 : SUCCESS in 48m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fc06dd885206446685b5bd83ceb674e3 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/054822cbb6ac473d8d84cb55ddf9bf67 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bae307956e642c5a2a8b804b3c79e2e : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/260c58cb5ff245858a6ebdb94e416ad9 : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ea2f3b5431a4c82a60dbd0bb42ff540 : SUCCESS in 6m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1825411369d94dabbda4ab592ec5356a : SUCCESS in 1m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1d4a4e80c07c40e0bf79bdee407833c3 : SUCCESS in 35m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f186c6b35c0541f992871f4e298b9772 : SUCCESS in 8m 08s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/fe78faddc7ed4af4aa38634a74d0fd4a : SUCCESS in 36m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"f6a5888e9c2d31c8bfa586631a3d5b0ef99a1145","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-10 10:57:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"a7c26d5b2dc93cb2e863405264c08f9051c0a5b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-10 12:15:15.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e239ea669dd49b7ac7ebf02f7cd5610\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/66d1b509e56048df89f21c11372750b2 : SUCCESS in 41m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6766ac2315994c60824e019dc1b5aed4 : SUCCESS in 52m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/827a5fe5bb0e4ecd91b87a5000a8ac8c : SUCCESS in 45m 50s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2f22d359de2e4c81b3d668aabfbe6361 : SUCCESS in 40m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dc962e1941754c09aa929c9ec73437ea : SUCCESS in 50m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ba51232d1c354752bb0398088bb2a4d6 : SUCCESS in 1h 07m 44s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/38cd634d14064d1c97b9d798fd10a6a1 : SUCCESS in 1h 10m 52s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cf5dcfaa48884b7fbdd454fe12055735 : SUCCESS in 1h 11m 31s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d8413610c62945039ae5a0698b33deba : SUCCESS in 1h 04m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3de1495edcda45f9b21c2f82a8856b4c : SUCCESS in 46m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8a341302fa9f4b0e8538d14eda40885b : SUCCESS in 49m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/61544c89a6654424b48ad9e79a75b840 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/398e175b271b4b469ad5233c289e87e9 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/be4958d0e202400f8a07ca07e30ebedf : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fe5b213af2c40e5be9cf83865279e48 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b901e2413764c7c83151752547d172b : SUCCESS in 6m 56s\n- requirements-check https://zuul.opendev.org/t/openstack/build/9fb3f280afb84d60b7e9423dcfd32cb5 : SUCCESS in 2m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2507f8db714f4491bb99e3a3d2a9739e : SUCCESS in 36m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/59fd51b7ca6a42f8b8e1b04ca2cadfbe : SUCCESS in 4m 22s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/44998249958b4de4881e02db9f7ea0ff : SUCCESS in 39m 30s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"32c1b674f1d0d5dbd77827d2e20027549e723efb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-03-15 17:57:47.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"8fb30b53ae08f5891f92ea19ae45c8c4251683cc","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-03-15 18:09:46.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4b1f5f313fa564c508571dd710b38acb6c6185c4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-01 11:42:18.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"419175de61a39eb32ce828550872849f9feb0868","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 12:47:42.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6d09dea6388d4f979d94c47e024e2d4a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/56e3fcf1c1f045a58f18cdd86679e05c : SUCCESS in 43m 05s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8fbf29928c624aabac6d30e753da92dc : SUCCESS in 52m 27s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/690d1b08c55942c583cd782f528326d6 : SUCCESS in 38m 51s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e670cf1b0da445f48ce0f21d91b3b41a : SUCCESS in 42m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f31764ad89ce48afb1e89aad89d5bd1c : SUCCESS in 44m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d5ed085192064adf89542fb69a66724a : FAILURE in 14m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e98ed0e4e6fe4e568517ec499b28df25 : FAILURE in 30m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/67860a1ca4704f81802bdbbc1180409a : SUCCESS in 1h 02m 31s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8eff62739d0e49cfa1b7de44b887fc94 : SUCCESS in 53m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e347cb5135e144dc91e020e7594b7cfc : SUCCESS in 44m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5971538f7055482abba0ddf66b4fc3c8 : SUCCESS in 40m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/39d667ae4141474e9827dbd1122923a8 : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e048576c0404f33a6183d0ccd021b3e : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6b0f973f66a46759efae88641ab008e : SUCCESS in 5m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c35238e1efbb49549b6db525be33d754 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/976b1562a0fb4e73a1d454f439ed7dc0 : SUCCESS in 8m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/acc08277638540b7aab8889d488ed86a : SUCCESS in 2m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf479faf496c48da9809410690b7a919 : SUCCESS in 38m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3abb3efcd04c499eb68fff0ccded98e7 : SUCCESS in 5m 38s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5ac207c3e6ab4e62b6e305a0b86046c3 : FAILURE in 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"b27a85c042d4e0ca56dcb295fad9fa1ec4f5f249","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-01 14:48:11.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7128b568dfe6990e0c39faf8ae139d124ddb1155","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"real_author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-02 00:17:30.000000000","message":"Patch Set 5: Patch Set 4 was rebased on behalf of \u003cGERRIT_ACCOUNT_31664\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"_revision_number":5},{"id":"ba2ddd47df68e190dfc0033089e4b4fdc243dce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 01:15:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e6b0ae2ee4244e549bc9fea0f407ddde\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/34d9c4fe37ba402ea5455945669a5665 : SUCCESS in 41m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/25a0ae0a7b1445ae8a15e0e4ec3ac8ba : SUCCESS in 41m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a9a3e5fbeed24f9bb0deaae1c833d4db : SUCCESS in 39m 53s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/7baf1dbf44c54f42a5e7261092b72a2f : SUCCESS in 39m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1105718489034953a4de63d997b18f95 : SUCCESS in 37m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/398b3d40e8c64d60bb4d45f8bdaa7747 : SUCCESS in 52m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d93a7f33be954bfe862b6eb5bb9f5c9c : SUCCESS in 47m 32s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a408071c57fc49f2b79b53c1cfd253b7 : SUCCESS in 52m 09s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/44a999a0f1d14b91990558b0c479a1b2 : SUCCESS in 50m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/15237db5b338434e98e6406566f4fc00 : SUCCESS in 35m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6d26ff37692149d4883c6f8c8614fa07 : SUCCESS in 47m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/941666f0553e4a75adfe22c74230c453 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8ca4a359a124c0dafed5a9353428741 : SUCCESS in 1m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4226bc840b3444ae81f51f6922481975 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9fb9ed4c82d0428597c45d51c6c6ef39 : SUCCESS in 2m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edaa2680ae934e2aa4effa9e0d44791c : SUCCESS in 8m 04s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6460772abf5642238f4727713558441d : SUCCESS in 2m 02s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c7d515966a634ca1a57a0f6efa4c91f3 : SUCCESS in 32m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cfc5d3e7d00746f0a1d050c4bbd84351 : SUCCESS in 4m 39s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/dccb8f62ddd9450b9bfe87874e29b117 : SUCCESS in 42m 14s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"51603cf404c7596333f0e977d23509c13ec9dddb","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 11:46:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7e3db0be5d1a764897a21c445c5c99ed8b97c6fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 13:32:28.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a21543a99f2c4d60829ffd09d8ffd221\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/06bf1529c5224c86ac59b2946698880c : SUCCESS in 45m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/43e1d29f60d043da91d82960debca612 : SUCCESS in 54m 22s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/849ef4ec621a4b4ab531fe803c0cd5e3 : SUCCESS in 47m 38s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/49c1b5bc223f46d484508cecd254ece0 : SUCCESS in 40m 22s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e51109c0472e4561a488d70e46f4123b : SUCCESS in 48m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5189d853603441f4a2586cbcdb0744f6 : SUCCESS in 58m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f71b6bf3a85040928d067de6fcba1c52 : FAILURE in 57m 22s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/841887f8072f4b4a8ff178d3e3895eee : SUCCESS in 1h 05m 42s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/632112647c97481aaca5996af633b0ea : SUCCESS in 1h 06m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/978b78c1045a45bfb85bc8e8f060c099 : SUCCESS in 43m 28s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/fc14600232394278ad042ae76b9384c9 : SUCCESS in 51m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c697ca380c004268ac30727dc145b155 : SUCCESS in 8m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fcecdf1a38f24231a6cf1c2d599e09a9 : SUCCESS in 2m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/64d99e2ceb824112a270f2c93677a931 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/65b30c010aa44ee7805ba835ffd700d3 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c30be2d122443f9b106272d3e1016f8 : SUCCESS in 9m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7d37ab4a924c46728298895345e06131 : SUCCESS in 2m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c5f291056ac54083976ea0b1ae319418 : SUCCESS in 30m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ff61db6a4bec4273ad1e3d873cd122e1 : SUCCESS in 4m 28s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c16c24f06fb14c67b153566aeec34cd2 : SUCCESS in 40m 44s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"9178943f9416b1f4525126bb455af34d00f1e5dc","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 14:13:45.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0525387a7f919c1b64dab530c3a40e47ca637e5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 14:14:05.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"31027788a4592cf4953af3dbd5fee7784f4ab805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 16:36:12.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8b26ad4a71440b5944d56a2f52b6b33\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9e73699c97cc40baba74f980678c297e : SUCCESS in 46m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b56cc7fac4bc44989f65f96ad569c090 : SUCCESS in 42m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/f7de650f7f534ac2a0f486ac0ff3dfea : SUCCESS in 49m 51s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/30b95d0e7da44f31a384aab81b8cdd7a : SUCCESS in 43m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/28991043ce0c4e96b77458707b582166 : SUCCESS in 46m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/66c00976a9ad468ca3a64c2edc73d565 : SUCCESS in 1h 05m 02s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9d5253ca2bb34368ae7233fb82e15e08 : SUCCESS in 57m 13s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/aa84ac99ca4243e391d6bf30233a291c : SUCCESS in 1h 00m 15s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/09d3f32f14ff4f2fa1c696d200f242dd : SUCCESS in 1h 05m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8fcb05572a5742f4ac26484f0cedc839 : SUCCESS in 40m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8450c225bd4146f684da5155448ef524 : SUCCESS in 41m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a0a032627154fc690da6b00c821f351 : SUCCESS in 6m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2379abad1dd24cd9a6518a45472eb5cb : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0bea667278844721ab649e70a9d59c59 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2992519031a34921bb53f94953b7cb27 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f65d7fc870c946e1bc8063b6a2f588f3 : SUCCESS in 7m 48s\n- requirements-check https://zuul.opendev.org/t/openstack/build/75acc6f70be74825a5a8e3b58e1e9e4f : SUCCESS in 2m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3664b41801994c15a8e4398eec9f9326 : SUCCESS in 32m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/02b9b386af2e4dc5b84282cbd3a218c2 : SUCCESS in 4m 35s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d12732166ab94d9b861f78336ffb747d : SUCCESS in 43m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"3393c5e7930ee27d659d2f24ab2d092777e0ea92","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 09:23:50.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"24ca242f25b263635f80d974a6a4507fef895cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 10:24:41.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/397a03a12af84dc3a68d7eeeedebd083\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5de6b84013324555b5304229609f0a71 : SUCCESS in 47m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/0183b701742f427d947e0d1b1b1107cb : SUCCESS in 45m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d40f8c5d904348f7a40bc18e899e08db : SUCCESS in 44m 29s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/61d08e3d549b43d7a870ae6aae3be5b8 : SUCCESS in 40m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7f046755136949dbbae30917a57c67e7 : SUCCESS in 38m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a219086fe1524a129df72b50f40ed265 : SUCCESS in 45m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/286f7bc3c2d84732ad2dc32e8fd26bff : SUCCESS in 58m 50s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/669bb1f5d7b44445b5463f97f82ea8d7 : SUCCESS in 46m 08s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/095beb35bdf64981a4a09567c08bf2af : SUCCESS in 56m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/52f21bd5c0144f6ba26e1e8eef0a7e62 : SUCCESS in 40m 02s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eb2090c16ded44acb75f49f60975edbe : SUCCESS in 44m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b58ba1ac5993402bb9b0ff2d2b3b16d7 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2e78d6d32e4b4e6193c8a777f8a276f0 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c9d84c8c4df45c2bf315d4d6fb78430 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e29bb16800f04031a1644238c55742d6 : SUCCESS in 4m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a52f58aec7174bd5829c095e0717495e : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/21df286b54394b40bedcc8951996d53c : SUCCESS in 10m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e593cccc2c554773a7b7677e5f36c2ed : SUCCESS in 3m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e6d8ace94af410bb4f8841d394703e3 : SUCCESS in 32m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7d4830f0e81543deb8eabe19ac03358b : SUCCESS in 5m 02s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/37f53d1be3a848b7b6ff4d47d2229e9a : FAILURE in 43m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"6f8091ec8fb6de31cee1508b85ddb661eca16d62","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 10:27:09.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"e59b2a51fe5124b11a6837eee5e807cca740933e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 11:24:30.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bafffebfa0464ec2ac883308a62965d5\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2aed309309604942969b008c962519ea : SUCCESS in 42m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5bd3d90efcc14279a3e55ed5b56016ad : SUCCESS in 50m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e583a0e4b18045cd852031034d2a82cf : SUCCESS in 44m 21s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/ea804b6d155041879fb8024e5be4da24 : SUCCESS in 38m 21s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3079f4049394494cb8ebcec9bb90b393 : SUCCESS in 38m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6dc7c3dc87804a47b9e40ec55c45a2cb : SUCCESS in 49m 36s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6ed0bd860aaf420980438feab6ff12b4 : SUCCESS in 55m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a2646f68afbb40b992b1945384af8526 : SUCCESS in 48m 07s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6b1bd3a0e0b34aae96dd2841c8647024 : SUCCESS in 54m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f3970ffa20d6445ab6ff45bba32e4ad7 : SUCCESS in 35m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/949b1a72c3d8442ba5528b5038da2e7a : SUCCESS in 42m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9866fc9075440e9b7c9ac071724da6f : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c7b66a5b62d4b088e4b042380c883e7 : SUCCESS in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf0e64df58744a099777f8b78e7247a2 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f34b11551a4b44068dcccd0a601dc229 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/12e5a517b3fd42359190294153fa3195 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09729ea2df7545888fbccc85a71e7f1d : SUCCESS in 5m 34s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1964f531733c4132817417865f0730b3 : SUCCESS in 1m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3087ce76d55b4deb946f12834369db6b : SUCCESS in 33m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/01ca98459c244b4eaee6015492154808 : SUCCESS in 4m 21s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ac82fbe77544475586017ceeae883ca5 : FAILURE in 45m 08s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2429a4c1ff9c2a03402374b958f46171651ca46e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 14:08:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7caee1abb265da415757ad3aa5d8714af00c3065","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 15:36:16.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8993ed2715f43909b487fc0588f9e74\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9ec360ae46344e9a84f219d3f2b2f593 : SUCCESS in 38m 55s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/780a4dbbdd7a473a8709662b25134c01 : SUCCESS in 42m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0194caa40d4444309843f8f57aedca71 : SUCCESS in 38m 57s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/facb9e0326314129ae9bc6754692b96d : SUCCESS in 39m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/833a2ec65aca4a4abe885f51c481c9fe : SUCCESS in 37m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0acb855796574b128819935956c58ccd : SUCCESS in 58m 00s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1a6b055691c0469b99ae43ba68387561 : SUCCESS in 1h 00m 48s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/601fbf77131c4c4db063a6e869aa8c99 : SUCCESS in 56m 26s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2249e2be83f3478493298311f3dd9c3f : FAILURE in 1h 12m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0fc8bc8700244253a4a8d01d204ab6d2 : SUCCESS in 42m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c9e85dd9b3894fd79ea13060f4d2bf84 : SUCCESS in 42m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1088cfd3503143cdafc676c4868e9104 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac601e5c469048579b6b17b445c4e34f : SUCCESS in 2m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5b03d3d241e24efb8cf6a2edb7823cd5 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b70e9491541a431dbb5f938db0b58d79 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/df38e57e15194f3b9a625333cdcf4d35 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eec970fb469c4b5987524606bd395379 : SUCCESS in 8m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a951b0ed7838499e998fdb36345ac086 : SUCCESS in 3m 09s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/197e156767b24a57a95f4ec7d12bc774 : SUCCESS in 47m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7c6c0f5f6e0a4a508d6a637022ab9b3b : SUCCESS in 5m 13s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0b27f4dca7ca4ccab0ba791ae0030960 : SUCCESS in 44m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"10fc489d3d81c1321bd7bce2699489f89004f79e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 17:06:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"1b24bbad3ba0874ec643466df80dfba6a9db2b3b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 17:07:14.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"27b40a1c2b66d14cd0c680f4cd6a70f6d33e9aba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 18:19:57.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7a64840926dd4f30a78f02beab55927a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e6a398e5845c4b24a9dbe8085e98ddee : SUCCESS in 40m 04s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4d0f3bc523774bdf9acb950567029e19 : SUCCESS in 42m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/53bd34c4d229418b995df5ed0f51b905 : SUCCESS in 53m 36s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/a47c1611be8c4293ad3301016a0b0fdd : SUCCESS in 39m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/de2c2c3349774bcd9721c5d786d3156d : SUCCESS in 39m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8ed67a5ec2fc4480ad3c814ac7b2c024 : SUCCESS in 57m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/766bc2cd5d134a1aa88cbebecd65bdf5 : FAILURE in 1h 10m 28s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d82c8834a11a42c4b24c363af7398533 : SUCCESS in 1h 00m 52s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/39be7c14faae4662a2fdb81735289077 : SUCCESS in 1h 01m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7c1f956b639145379beb6083db1ade73 : SUCCESS in 39m 50s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/70dc04eaa9b3439f89432d4b3aaa8c73 : SUCCESS in 43m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15c5ab2b59a0441797acc22dc2bbda43 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9814db7a8817473ea59e9f856576d688 : SUCCESS in 4m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e475e45122ff4a5ea8756bbf04886762 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8adb5756718e4b5eb0df3b4e6878a930 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b24b40be0bf84130960a77714586f722 : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fe8a001d2374e51840fb0e4e4bc9d81 : SUCCESS in 8m 19s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6e41528f340c45d7a8394a358801f0f5 : SUCCESS in 3m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4934b2d698344139990612139569f30f : SUCCESS in 41m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b5b08ade905c4aacb299877a46d76e48 : SUCCESS in 4m 44s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/14caf77759ba4109b067e6ce68f61932 : FAILURE in 45m 00s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"a055df9007ba0fd15425ec3335739e7ee81156a7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 16:07:59.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4e29966c4d1c0bb1726fb6b021dcd8d11377b394","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 17:17:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54ae91ac93cb4ecf83ea6c02f1b90c29\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1c17f01b7d3d4970bfdab55915e421f1 : SUCCESS in 39m 08s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/9e824f04829342d6a02a72499ac0c384 : SUCCESS in 48m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/5ac025ed5ca146dda770c302a244617e : SUCCESS in 41m 13s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d5e2435aae074b95bb94531aa3814ea5 : SUCCESS in 44m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5229ed485502400393e47200967604a7 : SUCCESS in 45m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1c9e97c42bc6425d894b77c78357d73f : SUCCESS in 1h 06m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9ecbe81aafe54e7dbc5cf8478b2d2d29 : SUCCESS in 1h 00m 43s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/95cad00e4d984e07b1d0d184dfa983c2 : SUCCESS in 1h 03m 14s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/44e817fb8f834fb0b38d0e39aa797c5b : SUCCESS in 1h 04m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/776d75cfb9494e69a9eec365f5fcfc59 : SUCCESS in 35m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d0bbefb4cbf94d6dbef28931c4cddbcd : SUCCESS in 44m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d214aa81425445bfb37e085cc7d42a29 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/465e3b296269444383e584244c7b02e9 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/21e216d568204955876c376119fba57d : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4aaf6428831545e285e0e0c3e8f85e3b : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/896d1f3afda14fc0b29a54efbbebe955 : SUCCESS in 9m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a6f217150d9e4998855ed1e3b0ff10a7 : SUCCESS in 10m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/df939c9af5ad44519354bf2263711b9f : SUCCESS in 3m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3b19a1deac0d491c9e133914dcd973ed : SUCCESS in 33m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/40db8d84488c4583a5faf09bbccdea3d : SUCCESS in 4m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2d01b7dd5ef84fc7ae2620727ab24672 : SUCCESS in 41m 02s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"58d9d9a323237309d1ef2ed258b031223e1b538d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 17:24:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b0fdb153036123957229ceebc2f88dbc59300a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 18:28:25.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c9ef01e23f04a9180d22804d83512ca\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e62bdf35db8f4f8daf5bea5fcab63eec : SUCCESS in 33m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4c15ec68a15a48d3b6698f1dd692cd36 : SUCCESS in 49m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a2cf929afbdc48ed989da6c8a747e1fd : SUCCESS in 49m 32s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/378a8a676db14c4c9e32b6adb3f8b6e0 : SUCCESS in 36m 49s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd374fe20d4b4b259086e5e72c58e933 : SUCCESS in 44m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e9a6158737a44fb7841c47267a6fa328 : SUCCESS in 55m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6ca07eecf02f4c5f99f11b5a87cc26d2 : SUCCESS in 58m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5ce0439ab1f44383a317f88f5dac9346 : SUCCESS in 55m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/93fdfc265eaf44838dc4e5be408167a9 : SUCCESS in 49m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3e39471b556946b582df871ee6a2ad1d : SUCCESS in 47m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/58a42c8309cc40c897077a636bd40a86 : SUCCESS in 47m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d285fd1f306347498e22c559f3c12728 : SUCCESS in 7m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c205571f76794c8fa53d5990f024dd5e : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67601abe3e8841178d544cdce6d2ac13 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44aa0888a987421fb8a97a47d1462fea : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/208ff180c916415d8b1779937cf5cf67 : SUCCESS in 4m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6bc06b9f0c64250a56515ff75c2c84f : SUCCESS in 8m 35s\n- requirements-check https://zuul.opendev.org/t/openstack/build/63b9beee3a4c4df5b1d21367216dcd03 : SUCCESS in 1m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcea1dfbf94c4b1aa8232eb3afae65ea : SUCCESS in 33m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a76de797e72b4f72947f8113c9a8fea8 : SUCCESS in 4m 24s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5b92721c0dc34e66909891d2cf02db91 : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1615d7dd8331c000371c90ecdc720ccdb25d2a53","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 18:40:46.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9d9da94f8b08e5abcd219bbebb05fd636373cd30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 19:34:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae34469b625e4ac4a74c1ee9ba72c38c\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/54d9c19710a1465eb8c2550b44c34ecf : SUCCESS in 37m 25s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5c02ee8413d541f9ae06ad02ec3644a7 : SUCCESS in 42m 03s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4535285d623844d6b81d39e95c3c07fb : SUCCESS in 40m 41s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/bbfe96e0d06d44a0a529a26b01c94f82 : SUCCESS in 37m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0652a15c72444db2832592feaabd76fd : SUCCESS in 36m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/267f63a9eb8b479bbdf1cddcfb46477a : SUCCESS in 53m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e9339794884b475593b997e65a0d6d7a : SUCCESS in 49m 02s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6297f303ffd945a697f8f4c915466e5a : SUCCESS in 46m 19s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bc864faa168a4945a3ad0dfdee826416 : SUCCESS in 44m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a1b8fa3037294cef936eee88c4a12c6d : SUCCESS in 39m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d59f5bb03e35480890e0cfb2a02eea30 : SUCCESS in 35m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d552c69843a84a6da3972968fb5dd2d5 : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de2a23bfe2a840658054d387534b710f : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0f4dedbc61d4964a1c84b9752ffb1d3 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a726e182e8f7415e83f328ad443d0d24 : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/00a35f3caccd4e6cb4c92e2a9a1fa241 : SUCCESS in 5m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dcd8bec4ae340c28fa5696bdbfeed19 : SUCCESS in 6m 39s\n- requirements-check https://zuul.opendev.org/t/openstack/build/15d7f2b01fbd4622a20ce74f850424a8 : SUCCESS in 1m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ec4f0c95916457b92181faabad2fb86 : SUCCESS in 30m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7c6c5f59043d4b349bbfb9850d0747e7 : SUCCESS in 5m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/96e035a873df4b66824bc1532168f9b8 : SUCCESS in 45m 16s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"9547302c0c6366bac2eedf97c02b348c475e66ca","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-22 23:07:59.000000000","message":"Abandoned\n\nDuplicate of this chain of patches https://review.opendev.org/c/openstack/designate/+/964385","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b5a78111ba219deb837ae74f5e346de3d6ae9648","revisions":{"f7e8fd9720c042cf1bcca9efa06f408dc2abfa08":{"kind":"REWORK","_number":1,"created":"2025-02-26 10:33:05.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/77/942777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/942777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/942777/1"}}},"commit":{"parents":[{"commit":"c3d38dda8aae058fc46da2c809aed654a06be0b0","subject":"Remove eventlet from designate worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3d38dda8aae058fc46da2c809aed654a06be0b0"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:32:13.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:32:48.000000000","tz":60},"subject":"Remove eventlet from Central tests and neutron-api","message":"Remove eventlet from Central tests and neutron-api\n\nChange-Id: Ic5d334960fa1f7585e2a8490131d9c11a44f100f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f7e8fd9720c042cf1bcca9efa06f408dc2abfa08"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f7e8fd9720c042cf1bcca9efa06f408dc2abfa08"}]},"branch":"refs/heads/master"},"cec4635c61e96bcd4c5fcc277eeba2feb4d97adb":{"kind":"REWORK","_number":2,"created":"2025-02-26 12:40:55.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/77/942777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/942777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/942777/2"}}},"commit":{"parents":[{"commit":"c3d38dda8aae058fc46da2c809aed654a06be0b0","subject":"Remove eventlet from designate worker","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3d38dda8aae058fc46da2c809aed654a06be0b0"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:32:13.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 12:40:44.000000000","tz":60},"subject":"Remove eventlet from Central tests and neutron-api","message":"Remove eventlet from Central tests and neutron-api\n\nChange-Id: Ic5d334960fa1f7585e2a8490131d9c11a44f100f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cec4635c61e96bcd4c5fcc277eeba2feb4d97adb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cec4635c61e96bcd4c5fcc277eeba2feb4d97adb"}]},"branch":"refs/heads/master"},"8391fde83c6c0a5bdf610b4d64493aaf56b14807":{"kind":"REWORK","_number":3,"created":"2025-02-26 14:35:19.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/77/942777/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/942777/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/942777/3"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 14:33:45.000000000","tz":60},"subject":"Remove eventlet futurist","message":"Remove eventlet futurist\n\nReplace usage of futurist.GreenThreadPoolExecutor due to its eventlet\nusage. Designate used it because of eventlet + py37 issues:\nhttps://bugs.launchpad.net/bugs/1782647\n\nChange-Id: Ic5d334960fa1f7585e2a8490131d9c11a44f100f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8391fde83c6c0a5bdf610b4d64493aaf56b14807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8391fde83c6c0a5bdf610b4d64493aaf56b14807"}]},"branch":"refs/heads/master"},"ddb1e3febbd71fc445083239fd3be8056ae9547a":{"kind":"REWORK","_number":4,"created":"2025-03-04 16:08:08.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/77/942777/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/942777/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/942777/4"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:17.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-03-04 16:08:01.000000000","tz":60},"subject":"Remove eventlet futurist","message":"Remove eventlet futurist\n\nReplace usage of futurist.GreenThreadPoolExecutor due to its eventlet\nusage. Designate used it because of eventlet + py37 issues:\nhttps://bugs.launchpad.net/bugs/1782647\n\nChange-Id: Ic5d334960fa1f7585e2a8490131d9c11a44f100f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddb1e3febbd71fc445083239fd3be8056ae9547a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ddb1e3febbd71fc445083239fd3be8056ae9547a"}]},"branch":"refs/heads/master"},"5b2f7cee9f34277155f33c580c52fd1747363532":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-04-02 00:17:30.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"real_uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/77/942777/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/942777/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/942777/5"}}},"commit":{"parents":[{"commit":"595894be2202bfd98805a0ed87281e69c59b5c1d","subject":"Switch the devstack upgrade to use the member role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595894be2202bfd98805a0ed87281e69c59b5c1d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:17.000000000","tz":60},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-02 00:17:30.000000000","tz":0},"subject":"Remove eventlet futurist","message":"Remove eventlet futurist\n\nReplace usage of futurist.GreenThreadPoolExecutor due to its eventlet\nusage. Designate used it because of eventlet + py37 issues:\nhttps://bugs.launchpad.net/bugs/1782647\n\nChange-Id: Ic5d334960fa1f7585e2a8490131d9c11a44f100f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b2f7cee9f34277155f33c580c52fd1747363532"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5b2f7cee9f34277155f33c580c52fd1747363532"}]},"branch":"refs/heads/master","description":"Rebase"},"b5a78111ba219deb837ae74f5e346de3d6ae9648":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2025-04-02 14:14:05.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/77/942777/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/77/942777/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/77/942777/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/77/942777/6"}}},"commit":{"parents":[{"commit":"595894be2202bfd98805a0ed87281e69c59b5c1d","subject":"Switch the devstack upgrade to use the member role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595894be2202bfd98805a0ed87281e69c59b5c1d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:17:17.000000000","tz":60},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-02 14:14:05.000000000","tz":0},"subject":"Remove eventlet futurist","message":"Remove eventlet futurist\n\nReplace usage of futurist.GreenThreadPoolExecutor due to its eventlet\nusage. Designate used it because of eventlet + py37 issues:\nhttps://bugs.launchpad.net/bugs/1782647\n\nDepends-On: https://review.opendev.org/c/openstack/designate-tempest-plugin/+/946157\n\nChange-Id: Ic5d334960fa1f7585e2a8490131d9c11a44f100f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5a78111ba219deb837ae74f5e346de3d6ae9648"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b5a78111ba219deb837ae74f5e346de3d6ae9648"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
