)]}'
{"id":"openstack%2Fdesignate~942779","triplet_id":"openstack%2Fdesignate~master~Id85c839385cf51261fb0cb9bbfbcf27741f16525","project":"openstack/designate","branch":"master","topic":"eventlet-removal-designate-eventlet-debug","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-06-02 10:16:15.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-03-18 12:49:32.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Id85c839385cf51261fb0cb9bbfbcf27741f16525","subject":"Remove eventlet debug","status":"ABANDONED","created":"2025-02-26 10:47:38.000000000","updated":"2026-03-18 12:49:32.000000000","total_comment_count":11,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"efdb4b7610bf2a35baa310542269af0642dfd78e","_number":942779,"virtual_id_number":942779,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-04-07 13:04:36.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-02-26 15:23:21.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-26 11:57:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-26 15:23:21.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-08 13:22:47.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"CC"},{"updated":"2025-06-02 10:16:15.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"c738659f9e73447e345ede878aa5de43c6290436","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 10:47:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6863e0db650252b9634bb718bfa71e1e0fb6645a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 11:57:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa943d809cd04dd49d6f27958e1716af\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/efa6056c65a74364ae60fe88a1aa2ce9 : SUCCESS in 47m 43s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/596b3f42cd1744659cb6fb71403bfdc7 : FAILURE in 56m 54s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8f3b79e22fdd4a119a11266c83310d47 : SUCCESS in 50m 00s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/139b26dd60074460b1a40ea5402deafe : SUCCESS in 41m 30s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3ef600a930154a54a7646614e338f961 : SUCCESS in 40m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3441d15247fe431fa45668fe19f19d53 : FAILURE in 1h 03m 01s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/54c801c8f64741e1a26c1d1e70849df5 : SUCCESS in 1h 07m 47s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/160b554d38694397a27eed096f2caa84 : SUCCESS in 1h 06m 55s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/32c0cadb07f441b5a2dbf3c3f8d5f39f : SUCCESS in 1h 03m 44s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/43837c84cb42438294f21adf6b847fcf : SUCCESS in 42m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c1236915460d4b7785a4c5f33bb245fa : SUCCESS in 40m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d8600d29a25b4a7e9a2259f894219461 : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fb0832e0d9046c8aeb57121894f38dc : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b7700434f6f94e2a9356648bd3f94cd2 : SUCCESS in 5m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae2e240c47b34f93ba642b1e73bacc19 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83f8cc60383f49c8b5dbd2a2b07519d0 : SUCCESS in 4m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1884fcd70094ab19d862069786bfb57 : SUCCESS in 41m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e647803d362b41ce8afeba98e4161599 : SUCCESS in 5m 31s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/68d5dfc596364586933a178ba95d5c2e : SUCCESS in 43m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9aaafd05b3c7c0d53e8fba37b3fa992a6071a853","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-02-26 15:23:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c2db275733de35ec3ed7d246a24b1e918e7d2026","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 16:30:44.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3ba613ec6b79d96c7b67a32e662aeeb191947186","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 16:30:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ceb63a6003af1fa365dffa1106e2b73afce939e9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 17:56:23.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/471c73da84d341fcb2d5b29d88b3daa6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fb22714a9c1e48878d6e07831dcf1edf : SUCCESS in 43m 37s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/11884ad4a4704de5b4544fd3e871bd6b : SUCCESS in 1h 10m 29s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/08c9d8c53995412bbd378edca9e3f14b : SUCCESS in 47m 36s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f58a0b20fd2c470eb2fff8f55d5495b1 : SUCCESS in 48m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/77c22f4c661b417b87ae1da14bb35520 : SUCCESS in 42m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/774a4292af7042c3b7a2c111b9973b52 : SUCCESS in 1h 07m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/643cf408967d49b2a406b32b95186f5f : SUCCESS in 1h 08m 50s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8ecba8b039ef4e948d82846c298d0a06 : SUCCESS in 55m 06s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/11e50267664e44fa8abc256143142f88 : SUCCESS in 1h 11m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6f6b43fd67744974b98d85ac0728744f : SUCCESS in 47m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a378bb098c59464f8b0f4090936e6801 : SUCCESS in 41m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72229ab08d49415fb42bd93f7c43e3a2 : SUCCESS in 7m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b30c1f486644e65867d5a128c5a398b : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a5ee51698d8f4304983bea4f9e0c2aea : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a5e370c734bd4490800f49674d5060b4 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1f42b8ab2e6496ba7b6196466d4296f : SUCCESS in 6m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a3a8fb6f4ea740a8abb49c14d78bc923 : SUCCESS in 38m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2f5693f7ff4840d8aeb0990536e582dd : SUCCESS in 5m 12s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/39765985bfc741f88a672a5840efc400 : SUCCESS in 41m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"14694ec7c780ead66e643a56a6023f2ee0acfe72","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-02-26 22:06:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9d874f238dab77ab8bd9b32fea94c25feb6aa92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 23:16:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/091485bb2ffd4088ae5c7d94cf06cde9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/40453b7e27984fcc98599aa16eb47b1d : SUCCESS in 38m 03s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/81b0cc34d38b4c4fba6b57e7565745da : SUCCESS in 50m 38s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9a99c52e8fb4ce5a393f735864f1ecb : SUCCESS in 46m 37s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/4957d7c20d5d4969a7d31977422aa17e : SUCCESS in 41m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3b98b7ae400a4fb6927389c8792bd6fd : SUCCESS in 46m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/69aa3c325c0247c28470708cdba49f7d : FAILURE in 1h 05m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/eb1092014fab434193b72c155ccd59a2 : SUCCESS in 59m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/60a1319d478e4668adf21a7a4059836e : FAILURE in 1h 05m 45s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6686fa74b9b44c67842795fbe53e1288 : SUCCESS in 59m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3880c8acced24e8497596895736c4529 : SUCCESS in 43m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c161dddb09b14b40a8f5611a3b8f70d6 : SUCCESS in 47m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05020826968f4fff8b733613a42f6453 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c12ed76e0cfa43618b91720ffe6bed36 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6506e8827414400aa2510b49a425883a : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a5be8409afc4408a2a37ace51fd33e9 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1678367836ad4727b3e489ca64604f3a : SUCCESS in 5m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e822f25ef4e44b528fa699e962134df1 : SUCCESS in 27m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6a9d2dc15b264f77a264aea0e151e7a9 : SUCCESS in 3m 57s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/660750ae335a4914946df86a2a4c1b2e : FAILURE in 50m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0901f598a1238865d6c79bc772e8c24e265e9da7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-03 14:05:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf5f33e7edf860f6ff2e13fbbebcb7af4efccde9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 16:03:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c944fef3381d489080450dfd2c136813\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4f3dce7f9c3d46bcbd2a387cce2c438b : SUCCESS in 44m 39s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/51cf2da4dba3487ca15d50db88f174b4 : TIMED_OUT in 1h 13m 19s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/8345aafe8eb240ff823b6fb8b3e47203 : SUCCESS in 56m 10s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/04491b185f0047dca4c7de62d1a0288f : SUCCESS in 37m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/43b2601f64494f7789e37daa374a8d2b : TIMED_OUT in 1h 12m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/41af0a46b81844a0982766ff299e18b6 : SUCCESS in 1h 10m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bd1df0275ee84ba4b8646d7a5ffdda78 : SUCCESS in 1h 07m 35s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8a3863afb1dd4bb7b3c03a1d13f81935 : SUCCESS in 1h 05m 43s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/54e0ccf0f3cb4267ade93355e0d61898 : SUCCESS in 1h 52m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c7aa5fd4a02740ab9f4205affb822e27 : SUCCESS in 42m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f6f011a7534849fe81034eef236af658 : SUCCESS in 41m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3754ced7668f43d9ab7bfaf7fc56f0f8 : SUCCESS in 7m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/42006bc13e59458792932835f2908eb5 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fdd936ab23a46cca0b225f0708e0d24 : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6cbe7443f08d4825bd2f8ffee2d979df : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a1734ba8ff44a3957ef92028752ef4 : SUCCESS in 7m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d8d96cc6a668426792498415158e07f8 : SUCCESS in 33m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bcfb589f72d546e68203b10fbdeaa91b : SUCCESS in 4m 51s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6e7e9dfa792a4be2961d523788108af3 : FAILURE in 49m 36s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d8e22b77f69cf78aef85ca8cdea70e2e74a967d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-01 11:42:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d145857fad255c9202fb81a9371776bbec8ab985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-01 12:50:49.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/81b7ef9924b34084bb54ccebec496c60\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ab9308a728974a65850da4993dca3f83 : SUCCESS in 42m 07s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/24b9d7056eb94fa8903edf092f36ce1e : SUCCESS in 54m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/101d127bf79e40d6985ec48bf274e9dd : SUCCESS in 49m 28s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/244b9ec6e1ac4087b851e34678c382c9 : SUCCESS in 40m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/21ab15d5dc1645ac91b20621bb97d278 : SUCCESS in 36m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3af819cfde524da8b9e7a3f26109fdba : FAILURE in 23m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/38b0b86afa144d04bd0e0e4fc4b864f9 : FAILURE in 25m 13s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5d7ae592139d479296c45415005cd82a : SUCCESS in 1h 02m 11s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e1ce507dabe94e2fbd9c2c4c33df3a1a : SUCCESS in 54m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/4d69bb8f5c7b4fb8942735d0fde7213f : SUCCESS in 49m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/52b7a451fb79430b80fedcfc58e5dc4e : SUCCESS in 47m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99bb0914824c4889aa56354c802ad0d9 : SUCCESS in 6m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45282b63f3e1403ba175d8f9d83cb2f6 : SUCCESS in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/136c2b31d83a4588b30c3bfa4316b4e9 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/43873a210be44f6d82af319b959c917b : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1047c351e9f54f1583b865edf5984851 : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35a5a420ca89471d818becdaa4b800d2 : SUCCESS in 33m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8a1c8c6ee4f0445586228167bd856cd4 : SUCCESS in 4m 56s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b2ce312f48bd4efdbc5faa750a10d5b1 : SUCCESS in 48m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"669551f3d1274275ce151f639f4722fefe815da9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 08:24:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a531fb78abb9dae39af953d16647c512ffdaba18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 09:41:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a65a465ffcc34262ad1d3be700ab212f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a9d34fd8f5ce462fb0c3ff44a06f022e : SUCCESS in 47m 32s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4af02a89f3d9408f9f9355603352ced7 : SUCCESS in 44m 05s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/54b24a48c75946b0a35f84fe4dc68742 : SUCCESS in 46m 37s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/41136889216b4125aff2e99cd1bc7c0d : SUCCESS in 45m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/98cdd6219b5f4d56af88527d621bfabf : SUCCESS in 44m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3688ea0f934540a2be6af185d921a8ab : SUCCESS in 51m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78b96bc0c4654e228c4b8f4b972bb3f0 : SUCCESS in 1h 06m 18s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e205671be02646e0b75e2aa339512456 : SUCCESS in 55m 34s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e28151d55f8349e4b1ffa5d2775a7fa8 : SUCCESS in 1h 06m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0e6d8892c734eaf950be9a59ff0aa49 : SUCCESS in 41m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aa6a57dc5cca40289b5a7a75730c6429 : SUCCESS in 47m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6e07b5f89d8470fa9f8a8af5c090da4 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48ab9ecaf9604382904382ae50bb2267 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3bf1f125fdf14c9f96ad260f69a4d104 : SUCCESS in 3m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7619fc8f59224a25ae0162f21c7663f7 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cce656cbee2497a92c7fc97bc02c169 : SUCCESS in 8m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/02a596ecc6684f53a768cec32fc4154c : SUCCESS in 45m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/38b5df2de9a1465bac94c8162913e4d4 : SUCCESS in 2m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ba876c79640d45639f1497693fe6d6de : SUCCESS in 46m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0e0b0581667d88cfff22ec57012b0a7ad32dc09","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 11:46:06.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fb7152681c5ecb05ac21e47ba8d7215b8f53fe26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 14:28:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acfc91fc3d774fc9806d1053d5918142\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/48b4395f7eba40d3bc3d84a7963f5bd0 : SUCCESS in 57m 59s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7f3936aa460a4bd2aa54e403165d3367 : SUCCESS in 51m 55s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b547a4cd8c7f4bd493f5776fd3e4faa3 : SUCCESS in 41m 53s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2900db2efc3e413c8b1f4d8ec6b1c25d : SUCCESS in 45m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fd4308447a1946e09bc89498eafcd31d : SUCCESS in 41m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d6ed91d7d7954d94844cf8e1ea811eb0 : SUCCESS in 58m 22s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/44547b9894bb46498adef04c537196ea : SUCCESS in 1h 16m 41s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/54e14782eadb41529eb02a17e343b897 : SUCCESS in 1h 08m 47s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/18895f713fe14e57971a6ff99a1e28d3 : SUCCESS in 1h 28m 22s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/95f7a526c81d4f218579ba31adfe2544 : SUCCESS in 38m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f7fba3448684a5985d8d4af1cfdc9cc : SUCCESS in 1h 01m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/23514090f4d74663b575e400fc29ee37 : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a70fff539c63461fb590231bb93acf64 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd95b8a56921454fba97fd4aef35372b : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bff8214c8f1c4ef4b6a403689d223093 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/425bdf82320f45c992aade1ba78cd130 : SUCCESS in 8m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5898342551b647d093621c6dac4c2ec8 : SUCCESS in 34m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/33478118e6aa40c1972db58fc07f6b63 : SUCCESS in 6m 36s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/4917888d5b124a489cec5c00a35c4acb : SUCCESS in 37m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe9e5583c8bbf926a6b782fc4839d2e378b04e7e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-02 14:42:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4f77d9a91e91c842bdcf4630742f44bd11281baa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-02 16:44:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7434029032b34d2d8bf5519d13fb748a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/4c4efc98e18d402ca9e8bd8b6c53c6bd : SUCCESS in 41m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e7a2c2b7fdbe452c86723812fc7dced5 : SUCCESS in 54m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/29df258866184443ab7f0b4747bb3a19 : SUCCESS in 45m 48s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e418591edb644a51a6fdb6b0e2c306d3 : SUCCESS in 42m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2ebe65ead5974f60bad9b62d774c1b60 : SUCCESS in 37m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/043dacf16f4f4f6cab53574494f1b1c8 : SUCCESS in 1h 00m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a872e07b8b1941128a05aced43f37bfc : FAILURE in 1h 01m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cc792e95014446869785ce69673a3db6 : SUCCESS in 1h 01m 23s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/3c84cd918eb64257a3a800f166ca73b6 : SUCCESS in 1h 09m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b473b565bcec49ab8127494457da47e1 : SUCCESS in 48m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d47edbb46eb94bbf9eab1c2d2d8f3a9e : SUCCESS in 44m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15bb463d0ffb45e78c6a3ca2a84fc563 : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e7c7045d3c64422a385557ec0bfebac : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06f419f7149343ab8bc71685ed8460da : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d07ddff6bd6c4cebb38ee9a628224c3d : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5a4ccb47fe0410e8b7b6a211dcf9db3 : SUCCESS in 7m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f9419d0459794f62a5abfbdcfbcc80b0 : SUCCESS in 31m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fcb9c1ad0b8044a181a2cb79994fd790 : SUCCESS in 5m 42s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/576b07fabd794318955b7130a18238a4 : SUCCESS in 40m 28s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c20decdf6e1a3d6ec282f24b2b6511728c9303f2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-07 11:46:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41102ab8a63fad0a8c571af5ec7510f9b71c1bf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-07 13:04:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65fd52270bf448bfa56d57f0109ff329\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3f79f0296e5648edb09ea20f18143e10 : SUCCESS in 49m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/7d52fe118c174b879dffaa8ca87dba18 : SUCCESS in 54m 02s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e99655ffbc5a4cc4aff89c9e7fd8f50e : SUCCESS in 50m 30s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e1122c56ffda433a85fdd2ac686fc4c2 : SUCCESS in 37m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0797f0b643b9495591f5721287b3f408 : SUCCESS in 45m 44s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/844ed2562c034bd6931f4217626c44ef : SUCCESS in 55m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78f2e652604f4b9b9836e4000a5075fa : SUCCESS in 1h 02m 26s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e5c2bfd557bf4581b5c862b0e3a80f45 : SUCCESS in 1h 04m 50s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1e38e38871774047a31ae9af88321918 : SUCCESS in 51m 03s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3510326bd4dc4829b0b88359852cba65 : SUCCESS in 42m 15s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c7c1e06ee671427faba091be6a95f32e : SUCCESS in 41m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d9fa0e2d0c04990ae6b7668c8cc0e55 : SUCCESS in 6m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14dfcb37b6124af48d3dc71a2ea926d4 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6470374042ac498f92bdea880eb76c39 : SUCCESS in 3m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/820b07c69cdd495491969bb22cac3645 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ee781c09ff84ffc8e151d7ab4d62dc1 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8dac7ebdf0b64e86be6a602794f0dd0d : SUCCESS in 8m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/025cf43293a7456892dca1dd0cb89f85 : SUCCESS in 30m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d7566ea99a924dd099e8b997fadd3c2b : SUCCESS in 4m 50s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/07e4df9995d24e6cbde55cbec7de707a : SUCCESS in 43m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"260ce887f1f5eb0316e650b9ffee5d4d080d05b1","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2025-04-08 13:22:47.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c737e1936fa0c0f2b2f58d138983c3b06a90918b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 10:16:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efdb4b7610bf2a35baa310542269af0642dfd78e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-03-18 12:49:32.000000000","message":"Abandoned\n\nNot relevant anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c9acf89d472ffc6749ea1eccc8ab8ead9d434e21","revisions":{"c9acf89d472ffc6749ea1eccc8ab8ead9d434e21":{"kind":"REWORK","_number":1,"created":"2025-02-26 10:47:38.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/79/942779/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/79/942779/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/942779/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/942779/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/942779/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/79/942779/1"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:45:05.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-02-26 10:46:03.000000000","tz":60},"subject":"Remove eventlet debug","message":"Remove eventlet debug\n\nThis patch may make eventlet to complain about our use of multiple\ngreenthreads reading/writing to the UDP socket at once. As we remove\ndesignate\u0027s eventlet usage, at some point we will have to merge this\npatch.\n\nChange-Id: Id85c839385cf51261fb0cb9bbfbcf27741f16525\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9acf89d472ffc6749ea1eccc8ab8ead9d434e21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9acf89d472ffc6749ea1eccc8ab8ead9d434e21"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
