)]}'
{"id":"openstack%2Fdesignate~943192","triplet_id":"openstack%2Fdesignate~master~If23345e24430d1d7fd65b9c02840c844fc680dcd","project":"openstack/designate","branch":"master","topic":"test_designate_gates","hashtags":[],"change_id":"If23345e24430d1d7fd65b9c02840c844fc680dcd","subject":"DNM: Testing designate gates - eventlet removal","status":"ABANDONED","created":"2025-03-03 16:23:44.000000000","updated":"2026-03-16 16:52:24.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ff582345425e39dc3e6e89ada17f37d64f5b310c","_number":943192,"virtual_id_number":943192,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-03-03 17:45:46.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-03-03 17:45:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"7be1cf763f35260417c44b9d88f72c0ff4ed1ee7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-03 16:23:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e310973a2ff95de1d8bd49d14e3c229b7e6b724","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 17:45:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c6e35fc46ed44a68de27a8f627549df\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/61d6304de97b4fc290d56984bbebc7a7 : SUCCESS in 41m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2c0d83b093804fdea7239f39d3ab6ffc : SUCCESS in 1h 11m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e63d23d20c7a4b2ea1266523b969d0e0 : SUCCESS in 1h 00m 45s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2eed944badf047f9853974c09e5d92a6 : SUCCESS in 49m 02s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/591a094b70964133a84a38f52697e1bc : SUCCESS in 1h 05m 48s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e7af5014e3fa48a58ab205c502c5abdf : SUCCESS in 1h 15m 14s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/46820bcc8053435db5d69e626db7f583 : SUCCESS in 1h 08m 23s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3a11f3d574b34f62ae29a111ffeb1d02 : SUCCESS in 1h 07m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a844bcb231b34794adfc20f202d1188c : SUCCESS in 1h 09m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/887f4e9b99aa4f0996f41cf53b4ece5f : SUCCESS in 59m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7e9da9b846814a19b37ccffd97f45993 : SUCCESS in 53m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ac336d136ec43568741c304459455fe : SUCCESS in 5m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/13a0ada7d76245aa82f8e379fe273b58 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3842ab54eff949409e846d14f45f8fd7 : SUCCESS in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fbe466087f8e4939a278fe7c8e025016 : SUCCESS in 6m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dadfa1a3a674558b7bcfd298858fb4b : SUCCESS in 6m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d34c0fcb0c004aee9c92b578cf36465e : SUCCESS in 37m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f76aab91bd744ea1bab65d3f533aeca0 : SUCCESS in 5m 51s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ea777b87a6014f2dad27bfcc3c9c8a6b : SUCCESS in 52m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"340165658b786a52b0a9b0b5c99a0db2ebd9619e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-03-03 18:15:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9d626558ddfd2fe67388624bfab84eecbadebacf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-03 19:54:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/585d19b5fcd74cb58c0224cbe343cf0a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/909c0d4ec6434b48b6197dd24d076bff : SUCCESS in 52m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/eade7470bd95439fa073623a028ca491 : SUCCESS in 59m 56s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/db06e8ab19bc428ebdb701ae1abfe18a : SUCCESS in 1h 01m 08s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/bcb7abdb3a024f82a0ab0a70a00a1951 : SUCCESS in 41m 37s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3b8aa179a6e446039dd8e23450cad411 : SUCCESS in 51m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b26fbd4c251b40cd91717830d19b05f5 : SUCCESS in 1h 15m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1b273ccc8963405b96ef785a54d929d2 : SUCCESS in 1h 14m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a6fb9446e0c74c9cb0ddfe238cb1353b : SUCCESS in 1h 12m 17s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/13c29b1d19f04f9583a7a08748e46d71 : SUCCESS in 1h 14m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c552f8880ce046a6a108b9a271d74715 : SUCCESS in 50m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e47ce0df952d44858096cc704ea06c74 : SUCCESS in 53m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e94af2ccf0b4247b5661ddf2eac18e1 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dd736fadf1c4a549d530ac268ec83f2 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b722a7709fb45b390871c9e0fc2d825 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dff5b81c956645d498c0235ac8c2bc05 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2990a3779aa44a3eb4ab60e00656f9ba : SUCCESS in 6m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/30b0c19df2f24b27ace219241143fa16 : SUCCESS in 36m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1994a568d8e04f4c96144f04a95d602d : SUCCESS in 4m 36s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3f518654800942daa9b4f1eada92a26a : SUCCESS in 1h 08m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff582345425e39dc3e6e89ada17f37d64f5b310c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-03-16 16:52:24.000000000","message":"Abandoned\n\nNot relevant anymore","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7b697bdf12f51dad4aa4df38475aabb0b1f72286","revisions":{"7b697bdf12f51dad4aa4df38475aabb0b1f72286":{"kind":"REWORK","_number":1,"created":"2025-03-03 16:23:44.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/92/943192/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/92/943192/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/943192/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/943192/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/92/943192/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/92/943192/1"}}},"commit":{"parents":[{"commit":"05ffe294ec8ad7810886ee6fb51a379b9edb654c","subject":"Merge \"Support \"-\" in region name for FIP\u0027s\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/05ffe294ec8ad7810886ee6fb51a379b9edb654c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-03-03 16:21:33.000000000","tz":60},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-03-03 16:21:33.000000000","tz":60},"subject":"DNM: Testing designate gates - eventlet removal","message":"DNM: Testing designate gates - eventlet removal\n\nAs I am testing multiple eventlet pieces removal, I would like to\nhave this kind of patch where I could run the gates a few times and see\nif I am getting random timeouts without modifying eventlet parts.\n\nChange-Id: If23345e24430d1d7fd65b9c02840c844fc680dcd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b697bdf12f51dad4aa4df38475aabb0b1f72286"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7b697bdf12f51dad4aa4df38475aabb0b1f72286"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
