)]}'
{"id":"openstack%2Fdesignate~946263","triplet_id":"openstack%2Fdesignate~master~I1c00c3789b3411ee657b2fef53eb5c1aa32083b7","project":"openstack/designate","branch":"master","topic":"restructure_ci_jobs_def","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-09-09 09:28:27.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2025-05-07 17:39:25.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_13252\u003e by clicking the attention icon","reason_account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-09-08 17:18:51.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-09-09 11:04:06.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1c00c3789b3411ee657b2fef53eb5c1aa32083b7","subject":"Restructure Designate CI jobs","status":"MERGED","created":"2025-04-03 14:45:25.000000000","updated":"2025-09-09 11:06:41.000000000","submitted":"2025-09-09 11:04:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"946263-restructure_ci_jobs_def","meta_rev_id":"ca3e4e4ac1508b4577cf88715e83d8fa043c8b0f","_number":946263,"virtual_id_number":946263,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-09 11:04:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-09-09 09:28:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-08 17:18:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-09-09 09:28:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-09-09 09:28:27.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-03 14:45:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-08 13:05:27.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2025-04-08 13:05:27.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-04-08 13:05:27.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ec403ecd8d8f40551748b2aeb32f47b020a4e5b3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 14:45:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"85d4722b10cbbddc53df3381f9f0821e34632d32","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 14:45:47.000000000","message":"Patch Set 1: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/designate on branch master.  The\nproblem was:\n\n  found undefined alias\n    in \"openstack/designate/zuul.d/jobs.yaml@master\", line 80, column 23","accounts_in_message":[],"_revision_number":1},{"id":"41a4e766085ffcbce319a75cedf83eebdf636bb5","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 14:48:46.000000000","message":"Patch Set 2: Published edit on patch set 1.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"745e47e27a7d60b93e23fed426ad9db21771a53f","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-03 14:49:54.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"07c014d0b9027d5248a1e73f383510b41ab9c4df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-03 16:03:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43d2493738df44d18883814f6523a9fd\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d82155d6c45d4dc8915a9202c741ba4e : SUCCESS in 1h 06m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4c885dffd50a4fe5a5fea349ff1e4bef : SUCCESS in 59m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/03329097d2744b6b8d2ae4c2f25472da : SUCCESS in 1h 06m 51s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/83f08e9a464e4619a8c2dec25d6c90e0 : SUCCESS in 55m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ecf6607839bd4ae19b7093599a92d01d : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f97f7b41b3743a5b64896559ff06203 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0e16d423184e4104918f32d760f98039 : SUCCESS in 2m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57e6c853a6dd4a8aa83d4915cdddee5f : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b5f2ef71bf7d497a83d4078816e9a13d : SUCCESS in 4m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2210c03f956d43c29ff8a1cf7fecbcbf : SUCCESS in 9m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/26ae99c26e33457bb080dd672a6a0fbb : SUCCESS in 21m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5ce6f31d516c496bbfc389000602f36f : SUCCESS in 7m 07s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"1aa4d7aac65d80e1172e356496df01beb83a6ff9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 10:38:06.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"e3ff735bdfe5a5d16c44e13bab2d227b20bdb86d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 10:38:14.000000000","message":"Patch Set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5d96ef1a8f6f781fcfc2a2a035a7dc30ce930514","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 10:38:16.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"6de1187039c28910ee3b97d40caf248c68e1b155","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-04 10:39:39.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"167e2923a1cc78f2d47e5a6b5ae1b762771a249d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-04 11:53:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29635dfecf7c424393decde7b56ad21b\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2e07c980e96d462280c33f6c97f30db3 : SUCCESS in 59m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/636ed8ff243f48bca74331564ff84384 : SUCCESS in 1h 09m 07s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cf7e8823b8144b3d89dba983fe7da0e2 : SUCCESS in 1h 00m 30s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/33511e4f15c84566956187529da14e06 : SUCCESS in 1h 02m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52dc53615c384766a6d31884e9d90828 : SUCCESS in 6m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb24b3e824d746daad86835619db02be : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/db264d7a165e4f49ab4353c1172eeee6 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/045e2c914af7427d993b45e909616b56 : SUCCESS in 4m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/562913797a984cd196da6a46492adb27 : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1219c04c70f41d889daad039b7c2d3c : SUCCESS in 8m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c1b890d3f5c844edbed5b39d9a183ee9 : SUCCESS in 32m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/da3c0382121a448ca2c8514ffdada503 : SUCCESS in 4m 06s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"853a1486537f70509250e2a64b9378182ade0b85","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-04-14 17:38:49.000000000","message":"Patch Set 4: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6a3076e845b770b53aea66e7850348bba2bd1cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-07 11:38:15.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"3dee9b3b3e75488bb6e2f9329b2d6e368c5cb0c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-07 11:38:49.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"a4d69818c08d8a38d24bf083b341f87953ea29d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:47:23.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d526691a775d4c8fa62d9a124fdc0eec\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/36fcec19f26c4e7d929664082ffe7015 : FAILURE in 52m 27s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/20b318ec0b31467baf83aabd1f866ac7 : FAILURE in 46m 25s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/645b67a5e4b647ee9dbbbba9ca1d0a50 : SUCCESS in 1h 07m 45s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/79b0463dbd584cc18848234eb8b6912b : SUCCESS in 57m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d17d7ca7c104bf4be7c203c7af83df8 : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f53493196331431899c84fef43c11c42 : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5833b663bf17432088c729c734bd8f09 : SUCCESS in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4cbeb09f73f54a65bca301544601b6f6 : SUCCESS in 6m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ca47f422c5046b8bfe0ceb064e06ea0 : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1edf41628715442780ed90293d5bd9e8 : SUCCESS in 9m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1b4b4b8a000a45d99c5c1b8943614fff : FAILURE in 25m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0f9496b714a6497dafe6adb34e39de67 : SUCCESS in 2m 52s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"1c0bc582a40e41d4b41dcd05259d11602b1b7190","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-07 13:31:21.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"56a965dde0d0c39c3d353c52ec1d99818e9e22b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:34:05.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/617d4e840d9546a7b81fd491a72d59ca\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/03ad70368bd247359c6f63ceaa1bf107 : FAILURE in 37m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1830cbff5ab54f08bc973f704c5bf145 : FAILURE in 39m 49s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4a174ddf73e84287a085f6e470b5091c : SUCCESS in 57m 59s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6cf30cbf1be3480f9637c294f0c0b531 : SUCCESS in 53m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6843e20041a744059bfecb1a9b71d784 : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5cb0c4f1c8b42fa9e2316f8e47c95d9 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/70de8d3c30364454a4a5bf16074dac10 : SUCCESS in 5m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de0bd52759d143208bc3d1235470b8bf : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2deeb9c809cf4ebbbbd410cd27dcf2b1 : SUCCESS in 4m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1bd4bf7e0ccf4adfb8343b33143387e6 : SUCCESS in 7m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/988f81889f624802a770889c1d878d7d : SUCCESS in 40m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fa568414388f432ab6a0624b68a9cfea : SUCCESS in 5m 11s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"87d253696227f7dbf9464e9bf66ac0c00b4aab1e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-05-09 09:17:33.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"d9fa035ee3a941bc26cb8a158f857b6c4917a36e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 10:25:55.000000000","message":"Patch Set 6:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c975360170a2421d98b71c969e7d08b3\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b2dd793d1edc4593a81a12c4eddf6972 : FAILURE in 45m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/896ad53bad19480283b7b23657ca77bb : FAILURE in 20m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/588da741b57242a589945647e8808838 : SUCCESS in 1h 02m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e4a4497abe8b42f2ba725226be80ef49 : SUCCESS in 50m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/add0a35ddd0e465b9e9f1ad82c0ed58f : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0fdbd942caba4809bd7e1b55aaf7a016 : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0995cc3bfabc4ae5a082a5bc324836ab : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b77b9522be214307923ce4274331919a : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a572088d7ab04a7fa6b8e3d2ddafffb2 : SUCCESS in 4m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a23145041d346d4bfad138bd73e6f24 : SUCCESS in 8m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ec49f9e4dce44601a2a4a9c74fb33819 : SUCCESS in 19m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/22914c63be084e0b926d9c0bd7891a56 : SUCCESS in 4m 44s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"4d76d7d07e4fe4f4aeb45ed4a3007cda59723379","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-19 09:59:57.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"36247ed5848eb844520b5c830355f8001d7c4a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-19 10:06:23.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c7760e6acb85346ea7804e20ecfca4266fb7277a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 10:57:23.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/35c1686569b64b599e49e570fc2b3140\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e0c5330d2b6744c6860da334f39ab080 : FAILURE in 43m 16s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8ce245f627b74b09bf46bc2bb0b8730b : FAILURE in 43m 16s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b85d06ad61e643b39e067217a64219f2 : FAILURE in 21m 48s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1ec15cc9181c47c88d463f8d2473b4f6 : FAILURE in 21m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac559ce04e124261bd643b43d1b77c4d : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6d99e2da04ce4ee79a9f38a981c7a126 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/522ab4d8f04849c995d2e9e2080d3d76 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/74ce5ec9a52a44e884e936333c570997 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3394822e625848d0be27a8f572ca9a1c : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a932fd161fc046abb3f395b7ba52d639 : SUCCESS in 8m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f8ce6ef2ed2d4a47823dae4de11d4af9 : SUCCESS in 30m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a42628103a534130b5c1faabbd16e605 : SUCCESS in 4m 36s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"0291b7c6338ab12d54718ae35a04313c6f073247","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-19 11:16:47.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a80fefe98ee04143a0f93fe515b5c3aacc8a728a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 12:10:05.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5a0d68285cb4271ac26c73733561dc5\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/49aa3ff4569344979d2997fb1d0dc0d6 : FAILURE in 45m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c8064ad680c54685987afa95bec37d52 : FAILURE in 35m 51s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/7f118959dcad422b80c541f32b6af0d3 : FAILURE in 24m 28s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9ba16620cac348cfae66b8f589f6d909 : FAILURE in 17m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/73a22172d4874bd594f6b1f9b1a7c694 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a52c95313f4733942a2d20dfdda0f9 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/acd671f1f3174e8fab25844885580c2d : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/42b8d53bfc434656bba8d2a3547e986f : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/599bc543eaf843d68e8f28967ec94b36 : SUCCESS in 7m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cd04490b20f4b1ca4e9b8faaa36b033 : SUCCESS in 8m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e2947ccadead46f8ac2b895cc1285850 : SUCCESS in 36m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5ef83a47b7ff4dcc92338ed32ea16190 : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e40252a7d25df5a531c8d6ed71504ee046cf90d8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-19 15:50:45.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"c10ef1206d4341fbf469a58fe7a525a49221874b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-19 16:12:05.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"4b28b25ca232f34458968f8bd3915311a1ac970e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 16:56:51.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3287243ee6aa478dbd153734b0858dda\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/53b4d60ca8e9489386a04c743a52a729 : FAILURE in 35m 11s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5ec537bf03ff4caa8a3aaed2133844e7 : FAILURE in 42m 47s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2e5b246cab6748fe9bcaf8bddd782cb6 : FAILURE in 21m 48s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cbdd0014ff034eb985627b94b2d12a42 : FAILURE in 20m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a954285211c8457fa0c22088f397e602 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f8b79b6421d64db4bcd8584e912d0ae7 : SUCCESS in 5m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e95857d67cc24c1eba606dff6162860e : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/38a0551f6a5b4e73bae346026a457968 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a39e47226d84c22b55e469ee407f17a : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8233b952a107451d8abc6184b7303946 : SUCCESS in 9m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/163302ee78b24531b541e4dc503de05f : SUCCESS in 37m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f8dba9efd759454d8a2b60b1a29108a7 : SUCCESS in 4m 37s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"755b5223c86c7b54a7ba421898cd03b41abcac0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 12:50:10.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"91caf48ba0bd9b131b61a0167489e3de3669adf4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 14:34:33.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4513e717f2be44f6825ab3750da49cda\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/87568128e6b641c597427dc83a908183 : SUCCESS in 1h 10m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/857ea4b2c869443cb1b8ff6e7c980102 : SUCCESS in 1h 03m 02s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6153fa35347e46528df5ffbd320237dc : FAILURE in 18m 18s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/3b34733ab0894a9ca87ae1241842b824 : FAILURE in 18m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a772a9da31045968da63f30d2bf7e05 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33d0fbcb05c848f5bb459bacd05499f3 : SUCCESS in 4m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5a19771a8c68438a8d4a9662e40b670b : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fd20182dfb9400d9657560aafbf61b6 : SUCCESS in 6m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/02990905d7f849e4955264ee22436707 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3642902e8cc54cf287c9a7b8f525aab5 : SUCCESS in 10m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/05f0bdbeb4d8491fb29b8783b6085fe1 : SUCCESS in 42m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/080d9b5532da4236a20e3ad4e61491bf : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":10},{"id":"e2f098f51b7ee1081189ba400b989fb569c74573","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 14:34:59.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"7bb304b4cedf0ff9bea4a198abeeb6aaceb37162","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:55:23.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/337d844393444a129fefcdcc6b8cba51\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5ca365c2ae084a54958fd0b75d066617 : SUCCESS in 1h 05m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2e512827be554b6e8f582439fddfd424 : SUCCESS in 1h 03m 01s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3998ebda508348bba2a7eaa0f534a7fa : SUCCESS in 1h 00m 39s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a7250649796644dfac8130159a95e3c9 : SUCCESS in 1h 03m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/85919eced5e3455097bd4123e80d8e9d : SUCCESS in 6m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f3c3b58a2e545a69a070fad99f470ba : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8b323335bc1c4630b4956828ccb7784c : SUCCESS in 3m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d42babe743b4761890629f8660c50d0 : SUCCESS in 4m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2970e9651f0d4fccb2194b001e7db1b6 : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/83c4b7ac140e494088ab0df76ceafacc : SUCCESS in 10m 07s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7e68badbc5941fbaf1766281086ef3e : SUCCESS in 34m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4cdd8c249fb945e9b7f68fc0f826d7f1 : SUCCESS in 6m 30s","accounts_in_message":[],"_revision_number":11},{"id":"d66cb55665c19aed662728e42b44f1e3fbb17c3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 16:58:27.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"76080a0a77b97fa81b35d83eaaf00f77a3825dfb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 16:59:07.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c0e33bebdd82985610f3c7c2e028ac3ce227eea7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 17:12:12.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"9e50864d48d26451ab47f01608a9f9ea7eb8c197","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 17:12:56.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"8ddb07fcb6be0345e136f9776735587b11e84c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 18:39:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/87f76f0673764084af86a1feedbc8634\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b3b90e25a85f465d9ee42975e2121f44 : SUCCESS in 1h 01m 37s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4356e99f45884b139a8eb6bbae2ae44a : SUCCESS in 1h 10m 19s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/528da2378a41481f8a77a94b9071bcf3 : SUCCESS in 1h 06m 49s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/006771e23728433db07d124dbb1352f8 : SUCCESS in 1h 02m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/867d0317521b41fea6a7cd7afcd96466 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fc01dda18d64a48a16c193f6d013b1a : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d95f60e17f4f5ca263947eb41314a9 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cde7fd1d31334459b347001997807bba : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/87054fa5cb114699ab7a531f17dd75c9 : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68cce43e992148aca07ffaabe2ef8220 : SUCCESS in 8m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e2e9acdd463e4b359aecbd6378f5fa21 : SUCCESS in 31m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1ce4cb61cc7b47d982f1a4b4839a2b4a : SUCCESS in 4m 50s","accounts_in_message":[],"_revision_number":14},{"id":"ae015e2f4b9976d510c7cfebdf1bb19e8dd27dd5","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-18 09:21:47.000000000","message":"Patch Set 14: Code-Review+2","accounts_in_message":[],"_revision_number":14},{"id":"0222dfe9d4d5fffd571bab476afc2290d0ee3ece","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-08 13:44:06.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"12a9d06319c627b0316aeeb5319b7938509aef11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-08 13:44:39.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"56b4917ee39bd5f61da523bbe9f1c123bd97a68b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 15:09:51.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c73063e6b224a14a2f78dddd3967771\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9d3e35a21c5e48acaa64fa7369a4bf03 : SUCCESS in 1h 04m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b2cf9ebf95f04f8aab39ae7d07fd7740 : SUCCESS in 1h 07m 44s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2450092782144c23bdd99ae3b95d7b76 : SUCCESS in 1h 00m 16s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cce45042c6214fc991fc3e9f93d4c443 : SUCCESS in 52m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f505727a19914457bd014e23bd0114d0 : SUCCESS in 6m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad4f32d864de495990cd3ef26c0a2a27 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/93104ea30516414eb53be1bfccf27670 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb488e4559d64f02b0d63a8280dfa43d : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f9226a127a34d2a8ac7d5ce59f7c7d7 : SUCCESS in 4m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9498c9196034289af7555c5f247c94b : SUCCESS in 10m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8b2e07db23494dceb0292e198ab913f7 : SUCCESS in 30m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8b9b907942a8440896597dabf5be6c5d : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":15},{"id":"387203a3c9c426be9dd038d9a2e5e53eb7c7003b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-09-08 17:18:51.000000000","message":"Patch Set 15: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"bdc58d6594d9445ca3540055b51525cc0885f469","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-09 09:28:27.000000000","message":"Patch Set 15: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ae3c513f6ea27b9e57a57f002283af27d6f1769b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 09:29:04.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"b88d440be72f7f312c721db3101703e8b2b8f10b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 11:04:06.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49ff95b635b54b2ca1e35020e5506440\n\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5ce574d89bd14926895a816bf5f6a7bb : SUCCESS in 1h 01m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4f94e0e1da5c4721bf58c7e06980d75a : SUCCESS in 53m 16s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f04c12e8ddfe448fad471dec85928f1a : SUCCESS in 1h 29m 10s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/862cb2154c9b45adb9a22de8859670b0 : SUCCESS in 1h 05m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cecaf8e191724df080a4aca99c6df042 : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f59339e379e94fd1b3c089980a2d9182 : SUCCESS in 4m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9507224226140a0b47d902443959cb1 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25bc7bfccfaa424eadabf2b9e4704559 : SUCCESS in 8m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/852e33f836dd41748e4847d533fb6394 : SUCCESS in 30m 30s","accounts_in_message":[],"_revision_number":15},{"id":"50e962666685927aac100a68995cab093da0dafc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 11:04:06.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"ca3e4e4ac1508b4577cf88715e83d8fa043c8b0f","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 11:06:41.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ecc0ba0cbc14142940f71c5808af23d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e75acb36d4eb4ac1acbba7ae2c2b92c4 : SUCCESS in 1m 45s","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"035453ffd1ba0f574a1d7f23b458222c0fd8fc68","revisions":{"3fa1f7530a9b5f98bc3f6ed806f6e9e34b5b19ed":{"kind":"REWORK","_number":1,"created":"2025-04-03 14:45:25.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/1"}}},"commit":{"parents":[{"commit":"595894be2202bfd98805a0ed87281e69c59b5c1d","subject":"Switch the devstack upgrade to use the member role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595894be2202bfd98805a0ed87281e69c59b5c1d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:44:32.000000000","tz":120},"subject":"WIP Restructure Designate CI jobs","message":"WIP Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fa1f7530a9b5f98bc3f6ed806f6e9e34b5b19ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3fa1f7530a9b5f98bc3f6ed806f6e9e34b5b19ed"}]},"branch":"refs/heads/master"},"1aff22acfa46bedc5c87e69a78a28e6771de3c64":{"kind":"REWORK","_number":2,"created":"2025-04-03 14:48:46.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/2"}}},"commit":{"parents":[{"commit":"595894be2202bfd98805a0ed87281e69c59b5c1d","subject":"Switch the devstack upgrade to use the member role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595894be2202bfd98805a0ed87281e69c59b5c1d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-03 14:48:46.000000000","tz":0},"subject":"WIP Restructure Designate CI jobs","message":"WIP Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1aff22acfa46bedc5c87e69a78a28e6771de3c64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1aff22acfa46bedc5c87e69a78a28e6771de3c64"}]},"branch":"refs/heads/master"},"a1823ccf2128ff6a36984643cb413bf11489218e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-04-03 14:49:54.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/3"}}},"commit":{"parents":[{"commit":"595894be2202bfd98805a0ed87281e69c59b5c1d","subject":"Switch the devstack upgrade to use the member role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595894be2202bfd98805a0ed87281e69c59b5c1d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-03 14:49:54.000000000","tz":0},"subject":"WIP Restructure Designate CI jobs","message":"WIP Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1823ccf2128ff6a36984643cb413bf11489218e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a1823ccf2128ff6a36984643cb413bf11489218e"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9f56d92c49c9bf2259014a14fa4da9a5d0179414":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2025-04-04 10:38:14.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/4"}}},"commit":{"parents":[{"commit":"595894be2202bfd98805a0ed87281e69c59b5c1d","subject":"Switch the devstack upgrade to use the member role","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/595894be2202bfd98805a0ed87281e69c59b5c1d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-04-04 10:38:14.000000000","tz":0},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f56d92c49c9bf2259014a14fa4da9a5d0179414"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f56d92c49c9bf2259014a14fa4da9a5d0179414"}]},"branch":"refs/heads/master","description":"Edit commit message"},"3408200073c80a9c3e849f25d4d1ab03f632acf6":{"kind":"REWORK","_number":5,"created":"2025-05-07 11:38:15.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/5"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-07 11:37:36.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3408200073c80a9c3e849f25d4d1ab03f632acf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3408200073c80a9c3e849f25d4d1ab03f632acf6"}]},"branch":"refs/heads/master"},"a5bbac910897e563d02587914323d28f5c9f4d43":{"kind":"REWORK","_number":6,"created":"2025-05-07 11:38:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/6"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-07 11:38:42.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5bbac910897e563d02587914323d28f5c9f4d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a5bbac910897e563d02587914323d28f5c9f4d43"}]},"branch":"refs/heads/master"},"92bea4fa3d291c3cab3ca42bafe6f9576f1b9bc5":{"kind":"REWORK","_number":7,"created":"2025-05-19 10:06:23.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/7"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-19 10:06:14.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/92bea4fa3d291c3cab3ca42bafe6f9576f1b9bc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/92bea4fa3d291c3cab3ca42bafe6f9576f1b9bc5"}]},"branch":"refs/heads/master"},"629bf1c594b411bd4d81372c4c028c900a900f52":{"kind":"REWORK","_number":8,"created":"2025-05-19 15:50:45.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/8"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-19 15:50:38.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/629bf1c594b411bd4d81372c4c028c900a900f52"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/629bf1c594b411bd4d81372c4c028c900a900f52"}]},"branch":"refs/heads/master"},"3849c35769f1af9d9784dbd323ddc6feb115a72a":{"kind":"REWORK","_number":9,"created":"2025-05-19 16:12:05.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/9"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-19 16:11:26.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3849c35769f1af9d9784dbd323ddc6feb115a72a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3849c35769f1af9d9784dbd323ddc6feb115a72a"}]},"branch":"refs/heads/master"},"f5fd942fcafa428fd3282dd35176b2946c6bc8bc":{"kind":"REWORK","_number":10,"created":"2025-06-03 12:50:10.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/10"}}},"commit":{"parents":[{"commit":"a395fd3ede778f139c3588958473b4c07c98593f","subject":"Merge \"docs: add OpenStack-Ansible as an installation method \"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a395fd3ede778f139c3588958473b4c07c98593f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-03 12:50:00.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5fd942fcafa428fd3282dd35176b2946c6bc8bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5fd942fcafa428fd3282dd35176b2946c6bc8bc"}]},"branch":"refs/heads/master"},"61658eca0953a75b9e515f1247ed2331485444b9":{"kind":"REWORK","_number":11,"created":"2025-06-03 14:34:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/11"}}},"commit":{"parents":[{"commit":"a395fd3ede778f139c3588958473b4c07c98593f","subject":"Merge \"docs: add OpenStack-Ansible as an installation method \"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a395fd3ede778f139c3588958473b4c07c98593f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-03 14:34:52.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61658eca0953a75b9e515f1247ed2331485444b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/61658eca0953a75b9e515f1247ed2331485444b9"}]},"branch":"refs/heads/master"},"2c3f95ca444bfba4fce519637939d668437ffa04":{"kind":"REWORK","_number":12,"created":"2025-06-03 16:58:27.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/12"}}},"commit":{"parents":[{"commit":"82652559ea8641b11c4227f0e9cf273b4c567b96","subject":"Merge \"Drop explicit executor argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82652559ea8641b11c4227f0e9cf273b4c567b96"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-03 16:58:16.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c3f95ca444bfba4fce519637939d668437ffa04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2c3f95ca444bfba4fce519637939d668437ffa04"}]},"branch":"refs/heads/master"},"9f62ce6ce3ac5fb07bea2560414db9d4e2029c1e":{"kind":"REWORK","_number":13,"created":"2025-06-03 16:59:07.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/13"}}},"commit":{"parents":[{"commit":"82652559ea8641b11c4227f0e9cf273b4c567b96","subject":"Merge \"Drop explicit executor argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82652559ea8641b11c4227f0e9cf273b4c567b96"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-03 16:59:00.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f62ce6ce3ac5fb07bea2560414db9d4e2029c1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f62ce6ce3ac5fb07bea2560414db9d4e2029c1e"}]},"branch":"refs/heads/master"},"6e6a996bf170e29a7bcb7286c9065bedca5e2f2f":{"kind":"REWORK","_number":14,"created":"2025-06-03 17:12:12.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/14"}}},"commit":{"parents":[{"commit":"82652559ea8641b11c4227f0e9cf273b4c567b96","subject":"Merge \"Drop explicit executor argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/82652559ea8641b11c4227f0e9cf273b4c567b96"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-06-03 17:12:03.000000000","tz":120},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e6a996bf170e29a7bcb7286c9065bedca5e2f2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6e6a996bf170e29a7bcb7286c9065bedca5e2f2f"}]},"branch":"refs/heads/master"},"035453ffd1ba0f574a1d7f23b458222c0fd8fc68":{"kind":"REWORK","_number":15,"created":"2025-09-08 13:44:39.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/63/946263/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/946263/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/946263/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/946263/15"}}},"commit":{"parents":[{"commit":"3bf4115271f99d87066d75cf99d3c3b28fa18748","subject":"Merge \"Migrate designate-bind9-keystone-old-default-roles to Noble\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bf4115271f99d87066d75cf99d3c3b28fa18748"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-03 14:27:29.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-08 13:44:31.000000000","tz":0},"subject":"Restructure Designate CI jobs","message":"Restructure Designate CI jobs\n\nSo far the Designate and Designate tempest plugin CI jobs definition\nwas different than most other OSP components. Some of the jobs were\ndefined in the Designate repository and other in the tempest plugin,\nwithout any clear indication to the correct repository for the job type.\nBy adopting a modular approach similar to other OpenStack projects, we\u0027ll reduce maintenance overhead and allow contributors to focus on\nenhancing Designate rather than debugging CI structure issues.\n\nGoals of this restructure process:\n  * Create a clear separation of job definitions between designate and\n    designate-tempest-plugin repositories\n  * Organize configurations into multiple files instead of single,\n    monolithic one\n\nCloses-Bug: #2106143\nChange-Id: I1c00c3789b3411ee657b2fef53eb5c1aa32083b7\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/035453ffd1ba0f574a1d7f23b458222c0fd8fc68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/035453ffd1ba0f574a1d7f23b458222c0fd8fc68"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
