)]}'
{"id":"openstack%2Fdesignate~946752","triplet_id":"openstack%2Fdesignate~master~Ie6526675f28e599a4b7487039f85bab75031da29","project":"openstack/designate","branch":"master","topic":"ptr_records_use_designate_db","attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-09-10 12:09:12.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-09-10 11:49:38.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"Ie6526675f28e599a4b7487039f85bab75031da29","subject":"Add a producer task that removes stale ptr records","status":"NEW","created":"2025-04-09 10:37:39.000000000","updated":"2026-09-16 11:36:15.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":18,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"a216357cf9f5b87303121ae7bb6c0fe33c7a809b","_number":946752,"virtual_id_number":946752,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 11:36:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-09 11:48:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-23 13:32:33.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"81b782c15fa6214ce65df3274e423eb0081de873","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-09 10:37:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a1d2de9280cfbe6f2475f7386140b6c1333826ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 11:48:42.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/77a643b9f6c341c1afd700a9155075a5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6732ecc28f2c4d3bbcf7789ac1c8f69d : SUCCESS in 2m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9331658d1eba43d69473591e16795526 : SUCCESS in 39m 48s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/6620b841e15343f4aa183c2a35748083 : SUCCESS in 53m 34s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1067ba2670ae4bf29deadf952e903ae1 : SUCCESS in 48m 54s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/14bd52f656db4cc2a8a76d9a2ad4baa1 : SUCCESS in 43m 36s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/30a9644b638e4047b6a9d72872ae4709 : SUCCESS in 47m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d62b98f084734697996b0eef2698494e : SUCCESS in 57m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c498b51bb07d4beb9022c9d492720c0a : SUCCESS in 59m 09s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/615e7098c90d4e599506afc7c41d6421 : SUCCESS in 55m 52s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2863db55e69d42a2aaa7b3309e7fb959 : SUCCESS in 1h 06m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/95b4ceaebc744887974ee7aae5f906f5 : SUCCESS in 41m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9d22b275c3294d2b89dbc697db75e0ab : SUCCESS in 42m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4aaa493468f04d1e9529749baac60364 : FAILURE in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3bd17a85fc044ad825e611e445418cf : FAILURE in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3100e7192284d29a1c75b1e19dc3d30 : FAILURE in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d8fb47a72dc34669aa5766446f5e8e00 : FAILURE in 6m 26s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f13b493cc444944b55e6916e9182f6d : FAILURE in 6m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4aebe23ff644fa486d38620e488d6e1 : SUCCESS in 9m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b73597deb5964ae09a7b3fd86403ac5d : SUCCESS in 41m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/aa359f84144348bc880db394d885fd60 : FAILURE in 5m 10s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b33799cb1ee14531b429bdac7986e542 : SUCCESS in 44m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b4e4c9bdcd7473f4bf9029dd529aa8f82bb910d4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-15 09:59:50.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"f8140d535486a398e0eafb63f4a91921142d2cf6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 11:12:03.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/839f09a9bad449fc95dd8b674dc9ce1e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5ac80ec78424bb79a7ae49200d69d46 : SUCCESS in 2m 35s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d6552d32bb0f45868a888c640ad37b3e : SUCCESS in 40m 00s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2476ba50ddc947cc8368f8fd05b2c66e : SUCCESS in 49m 59s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2e62eb9f17f4435da04fe0ee66e20e6d : SUCCESS in 48m 38s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d792ddaf618d401bb9e7a07ff1c1da8f : SUCCESS in 44m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0b8e8709deb444d784d105fa3ae0a6dc : SUCCESS in 42m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e36f5171414f4c8b9c269b2836d46b23 : SUCCESS in 1h 04m 53s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6faa11b147984a4a9afceb9ac53deb6a : SUCCESS in 1h 06m 58s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e3f3fdab2ee2474594a0ba08490000d0 : SUCCESS in 1h 01m 01s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/712935e97127411ba7490bdb74538ae6 : SUCCESS in 45m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/48e84b3b06e844b2a5f035795bf76341 : SUCCESS in 45m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/880b2f78ed0e4a6498b36ad4ffc06c09 : SUCCESS in 44m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa30d0d2fe494331acdfd255b9da667c : FAILURE in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6316311407344480acbea2cb54c60e7c : FAILURE in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e70322187854407cad9da81b9d38de74 : FAILURE in 5m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8918e21d06ce4a718418b501e866a51f : FAILURE in 7m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af064423f04f4344816a9d815a34d780 : FAILURE in 10m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fe660caa88f4c20bc44ca6b2ddcb640 : SUCCESS in 10m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c62253d0d44a4df0a8b6d1e001dcc7a4 : SUCCESS in 45m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7697c287d4544083bd0638a8a24790f2 : FAILURE in 7m 06s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/74f12cd778b34253934ca8d0095425ab : SUCCESS in 43m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8800ad01b1055cf801da68f3fc2232db31dae4b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-15 16:44:58.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"56f25568537908bad7c5e3ae115512e0f08a3fe2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-15 16:46:46.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"7fad8b84d449eeefdd689b40409ad15f6ce42b4a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 17:57:14.000000000","message":"Patch Set 3: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3b26eb85695547ada681c0617d4b65b8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3ff93c62298b4ccab89b60bc4a4e5826 : SUCCESS in 2m 41s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e55d34a0afaf4ca4b01b0ee3404957c7 : SUCCESS in 42m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/22fdde3c1d5849baa40c1d16e526604e : SUCCESS in 45m 30s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d2c3e8fb37ab43b594b283caf64894ec : SUCCESS in 52m 58s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d8b6ed8f9ff84a1b8bfa0a209ff97e04 : SUCCESS in 40m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ee8013c4fbcc4ac8a47d195a8b773755 : SUCCESS in 40m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ea4f00f7419c4e32a5de4832393876a0 : SUCCESS in 1h 03m 40s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3aac1b9712c2407c9817ac772b3b7531 : SUCCESS in 48m 50s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d9454b22f3bf4fee8b1ba69156c93ad0 : SUCCESS in 59m 59s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/04bc73e749404695bbec8f173ad398a8 : SUCCESS in 1h 10m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/8aa8b878742d4eed80092495fde60873 : SUCCESS in 45m 58s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3b40568d3c434692a2abcd1a5d0be842 : SUCCESS in 38m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/274f4c48ff0d4d81affba46906107edd : FAILURE in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8479497e66284896b0212487b18696c8 : FAILURE in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8eb4d3d1e2db4405b7dfbc07d1b752cd : FAILURE in 6m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fb61b0cfd8ac4f57ad51a7bb3b1ecb36 : FAILURE in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9a0c67b5190f46e9add70b9621f12bd4 : FAILURE in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdfae24dbf5f4e47b06786cdeaec12c4 : SUCCESS in 10m 32s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f91e8edb72ce415ca0bdcd66c5ca090f : SUCCESS in 43m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6fc58396c1fa4299a43f33e5d97cfd1c : FAILURE in 4m 53s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5a0199ef674046aeb5ffc48614307e2c : SUCCESS in 48m 47s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bf01d0f42e4f4da65356314d97ac5f08d45acaba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 08:41:01.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":4},{"id":"83979b8dadd13e453996026d609f83e6720f1cb2","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 08:43:49.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.","accounts_in_message":[],"_revision_number":5},{"id":"462b783c500c42442c24514b67f9a553b33208b4","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 10:12:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"57c940097317787dcd0e14f98e0e1aaa20f0211d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 11:08:29.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/62729503ac99403b960b64d167becaf1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ece1c04ed5a24736a394f834b7409bed : SUCCESS in 2m 20s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d3a60d2a96c3443eb687e8afa23ad3a4 : SUCCESS in 39m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/abd9dbc9db83417c88eb9e4d83c9531d : SUCCESS in 47m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c5b6503f87624bd2a2e183041b2979e8 : SUCCESS in 44m 22s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/09288d4388ed4eba80914b6de237bd42 : SUCCESS in 45m 25s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f68a2c39b16c411eaea237d87f393fb5 : SUCCESS in 40m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/451c8a921e9d4da581f9d51b2b1c9971 : SUCCESS in 48m 58s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0f7fde0bc66548589e23413a7a908efa : SUCCESS in 53m 37s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ab841312a1df46fca1aa90b75f524751 : SUCCESS in 54m 42s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d2ae8d0c6c82436e8464a6b77ae6851d : SUCCESS in 53m 14s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2fedd0f621184d65a77971e92373fc4e : SUCCESS in 40m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dd7cc97bdb934239b2fa69596b9e69b4 : SUCCESS in 43m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30e0c55687f845baa2c8b7aad77a93c9 : FAILURE in 6m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/60fabef549c04c30aec65bf57b10cec4 : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0272e1c923149ea8cc5b9a9863a040e : FAILURE in 3m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8593780a5e8418a910551084d790dbe : FAILURE in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/57b844ec14a24a40a54fb5f156eb8dd6 : FAILURE in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fca1017e5ee4ef78573f878457b4017 : SUCCESS in 9m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/32f416d04b0c405e97df7aa0d1ce4118 : SUCCESS in 31m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/52fd3f3c64334526a50474700fe71113 : FAILURE in 2m 55s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ab50c404712447c3b146d27d266aad88 : SUCCESS in 51m 10s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"2769ca63ab4a944978b831a72d845154d545eeba","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 12:27:00.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"3c3c1de440b9e9483fe06e4f2dd6e749035ba755","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 13:43:07.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"0acd8317826e618abb7f65e6de7459de9b96759e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 13:43:30.000000000","message":"Patch Set 8: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"a48f63da992e6baf08bde747172177dadf41d52b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 14:55:58.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/023e3b3d8b144cf6b238d9efb4dc04f2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cd14efd9f4a64d73ae3a28dd94c6567f : SUCCESS in 2m 49s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/334e56413c6249ef89ba176e422e7702 : SUCCESS in 41m 54s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/edc8b1b3448c442ebfc139f1c5120077 : SUCCESS in 48m 21s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f724bd86d90462fa1651e0b4e0130f1 : SUCCESS in 44m 23s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c98720f1ecdb4bd48a115796826090a7 : SUCCESS in 42m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/900146b7dedb405ea702abb66a7617b7 : SUCCESS in 39m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e774499603ba4cbcb682a9991fad7c8b : SUCCESS in 1h 12m 03s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f9b7db50c7434b3abb80dca6458c9ab9 : SUCCESS in 56m 19s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ab7b84ffd59c4d3888c1af9638ab1cf5 : SUCCESS in 1h 01m 13s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/80022f5615524758b619cea7a8dd977a : SUCCESS in 1h 00m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f27948a7308949d5a87c3f769d0ca14e : SUCCESS in 40m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a1812e3d8f3445f19a2f9ec35994ecd4 : SUCCESS in 42m 46s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba3c04a8c2584e90aaa043d8d6ec0306 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3bbc0dc6e3342778fc4eecd581d1120 : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/319789761e064e61a1d0d7a58d3aa1bc : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e3a5d92895c45db8690ffb00dc96514 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48746f6ba1194209aa87204475b932ed : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2fae71d84584577a66138bd8e841e01 : SUCCESS in 9m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e3141c1c6db44da8c5e2abd5201694f : SUCCESS in 31m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/16dd1d3897124f0ab6f8427d7b6c905d : SUCCESS in 4m 07s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/522584b31c5448db86fa7b9c8cd6c6fc : SUCCESS in 50m 35s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4c44959e63903d2ca7fa89aedcb032f5d65b51d6","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-21 16:02:17.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":9},{"id":"93961ddc1cadb440613e3994a23a65a2b13c1177","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 16:03:22.000000000","message":"Patch Set 9: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update gerrit/openstack/designate-tempest-plugin","accounts_in_message":[],"_revision_number":9},{"id":"f5d5e94cd91b497147580bb3be12843a07454ed3","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-21 23:26:47.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":9},{"id":"c7280d67b24a850de3312a5663b35a03a2b0a478","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-21 23:26:50.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8b8e137034e4514dbbe019e014545ecd4e9966fd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 00:33:40.000000000","message":"Patch Set 9:\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4362b30c1cf7446c983ee53f76aa86ec\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/55633317de4a46f9b0b4f346fdf0d903 : SUCCESS in 2m 44s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aaaa1d43f3ff41fa84bc60d142130a8e : SUCCESS in 44m 29s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b90db352ddd1443c85ea58d2fe039a42 : SUCCESS in 40m 14s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/22b2cf8919c449b79564d6879f03a22b : SUCCESS in 42m 08s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/4c96ebc190684767b91595560610259e : SUCCESS in 45m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0031b2cfa2634b38ad3e52b758d76368 : SUCCESS in 42m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f0a6f97c3f444a05914d66fbc72e7b5f : SUCCESS in 50m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9349e1f7ed0d463e96bd4a082160fa06 : SUCCESS in 59m 56s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/72246e191f3a41ea8d466cfc6e5169ea : SUCCESS in 50m 11s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8f3613746a7d486aaf45f5bafea05fd2 : SUCCESS in 46m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b542e0d4e31046528a72914436029a54 : SUCCESS in 41m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6fdad51e5acc443a9f5a3da094c01b78 : SUCCESS in 32m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26ed40a06b0d4773a68c0603c4c3dbce : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc71649f717f45a498eac5371acd55ce : FAILURE in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cec8eaa223dc466b83d6f28f5b6f3656 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c024a7c6e9a14bf1b3b2031bd7e12d88 : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/04fb8ad146a3494bac12477e757e68c9 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4dc913535634fcd87ccbefbac979144 : SUCCESS in 7m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cc2561a406d4cd6af0661e2bc9dc78d : SUCCESS in 2m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee05663248d6462a8ad899dd2a297771 : SUCCESS in 18m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/27da23eca616465ab275f65acc077f90 : SUCCESS in 7m 52s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5e277d92b08748eb9ce50af5d09510e5 : SUCCESS in 40m 14s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"a85538eba60161d53729dd403ee7938af6787472","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-22 09:57:25.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"ca89f6bcf4fc9e4b94d6890f027a4f9ad390bf71","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 11:11:33.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd4d1e0c48694bf7b9d7aba40d58d74e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4f21a36ba375415d8c1b474373a5498b : SUCCESS in 2m 11s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f172359a940145c6aa0ed3634294211f : SUCCESS in 39m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/629ef130fb724e11a7a8882ddbb99cc5 : SUCCESS in 49m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8ef9ea0370c4fddad916b59a7753050 : SUCCESS in 44m 42s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f65f9004f4134b9c82a0197118c00388 : SUCCESS in 35m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/48d8bf6128134fe499c1bb7b858e0f41 : SUCCESS in 39m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a456a85da2e14ef49d602d870d8d010b : SUCCESS in 1h 01m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/20bf5eda1a6545759c220d70b0079840 : SUCCESS in 57m 20s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b197cf4ef0b84da7905bd60a084ed31f : SUCCESS in 52m 16s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8618e155e8da4cb8afdecea9a62ed232 : SUCCESS in 1h 01m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/78522e95ae8040e9ad6297180f403c13 : SUCCESS in 42m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7be171b6efbb4e73b82b5ee2e096181b : SUCCESS in 35m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f562cb66dcc4049877c7c76b0412674 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef03d32c7a7a49f8aa501886a125af52 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23f273efa7ea42f99d2c50090a79ae4c : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb71bf53015f447fa21f6c9b99d5488d : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8474cac02a8a4bbc80608a311f3f112c : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65263f78100489aa08f94535b060f8c : SUCCESS in 10m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6354844655f4cdcbdda8fe33161601d : SUCCESS in 4m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cd759e226403416dbd50369bb925e05c : SUCCESS in 42m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/29f69589c62449dfb0d8bcee13464526 : SUCCESS in 5m 55s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c3caee8bd09545ecb5297e5b9d0924a9 : FAILURE in 44m 10s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"6d78489456d1fca15a8b454db69f72222748c237","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-04-23 13:32:33.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"2becf12b5a59808c6e0c4afc14538e5b96894fd9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-23 15:07:34.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"2f152f21f5bd71bf82c81cd57cf412f5af467109","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-23 15:07:59.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"8297e864b2d60a1cca511fe6631bcf091acd8eb3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-23 16:26:47.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/feaaf8b3687e486baf9922496b914e16\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7eb25dd1a4e46b69b1e1aa4836973ef : SUCCESS in 2m 46s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ee0b2eb58f0a479c87ece63bb0b31e00 : SUCCESS in 47m 42s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1d2d11bfa31d4619a20e1e4c596d5737 : SUCCESS in 58m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/2615d6e6dfdb428fb2f29769793b0346 : SUCCESS in 55m 07s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/717ac62b907945e5918532a8cf3f66e4 : SUCCESS in 40m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2f39900560c64d6db781ad1f26a82c4b : SUCCESS in 42m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7d16de2ad2d14cd48e2a6383ac54121f : SUCCESS in 1h 01m 59s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fd481a736ca0474c820c046578fd1930 : SUCCESS in 1h 05m 53s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/364f89b97c394208b9c86c20c68178e7 : SUCCESS in 1h 07m 42s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e6103836c17b4ed093fedc545fb4fe13 : SUCCESS in 46m 19s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e73651021d274edd8ed3bb04cad715a2 : SUCCESS in 37m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/44b71e00837f4377ad53abbd788bf528 : SUCCESS in 45m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c715de609b234ac69b07d6f3b13e1e84 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e9c4d9fcb3346a7a83c8cc7c9c254b6 : SUCCESS in 5m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d9a0c330afa74411acf65146a9ea541e : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5a09d801f3714d06a187f00a2563ea4c : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5b5c96689a8b488ca5af170f0117e783 : SUCCESS in 9m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bce6dbf476504edc9a5d5ff31a7d8318 : SUCCESS in 9m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c0db78137aef431f8eb29c99d8faf33c : SUCCESS in 3m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/80f0cbcb37ce4f9aa3fdac7a11cc4793 : SUCCESS in 48m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/76b8ca4f0daf40a29f9bf49b46959e87 : SUCCESS in 5m 17s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0e093981074a426ca0d0d894fbf7eeb0 : FAILURE in 42m 28s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"95e1f200ac1d57da328bb323404a010822873e2e","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-05-09 09:05:47.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"70c320e3f373c597e5de985be1f80e6ead8868ef","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-22 12:23:52.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7cf3540831ce7079302f64ac1d15f1ae010044ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-22 12:26:34.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"e794da436afe24be69b8b1f894ff98091456ae40","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-22 12:30:24.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"720ebf695cafc63148a1407722a8fa47fc4f1616","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-22 12:41:59.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":13},{"id":"b13e3f5dc87ce0943464c88eaca1ccbfb4228992","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-22 14:02:03.000000000","message":"Patch Set 14: Patch Set 13 was rebased","accounts_in_message":[],"_revision_number":14},{"id":"4d083819c2fd10ed643ebfccd1644f1461de8864","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 15:19:24.000000000","message":"Patch Set 14: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09ac6613916541a7a4f884f5866fc222\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ad9c4fbad005452dbd14119eef022023 : SUCCESS in 2m 30s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/19708688ab284f1bbada4a2e7240b696 : SUCCESS in 48m 58s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3c64d195306144f1ada1a877678e9a0b : SUCCESS in 38m 20s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/d5479f1044344a3fa42f16fa2528260c : SUCCESS in 42m 29s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/9d175e2de10540248aefcae1013edf3f : SUCCESS in 24m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ad9ba761bff441388ff49e63e57c3ed9 : SUCCESS in 1h 06m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/944b0f76c40245298d73cf394326cf81 : SUCCESS in 1h 08m 46s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3c1821e02bc5460e9b73fe9d3132f0c4 : SUCCESS in 1h 10m 49s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/15171ca3c5f74844bb400459122f6c4d : SUCCESS in 53m 28s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f12e959f839f4667b56682e79ee46131 : FAILURE in 38m 54s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/aaa29231942f489e93615555c153d5d9 : SUCCESS in 47m 50s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/e625d89d21f2432099cd9904b7d67534 : SUCCESS in 42m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c8fe55da6e14a54bbda1498607f599c : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dcdd172a0f8b4a9a99557be20e37367d : FAILURE in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4e4919c3a414466bd50b9fa3a400d73 : FAILURE in 4m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6efc266c0c834ed18cf38f65bb9526ed : FAILURE in 5m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/42fa6fa8806c4412b8e5e8c476374aeb : FAILURE in 7m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d20d845f624bd68983246cfb01dc2e : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c62bfc9dace44ae9f3b95c8ce9a6b21 : SUCCESS in 4m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dfa15cd985d1481cae76c3b84396d91d : SUCCESS in 37m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b5eca27265234c63a046fd43419cce32 : FAILURE in 7m 21s","accounts_in_message":[],"_revision_number":14},{"id":"e5c2ceaaf9ad44be09130923c415683c7c949e0e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 12:45:04.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":15},{"id":"bb3cf9669f37f9e90f575137244fef67e1f0831d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 12:53:07.000000000","message":"Patch Set 15: Workflow-1","accounts_in_message":[],"_revision_number":15},{"id":"2a8bede2e4a08daddb9e130a663b1c371522947a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 15:22:35.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":16},{"id":"a3cb5929e2ab07a0ab875e50583700bc449d2150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 16:36:13.000000000","message":"Patch Set 16: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd9a6bab28284b54aa6c8ac473fbb9a5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8d73f85055894962ae633f0a5ed4b919 : SUCCESS in 5m 19s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/6594bf6dabc94b09a7cdf49af6bef013 : SUCCESS in 47m 19s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/61aae9d00bfb4551b7bf2621e4bd6509 : SUCCESS in 43m 30s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/36a6f26f20c948d291895cfb361e21b6 : SUCCESS in 44m 34s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/dc9a27b09e8842c5b0f5e52eb7a373b8 : SUCCESS in 15m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3657237e656844b29ffe70d48033d1ad : SUCCESS in 54m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b45700bfa9b4462d898ed0b3b3b0677a : SUCCESS in 1h 04m 19s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3217ee1036c14ff1ae7a98e9c8a075cc : SUCCESS in 1h 08m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/874e59db00cf4b3fb9bbf79aa7c80fe4 : SUCCESS in 56m 41s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/19a8e2d5e2854620bf5e4a084dc92615 : SUCCESS in 50m 01s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/c2a694d9229d48b584d590ee52269d87 : SUCCESS in 46m 16s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/dcb5357c9f094a4d9b0c4108593810b2 : SUCCESS in 45m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f27078c00e074eafacbcdc0a775069e6 : FAILURE in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2560436910704acdb969246b53a8e142 : FAILURE in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72fdcf6f1f264aa0a45928f45725ab8e : FAILURE in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d011033d0e4449485ecb5b8c0450519 : FAILURE in 4m 45s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f933bdaa6537445aa92a22cd85ef4f27 : FAILURE in 5m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99bd40e78f004b8385318a9998c0abc1 : SUCCESS in 9m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d8441e9a302b41d38612d7379c7cbbb6 : SUCCESS in 6m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6e405046e6834ee79b13cd3294c4436e : SUCCESS in 31m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/648cdad2e2f74542a2787abb20d9b119 : FAILURE in 4m 34s","accounts_in_message":[],"_revision_number":16},{"id":"9186ca33533efc5b7fffc4c6fce690d8e051b66c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-24 09:31:48.000000000","message":"Patch Set 16: Workflow-1","accounts_in_message":[],"_revision_number":16},{"id":"1f838afe6d0b94718c5c2cb013c146e461caf0ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-26 11:22:07.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":17},{"id":"c66dbde833e9e5aa68498bd9f428e00cc1155a1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-26 12:01:55.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"c4f30b8e37baca8554ae1bdd357bb096b4e2c839","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 13:23:51.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37f7b57fd5a24faba98c4bff81899662\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b195ad55fc7c40a683f3f9d714017c38 : SUCCESS in 2m 48s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/c7646123e66a4a948a91d644e100bcbb : SUCCESS in 33m 09s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6d6a80b37ec94057bffbf1c4fef313b4 : SUCCESS in 43m 57s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/de50ef38a0584e56885440b35ae6a5c6 : SUCCESS in 50m 25s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/0221aab07ada4478a460183cd6cfece6 : SUCCESS in 27m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bfebea483c42482ab08664b47e953ffa : SUCCESS in 1h 00m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0d437068592348e0b6522fd18bc72a42 : SUCCESS in 1h 03m 28s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4083cbc9373b4367af46cd00494e5db9 : SUCCESS in 1h 11m 20s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b4fa99ed8fd448dbb441cd930eacd405 : SUCCESS in 1h 12m 52s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/62219285306c40d8a4241f0605069f55 : FAILURE in 50m 55s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d62a146c834b48899f56354f2f368351 : SUCCESS in 46m 57s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f9a4eb20c30f4b7ca52a41004a221b51 : SUCCESS in 47m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b44a378ae6f143ec844a21ea59ed2bca : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f89c5c4c09b453692f6163d179b8801 : SUCCESS in 2m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81fb45a68dd544609df1a736897277da : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c127e19f7a4948ad822fdc7b373e7594 : SUCCESS in 6m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e457f1bfa7e84022bd7317c57bcdb8c7 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ec8cb9b0bc84d4f8ff45e8c06a09804 : SUCCESS in 9m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b48f081735b41bc8c2e2d0fb35bf905 : SUCCESS in 5m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/80af2156c1194e8db66bd94e48081b66 : SUCCESS in 32m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6f056cc701a94b03acbf3cbee17c1055 : SUCCESS in 5m 42s","accounts_in_message":[],"_revision_number":18},{"id":"009d7966bda181fbd5d58ae1671ec38af327f340","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-09-26 13:34:45.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"2d269cbf8af8f0fee427f26d624c002d0cd1bc4d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-26 14:47:45.000000000","message":"Patch Set 18: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"cffb4835de71f35894b46dd1c8faff2e3565ca1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 14:28:30.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":19},{"id":"404ec6c21b20c7d21c9a31d73c6b04ae9b1b8566","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 14:37:32.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"1c2a589da549927b816d4be63f70b6ae0e73d316","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:46:39.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b4c63c894d244193bcba1ec7624d4d9d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e11375f8c52241d98126e4ebebed48ef : SUCCESS in 2m 51s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/b4b8a004d4e141a0a3a5838a30119117 : SUCCESS in 40m 22s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4e2ed064938c498bb7e09d1fafa43d6d : SUCCESS in 42m 45s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/22355a0ada20473e977ebb9030d0b40f : SUCCESS in 40m 00s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/a333e4e96bb242f38c2e37af1a2c0b20 : SUCCESS in 30m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e04b23c9dd5a48ab8da3424339292558 : SUCCESS in 1h 12m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0543c7e2931344b6be3394320d929ad6 : SUCCESS in 53m 36s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/076c352576234c73866c21194ef06d3b : SUCCESS in 52m 40s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4e1fe0d452f94099a84437f5b7b3ecc4 : SUCCESS in 1h 04m 54s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/0eb6a682508d47659a3cae91698c06fe : SUCCESS in 43m 06s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d9cdb03b0f3a45888b900e0a1d59152e : SUCCESS in 44m 27s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/9564b01c7fde42ca8ce3f13896d950fe : FAILURE in 9m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d47e32b4e9644f60838baeea0b28822a : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f80d9b65c8644977bb7ccb90b6480898 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/57453b2ac83b4a7db50c989f8a03cd23 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9170f3d77e249daba0b0c2a0418472e : SUCCESS in 5m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ac4c1a4757a74b47b648c3b257b44248 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98989bfec6b347dcbb62888db4cc04bd : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/271259e67bf04215b6e203051957a87b : SUCCESS in 4m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ddb9a7ecd73a41668f3edc918284114f : SUCCESS in 30m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3c50919abea54015b2f9002fce9eb8a8 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":19},{"id":"ddcab4d0a9be19d3468a10489f95e3c8c0f78ea8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 16:03:36.000000000","message":"Uploaded patch set 20: Patch Set 19 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"77561aa24d8af759c330632bfee131b5d9867ed6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:17:55.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/638ca5229eb7495e9361ad95c836319c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a37aad0af70f49c0b401c85660e04929 : SUCCESS in 1m 38s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/561567f2e49045b5bf5d2c3ca0537f17 : SUCCESS in 39m 24s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/8cc9fd5c839a441cae440fdad7bc58a9 : SUCCESS in 48m 03s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/88e15d1d44a84bceb70f88052d850783 : SUCCESS in 49m 02s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/16f52fcf031d4dafae3304d4ce303d61 : SUCCESS in 23m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8d3546f5dd944ffb9ebd1e54877a9f09 : SUCCESS in 1h 12m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/67b8af878bde43c2acba559899640dba : SUCCESS in 1h 05m 24s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3a0a3b71b9204efa92c878ff35f2fb33 : SUCCESS in 1h 07m 50s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e9b62342e3ee4db0928894172dd5c651 : SUCCESS in 49m 48s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/b8db22a4d5034c6c9ddf2dcbc74ba057 : FAILURE in 38m 11s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/f8fe3958f731441aac18669b0403ee88 : SUCCESS in 47m 27s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/0b910c8c245e4d649e0750a7f851756c : SUCCESS in 46m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8005d444d9834717805d533b08cb3c9d : SUCCESS in 7m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3b0f4f7e1a24a6eb2264a4f29c62fc3 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e3ceb9d7e354d788c2aa1e0272f5b91 : SUCCESS in 3m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8b27d51d8d44a92aba113832e27509f : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5e833bc04c2b40e9ab60bb82ea4c2af8 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3feb26c5a0d418ebc1d521fba02715c : SUCCESS in 9m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/615bb05497074e8ca56f4be30cf8da57 : SUCCESS in 4m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1c8dee8f1bb648d9b29aa7a3206ef47a : SUCCESS in 31m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/91d6da019ece4c838d9d6ef4fa83be0b : SUCCESS in 7m 10s","accounts_in_message":[],"_revision_number":20},{"id":"4fed4198ce32411c7a57a33ca06427e28bab94e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-22 09:26:11.000000000","message":"Patch Set 21: Patch Set 20 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"28d848fa24ce8a82339f953baf6402b44adafc63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 10:44:55.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99234afc9d8c49af9df89746bcec7f4f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2e5e9b46593445eaa55213862e87c558 : SUCCESS in 2m 47s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/7148b52c30d24d5482be8328933bf522 : SUCCESS in 40m 31s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/56093a3fa40f495c84e6c864d842e1b5 : SUCCESS in 47m 35s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bff824234c234052848f667f5669c4ae : SUCCESS in 45m 51s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/bf39bcbce80c42a4a3f08d61db5d5513 : SUCCESS in 31m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/34d4cb3728e8433e95ae40e386b09036 : SUCCESS in 1h 11m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1e93475a856f48209e8ed2ab47f65f71 : SUCCESS in 1h 05m 24s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/36e6777e5f264faf98a9584cf1b76762 : SUCCESS in 1h 13m 30s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f207e36e3e96427aa07f0b1d046a485d : SUCCESS in 1h 05m 22s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/8203dc62acbf42229d4e69171a29373e : FAILURE in 53m 07s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ab48974bf3e44d29a7aa5ae1fd86ed00 : SUCCESS in 46m 37s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/9ec5b28b7f99408ab04e64b545e13909 : SUCCESS in 47m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d2edf1284c748f4a0958b8b7f3ebaa0 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/479af5db5c40402c8d554ee369800bbc : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/02e64555d49e47478497a3c9f647988c : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21e6476cbfe74408b7fa0456e696a0a6 : SUCCESS in 8m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70d01776cad7406b9a049fe938b18f04 : SUCCESS in 9m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1ec0c14c93a49e58d2468df48507771 : SUCCESS in 5m 06s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d1908276dd844d7fa3a183ae72dbd209 : SUCCESS in 31m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2b7dda3c3052481f8be28f5fb00ab439 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":21},{"id":"9dc6d05e70e09462bf79e418db6366f41e589c13","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-23 09:01:15.000000000","message":"Patch Set 22: Patch Set 21 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":22},{"id":"dec0c86fc989fc4d5b144364e369955c18236008","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-23 10:22:40.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9686f353b97e41fdb236ef373b7fb3ea\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0932b414177a4a3c876aeb214e365a50 : SUCCESS in 1m 30s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/d3e35182144f477e9b8873d312163616 : SUCCESS in 37m 23s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/85dfc918981b47408612ae35404d0f81 : SUCCESS in 36m 29s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/fc5377212b264c76bbfa412eac92d31c : SUCCESS in 35m 53s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f868a36cc3ab42b0b0de3c0fab68cfb7 : SUCCESS in 31m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/625509513f804800a02efd024a0997a2 : FAILURE in 25m 11s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/18c583a2d0194931a902e01acdedcbeb : SUCCESS in 57m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c0cd41c99bf64778b4ac2f077930afb1 : SUCCESS in 1h 09m 54s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/333bea719a504e07ba4c2a7b9bcf43e6 : SUCCESS in 1h 15m 40s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/584c19ddb35a40dba8aa6a568168a991 : SUCCESS in 47m 23s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/26e6176546074f08a787323c81a46a0a : SUCCESS in 44m 22s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/4824579e03264ed28f08e79b9e3e5ff7 : SUCCESS in 43m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c016c3f399054332b1ac5222b3965c63 : SUCCESS in 5m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a238ecf24d4c6682cc0ca5b5e2aae0 : SUCCESS in 4m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29be5b5d9579415393fe86359440870b : SUCCESS in 4m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eac9930c721c4de4bcadb8a7f40ce5aa : SUCCESS in 9m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afa5c68155c645cb99cbf77363192a30 : SUCCESS in 7m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6438c6c079645238f4d30e522274bf1 : SUCCESS in 4m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68f9ec9bda0443eeb10f13da94079b7c : SUCCESS in 34m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b243201a1b0b43b2a36f7ed8cc9950d8 : SUCCESS in 5m 29s","accounts_in_message":[],"_revision_number":22},{"id":"4a071e7315984ff4169b242cb37e81f0c5f12f50","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-24 16:58:05.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"0ce9d01cad7c8807f8606767a368655817493357","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-24 18:12:06.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/338e4caba16c408289daf4639fb9327e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f75d74909dc043cdafa8fd0407a6ac8a : SUCCESS in 2m 02s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/b70ac944efd54af4837c05c12e4b026b : SUCCESS in 40m 16s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/305ec1a366a2458cb9c445f94aca8278 : SUCCESS in 47m 15s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/d782ed24a9c7400884b04bde7d17f2a0 : SUCCESS in 41m 30s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/c352e2d23145439b94c6ade6d15ce045 : SUCCESS in 33m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/286843eb3f204f0db29b2390eff726b3 : SUCCESS in 1h 06m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e2b07af74a3641cca58ed3ebe6bf9726 : SUCCESS in 53m 37s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9b7cc31cedc24670b4164da594934ba5 : SUCCESS in 1h 08m 59s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5cf9cb3625b8475ea2bcfbab51c13412 : SUCCESS in 53m 59s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/024ccd12d2ee4cdead99957b38f14157 : SUCCESS in 41m 28s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/750ec69c79b447df901ea2a2b5791a7b : SUCCESS in 46m 19s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/588817dbfb5a4186badba65227425370 : SUCCESS in 44m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2b76e9c6b634c78818d76e050aec309 : SUCCESS in 7m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ff10b9ef758424e939d7a3334910572 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2e2e362e679d464f95a60e3d71fca3bc : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d0ad8235cbaf4a488c31465b2cdb5de0 : SUCCESS in 10m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdf7550ac42546f7bb9610a499a6c26c : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f21c0e8e091478bbe7d897429162696 : SUCCESS in 4m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/fe35bd5eed4448aa9474227693207779 : SUCCESS in 32m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9e37ccca020a42a8b1d7ff2a08ab99a9 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":22},{"id":"d16c7f3b2d1a4988e7d79dc69ff4a2a053219a68","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-11-07 00:54:35.000000000","message":"Patch Set 22:\n\n(1 comment)","accounts_in_message":[],"_revision_number":22},{"id":"40dff6ea4520e9864a872c1519c861df22e8321b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-10 11:48:19.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"cbb50cc9b5047d29d40744f78aef0541d75f2347","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-10 11:49:38.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"a2cbeb87455e44efa278c21765d3bd4876a08d51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-10 12:08:00.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"829d4cec4c1355a47b459fdb25b0269d2723f4ca","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-10 12:09:12.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"268636ea263b040db13534b2624287270063fd34","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:29:48.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c687f57e308405ca26c09aca28263a0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/60c012cecbe24fc7beba5a657ecea460 : SUCCESS in 3m 04s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/d66468cbf3184e8a836a0856c0fc0c7e : SUCCESS in 33m 44s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/832bb2055d234a33b3fa36b5a4274ff0 : SUCCESS in 46m 56s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/372aa88659b24987adcc742ae1cf2310 : SUCCESS in 36m 32s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/4301bfb759534ed29c1f5a573a85cbd5 : SUCCESS in 35m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f53e46c4242c4b94ae284885453d3bc8 : SUCCESS in 50m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/353749bbb1c048d88aae11fc7a171f49 : SUCCESS in 1h 05m 23s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/f5dc90875ea843e791e1f75696543521 : SUCCESS in 1h 11m 07s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1cc30fe063e3461bb59ae73bf4f95999 : SUCCESS in 1h 14m 25s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/b691045a17ba485d863212c0a3cc7805 : SUCCESS in 48m 21s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/af20d7509429482382880cef9061ecb4 : SUCCESS in 53m 41s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/da8a95b3663047aeb327244536228fa7 : SUCCESS in 34m 40s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43c4f44b14374fff965ecd64c1962f91 : SUCCESS in 7m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/88c316000230408cbbb424b1e88db735 : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7374e17f0de43d983643406568d75c4 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/207c75b3dcb04023acb9acce8948935c : SUCCESS in 4m 52s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/5e32775f97e640caa87b36edb2352b80 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ecfd6e1316c4b77a1c5d7d70dcb6038 : SUCCESS in 8m 40s\n- requirements-check https://zuul.opendev.org/t/openstack/build/a469b333975642009943af7f315793dd : SUCCESS in 3m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f7ad86e24fe4339af9d5da4eab5c4df : SUCCESS in 6m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d86281bd14394728aa6d286aaeecfaab : SUCCESS in 29m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/29d1b39b6c28424e8d1bdf745438a0f9 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":24},{"id":"22a89e7e14291d5ad7ab277d52f7115831515c6e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-14 16:58:49.000000000","message":"Uploaded patch set 25.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":25},{"id":"6e2955d012ac7c5872fa00a5429b0ef78e97acf2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 17:56:00.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e681cf6eae9b4b738eec1f8b81ffd226\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5f3033bca03d4c4db97ff412c05ea2be : SUCCESS in 2m 56s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e066760420bd4cc88b33f3fd1b55728f : SUCCESS in 33m 47s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/597773bb6a2b42eaa9f9146cc59660ad : SUCCESS in 42m 08s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/e575b34041c04c1084f770fcff496c1e : SUCCESS in 39m 49s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/28453896a9ed462ebb86688e669b3546 : SUCCESS in 30m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/55c1605bd65447a3970c5d54919436ad : SUCCESS in 46m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8b10012970bb45b6968581fc781f2070 : SUCCESS in 48m 56s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/637fa88cc54c46f18601f7a9481fb0ab : SUCCESS in 54m 29s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/89b063049ec34b09845cfce96ffb6a4e : SUCCESS in 48m 06s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/5442c801c33947279180f0591b9d170a : SUCCESS in 35m 15s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/a1a23755bf774dc599c7f8c1f9b5217a : SUCCESS in 39m 55s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/969f8eee9fa3452ba76566952b9a7363 : SUCCESS in 40m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d85b439774e409e86674196e831564e : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/25c32d05f0434b269722d8a5f9f8c930 : SUCCESS in 2m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1566f3a36b040679923c61c8c4c57c4 : SUCCESS in 3m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a6e1c1c8cfe43f3a8fce5ebdd6b5023 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ded38973a72411d973857fc8a9b5c86 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8424964f821f4251a574ac562aa82f98 : SUCCESS in 5m 36s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1ea3313739f244bd8cad2db05b0bb10a : SUCCESS in 3m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd05a02f53824c8193f455c37dfe9ce1 : SUCCESS in 6m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c3e5f6723f514915898009cffcecd8ca : SUCCESS in 28m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6f67f13d19424c45a005642d27572db3 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":25},{"id":"79ccf51db193509884dd2350a6d185898dfa9103","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-15 13:02:03.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":26},{"id":"6f446c53b1f160757a9e384a499181ba67d66157","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 14:19:45.000000000","message":"Patch Set 26: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c0fbce547e244428327646ce62e63a0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ce8a9e2e66c24c3bbbc12fb728631f6f : SUCCESS in 4m 09s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/6662fa715f944711a7bf7c88a9a8ffff : SUCCESS in 36m 12s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/420d22d98d0e4649975313c3c67ed3d7 : SUCCESS in 42m 21s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/3d4f1c2b12ea4587881ab86c2a31d510 : SUCCESS in 42m 49s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ba1f096c8ceb4f2eaeabd2b7a706ba3e : SUCCESS in 31m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5ecbf308f12e42c18fe31405002a452a : SUCCESS in 1h 08m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/900eaf5e3d9d4d69b593d972c02953ff : SUCCESS in 51m 11s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/105e5349989e4f16843754ae2ebb4141 : SUCCESS in 1h 06m 25s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f6b65314a0e349189fed20978047fe3d : SUCCESS in 1h 05m 33s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/29c2249a54744f51b532a211d188110e : SUCCESS in 45m 04s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/21d479fd7f184b5eb90f10df5fe512ed : SUCCESS in 41m 50s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/b0601ae33e864e61816fe35cadf38e71 : SUCCESS in 43m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74c03ebd5edf40e8a1ca51422da5e6f9 : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc8051de04c94af1af6d84e533ca5cd4 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a93b4f38a4004f159a9615843c0e5a23 : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/56a8fa559a514c53806cba06851fbe1e : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/aad11aa3696343c8b78811ee1c2af278 : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c7baa95a2d4e482693fd9c90e0a60766 : SUCCESS in 9m 24s\n- requirements-check https://zuul.opendev.org/t/openstack/build/5b0b4c0253384a248cf238056ec6c435 : SUCCESS in 2m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6221ed4482384ee281fc5c8d47871299 : SUCCESS in 6m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4571bc35333146dd8809d0b30c54e628 : SUCCESS in 29m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0d2b2bc4d0ec4bef86ba4e2fca78d2d3 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":26},{"id":"4ec5c9a7d5ed5a523ac6d487e6d0945d56381998","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-16 10:24:43.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":27},{"id":"a216357cf9f5b87303121ae7bb6c0fe33c7a809b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 11:36:15.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/635ea9100492432cadc170afc9f71e28\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f2429d5b1d943f28d04d65b9495c91e : SUCCESS in 1m 54s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/b6bfde3fbfc544d09470968bf899cad3 : SUCCESS in 35m 43s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0813dc8d69044d01aded7f70143ee547 : SUCCESS in 45m 21s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/56c1fa453cba489ea89b2ff4e2941670 : SUCCESS in 44m 12s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/70d5aad70d0341ee94ad969a455a27c5 : SUCCESS in 16m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/072db06e545b48ab8f552b6c22b77380 : SUCCESS in 51m 16s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5364fb5af94046ca9dcef759d6431bb2 : SUCCESS in 1h 05m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/79550655e0884f34885f2b47d75ade8e : SUCCESS in 1h 06m 19s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/21572f11bb934ca6bafb0db217d5bfc6 : SUCCESS in 54m 28s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/65d37efa45d345c9b91cf37a6d856356 : SUCCESS in 38m 36s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d45e4125d26b407096088894f3d8c751 : SUCCESS in 42m 27s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/fd4cc499c7274c19adfb4104327c7572 : SUCCESS in 33m 29s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6632e7b714624918acadfd5234b17e7b : SUCCESS in 6m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68917c801f8348a295836ca60c17ea0d : SUCCESS in 3m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ab70f4c50d024fafa774ea31cb5d7bea : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efbec4dc29064ed7b02a69e33046eb94 : SUCCESS in 4m 50s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/ffb917ebf0da461097ba6da72e0de0d5 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9353354f34694a58b7bd5ca93d8749ee : SUCCESS in 8m 38s\n- requirements-check https://zuul.opendev.org/t/openstack/build/175957fd286e4500afc2cdeb474af326 : SUCCESS in 2m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22ad69e41794473c85a14e72673f9bd1 : SUCCESS in 6m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3e5b7c66fe354e549c9df7fe6d742296 : SUCCESS in 30m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0e3519489da646d68335153c4288de9c : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":27}],"current_revision_number":27,"current_revision":"0a88693e193be0f226c0d2296150c2f28a6a9580","revisions":{"6622963e903f97a96e8a429a654607dc8a57f43f":{"kind":"REWORK","_number":1,"created":"2025-04-09 10:37:39.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/1"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6622963e903f97a96e8a429a654607dc8a57f43f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6622963e903f97a96e8a429a654607dc8a57f43f"}]},"branch":"refs/heads/master"},"146bd171cc7677e1a92d5c8ee50becb235610f24":{"kind":"REWORK","_number":2,"created":"2025-04-15 09:59:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/2"}}},"commit":{"parents":[{"commit":"ad8701b64dd29abd2d2849e2b562b5c705e1340d","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad8701b64dd29abd2d2849e2b562b5c705e1340d"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-15 09:58:38.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146bd171cc7677e1a92d5c8ee50becb235610f24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/146bd171cc7677e1a92d5c8ee50becb235610f24"}]},"branch":"refs/heads/master"},"9232613c1c228479a7592a4afc99ebbf4a60c2a2":{"kind":"REWORK","_number":3,"created":"2025-04-15 16:44:58.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/3"}}},"commit":{"parents":[{"commit":"f165f956b2a83e42c3d91a664c6d17ca6e3742ad","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f165f956b2a83e42c3d91a664c6d17ca6e3742ad"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-15 16:44:42.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9232613c1c228479a7592a4afc99ebbf4a60c2a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9232613c1c228479a7592a4afc99ebbf4a60c2a2"}]},"branch":"refs/heads/master"},"9ffa8884610aa62c6ce248710913bb6a21df8953":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-04-16 08:41:01.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/4"}}},"commit":{"parents":[{"commit":"34cc4cf6b8d4ef70ab0ba020f5950f885b43f484","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34cc4cf6b8d4ef70ab0ba020f5950f885b43f484"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 08:40:55.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ffa8884610aa62c6ce248710913bb6a21df8953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ffa8884610aa62c6ce248710913bb6a21df8953"}]},"branch":"refs/heads/master"},"196ebd30e4e37979dce2956a39ef705fd51ffa7c":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-04-16 08:43:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/5"}}},"commit":{"parents":[{"commit":"af39c1b07d9515639221130b939c42e5232e9662","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af39c1b07d9515639221130b939c42e5232e9662"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 08:43:41.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/196ebd30e4e37979dce2956a39ef705fd51ffa7c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/196ebd30e4e37979dce2956a39ef705fd51ffa7c"}]},"branch":"refs/heads/master"},"f23c850b6de3e0a3c1ffe59d5270f52f758eb9ac":{"kind":"REWORK","_number":6,"created":"2025-04-16 10:12:27.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/6"}}},"commit":{"parents":[{"commit":"4d48bf1c71216e45da94b44e0fa147a7a220c38f","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d48bf1c71216e45da94b44e0fa147a7a220c38f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 10:12:20.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f23c850b6de3e0a3c1ffe59d5270f52f758eb9ac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f23c850b6de3e0a3c1ffe59d5270f52f758eb9ac"}]},"branch":"refs/heads/master"},"f36873a99089f04d9cf44cc3217b152c265dc807":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-16 12:27:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/7"}}},"commit":{"parents":[{"commit":"62a56797631d3d04c3d6eb8a2f90d23d5462ec09","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62a56797631d3d04c3d6eb8a2f90d23d5462ec09"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 12:26:54.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f36873a99089f04d9cf44cc3217b152c265dc807"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f36873a99089f04d9cf44cc3217b152c265dc807"}]},"branch":"refs/heads/master"},"ced2771c948c689be8c9dd0ef60227c6c9392788":{"kind":"REWORK","_number":8,"created":"2025-04-16 13:43:07.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/8"}}},"commit":{"parents":[{"commit":"62a56797631d3d04c3d6eb8a2f90d23d5462ec09","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62a56797631d3d04c3d6eb8a2f90d23d5462ec09"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 13:42:59.000000000","tz":120},"subject":"WIP Add a producer task that removes stale ptr records","message":"WIP Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs)\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ced2771c948c689be8c9dd0ef60227c6c9392788"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ced2771c948c689be8c9dd0ef60227c6c9392788"}]},"branch":"refs/heads/master"},"d62e36a6b6f056ca9218bcaca7f907d29e4418d6":{"kind":"REWORK","_number":9,"created":"2025-04-21 16:02:17.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/9"}}},"commit":{"parents":[{"commit":"654798ec82c88f8477a9840c51bed41e7262f894","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/654798ec82c88f8477a9840c51bed41e7262f894"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-21 16:01:59.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d62e36a6b6f056ca9218bcaca7f907d29e4418d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d62e36a6b6f056ca9218bcaca7f907d29e4418d6"}]},"branch":"refs/heads/master"},"b13564f6dcd0a3c6a2954df84ee66aa5d163e65f":{"kind":"REWORK","_number":10,"created":"2025-04-22 09:57:25.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/10"}}},"commit":{"parents":[{"commit":"654798ec82c88f8477a9840c51bed41e7262f894","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/654798ec82c88f8477a9840c51bed41e7262f894"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-22 09:57:17.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b13564f6dcd0a3c6a2954df84ee66aa5d163e65f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b13564f6dcd0a3c6a2954df84ee66aa5d163e65f"}]},"branch":"refs/heads/master"},"e3cba95124f240ee3dd242c2e161b5e0f8554ed2":{"kind":"REWORK","_number":11,"created":"2025-04-23 15:07:34.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/11"}}},"commit":{"parents":[{"commit":"654798ec82c88f8477a9840c51bed41e7262f894","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/654798ec82c88f8477a9840c51bed41e7262f894"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-23 15:07:25.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3cba95124f240ee3dd242c2e161b5e0f8554ed2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e3cba95124f240ee3dd242c2e161b5e0f8554ed2"}]},"branch":"refs/heads/master"},"918d7168e4b3b97370db10a0b3c9d2aecc0caff2":{"kind":"REWORK","_number":12,"created":"2025-09-22 12:26:34.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/12"}}},"commit":{"parents":[{"commit":"6580ed966ad5dc2d8322b1d68bb5a7747de585de","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6580ed966ad5dc2d8322b1d68bb5a7747de585de"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-22 12:26:13.000000000","tz":0},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/918d7168e4b3b97370db10a0b3c9d2aecc0caff2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/918d7168e4b3b97370db10a0b3c9d2aecc0caff2"}]},"branch":"refs/heads/master"},"dd10e79eb2934d2e3eac095fb24a703796d34d17":{"kind":"REWORK","_number":13,"created":"2025-09-22 12:41:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/13"}}},"commit":{"parents":[{"commit":"6580ed966ad5dc2d8322b1d68bb5a7747de585de","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6580ed966ad5dc2d8322b1d68bb5a7747de585de"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-22 12:41:21.000000000","tz":0},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd10e79eb2934d2e3eac095fb24a703796d34d17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/dd10e79eb2934d2e3eac095fb24a703796d34d17"}]},"branch":"refs/heads/master"},"86230cca11bc5a8e00426f309cde75b828ca7bdb":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-09-22 14:02:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/14"}}},"commit":{"parents":[{"commit":"c9d15e31220aed3a0497aaeaa70707b40fba11fd","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9d15e31220aed3a0497aaeaa70707b40fba11fd"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-09-22 14:02:03.000000000","tz":0},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86230cca11bc5a8e00426f309cde75b828ca7bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/86230cca11bc5a8e00426f309cde75b828ca7bdb"}]},"branch":"refs/heads/master","description":"Rebase"},"79f5b981d20edd27596f97192c522839e1c8236f":{"kind":"REWORK","_number":15,"created":"2025-09-23 12:45:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/15"}}},"commit":{"parents":[{"commit":"377c063d5ebbfb493b9a3e6cb7fcc90ea527c5b2","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/377c063d5ebbfb493b9a3e6cb7fcc90ea527c5b2"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-23 12:33:23.000000000","tz":180},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79f5b981d20edd27596f97192c522839e1c8236f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79f5b981d20edd27596f97192c522839e1c8236f"}]},"branch":"refs/heads/master"},"1cabdeff1187c410c180010947062181485ce86a":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2025-09-23 15:22:35.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/16"}}},"commit":{"parents":[{"commit":"2b655fee647d7775da0d333b14154d9b1e60c75a","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b655fee647d7775da0d333b14154d9b1e60c75a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-09-23 15:22:35.000000000","tz":0},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cabdeff1187c410c180010947062181485ce86a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1cabdeff1187c410c180010947062181485ce86a"}]},"branch":"refs/heads/master","description":"Rebase"},"496a24305ea0f09c6df277903c49b5fe71817216":{"kind":"REWORK","_number":17,"created":"2025-09-26 11:22:07.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/17"}}},"commit":{"parents":[{"commit":"2f42c150a520f3a7dbfdab9a89b1c76e74edcca3","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f42c150a520f3a7dbfdab9a89b1c76e74edcca3"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-26 11:21:50.000000000","tz":180},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/496a24305ea0f09c6df277903c49b5fe71817216"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/496a24305ea0f09c6df277903c49b5fe71817216"}]},"branch":"refs/heads/master"},"1db3391f7505b37bdc2b7e4e6f48169140da441f":{"kind":"REWORK","_number":18,"created":"2025-09-26 12:01:55.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/18"}}},"commit":{"parents":[{"commit":"2f42c150a520f3a7dbfdab9a89b1c76e74edcca3","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f42c150a520f3a7dbfdab9a89b1c76e74edcca3"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-26 12:01:36.000000000","tz":0},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1db3391f7505b37bdc2b7e4e6f48169140da441f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1db3391f7505b37bdc2b7e4e6f48169140da441f"}]},"branch":"refs/heads/master"},"e4b24926b0f66c983c3694b63b3fae6b8aa623ff":{"kind":"REWORK","_number":19,"created":"2025-09-29 14:28:30.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/19"}}},"commit":{"parents":[{"commit":"0340adfb784e8685786c5c85bd7da36818847cf6","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0340adfb784e8685786c5c85bd7da36818847cf6"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-29 14:22:57.000000000","tz":180},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4b24926b0f66c983c3694b63b3fae6b8aa623ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4b24926b0f66c983c3694b63b3fae6b8aa623ff"}]},"branch":"refs/heads/master"},"c98452d8ab31c6820f524015f061383425dabc75":{"kind":"TRIVIAL_REBASE","_number":20,"created":"2025-09-29 16:03:36.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/20"}}},"commit":{"parents":[{"commit":"cc3d03e3a93841926718ea4a007942b3827076b1","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cc3d03e3a93841926718ea4a007942b3827076b1"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-29 16:02:55.000000000","tz":180},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c98452d8ab31c6820f524015f061383425dabc75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c98452d8ab31c6820f524015f061383425dabc75"}]},"branch":"refs/heads/master"},"db43ceee87e6d636990fc7e07dc801ab44a8fd98":{"kind":"TRIVIAL_REBASE","_number":21,"created":"2025-10-22 09:26:11.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/21"}}},"commit":{"parents":[{"commit":"4ee626ac54b6e530542a5214fc4a53a8a8987e33","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ee626ac54b6e530542a5214fc4a53a8a8987e33"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-10-22 09:26:11.000000000","tz":0},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db43ceee87e6d636990fc7e07dc801ab44a8fd98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/db43ceee87e6d636990fc7e07dc801ab44a8fd98"}]},"branch":"refs/heads/master","description":"Rebase"},"c82aafbfef74dce759fd7ba91236ea918fc15c2f":{"kind":"TRIVIAL_REBASE","_number":22,"created":"2025-10-23 09:01:15.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/22"}}},"commit":{"parents":[{"commit":"d15e3a62b90cee39109c87968e3612c68eeb2503","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d15e3a62b90cee39109c87968e3612c68eeb2503"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-10-23 09:01:15.000000000","tz":0},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c82aafbfef74dce759fd7ba91236ea918fc15c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c82aafbfef74dce759fd7ba91236ea918fc15c2f"}]},"branch":"refs/heads/master","description":"Rebase"},"e62db69ef8d2f9f04aadbc9f0ab458e9f9137762":{"kind":"REWORK","_number":23,"created":"2026-09-10 11:48:19.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/23"}}},"commit":{"parents":[{"commit":"ed52e8751455525ab38919c1d969ef99c057a08a","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed52e8751455525ab38919c1d969ef99c057a08a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-10 11:35:11.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e62db69ef8d2f9f04aadbc9f0ab458e9f9137762"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e62db69ef8d2f9f04aadbc9f0ab458e9f9137762"}]},"branch":"refs/heads/master"},"63b31846ff33aebd05340b27eb006f9b20403e87":{"kind":"REWORK","_number":24,"created":"2026-09-10 12:08:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/24"}}},"commit":{"parents":[{"commit":"ed52e8751455525ab38919c1d969ef99c057a08a","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed52e8751455525ab38919c1d969ef99c057a08a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-10 12:07:52.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAsisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63b31846ff33aebd05340b27eb006f9b20403e87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/63b31846ff33aebd05340b27eb006f9b20403e87"}]},"branch":"refs/heads/master"},"bfdbd89c9baf54df9b3b2f2493b9f15ad0f8a8ff":{"kind":"REWORK","_number":25,"created":"2026-09-14 16:58:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/25"}}},"commit":{"parents":[{"commit":"8c70164f20c6a53aedb9d1ca2dd08674bf1c30ff","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8c70164f20c6a53aedb9d1ca2dd08674bf1c30ff"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-14 16:55:13.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAssisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfdbd89c9baf54df9b3b2f2493b9f15ad0f8a8ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfdbd89c9baf54df9b3b2f2493b9f15ad0f8a8ff"}]},"branch":"refs/heads/master"},"fc8332dd2cf16245c9723635a39a861a3da471f6":{"kind":"REWORK","_number":26,"created":"2026-09-15 13:02:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/26"}}},"commit":{"parents":[{"commit":"4bfaab14ef588c60a735392bcf1333a356a55634","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4bfaab14ef588c60a735392bcf1333a356a55634"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-15 13:00:32.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAssisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc8332dd2cf16245c9723635a39a861a3da471f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/fc8332dd2cf16245c9723635a39a861a3da471f6"}]},"branch":"refs/heads/master"},"0a88693e193be0f226c0d2296150c2f28a6a9580":{"kind":"REWORK","_number":27,"created":"2026-09-16 10:24:43.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/52/946752/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/52/946752/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/52/946752/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/52/946752/27"}}},"commit":{"parents":[{"commit":"8ac7e8b6fe3f5d24c93fae369716d714c8ac61d7","subject":"Add a Neutron client using openstacksdk","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ac7e8b6fe3f5d24c93fae369716d714c8ac61d7"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-09 10:34:25.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-16 10:23:18.000000000","tz":120},"subject":"Add a producer task that removes stale ptr records","message":"Add a producer task that removes stale ptr records\n\nAfter we adapted the reading-permissions PTR records APIs to use the\nDesignate DB tables instead of calling neutron APIs, this patch adds a\nproducer periodic task that removes stale ptr records (that is, ptr\nrecords that are associated to non-existing FIPs).\n\nCloses-bug: #1935977\nRelated-bug: #1746627\nAssisted-By: Claude Code\nChange-Id: Ie6526675f28e599a4b7487039f85bab75031da29\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a88693e193be0f226c0d2296150c2f28a6a9580"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0a88693e193be0f226c0d2296150c2f28a6a9580"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
