)]}'
{"id":"openstack%2Fdesignate~947206","triplet_id":"openstack%2Fdesignate~master~I43b2e511542566e80698fb55ab24aff3ae6877a4","project":"openstack/designate","branch":"master","topic":"ptr_records_use_designate_db","attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-09-10 11:48:19.000000000","reason":"Vote got outdated and was removed: Code-Review+1"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-31 21:34:01.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-09-10 11:49:32.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I43b2e511542566e80698fb55ab24aff3ae6877a4","subject":"Add a Neutron client using openstacksdk","status":"NEW","created":"2025-04-15 09:59:50.000000000","updated":"2026-09-16 11:21:09.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":24,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"22718f1c9795bfd159e3f3ca4dd9e56b1fffe55a","_number":947206,"virtual_id_number":947206,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-09-16 11:21:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-15 11:13:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-31 21:34:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-31 21:34:01.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-09-26 14:32:45.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-09-26 14:39:04.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"70ea3e08530550000b3359ed0a1fe0292484f79c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-15 09:59:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0270b114a5fa86b7021bcfa223b11087e71ed5b2","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-15 10:08:04.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"1ba6e5f8c1d28d1a06e1c56d8916afd904f980c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 11:13:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/83685edefe04492c872474b69dc7b006\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5ae94753164244a98a4a4131efeff734 : SUCCESS in 43m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/44389c36fab443a3bb8cd84bd7570c53 : SUCCESS in 50m 49s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cec9c9d7f0e546e59b482de331d05867 : SUCCESS in 49m 10s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/b68fd18df4bb4aea98897500cb421cbf : SUCCESS in 39m 55s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/06b6ddf5e1a645019e4be5e9697c908e : SUCCESS in 41m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7bdb58db51ec46f98a5acc445c88541f : SUCCESS in 49m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d0c21deb49534070ac1788b8db82b04c : SUCCESS in 47m 00s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/73e602694b2e490894710d9bb2bc8d06 : SUCCESS in 1h 08m 08s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9a5b22665bdf4226a20052b67adcab42 : SUCCESS in 55m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/462a993bb422441096a34da3c35d2eca : SUCCESS in 45m 11s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/73a6bac72567470b80cafe36d8305d03 : SUCCESS in 45m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/05ecd4ce61fa4fb09896f1f83cbf5b0a : FAILURE in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1b709131a524319a7f0c94e87c75706 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/68a4e08f8ee64e408dc98bcd90fd9c3a : FAILURE in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5a03496c9844da78efdabf99446f950 : FAILURE in 5m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bd83e9f76a21491dbd2f0bc91b3fc5c7 : FAILURE in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4d96c6b60246499eb73f387e40d6ccb1 : SUCCESS in 8m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d54470980a341178945a9cf0af16189 : SUCCESS in 4m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e1d1c94e464d4016a51d7081225bafa5 : SUCCESS in 33m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f2045090ee1d42fa82c6ec013a059b1e : SUCCESS in 4m 21s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/10bed900d52e44fe961fd98cc3b5e625 : SUCCESS in 43m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1f9372a711a03087388352dd7ab98b77f289610","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-15 16:44:58.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d1f23690b76d0d5f3978dcc714132b4a8a813f57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-15 17:59:38.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cde4187154946faa5d94198ddf6b41f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/616324616aaa453d83535a9e9a825de5 : SUCCESS in 35m 30s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/881f9f39071740859eaa2a132ec7d5fb : SUCCESS in 49m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6e37107448ba4c3895e0ff0379927a39 : SUCCESS in 53m 53s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/90891b37bd104f86af4c335b6e31fd81 : SUCCESS in 46m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/428d48dac69c4a08b12379fa753984bc : TIMED_OUT in 1h 14m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d036f0afb20e489a94b55150d0c0da2e : SUCCESS in 55m 19s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d52d2f0ca5124c189dabe040967635fa : SUCCESS in 1h 01m 36s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5432b404fa894e4d8a228f8fd7b64ced : SUCCESS in 1h 01m 05s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/700375234fbf43beaed4cee08d904fb6 : SUCCESS in 59m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/63e5a5abff89480a90b4acc54008058d : SUCCESS in 39m 54s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d41f62999b32453fbbf37b47ecfbfd4e : SUCCESS in 44m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4c1a5b15a8b64191bc44be1347fb6d6f : SUCCESS in 6m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f0047c9bcc046fba75e3b9e1deb2cf5 : SUCCESS in 2m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d3be867284c436b82d8c976f0bb21be : FAILURE in 2m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/472ffeb5c6ec43a1a94a7eb5f187b4f5 : FAILURE in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/68450cffd9644f45b3f29e1817019f37 : FAILURE in 8m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9f9fe46e6f1431084f1aa0ce9bf77bd : SUCCESS in 8m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c601340e484b4f4086eecb1ba715608c : SUCCESS in 4m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e00e562687649b18402721c584f3c6d : SUCCESS in 45m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/8cf2ee552a6844ff8851cc38044432d4 : FAILURE in 4m 26s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8d27032ca181420d8147ba029e70de06 : FAILURE in 42m 08s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"ce3f359aabc0e881cf6c3e02909ad68ad442666d","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 08:41:01.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"eeecc5e2018a7f0d6b4fa787d2e641333b0b4848","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 08:43:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b2fd7e967af56a19577afff7427b8147385482c3","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 10:12:27.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"82e30a451fe44de54ca24d0fe45dbf08fea23c4f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 11:11:49.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/548907a3db0740ef9d4e22c1de63046d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/137fe3b029b74c5699eb260682af6d47 : SUCCESS in 46m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/5979e18a2d4c4c96934d7f92114aa8c2 : SUCCESS in 57m 28s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/9fbc366b4b8e4c049ec8e0dc1f6161e9 : SUCCESS in 38m 34s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/31f4e5d0b5de4f44bf8b1e8c698bc8a9 : SUCCESS in 42m 45s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0d8052275298481c810e43dfc7f52989 : SUCCESS in 37m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b660e7d47b0d492db5c2aa578dc9bbb0 : SUCCESS in 51m 18s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a99c524865c9423498ccf0e701e24ea2 : SUCCESS in 51m 18s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a7fc22f87caa435db6dbccf1e66a9a08 : SUCCESS in 50m 16s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7cd61e7b818643409ce88dd3752fa19a : SUCCESS in 50m 00s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c09dcf3e81dc4376b25cfa07f590aca1 : SUCCESS in 38m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a2950d9d77fe450c9d703397ccbd1d9a : SUCCESS in 41m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ba794b5fe7704ae29df568d4bb384b9c : FAILURE in 6m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba0f33a962ac428281b3f06acc2a1a36 : SUCCESS in 4m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/030d14bf801e4ca98402ee03f9df116a : FAILURE in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e52f273a56574affab34f535618881b1 : FAILURE in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa3a8d2245e14b10957b0d4cca095de3 : FAILURE in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/121db9f510c14783996330c0c51becff : SUCCESS in 9m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e19117051d3f467585e65129b0564f48 : SUCCESS in 3m 43s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/da8d4bd44a9c484b8f0a5d78b682a517 : SUCCESS in 32m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9d86c75ccd7643d0917b180ca2ac7531 : FAILURE in 4m 43s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/05f2fb495cc74021a6f3cde377ebc77a : FAILURE in 45m 29s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"1a2c5778d5ced78a24d2e536064346e7f019ac2c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 12:27:00.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"a3b96a048b535ba65f3465220a3d89bce39321a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 13:31:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2665bd5c2e4b40d3b262fb2128cce706\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8cf70c2c34f4ab2930b05995aeb2022 : SUCCESS in 37m 10s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/83bb9deb87434f249bd487b05ad8edf4 : SUCCESS in 42m 48s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/c6de90b5dbcc4080ac7520ca83b73521 : SUCCESS in 49m 23s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/67d7eea6f0b04128a50228223e4cf6e1 : SUCCESS in 38m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d0d2dbea78ab49ee914e068cf4acfb35 : SUCCESS in 47m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9156ed22b72a486185f7adfe5f0021cc : SUCCESS in 51m 25s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2f8e7265d02e49b2ba402a8c18d18ce7 : SUCCESS in 1h 04m 07s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/69ee893f9a024dcbad3da321f1a8dd9b : SUCCESS in 53m 32s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7d0c631ec846483ab2278546e5cee642 : SUCCESS in 44m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f0d4ef8d38d54392a44e1733a23b3031 : SUCCESS in 38m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0727b2cad77b428c99c66da502276dea : SUCCESS in 40m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa1196a1ef074e0f8da470e054c4081e : SUCCESS in 3m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b40e84cffb4a454abd3d7241ca7b0f17 : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3fc1ec5aab53495f9a1467152a1833de : SUCCESS in 2m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/94e9d1340c434ef0bd62a58b6dafa639 : SUCCESS in 4m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b4dc44a46024b24a15147a581efeed3 : SUCCESS in 4m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a589b4b093994d679823e84d342cc45e : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d17b26edd7f344b7a34969d89f754b69 : SUCCESS in 2m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/47fde6a0362443808c869b1043cf95f8 : SUCCESS in 18m 27s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3d22cd4b71c24c9991e280a13216a837 : SUCCESS in 4m 53s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b095ad98e7344e488c393ed7643cb158 : SUCCESS in 42m 59s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"543da8fb1c337a6cd65b40db91c4f7a9fac306c7","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-21 16:02:17.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"7ffd292c2d69b967734598d1afb8a38bebf7c746","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-21 16:03:22.000000000","message":"Patch Set 7: Verified-1\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.\nWarning:\n  Unable to update gerrit/openstack/designate-tempest-plugin","accounts_in_message":[],"_revision_number":7},{"id":"42c71f09d4f808cd52069fdf1b73f1d416431923","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-21 23:26:00.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":7},{"id":"d8423b294e235ec9d93a7ce5ccb29d7ad4352ed8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-21 23:26:44.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"122834857553613e67a517c7375c954a73592791","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-22 00:26:36.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8e61cad5585849ff9c32bc2f7385c26b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ebbcba86314948a6bf1e643a4ab68747 : SUCCESS in 40m 49s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/8ff7f179dedc4fafbe1fb9f6ce6a45b5 : SUCCESS in 43m 51s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/6d76195bf0d44e1ca866ad1aad782285 : SUCCESS in 49m 16s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/7864036ab7134017a41949910d135e3a : SUCCESS in 33m 08s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/fd9b157480d249eb80408c44ed86903e : SUCCESS in 33m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7702f846b27e4fc08de0a3678b6f2bfb : SUCCESS in 49m 17s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/163d39e71c534a00a6dac66b4ebca4af : SUCCESS in 59m 20s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/da1190c513354c63971492413a88ec4b : SUCCESS in 50m 48s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/482787e9cad14398945e6baa0b230a5e : SUCCESS in 50m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a6733a0abdb844afb3da0b15d2df7bd4 : SUCCESS in 37m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8b3614b1a807499e93075d517291ac8f : SUCCESS in 34m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc54b4a439c54df98866e3fbc13cc94a : SUCCESS in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d08de056154346bca2f1ee4e3760b74b : SUCCESS in 1m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/51ca353118da4cd6af1a7c437c80f8e4 : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1607522f8a13449db69868cd5571d83e : SUCCESS in 2m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8fa3cb24df314b93a8ddb442adfb7d9f : SUCCESS in 7m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac440b5c3597429fa1750082abb3e945 : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b41e9593379b4971ab1871f3b6ac59ef : SUCCESS in 2m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b77a83f0b2ea4c43b6c8f8aa10016d00 : SUCCESS in 33m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7df2f62e009b436d91c31e78b9926f89 : SUCCESS in 3m 30s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/6c38413dc0104553916beabef941b370 : SUCCESS in 42m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"64eef9e47e7d52c6af5b24be27a6032557d67cbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-22 12:26:34.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2dabc05fbf3805a34ef88d008c2902b998cd4ead","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 13:54:12.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c88af3c143c44edb9239b79bdcb5dfbe\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/295f276cc9b2427cb484c1e732c39e8e : SUCCESS in 52m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/06e23ba0c6fb439f9f5fb76a4157abc4 : SUCCESS in 1h 03m 57s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/89d6aeba01984a399250df0b06459f1e : SUCCESS in 1h 19m 45s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e6790704931247f4a255a6cc6fb24b67 : SUCCESS in 1h 03m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69bf47948b3349169fb60ea6c37e97bd : FAILURE in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39f7855dafb140af81632726547ff273 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae7d67fe989e404b80895cde9702c0f6 : FAILURE in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57f9de2629ea4b99936e7a639d9476ce : FAILURE in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aea008b129ce493a8e85e26ac0df5d71 : FAILURE in 6m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f03620e732604d448a7b5719b49660c5 : SUCCESS in 9m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e4e32699cb548dab7419feec4c85179 : SUCCESS in 6m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d8515c2eefd4aaebd601ed2105c327e : SUCCESS in 34m 43s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/55f11c454a71490db366ddca78db6910 : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":8},{"id":"7f7a7fe8eff2c881f9c4d4db7fe6ef26d0ab0483","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-22 14:01:59.000000000","message":"Patch Set 9: Patch Set 8 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"06a3b44f2c781824a68d492a60cb42e737638314","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-22 15:19:20.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d8eaa5fd0cba4307b66b0c2f3d6d3073\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/bb1d38e77cd845f39ee0cdcbd0de9c5d : SUCCESS in 50m 41s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a03306056f6a4a5294b2faea2212776c : SUCCESS in 53m 25s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/1d8bc5d627954e16b758409ac9373817 : SUCCESS in 41m 21s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/fc3badb94bad4b3ab3db8ebf5c4aad68 : SUCCESS in 32m 00s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a943b07627494df28ce8971a2ee4ad5c : SUCCESS in 59m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/748e7e6a33ca4c55a88b27a47182c738 : SUCCESS in 1h 08m 43s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9ce504c31b70492c95d4fd3672903e8c : SUCCESS in 1h 03m 44s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/69cb21b3879f4f2b847f6afcba3f7e1c : SUCCESS in 57m 51s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/3b7f0266de534ae7a74be91afaaf29d5 : SUCCESS in 54m 37s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/2e2aee1c51684888b50b842d5f54ba16 : SUCCESS in 43m 21s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/7b3cffc31ecc455caf7526892aad8a2f : SUCCESS in 47m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/211627dd1cdf44b0876b6533a145e00d : FAILURE in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44af72564c54486ab114eb68dae1b62 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af3ff6b3313b4abbb8482002a4973244 : FAILURE in 4m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9479c30188f4022b58d16f89b17495e : FAILURE in 5m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f686b7f438f4028bdaf97af39a26f6c : FAILURE in 6m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce53898c842a47dabea2401cd716bb0b : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71233be82ee34c009651761a548f6000 : SUCCESS in 4m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb4b47af7f1447b78ec86333e3d65129 : SUCCESS in 29m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f83b6113a09e49739476582451bbc7a6 : FAILURE in 5m 06s","accounts_in_message":[],"_revision_number":9},{"id":"8d697d21e775ba97a09a88ed3dac1ec6f79524aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 12:45:04.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"c4890a942bf8dbaa6f4de7f4357c55b775985ba2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 12:53:10.000000000","message":"Patch Set 10: Workflow-1","accounts_in_message":[],"_revision_number":10},{"id":"3497a09bc123825bac335eed57c1cbf58787e1f2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 15:18:04.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"c3189f4f1b8596a292da158ed11c4bf820ba1273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 15:21:15.000000000","message":"Patch Set 11: Published edit on patch set 10.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":11},{"id":"085b88528a7cf5642dd6fa1377ab2e0f017c613f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-23 15:22:29.000000000","message":"Patch Set 12: Patch Set 11 was rebased","accounts_in_message":[],"_revision_number":12},{"id":"7838ee633c66b0405ab64e1697349e43bdea3ea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 16:35:36.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3198d61231954a828cb4b266878e6b28\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/c5116729bb6d48acb147299b45e9e907 : SUCCESS in 43m 19s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/150c333e449d49e198e538ef3d858602 : SUCCESS in 50m 43s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/f2ea5834814848e490cc5bc06dc9d679 : SUCCESS in 42m 57s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/4aa69ab993f74b6c9b8ba5c242eeb142 : SUCCESS in 32m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b586852f84db454882a61a0f9f1d72b4 : SUCCESS in 1h 07m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7b444aafe17345c1af751659599ea818 : SUCCESS in 47m 37s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/16e81c98961b4fba988aa7cb6486f305 : SUCCESS in 1h 03m 25s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/82b5e50eaa6a431e82d1954dfa243214 : SUCCESS in 1h 08m 07s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/54db5e1dd3a841228215c9cd78473223 : SUCCESS in 39m 42s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/61064dfe879b4377963a331efd861fbb : SUCCESS in 45m 02s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/fc5bcaef562e483e9bc39ee73936e286 : SUCCESS in 40m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0810d68b5ec644f38b6538ad22c348d9 : FAILURE in 7m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed399c079fba444f9f9434064fe36d9f : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e498a74a68845b98d4daa4611f8baea : FAILURE in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a79cb9c0fb52487f93a50adcfd599468 : FAILURE in 7m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/27e55879d79745dd85c2491728e43d44 : FAILURE in 9m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f731fb9c8174b7ca1a6c3a22ecac7e9 : SUCCESS in 9m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4c2cae360034ad4af5c5fab8fa4e963 : SUCCESS in 5m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6b8250489f5c4ba2993304ae8f50256a : SUCCESS in 31m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/01445d875bf24407980341382598edb1 : FAILURE in 7m 17s","accounts_in_message":[],"_revision_number":12},{"id":"baa48c4d7f936a7c1ba4ec26e635880204bb6ec1","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-24 09:31:43.000000000","message":"Patch Set 12: Workflow-1","accounts_in_message":[],"_revision_number":12},{"id":"bdf17e45aa678bd1689b6361d6caa9ffca35ec89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-25 12:09:39.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":13},{"id":"47f9267e12d839a122c68fd4304b13aca824a185","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-25 12:43:47.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"4b987bf9625801c29faac84cee51efb20aa1beca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 13:59:37.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04d0b6a3539948a5a390307a7552e15c\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f8b4ebe8dd104bdab71b6b53bc2c1390 : SUCCESS in 50m 30s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f44f3f010e514284a6f1acf8931b63c4 : SUCCESS in 47m 59s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7e1956fa05804721a5fa612b46324a74 : SUCCESS in 44m 14s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/10d70f6ae174404199d79a015c1d27b1 : SUCCESS in 21m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b0c657fe58614a188e753722e2b00c35 : SUCCESS in 1h 08m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b4babe4ec5244c4a87dbe83b5d6ab3b7 : POST_FAILURE in 1h 07m 54s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4967f32b042c40f5bd64cee2ce472559 : SUCCESS in 1h 00m 18s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bbc9e3639c1f4ba7b3b9bf5493129f0e : SUCCESS in 1h 01m 55s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/afdaaeec90b64edaaa4112c4f13cc2bf : SUCCESS in 45m 45s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/7b77c85caffe4bf18de95d6503c0f0c2 : SUCCESS in 45m 49s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c143b615cfc945a2966719a191b66feb : SUCCESS in 50m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9d94932af854cecafdb45ec8761af3d : SUCCESS in 8m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1ed4c8d6d12453f88cce9d97cbab1bb : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/309c0ddbd5e94a68a8039413c8e82593 : SUCCESS in 5m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bc2276804aaf4addac734ab48dd7e480 : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/50b6bb51da2e44018b27d4bc0e78b87d : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d652c7b43d3b408db5de86c3956be502 : SUCCESS in 9m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ce64dc500ba41a8bf258ecb45920410 : SUCCESS in 5m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/efd2a85c469e426f9229f5fcbabb2271 : SUCCESS in 31m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4385136eaf72474a92826e029244c9eb : SUCCESS in 5m 16s","accounts_in_message":[],"_revision_number":14},{"id":"cd48b0768e7a5ae36a67831fbf7f4febb539e889","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-25 14:17:55.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"ce31eb77194d325420565e962b6686939e5f4eed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-25 15:36:12.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3775b99b7af04a62a87f9a0f2cb32343\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/6011f7d114b7468788b29636fe106f1c : SUCCESS in 40m 30s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/370d56fb37274da0b90afddd25122abc : SUCCESS in 49m 11s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bdb4d4d7779147cda4e07e5234da8497 : SUCCESS in 53m 33s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/4ea4a2ab6a3645a0bf8d8eea93dd60d3 : SUCCESS in 19m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/27fc9d7477344d0fa78a53fc3fc33756 : SUCCESS in 1h 03m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/bfbc56468dab4dd08a9884b959fae584 : SUCCESS in 1h 04m 34s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/54d374ec0cd846aa8c6ed0b26b6ca73d : SUCCESS in 1h 12m 48s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/50cc2e6b92f1449abee95514e894c8a3 : SUCCESS in 1h 02m 26s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/251b7d718e884dbb88378dbe085ddbab : SUCCESS in 51m 27s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/5e6657a797fc4e698d1d3c0b47378ca0 : SUCCESS in 49m 13s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/82466d04ed46465f99b685dd9e1cd293 : SUCCESS in 46m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96fe1fcb7ce34561a9fe52b4a0da475c : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c8c0647da4f4f418ee8b6465eaad5eb : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dc20ec1f7114a959e45af68b1006b92 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f32f7290161c4654bd0af051d82f447f : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/afc939cdf4884619b83ba0e2e806936a : SUCCESS in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c9a5f76938245d7bdc97cf8f55a2ef1 : SUCCESS in 10m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84d5447185304ff18a598abf0a5232f9 : SUCCESS in 2m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6dfc14d3cf2c460fa2d0e344088d525a : SUCCESS in 30m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/126388d46fad46a9a3fbda367f1ae06b : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":14},{"id":"4380b5703161c9ffb6e3a7d86c5b3457da50bae4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-26 11:22:07.000000000","message":"Uploaded patch set 15: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"26eec2bc25a43d3ae864a5923a6de77c406ea40f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-26 12:33:09.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbde82e61b3d4d45a637a6f4a048699e\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/93eae1e77f454eb2a426bcf282b5e0a4 : SUCCESS in 38m 16s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0a8043bc3fd94e6c80039ed3c0bce40e : SUCCESS in 44m 02s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/36c608c48722401bb0c4367b514fe00b : SUCCESS in 47m 29s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/a2515e9da05541a78587afe3fbcc1b41 : SUCCESS in 31m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2f45da8d1a2a4a67a0dcb4fc843c3103 : SUCCESS in 55m 09s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/752b8a829e524f1fbab6eceab43e853c : SUCCESS in 50m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/2ed1427ce2634d46a4921233796c0799 : SUCCESS in 1h 06m 01s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e2f20258dd844aa7835d843110c0adf5 : SUCCESS in 1h 05m 14s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/2399f4cedde544eba6f615af0ae14c02 : FAILURE in 41m 49s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d6d86b02b6d148648e3af14d16757531 : SUCCESS in 45m 41s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/dbe9409d1dfa4007a0a77a1bc2c8a1e9 : SUCCESS in 41m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f458a03338a04f45be4f46291b48b65b : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdbaccfa58de4042ae10a63801ece479 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a2e5ffde51004df4bd39552a512ae81c : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6e8221fd663647c38ab63981f7e1ce49 : SUCCESS in 7m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c4ecc7649b34908b938420eb3453e06 : SUCCESS in 9m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c79cf91e27f44a028f7fb72ae9fb367f : SUCCESS in 9m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97e8ee881ed94fab8fe836460c4fc656 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ba8224c0669549dbac991d5e989cb2b1 : SUCCESS in 34m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/79d9768449584a2fa1f5466e3f03589b : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":15},{"id":"f1c175dbe0ba62af8daa6a3285a59f91d5146db2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:32:45.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"79503c36d366e838e456daf530867adfe7fa4363","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:38:13.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"f916c4c7007c48a47ebd80ca1bd1a965ec5367ee","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:39:04.000000000","message":"Patch Set 15: Code-Review-1","accounts_in_message":[],"_revision_number":15},{"id":"a34cbd62644780de76e9472df8b10e5019d450e5","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-26 14:49:15.000000000","message":"Patch Set 15: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"69902e09ec1c8af8ecfbdf7840debb649b0ca39a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-26 14:51:22.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"5d306cbd10bf9f4e7c8c793614e5a50a6a6ba967","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 14:28:30.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":16},{"id":"6766164ee28baffeb4e664688af520f694736a59","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 14:32:03.000000000","message":"Patch Set 17: Published edit on patch set 16.","accounts_in_message":[],"_revision_number":17},{"id":"1037472b36817c0a3080d3b778b59f2437726714","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 14:37:11.000000000","message":"Patch Set 17:\n\n(3 comments)","accounts_in_message":[],"_revision_number":17},{"id":"41c651d13f03e245f163663fd96bda1c694b812a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 15:36:18.000000000","message":"Patch Set 17: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a08938e75af54b0c9a6d985ca90be884\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/3955f2910e8f4839bdf9a7c4f74fcac0 : SUCCESS in 36m 25s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/17abf8c24df24f94a1073c0b6332077a : SUCCESS in 43m 21s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bfa1485d0be74aa2a56725e938d34490 : FAILURE in 9m 07s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/abe053987b9749c1b6568ef98cf0eb8a : SUCCESS in 33m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/44859559345c4ec891234078a60aa810 : SUCCESS in 1h 00m 25s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7a3d6442811042578a35630709dc1ef2 : SUCCESS in 49m 51s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/86607ee987ee44689e1a18cc3f42e2d7 : SUCCESS in 1h 02m 50s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/35e9006b47bb449e803438eac6f5a3b0 : SUCCESS in 53m 21s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f0c8772772c84593a0d01ef5a6ee1daa : FAILURE in 52m 36s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/02f55a8c9f1e488b95e7319bc44aa473 : SUCCESS in 44m 55s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f27fddc0c82242f79e8f7cc07a98c682 : SUCCESS in 44m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/83d0572b287a45888f1da3dc9b56a494 : FAILURE in 4m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b43fb2ab9b05425eb3ce5dd43564c24a : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/31a54bf86dcc4676914b7dcb0dac23a6 : FAILURE in 3m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/862ed6499eae481e90438bedfa07fd25 : FAILURE in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/124223dce4614c609616a285b84a736b : FAILURE in 8m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/266cd41e52b5441a8b7b4c64bc5c67e6 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ea3281466ee348c9830bfc26d2fdad4b : SUCCESS in 2m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/781a348583f24a379b9cf7aae2c92097 : SUCCESS in 30m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/003962b8ae3743aba2c379444f377600 : FAILURE in 4m 44s","accounts_in_message":[],"_revision_number":17},{"id":"5692d96c1c0f62f9e326ff39bf458ee4226ae22a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 16:03:36.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":18},{"id":"8bba21702c728a5d7acd193643dd74b84daf9e61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 17:19:23.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94de079a9e854963876cb6ef7dd187b7\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/fa32d0ccc8a9466ebc1062bd3d7ff727 : SUCCESS in 43m 48s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cd105956e12d4c1291615c0421e37479 : SUCCESS in 50m 32s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/756e29dde30940099561cf02981bf8f1 : SUCCESS in 46m 21s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/0672fb2aab7a4b59940160ae3826e490 : SUCCESS in 26m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c6ce3bfe49d24e2db02baa4cf58a15df : SUCCESS in 1h 04m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b2588f9409974e058a0a83237b461119 : SUCCESS in 1h 10m 44s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4452fb1f69fb481ebf3c5f3873d6bb8f : SUCCESS in 1h 12m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/86e30e506f844927b81675682fa281f7 : SUCCESS in 1h 08m 38s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f46e0c2d94fb4443a43ebc5f0ce9eda5 : FAILURE in 56m 47s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/fc2f77a6deb6455eabb2ab57b0915d7f : SUCCESS in 53m 39s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c575d30251694a22b52e86de4c38fee4 : FAILURE in 7m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3d71a237173e4a8580e6d0095de82ac5 : SUCCESS in 6m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b2f72aab8de45bb831ed9851f097009 : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66b65ff59ded454389ff4c5226f594ae : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a0c5e9211cf443e98ba9d861ac32621d : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88daf4640c4d49d3bcd518ae4c6639a1 : SUCCESS in 6m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2c730bb24f5b4240a9474790a77fc7aa : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13ba0e452c9e4cbd9dfcb281228c6ba0 : SUCCESS in 3m 31s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/17b6d3c90cb54456816f496f3d52683e : SUCCESS in 30m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/72a5f229d70a4c77a1c228ef9ea0a7a6 : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":18},{"id":"ccf0b634e0852cf9ed2552df817febea441bd0a2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 21:44:24.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"667a56a2715043fdc250686731824bb57bdc9b27","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-29 22:52:59.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0418d8f003b34f5bad9d37c2385985c4\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/2f799b502a234464991510098de289a3 : SUCCESS in 36m 51s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0c7de41923c04dc2828fd9ed44524f37 : SUCCESS in 47m 58s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c5014cd4beae488791cd9d6a3a59bdff : SUCCESS in 43m 07s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/66ea22fbb90e4ee7bfc2971b0528a437 : FAILURE in 8m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7be77aa9d4aa4a15a45b884c9d07179c : SUCCESS in 48m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2ea7c18e748342ce9b4a8da653e28f8e : SUCCESS in 55m 04s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9f18fbae756741479148563d2003f9b7 : SUCCESS in 57m 40s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0b16b74024e74ddf8a5aa7bd696b102c : SUCCESS in 1h 03m 43s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/5539e4bbc2c14fc1af48cc673ddc3347 : SUCCESS in 46m 22s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/648a18a8e5a7438aa4616a5ca9fb023c : SUCCESS in 48m 16s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/486b4e529af94dfc8490a0e8711cd576 : SUCCESS in 38m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/741577e3da854dc09bb9c9c44384db0b : SUCCESS in 8m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9e906df0a3f43f0ad81a18bb2976ace : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/be2dd3b85201493b9fdb7511e1da94ae : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f180d8a34b5a47cc8f07632755f9f6a4 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8c1574e198fd4c2e87135d040eecb1ec : SUCCESS in 4m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/460b5aee96fe44ee81c65b054d96dc29 : SUCCESS in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f374cddca89d4d5090656e51cb220c94 : SUCCESS in 5m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/61b1119821ca422e80d6994069d3f49b : SUCCESS in 30m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/be27c2d3615f4176a68668ee619deeb5 : SUCCESS in 5m 05s","accounts_in_message":[],"_revision_number":18},{"id":"d40449de5fd82adc8e75d104d1685771dce06cb6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-30 08:14:17.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"3597f25a0055738b3bf316b0c027d747165a2ef8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 09:28:04.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1a0522196403437fbce51bdea8f4be68\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/ca44430ac98c49ceae37ad0e104ad03e : SUCCESS in 45m 58s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9fcf82bf95a34b3e9e8a9f3d40e56b54 : SUCCESS in 47m 14s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/a7fcecb1d17f4dd18b9d0505cc31eb1b : SUCCESS in 37m 37s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f660feedb2e248039e17068b73fa4fa8 : SUCCESS in 30m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/dfe2676335e3406d881a9797911a5efc : SUCCESS in 1h 00m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3964f33fe55d4be5925d421bdf065d79 : SUCCESS in 1h 08m 49s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c84df964a7f046c98d995f6e9d0aab4b : SUCCESS in 1h 06m 41s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/dbdcf344b1cd42df876735ea1669920c : SUCCESS in 55m 23s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f87be5b7f58848fdb04f4f206c02f431 : FAILURE in 49m 53s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/e79c75aec9214abeb44a8ae0bde6716b : SUCCESS in 39m 53s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/e81be7e7f51a482297e6b7cf9aa1bb6a : SUCCESS in 47m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1140ab79d33425e841854738ac8b928 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/049cba94fe164932b0693c2506029dd8 : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45c21ceff26c4d4f9ec72b9611c203d1 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/77e23f61049246cea3e137e7cad45ce9 : SUCCESS in 6m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da2c3c7fc63242149579502f80825a4a : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16427b000453452dbba8ecd21c3138a0 : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/43fde3924d2f4b978e3fcbc674f1281d : SUCCESS in 3m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2de4c2e6749410c85a7d0da47a82a1f : SUCCESS in 33m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bd64d7f063824277936bd09d1073d242 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":18},{"id":"02b51b668df8ec94fe4448150d4af93fd1594168","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-22 06:44:15.000000000","message":"Patch Set 18:\n\n(3 comments)","accounts_in_message":[],"_revision_number":18},{"id":"70984b28593c5c4f615137abcfdadad7d32dd911","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-22 09:24:46.000000000","message":"Uploaded patch set 19.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"da8b2b662e035a101478143113ec6ede454f8580","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-22 09:25:33.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"befcd8a51eaf035bd88ec3f1b18cc06335877d6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 10:39:53.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/36ac3d7cf670454e873dd065d96a62ea\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/56ceacc3a9264d71be952b4a502cdc8e : SUCCESS in 44m 11s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/559704b1a6904b449027867d493b6152 : SUCCESS in 46m 29s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/344b537ab2d54cfaa70a3b430b8fe52d : SUCCESS in 44m 37s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/66ac826147ee471a8191c3c4997addcf : SUCCESS in 14m 37s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ffd0d08238e74aec829fbb98ac94e695 : SUCCESS in 1h 04m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ed59fd3b0b00486b94ef5634a099f365 : SUCCESS in 1h 09m 36s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ec6e67859b6143d7991102c43ec52cdf : SUCCESS in 1h 10m 10s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a5066d74ef654b3c832653ba135332b2 : SUCCESS in 1h 00m 01s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/8cc3a3279b7a4f75ae80f9301a626797 : FAILURE in 53m 20s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/fe9265e24c14407b886d289f7e2ccc48 : SUCCESS in 46m 21s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f97b601f5eb74d278a1d3c7812dff8ed : SUCCESS in 39m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30e6f68f464543b2851b3b9f04fd1d7e : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7651b89c84274799bbbfe85f1785695a : SUCCESS in 2m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3b2e7a3f6cd4668b4a56da467aefe16 : SUCCESS in 4m 29s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4981350f83ea41f08c3a222afe5023f1 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e205693ecbd4bd1a6d50552f2a3c208 : SUCCESS in 7m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df69b72645af4408b45a6776a522c8bd : SUCCESS in 5m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9e24af94c43744699b533e72a04c8e8e : SUCCESS in 32m 19s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/508e128451994ef7b4f1631722e7e53b : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":19},{"id":"32c42ea412a06fff27dc23c32cbb59d8d26b5051","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-22 12:43:15.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"7456a5cb5227ccbc97b99f6cfce43ea1e48b150b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-22 13:24:11.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"c5166cecd8a001b8ca3cad32b6721356e8378676","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-22 13:24:20.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":20},{"id":"03f6d9cd0704f3d945385cd85e5859a8c7d60990","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 14:34:18.000000000","message":"Patch Set 20: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bea8cd7b5a1b47b5bf6b5b041ddcc27c\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/e78efd043db54d8eb68942319fe67be8 : SUCCESS in 40m 39s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/aa640998a5b34204b8cff4b8120f7966 : SUCCESS in 51m 21s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/4c4b8ae26e5d4164866c2c4ddbd4fdf9 : SUCCESS in 43m 23s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/3634ad858e7549cbbc44deb55b4ba264 : SUCCESS in 16m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d959388624304074b17a3b00662b88b1 : SUCCESS in 1h 04m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/74b61f2463ff4f6297f28c9b639104df : SUCCESS in 49m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ddb0ca41b98a4c2b8bf3266c3e4e7ae8 : SUCCESS in 54m 54s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/99b34ea198f944aba50fd8a112c091c1 : FAILURE in 27m 23s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/802472037a7f4b5f954fb75e23d7bfa9 : SUCCESS in 50m 22s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ec3ba32d961e43c5a6b9c244b77398dc : SUCCESS in 42m 52s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/e87403c6efe84b43be338477c3b0054c : SUCCESS in 47m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c1b7654425c243be9221e78972b9b70f : SUCCESS in 4m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d51ab3ed7246463eaeecffa7e17a7dc5 : SUCCESS in 2m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c13cf1c8b29b4d3988cdd877217eb556 : SUCCESS in 8m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a8410683da614f01b2078804acf035c2 : SUCCESS in 6m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b49976bc3a6547beae9db3228a1ce777 : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eab241ad994d45ff8cf9d1c5dff79e9f : SUCCESS in 4m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cb42a924246544e983ad35cda804c0b2 : SUCCESS in 31m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/995e6b425c314757b53dfdce8c375861 : SUCCESS in 5m 22s","accounts_in_message":[],"_revision_number":20},{"id":"3e3b2ad443ccc6a4a96994d3539ea445f3d987c3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-22 14:37:58.000000000","message":"Patch Set 20:\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"2f1ee942d7b2034f7604b301e78ed8558719f98d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 15:48:44.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da53e3c5bfc24053905aa7c8af2f114c\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/15662a7729144b38b13c6308a37e95fa : SUCCESS in 40m 08s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/45e260a7bd7140049cf91b0abbcb8bc8 : SUCCESS in 44m 07s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/61ce851349f44db593f3576bb4e94d8a : SUCCESS in 46m 54s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1b18c7a2dd7d4ae39516c2e565259663 : SUCCESS in 20m 20s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/617307828d07493cb9441ff5bb5fbbaf : SUCCESS in 59m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7dc3c3e368084b219d5fde1dc2e8bb73 : SUCCESS in 48m 48s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3316ac6982434749be16e62dba684b53 : SUCCESS in 1h 01m 40s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/15fd14efceb54471abe1a457461973b9 : SUCCESS in 1h 05m 36s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/6abad6010a474c40acabe6fd6f4933ae : SUCCESS in 47m 49s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/e0caada5afab4e438e2026a7a35cee14 : SUCCESS in 54m 44s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/0cf412e54a6f460192b92ce3494b5eb1 : SUCCESS in 36m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bdbdcee9127748d2b73eef2a90abd74e : SUCCESS in 5m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31e8f721bb4b4ecdb897c63259f12541 : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a6c5a14740074c2bafccb0df8fff12d9 : SUCCESS in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0805e1f3f2314a279f2aa7814876cf02 : SUCCESS in 8m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e16bf00e120e4c60b3ca4f0c0ffeba94 : SUCCESS in 8m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5fe7189e592541eca3416e2110869369 : SUCCESS in 4m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/83fed387ec6d4f09af02d5223cb7e461 : SUCCESS in 33m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0a8af5b8292741e6a63197519aca870f : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":20},{"id":"fa6a94529d3e034b03897b28ce18c96aa262716d","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-10-24 16:55:20.000000000","message":"Patch Set 20: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":20},{"id":"76e35ff4a379c7ad552b41e61bb131e8833d6da5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-10 11:48:19.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":21},{"id":"16689359d74b1f5b0832262f60e5aee32e39f700","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-10 11:49:32.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"321f1993defad20bc49568b7bc8688b90f7d6dea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-10 13:26:50.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/853bcd12bc824a07a892103e4b62e97e\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/841af4cb8c1e4390b85b94188bcd5732 : SUCCESS in 49m 05s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/33f49d2109d9451e8ddd9c31d1c773ce : SUCCESS in 32m 39s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/5005a21c790d45c8a97f182cb9aecbe7 : SUCCESS in 40m 08s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1ae3bfe10cd94de795d2467ee49e77e8 : SUCCESS in 31m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/729f02766acc4d07bcb02e0515103572 : SUCCESS in 1h 30m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c9af8f5fe6dc4631a15023707cf4f2fa : SUCCESS in 1h 16m 17s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d5f6ae11ad554cc0968d67a90869b1b1 : SUCCESS in 1h 03m 21s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d06cb3cd110c4522ab1db397baa5ceee : SUCCESS in 1h 09m 48s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/0661619dfc2d4941ad85931ad0a752ec : SUCCESS in 44m 23s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ca885c99ed4f405ca638b28f48cfe800 : SUCCESS in 54m 53s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/5e5ed4f2570042d18c9f892549cc4005 : SUCCESS in 43m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c21ceb12cdc4d5cb07b70c1dec2e8a8 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd2704fd48c44cf9b97c15914547fa72 : SUCCESS in 3m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25690dded96e4d0188966b4bd10b2c8d : SUCCESS in 5m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81f2577d0b444a8e8ee6b271da763066 : SUCCESS in 5m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d93496c6ecc0405c80ebef54a14f115b : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70c6c76ac9584baca5eaf47a4af613e4 : SUCCESS in 9m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/aefa95c5e174413aa0370b6ab72b5a39 : SUCCESS in 6m 22s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5e9f796fa70a44a1b0379088d4842342 : SUCCESS in 35m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/84281af996d44468ab707ace60072e1a : SUCCESS in 5m 14s","accounts_in_message":[],"_revision_number":21},{"id":"bace4fa314e760a564a66875cd8f319c2f6c7060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-14 16:58:49.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":22},{"id":"e0dc40e79bbe0f5e7d34c21821cb7afddf013b7a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-14 18:12:34.000000000","message":"Patch Set 22: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2fa17b0a68440ae8537147f6b1a2548\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1d9112db320e468bab343b5ba5cc3955 : SUCCESS in 42m 31s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/7fb077e2d31044f6ae0200e7c7e96bd5 : SUCCESS in 40m 34s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/5850de3f170e4bac8709cfc4a03a94b4 : SUCCESS in 38m 30s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/6d4335a11ab34bafba87ada2c9210201 : SUCCESS in 31m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e463792fd9cc4046910a5dcc5148ad7f : SUCCESS in 1h 09m 22s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e7aefbe4b7164f79b32530275d149d50 : SUCCESS in 50m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ba6a0b74081241ac88f43081f2c22581 : SUCCESS in 53m 43s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/382afbae9f1349399a08532eeaf36b38 : SUCCESS in 52m 43s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/2d9e176675ec477b8b787c42730c467e : SUCCESS in 40m 50s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/95086c7f7be4417c9a9aa04313b7083e : SUCCESS in 36m 23s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/bbd05c73271d4d73a0b82f611ecdfa15 : SUCCESS in 34m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/badf7f56b2134d40995c1f590c22f6df : SUCCESS in 4m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ed2ff0aeee8414fa2d99c921d13225d : SUCCESS in 3m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1921c98809ab48e8b8a485716ccc6f56 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f9d5ef56cf24489b8742a56cd7e92bd9 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/293bf6ba0a46418bac9b73efbec21abf : SUCCESS in 7m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c720ef26f3b542d784af13899ddda540 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc102ec9d429452695871c68aeb0bd5e : SUCCESS in 4m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00f1a84b0a874afca6d1e883050bc10f : SUCCESS in 30m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ef9de0b942314570b7b2110f9fc10e31 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":22},{"id":"250eb72601f4a4bd6d52e29357caa4b521c5a216","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-15 13:02:03.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":23},{"id":"2e0eff160888209469dad2cabf8bf21f75c3ce85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-15 14:21:58.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6992049b5d6e4e148266d287415b6d85\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/2289e4535acc46d9ac369faca572fbed : SUCCESS in 48m 19s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4a0380967b084efeba476cbb40d73a72 : SUCCESS in 37m 54s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/ff540fd86314441884957883c49bbdb2 : SUCCESS in 41m 58s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/a7e09c695c47484aa993dc07376680f8 : SUCCESS in 31m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e296cf0ad7334b29887974188f5debee : SUCCESS in 49m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1c798e36e1ed44dbbf0f0c09dca4572c : SUCCESS in 1h 11m 05s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6237858e893344d083e8a2c5699381a1 : SUCCESS in 1h 07m 29s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e9a09ba2a4844cd19b4ea46f40a3108d : SUCCESS in 42m 52s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/3930548777ab4ef3a00512c37c39ca14 : SUCCESS in 41m 19s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d0d230041d9e4d73a213cfc4fe2dc7b4 : SUCCESS in 44m 46s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/9f760ca476054ede88d5fa03f783306a : SUCCESS in 50m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46a72b032a9f4ad9a02f8609392c85dc : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cf19f56dba04da1b00aeefba3c4de2c : SUCCESS in 3m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/683ba332383e4e9b8c872f76348c5684 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/75a4d87da34841b2a4daad9eed90f51b : SUCCESS in 5m 20s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/442b49b6060246078d0f5149cdd39dfb : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8aaa611b86747ec8e18bb47dfc407c3 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac0160acbb034a858a5aad124c817688 : SUCCESS in 7m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5ec9b4d61d644d139dfa39dad734d102 : SUCCESS in 30m 53s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5e8c16409ed548b08af957cf9c0ccffb : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":23},{"id":"3c6a31a5c4242536a304cdea79a6b925a39a84b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-09-16 10:24:43.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":24},{"id":"22718f1c9795bfd159e3f3ca4dd9e56b1fffe55a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-09-16 11:21:09.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcbff7e5fb2142db87d28fe38cbf4d45\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/7cb46018fde0449ca0244abc03d5493f : SUCCESS in 43m 35s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2dca5a4b7a3d417796a85c8fcd51becb : SUCCESS in 37m 15s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/5f2ebadd0695442c9881563b637667db : SUCCESS in 41m 27s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/6794dd5e14d34e9a8f3687743d40ab64 : SUCCESS in 31m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f766100263264b128d24654dd6495d72 : SUCCESS in 54m 53s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a8e56638cb2645fb8eafca28bf6177d5 : SUCCESS in 52m 33s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6306aafc537241e6b8d65b029f6d87a0 : SUCCESS in 51m 34s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cf32d6876e854b4da66521968265a0c7 : SUCCESS in 51m 27s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/fe6602365ae44d0990efa9627ac6e72c : SUCCESS in 36m 20s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/97499844929f4eb8a4b176e89c89b267 : SUCCESS in 41m 40s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/9dff73c175634dc69e46207e563d5d4c : SUCCESS in 37m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f0f242881674fe4bbbc7e8f97f50890 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5db938264e114bf992ea7a98aea3e8b4 : SUCCESS in 2m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c8cce757968c40f496aa2cc401518b2e : SUCCESS in 5m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5bd0fd491587461c913f4ae7d569cc72 : SUCCESS in 3m 21s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/347af6fcdc5d47a58e5901858fa2278f : SUCCESS in 6m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e55efa61d7de4518926aeda57895d2f8 : SUCCESS in 8m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1bb75d4978948e897466b05e8e4a3de : SUCCESS in 6m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6ce03efd409c4a86a903fdcc02aa94b2 : SUCCESS in 30m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/328516f0928e4f46ac888bb1e6dcd2fd : SUCCESS in 5m 26s","accounts_in_message":[],"_revision_number":24}],"current_revision_number":24,"current_revision":"8ac7e8b6fe3f5d24c93fae369716d714c8ac61d7","revisions":{"ad8701b64dd29abd2d2849e2b562b5c705e1340d":{"kind":"REWORK","_number":1,"created":"2025-04-15 09:59:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/1"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:51:53.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad8701b64dd29abd2d2849e2b562b5c705e1340d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ad8701b64dd29abd2d2849e2b562b5c705e1340d"}]},"branch":"refs/heads/master"},"f165f956b2a83e42c3d91a664c6d17ca6e3742ad":{"kind":"REWORK","_number":2,"created":"2025-04-15 16:44:58.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/2"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-12 01:16:41.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f165f956b2a83e42c3d91a664c6d17ca6e3742ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f165f956b2a83e42c3d91a664c6d17ca6e3742ad"}]},"branch":"refs/heads/master"},"34cc4cf6b8d4ef70ab0ba020f5950f885b43f484":{"kind":"REWORK","_number":3,"created":"2025-04-16 08:41:01.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/3"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 08:40:53.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34cc4cf6b8d4ef70ab0ba020f5950f885b43f484"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34cc4cf6b8d4ef70ab0ba020f5950f885b43f484"}]},"branch":"refs/heads/master"},"af39c1b07d9515639221130b939c42e5232e9662":{"kind":"REWORK","_number":4,"created":"2025-04-16 08:43:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/4"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 08:43:40.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af39c1b07d9515639221130b939c42e5232e9662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af39c1b07d9515639221130b939c42e5232e9662"}]},"branch":"refs/heads/master"},"4d48bf1c71216e45da94b44e0fa147a7a220c38f":{"kind":"REWORK","_number":5,"created":"2025-04-16 10:12:27.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/5"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 10:10:55.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d48bf1c71216e45da94b44e0fa147a7a220c38f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4d48bf1c71216e45da94b44e0fa147a7a220c38f"}]},"branch":"refs/heads/master"},"62a56797631d3d04c3d6eb8a2f90d23d5462ec09":{"kind":"REWORK","_number":6,"created":"2025-04-16 12:27:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/6"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 12:26:53.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62a56797631d3d04c3d6eb8a2f90d23d5462ec09"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/62a56797631d3d04c3d6eb8a2f90d23d5462ec09"}]},"branch":"refs/heads/master"},"654798ec82c88f8477a9840c51bed41e7262f894":{"kind":"REWORK","_number":7,"created":"2025-04-21 16:02:17.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/7"}}},"commit":{"parents":[{"commit":"69b1dc2238282ee27782aed545565cce3bab88bc","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/69b1dc2238282ee27782aed545565cce3bab88bc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-21 16:01:58.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/654798ec82c88f8477a9840c51bed41e7262f894"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/654798ec82c88f8477a9840c51bed41e7262f894"}]},"branch":"refs/heads/master"},"6580ed966ad5dc2d8322b1d68bb5a7747de585de":{"kind":"REWORK","_number":8,"created":"2025-09-22 12:26:34.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/8"}}},"commit":{"parents":[{"commit":"3036a11142e23644714027783c05a52f80e3f485","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3036a11142e23644714027783c05a52f80e3f485"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-22 12:25:42.000000000","tz":0},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6580ed966ad5dc2d8322b1d68bb5a7747de585de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6580ed966ad5dc2d8322b1d68bb5a7747de585de"}]},"branch":"refs/heads/master"},"c9d15e31220aed3a0497aaeaa70707b40fba11fd":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2025-09-22 14:01:59.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/9"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-09-22 14:01:59.000000000","tz":0},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9d15e31220aed3a0497aaeaa70707b40fba11fd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c9d15e31220aed3a0497aaeaa70707b40fba11fd"}]},"branch":"refs/heads/master","description":"Rebase"},"377c063d5ebbfb493b9a3e6cb7fcc90ea527c5b2":{"kind":"REWORK","_number":10,"created":"2025-09-23 12:45:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/10"}}},"commit":{"parents":[{"commit":"3036a11142e23644714027783c05a52f80e3f485","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3036a11142e23644714027783c05a52f80e3f485"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-23 09:42:28.000000000","tz":180},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/377c063d5ebbfb493b9a3e6cb7fcc90ea527c5b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/377c063d5ebbfb493b9a3e6cb7fcc90ea527c5b2"}]},"branch":"refs/heads/master"},"d6bdc17d54b0b2661898b735fddb7ea153f87c05":{"kind":"REWORK","_number":11,"created":"2025-09-23 15:21:15.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/11"}}},"commit":{"parents":[{"commit":"3036a11142e23644714027783c05a52f80e3f485","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3036a11142e23644714027783c05a52f80e3f485"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-09-23 15:21:15.000000000","tz":0},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6bdc17d54b0b2661898b735fddb7ea153f87c05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6bdc17d54b0b2661898b735fddb7ea153f87c05"}]},"branch":"refs/heads/master"},"2b655fee647d7775da0d333b14154d9b1e60c75a":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2025-09-23 15:22:29.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/12"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-09-23 15:22:29.000000000","tz":0},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b655fee647d7775da0d333b14154d9b1e60c75a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b655fee647d7775da0d333b14154d9b1e60c75a"}]},"branch":"refs/heads/master","description":"Rebase"},"b66f63b76be7d375021e012f8fac37626e6b5209":{"kind":"REWORK","_number":13,"created":"2025-09-25 12:09:39.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/13"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-25 12:08:56.000000000","tz":180},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b66f63b76be7d375021e012f8fac37626e6b5209"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b66f63b76be7d375021e012f8fac37626e6b5209"}]},"branch":"refs/heads/master"},"e37af8d5db047f6777e3299c18c44121989f2702":{"kind":"REWORK","_number":14,"created":"2025-09-25 12:43:47.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/14"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-25 12:43:30.000000000","tz":180},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e37af8d5db047f6777e3299c18c44121989f2702"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e37af8d5db047f6777e3299c18c44121989f2702"}]},"branch":"refs/heads/master"},"2f42c150a520f3a7dbfdab9a89b1c76e74edcca3":{"kind":"NO_CODE_CHANGE","_number":15,"created":"2025-09-26 11:22:07.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/15"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-26 11:21:30.000000000","tz":180},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAsisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f42c150a520f3a7dbfdab9a89b1c76e74edcca3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2f42c150a520f3a7dbfdab9a89b1c76e74edcca3"}]},"branch":"refs/heads/master"},"0340adfb784e8685786c5c85bd7da36818847cf6":{"kind":"REWORK","_number":16,"created":"2025-09-29 14:28:30.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/16"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-29 14:22:52.000000000","tz":180},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAsisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0340adfb784e8685786c5c85bd7da36818847cf6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0340adfb784e8685786c5c85bd7da36818847cf6"}]},"branch":"refs/heads/master"},"bacd2785f7ecfbf78fee2c29fd542459497cc4c1":{"kind":"REWORK","_number":17,"created":"2025-09-29 14:32:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/17"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-09-29 14:31:52.000000000","tz":0},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAsisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bacd2785f7ecfbf78fee2c29fd542459497cc4c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bacd2785f7ecfbf78fee2c29fd542459497cc4c1"}]},"branch":"refs/heads/master"},"cc3d03e3a93841926718ea4a007942b3827076b1":{"kind":"REWORK","_number":18,"created":"2025-09-29 16:03:36.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/18"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-29 16:02:50.000000000","tz":180},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAsisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cc3d03e3a93841926718ea4a007942b3827076b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cc3d03e3a93841926718ea4a007942b3827076b1"}]},"branch":"refs/heads/master"},"4ee626ac54b6e530542a5214fc4a53a8a8987e33":{"kind":"REWORK","_number":19,"created":"2025-10-22 09:24:46.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/19"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-22 09:24:23.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAsisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ee626ac54b6e530542a5214fc4a53a8a8987e33"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4ee626ac54b6e530542a5214fc4a53a8a8987e33"}]},"branch":"refs/heads/master"},"d15e3a62b90cee39109c87968e3612c68eeb2503":{"kind":"REWORK","_number":20,"created":"2025-10-22 13:24:20.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/20"}}},"commit":{"parents":[{"commit":"ffa7363f5e3b5f48ff2b314d59843daaa70c1653","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ffa7363f5e3b5f48ff2b314d59843daaa70c1653"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-22 13:23:46.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAsisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d15e3a62b90cee39109c87968e3612c68eeb2503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d15e3a62b90cee39109c87968e3612c68eeb2503"}]},"branch":"refs/heads/master"},"ed52e8751455525ab38919c1d969ef99c057a08a":{"kind":"REWORK","_number":21,"created":"2026-09-10 11:48:19.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/21"}}},"commit":{"parents":[{"commit":"c1337b0de061725b527985b7e6edde7ffb964865","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1337b0de061725b527985b7e6edde7ffb964865"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-10 11:11:34.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAsisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed52e8751455525ab38919c1d969ef99c057a08a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ed52e8751455525ab38919c1d969ef99c057a08a"}]},"branch":"refs/heads/master"},"8c70164f20c6a53aedb9d1ca2dd08674bf1c30ff":{"kind":"REWORK","_number":22,"created":"2026-09-14 16:58:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/22"}}},"commit":{"parents":[{"commit":"99920604e7ddd21cf956e321ea529bda56c2a164","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/99920604e7ddd21cf956e321ea529bda56c2a164"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-14 16:54:49.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAssisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8c70164f20c6a53aedb9d1ca2dd08674bf1c30ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8c70164f20c6a53aedb9d1ca2dd08674bf1c30ff"}]},"branch":"refs/heads/master"},"4bfaab14ef588c60a735392bcf1333a356a55634":{"kind":"REWORK","_number":23,"created":"2026-09-15 13:02:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/23"}}},"commit":{"parents":[{"commit":"2fd490d6535866e4f4ca18dcba6fc2e3f520faa9","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2fd490d6535866e4f4ca18dcba6fc2e3f520faa9"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-15 12:26:42.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAssisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4bfaab14ef588c60a735392bcf1333a356a55634"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4bfaab14ef588c60a735392bcf1333a356a55634"}]},"branch":"refs/heads/master"},"8ac7e8b6fe3f5d24c93fae369716d714c8ac61d7":{"kind":"REWORK","_number":24,"created":"2026-09-16 10:24:43.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/06/947206/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/947206/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/947206/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/947206/24"}}},"commit":{"parents":[{"commit":"2fd490d6535866e4f4ca18dcba6fc2e3f520faa9","subject":"Use Designate DB when ptr record API is called","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2fd490d6535866e4f4ca18dcba6fc2e3f520faa9"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-14 16:43:39.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-09-16 10:23:16.000000000","tz":120},"subject":"Add a Neutron client using openstacksdk","message":"Add a Neutron client using openstacksdk\n\nSo far, some services (e.g. Producer) did not have a user context to leverage\nwhen calling Neutron APIs.\n\nThis patch adds a dedicated Neutron client that would authenticate with\nkeystone as the Designate service account.\n\nCloses-bug: #1935977\nAssisted-By: Claude Code\nChange-Id: I43b2e511542566e80698fb55ab24aff3ae6877a4\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ac7e8b6fe3f5d24c93fae369716d714c8ac61d7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8ac7e8b6fe3f5d24c93fae369716d714c8ac61d7"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
