)]}'
{"id":"openstack%2Fdesignate~947524","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~I9edef4a1f68d86c1d862444917b4fa3ebb3aa1cc","project":"openstack/designate","branch":"stable/2025.1","topic":"2025_1/change_neutron_job","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-05-20 18:40:24.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-19 15:13:03.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-05-09 09:17:44.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I9edef4a1f68d86c1d862444917b4fa3ebb3aa1cc","subject":"Fix neutron job for stable/2025.1","status":"MERGED","created":"2025-04-16 16:55:49.000000000","updated":"2025-05-20 18:41:48.000000000","submitted":"2025-05-20 18:40:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"947524-2025_1/change_neutron_job","meta_rev_id":"3306518c9c08d136db4a0be4cff435377e341dd4","_number":947524,"virtual_id_number":947524,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-20 18:40:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-05-19 15:13:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-04-19 21:51:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-05-09 09:17:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2025-05-19 15:13:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-04-16 18:15:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-04-19 21:51:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-04-19 21:51:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-04-19 21:51:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ac3de5b602e24cfbebb8ffc354e43129be488a53","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-16 16:55:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"40f4280de5bb42de54341dc018fda416721e9c6d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-16 18:15:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2e859d5ec0c840e5b4e09a4de8a868a0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8f08aef712b54797bae6dc3e21b2df1a : SUCCESS in 46m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d10aefcd5a56492eaa55e361c65551c8 : SUCCESS in 58m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/df72b6b087b34629b813a0f9cd185f41 : SUCCESS in 47m 31s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/ddc109466f894601bddefcc69f500813 : SUCCESS in 39m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e8cba798111e4e0787c77cad8b7387a7 : SUCCESS in 38m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/718418c27e174f75b40f63208229a350 : SUCCESS in 1h 14m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8a4ce721a6c5431d8eefe28dbb85799c : SUCCESS in 50m 53s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/15522ccc9f024650908e03f2c78c3493 : SUCCESS in 1h 04m 46s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/dd37c475abe640b49913ea31073f33e6 : SUCCESS in 50m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/09698c95401243429cf64275c94b681f : SUCCESS in 34m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/2ca2c7e6ebce4473830be44376916f7c : FAILURE in 45m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99c8b3d3d3864b86bac05008cba7dca8 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dae017443b0640c58aae06c800e1639c : SUCCESS in 3m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/199af8444a9241f2a5d9b388003f62a9 : SUCCESS in 3m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/727dd66d38d1419789425220f4dc61c9 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16abfc9997f04ef194270da404c07a18 : SUCCESS in 6m 53s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/3d254ec714e64367a3b3e850126917a6 : SUCCESS in 17m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9d9e68bb33f64c47a3e8ad45a0e02070 : SUCCESS in 2m 54s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/9a9e266452da4465987e7b5a515ebc24 : SUCCESS in 44m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"24ba0a7b67e375d914239c9b15edc7a2351f9528","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-04-17 09:52:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7dedf0add3d632def4772f0299e402fc3d799afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-17 10:59:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e38a5324e5e846a89d08b7f8281a517e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b5e5d471fbba468e9c8e8538218742b0 : SUCCESS in 44m 01s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/ec4c2dc532604d1cae64d7dd671d6fc9 : SUCCESS in 57m 23s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/cfdfcc730b1142b1a905839a2c2667f8 : SUCCESS in 48m 02s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/ccb6cfa81893421ca301462ea5ffce6a : SUCCESS in 41m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a8ad4aaba20c449c818697d3ad041aad : SUCCESS in 37m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/392490b3c0ab46deb524fca065dad37d : SUCCESS in 1h 01m 56s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cb83140f76914bd38e456b797de52064 : SUCCESS in 1h 00m 21s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e6669ff5b750474081f2d25a8b203cdb : FAILURE in 48m 41s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/11edadc319b9486591f22b82c8006ac6 : SUCCESS in 1h 03m 32s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c141e2a4cd16414a87d1aeae392d1da1 : SUCCESS in 36m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8f93ecc0091c4df1a588894756c8a314 : SUCCESS in 37m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f9f44ac104846d7bba3beffbac031f3 : SUCCESS in 6m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f7ce5e8975d413c92a1ef907bdbf925 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb58fce7dbfa49339f7799705f73a5c4 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/134043b3b9aa4532a2f15d67b4a2d949 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e92d946fe784ada9b259941dee8c6d4 : SUCCESS in 4m 04s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/a87441d3d8ca408db1e82fa615eb92fd : SUCCESS in 33m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/09c7f09f72af47b691e67bfb5a993ae0 : SUCCESS in 4m 21s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/fe53d9e0c1e44fbea97bbad936bb4d92 : SUCCESS in 42m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc069307439d1882a426c95722f9f63b1cd26f90","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-04-19 21:51:37.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"da709b94ce9ac3963a236856d8299a391bc51c23","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-07 13:33:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"66fb7eeb3c9902d914e70830de1eb1c3fd582176","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 14:39:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce4008d4fdda484bb5e1937928316b30\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/b3f4b70ef7f84bd19a754df5e963e8bc : SUCCESS in 40m 38s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d1fcd34df15d43b18feae6afe697489a : FAILURE in 31m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e0fe60965ecd4b588436ab443a416728 : FAILURE in 22m 30s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e4ab60d805e64325b7cd8052120a816d : SUCCESS in 32m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5599c070e27247839d73c5623dab8cc1 : SUCCESS in 38m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d41b5c3e6f6445d7b87218efda3d9d88 : SUCCESS in 56m 15s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3ca311b4c8e34dbcb673948d6905657b : SUCCESS in 1h 00m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/68619d3fcb0f49ccac5069f53e871bbf : SUCCESS in 50m 09s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5b4db01c2ecd4ff1b7efb197d9f832d7 : SUCCESS in 1h 02m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/91103b9957954a9fab89fa08cf7c4c75 : SUCCESS in 40m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d404e833741f4963a19771d00ab8c5d6 : SUCCESS in 42m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b928d3352174657b260c4b7b7b70b0e : SUCCESS in 6m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96dbeac44f9b46a5a14a587f2e899342 : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b58459de22f4d78a0e33af3f982d144 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e4cc1754ed2472c8a906d32ffa8dc48 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc8e6ddbd5614ee590c74bf407e4d46b : SUCCESS in 8m 35s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/91fc5ff1bb524a10813111b5b076faa4 : SUCCESS in 35m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/059bfc9803d64cb38b090b5829496870 : SUCCESS in 4m 41s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/13ba7bb1e6ee4a5382d9cc6e449e7d1b : SUCCESS in 40m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6279ee9330b0bb5fafa992bcd7e9972b768ffd50","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-05-09 09:17:44.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"4e1bea4d1f2e042a880866769ec50c61fc9e0230","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-19 15:13:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c5afa0d1c70d3dae75aa86dbc47cd7a29e8f2e4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 15:13:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0217113e0cacbfd046c7d0c4b650e86a29c4b15e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 15:37:51.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/78a7d0c1ac904a24b73ae1c186d25bba\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2550a4119d2c44f4b6bb3c688b1793aa : CANCELED\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f8faa81f8ecc4af88ceba7ae21eb2f63 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/50bc1968f2c9416ca1ad39a28ba336e4 : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3a3d22191f1e460c811ad04399a85a7b : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e14a49189a7a4aad9457d30b6ca3ba0c : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0924bb3cf9bd404bb678ccd0270acd20 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b95f41abf794105b75931f6cc7319e2 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ce97e48354247d9ba98b140631b7a3a : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/94e8ead50305404eb6c25860bf5bdb97 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e9dfbf306884e3ea65e5da2e2658fcf : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e430d68e12e49b482bed78aa928b19a : SUCCESS in 9m 05s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/1e92bb75686647b9a26d12a1ef53f4c7 : FAILURE in 21m 44s","accounts_in_message":[],"_revision_number":1},{"id":"4488b9232ece2f9df4180988ffcbb4577767e5d8","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-19 21:23:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1f3048183811fa2e284de2133edc9e7d654d05f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:40:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/065bb7a327ff4729bc06242cba3e255a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9cef8a237b8a4b729a39851d352367a7 : SUCCESS in 49m 17s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/d3f1fec15153432ea9e1604e44bdfeff : SUCCESS in 48m 24s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/883da1b40a3441ee9c1b707737ebe13e : SUCCESS in 51m 05s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/4e2e8b7fb2c94e8eb4cba31404928dee : SUCCESS in 41m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0e56423181574e40a66edfdd7024cb2e : SUCCESS in 42m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b61fcf7f95464f9d8cb0bdc78eb806e0 : SUCCESS in 1h 08m 21s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ff7028f835474fdea0e118d7f32b6ed5 : SUCCESS in 50m 26s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8e429302349946af9cc9466912347cad : SUCCESS in 52m 20s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/27ca1b26cf1f40569ddbea379436f22f : SUCCESS in 1h 00m 42s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fe2c52aa5fd04aeabcbad811f81e66d6 : SUCCESS in 48m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/05eab36f83fa40bba04f1ba11b681675 : SUCCESS in 40m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ff5dc23bb40461683ae6269d05118a2 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6afafd57dad54ede97d13923562353f9 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2adc0f20668d4cfcbd8df62f46458628 : SUCCESS in 2m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4289ffd02a5e45bcb28396296fdd5fce : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dff41f5c11b843d0940fb642a2430c31 : SUCCESS in 8m 46s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/a9511271cd6e457fbd594aa86e3d114f : SUCCESS in 32m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/05418117c5b2485d9ed12070e441b968 : SUCCESS in 4m 24s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/15c6e0920a6d49f9b936f28614fa3899 : SUCCESS in 44m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"23fe4f29be5573cbe0be933353f38b342b2aefa4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:41:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0ed60f85bbf5eb270e8a8bcb7d7cc3644ea6917b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-19 22:53:25.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2efb0656c48421f86fc43dd66a78371\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/2ab1660490e9490b9ee8ec362c8370c0 : CANCELED\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/0ae7b46a715144989c8131816640caaf : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cbd4d194e088443e82cd2e5b379a8acd : CANCELED\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b9d7fcf2665e419bb7af7d17f3180f37 : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/06a1cbc8f67b4a31a26defff63396b49 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7eac547069c1446380ee21244d04c514 : CANCELED\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ba8657040f2c46399c3abcd1f5a22e29 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1aff8d3104554b8ba4deff62f10f61d6 : SUCCESS in 2m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e7ba63d213e4890aaac6b39bdc906e2 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e3a4540dc6974a33a1ced28d8ccdd41c : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7929cdb3b0d4461bbd0f1756e73dc19 : SUCCESS in 8m 05s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/069f06b551114fcda148525a35caa6f0 : FAILURE in 10m 33s","accounts_in_message":[],"_revision_number":1},{"id":"230609b8984c81bbd18321e876ce7e8446ec02d6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-05-20 14:32:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d6c3c62f62ef7e92eb2b86af31ed4e7395b741f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 16:14:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c135f9df669c45b4a0fd42f774c7680b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9be2f7ccfe4141fa9a59414e81ba4052 : FAILURE in 41m 18s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/442ec6d9e850466facf612e284a16e7a : SUCCESS in 54m 16s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/d58653822bc141d3a49c6eaa0e137856 : SUCCESS in 1h 03m 37s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c4a9fa93a4bd40b4ac938a41aff8539e : SUCCESS in 48m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/acd4adafbbe14d4a930487efa0c498fd : SUCCESS in 42m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7d01d9badac245d88df179ccc4d815c7 : SUCCESS in 58m 57s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0ebfb4c914e491f8e49b1c350ba1248 : SUCCESS in 55m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e356ebb0c3a3428c82c672b48a1ff311 : SUCCESS in 1h 03m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/717776b512eb438aac9fc3c47855fae4 : SUCCESS in 1h 10m 49s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cf4a234a1ef64032a323dfcad3556ee4 : SUCCESS in 42m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9a7ea38a74dd44709f4ca6ee9f75ca21 : SUCCESS in 46m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/232eb823276b47939a1f90e01b31d277 : SUCCESS in 6m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/178bc0637a8d4f29986815376baae7d4 : SUCCESS in 2m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0ad2db844424edc84d7b6f99bb75008 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0887554a84934fec9256c8dc5b63417c : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ecc8f8a601b44cf9f7a40539900003e : SUCCESS in 8m 28s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/0499d6fad54f40abbcd78da77c623e43 : SUCCESS in 32m 13s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eb0c3f0dbba84effb3ccc29693c23c1c : SUCCESS in 6m 59s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/f2787cc39c8f46399211ae8cabf3798c : SUCCESS in 45m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e2d6f4a2a17d2a6932c4d15d2b53ad5a6180339","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-20 16:15:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"468b64ec90304b64ea667fd4ad2f364235f39e67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 17:29:23.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03524458ce7c48578ec0db5093901d7d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/735bfb5125944100aeeb36651c06f37b : SUCCESS in 40m 21s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e488398c76fe4406b759f00a92a3acf3 : SUCCESS in 51m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1e3d0f6741e849d59e60c52d60a3debd : SUCCESS in 47m 43s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/d1f3a96853f143b3a7fadffdc1c8a67b : SUCCESS in 44m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d67b1d20035e4563b3653d225b761725 : SUCCESS in 46m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5f3e0bf7a6e54945bab696a60a5b82f4 : SUCCESS in 1h 01m 54s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2d2aff93ad464bafaa57283eeada038d : SUCCESS in 1h 04m 24s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/44efbfe5a36a48d283c050ef88c82d3e : SUCCESS in 1h 05m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/22b74e4a9e4f4a8f951b94988226a046 : SUCCESS in 59m 57s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7943524cbc3741b5932c5049fc09f156 : SUCCESS in 44m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b2adebc65c42456f9500eadd24436296 : SUCCESS in 40m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/da7b843915ba4ce68d7bc5cb84ebb0e6 : SUCCESS in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f47adeccf61449e3868c4eb0490122a8 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3329ad6cdeb41c7a69fa8fb6a57389d : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1fe8bc6e588e4d2a824b93cd61c54724 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfd3e6a6ac9d43d4b0b887ce20c71e1c : SUCCESS in 5m 05s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/e353fe855dbe48ffa1c99571ebbbaa28 : SUCCESS in 32m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/31c8e5bdcdc04a28a121199f41c9b646 : SUCCESS in 5m 31s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/4e396b2e272a4773ad5bb27f3a0440fd : FAILURE in 59m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4c05a19c75aa3a15477f010f93fcfc2a2ae8abcc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 17:29:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8fb23332a5ccdfb0c091896373e0ea363ca6c968","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:40:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/632d8633cd054f4abbc1dbd6c75a7636\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0362c1809b1841db92409cd7461f51fd : SUCCESS in 42m 13s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/31619351ac094d11a5921149984dd183 : SUCCESS in 48m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/dce9a07caa4547048d4386352e0b919f : SUCCESS in 44m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f59495416d3c44b094fda9f756250062 : SUCCESS in 1h 04m 48s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/78f1ba762da34bba953ca2304a226cc1 : SUCCESS in 53m 23s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/003d88c74fcd404995544d058dfd1f83 : SUCCESS in 40m 18s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ce8e0732634e46cc9e01c673d96143d2 : SUCCESS in 41m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ab262745dee45afa2370ef414da24b2 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3389881166f24785a2217a553e0cd7cf : SUCCESS in 6m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b8217cbb81364dd6b4bd0befed8ac7f3 : SUCCESS in 4m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b7668414ac44168b8d44bb17b82e0af : SUCCESS in 9m 47s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/2e686b4eca2946b5bffdb3f62d69dcae : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":1},{"id":"cc81db1d6e934ed33366e915ede0a3c4f9b0bb29","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:40:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3306518c9c08d136db4a0be4cff435377e341dd4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 18:41:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e85b4addeb54e41afe101ca273529b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f3e2720f95ce4ff5abb0a97af6d69759 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0d250cf6bf3fa7a78520c76273ba25e843465375","revisions":{"0d250cf6bf3fa7a78520c76273ba25e843465375":{"kind":"REWORK","_number":1,"created":"2025-04-16 16:55:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/24/947524/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/24/947524/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/947524/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/947524/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/24/947524/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/24/947524/1"}}},"commit":{"parents":[{"commit":"eddaf86b36761f0a3a0ebad4470bd8b221b18931","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eddaf86b36761f0a3a0ebad4470bd8b221b18931"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 16:55:19.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-04-16 16:55:19.000000000","tz":120},"subject":"Fix neutron job for stable/2025.1","message":"Fix neutron job for stable/2025.1\n\nChange-Id: I9edef4a1f68d86c1d862444917b4fa3ebb3aa1cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d250cf6bf3fa7a78520c76273ba25e843465375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0d250cf6bf3fa7a78520c76273ba25e843465375"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
