)]}'
{"id":"openstack%2Fdesignate~948799","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~If90056ae1de4a6a3557fc870497e6233088bfb02","project":"openstack/designate","branch":"stable/2025.1","topic":"remove-wsgi_scripts","attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-21 16:11:51.000000000","reason":"Vote got outdated and was removed: Code-Review-1"}},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2025-05-07 11:24:50.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If90056ae1de4a6a3557fc870497e6233088bfb02","subject":"[Stable-only] Add wsgi module to Designate","status":"NEW","created":"2025-05-05 12:50:34.000000000","updated":"2025-05-22 11:03:52.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":902846,"cherry_pick_of_patch_set":7,"meta_rev_id":"0cfeefe5f9680bd8e2a1b9e50cd5e17c90fd35d8","_number":948799,"virtual_id_number":948799,"owner":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-05-22 11:03:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-05 12:50:34.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-05-05 14:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-05 15:16:10.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-05-05 17:00:05.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"},{"updated":"2025-05-13 13:14:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2025-05-21 16:11:51.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"}],"messages":[{"id":"c8b86a251d35ca39dc574ca5b2e5ac2f4a1a96c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-05 12:50:34.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"e7a152ff30675df4de55c8985bb3f6c9b2bd2ff3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 14:08:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c8b365da9acd46638675b730e3b8a494\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/753a45b15c44405d8d41d42b4f07da73 : SUCCESS in 2m 18s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cdf238846a3c475ab4f327d1a0aee8de : FAILURE in 41m 53s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/b71a3b62bbe84bc3ad3af53e3c6521a4 : SUCCESS in 39m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b24c4e7bbb3742ea9329492661e137b6 : SUCCESS in 42m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/739eefa01a2b4665ba1e9dabe2b373d4 : SUCCESS in 55m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/27139427208042e8b0905672fdf52b1d : SUCCESS in 1h 07m 22s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6c629ed0c24c43d7abff976ffd902ede : SUCCESS in 1h 03m 09s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fd907728e6264fd5b9cb8a0a6de36887 : SUCCESS in 58m 18s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ada0765d4fcb40a79add8ffd45198c75 : SUCCESS in 40m 33s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f1341a1664354a8eb5d51d15a2901dbf : SUCCESS in 41m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9bc56075c9ce4db4904fbdc4296029e0 : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/828151480b7b406fa8c1f347af1f6307 : SUCCESS in 3m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63132502dca04d00a764a2a5bdb00d4a : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/54574c81f08b49e9b3c8ca21aaf13a66 : SUCCESS in 4m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c78712bb8614522863b2204d9019180 : SUCCESS in 8m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a6a3e4568f9a42ae9217915428924f47 : SUCCESS in 3m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/04cfca06b4c8438488b8e50cfa8d75c9 : SUCCESS in 32m 48s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7993de61b3f54e4c850be6a093c80033 : SUCCESS in 4m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/938a81f03b2544e0a7c85c114282b0ae : FAILURE in 46m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c48ea74a51b0729e0cd9897688eaf14f6faeebd3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-05 15:16:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58dc753f279ffce1f97a86f15f59a0762b2484a1","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-05 15:26:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f6f7d80e27a3e5c3e7b2c72c4a9794074bf2fb30","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-05 17:00:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e42042584d1b1d8d040d804c32105b730cfcb654","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-07 11:24:50.000000000","message":"Abandoned\n\nThank you for providing this patch. I am abandoning it as Michael implemented this one recently https://review.opendev.org/c/openstack/designate/+/948963","accounts_in_message":[],"_revision_number":1},{"id":"c41ac8a3dc8b60454a0a3b84c045bfa89717ea40","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 13:14:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cc6df05b7c88756d26b21e9d1296d688cf7a3dd","tag":"autogenerated:gerrit:restore","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2025-05-21 16:11:21.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"7bac2f3df3345f0caa4586382d56cefd7eaf5a71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-21 16:11:51.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"32314aeac468aa6a7beed152b51159f3e62cc35e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-21 16:32:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5551f21652bf0ae36c6f6fb09a83b7d1702b0f5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-22 09:51:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"0cfeefe5f9680bd8e2a1b9e50cd5e17c90fd35d8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-22 11:03:52.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b93bfe12d793499c87b3285b01378356\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fb7107b9e410401e95ff8d249727c169 : SUCCESS in 46m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/44b29b90ca44485b88f91cc77f3fdf40 : SUCCESS in 52m 50s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fccd473652db42dcb106a3b3dc17b70b : SUCCESS in 54m 26s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/050f7f065eb24c57a566d44f765c6223 : SUCCESS in 42m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b128fc3c7fa04ef5ac569acde0cede15 : SUCCESS in 40m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e02e3ae40ad74e16b8b863ec0a403c09 : SUCCESS in 1h 01m 23s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/956895ac4b514a8785757cc7001d9ada : SUCCESS in 1h 04m 45s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/66a167ece7824e758ce6c25f241c4af6 : SUCCESS in 1h 04m 53s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/dd074dd3ba81400cbd8a366a831cff57 : SUCCESS in 1h 11m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/80c227224fab4fd8bd667d8e7cdc6836 : SUCCESS in 39m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/27cd1feb4d964867a80dd256da4e8f18 : SUCCESS in 50m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87ecda6b08ac4982aa1448b9060bf13e : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06cbecf07318441f9558e43dc00e0341 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea35d96c533c42e88d073ea50e2c0518 : SUCCESS in 4m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/99b41b09807747ecaa9e050fe40fcd70 : SUCCESS in 6m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55074d7b4ff24964a0ee5e37a2ff7472 : SUCCESS in 9m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd0649a86d2b468da6ac6780ee5bf8d2 : SUCCESS in 5m 12s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/6626e74989f34bb2a4d3e12f75fb67d6 : SUCCESS in 33m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a6e1c92a0a214e379dcb48726ab8d8d1 : SUCCESS in 6m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/adac5f9fe01c4e7d8fd5a635ced83fe7 : SUCCESS in 46m 15s (non-voting)","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"4f0185901f90630ec980ade00d8001da715d68da","revisions":{"9833f7660a32b0426c8fccb693bf8c9f41a13918":{"kind":"REWORK","_number":1,"created":"2025-05-05 12:50:34.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/99/948799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/948799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/948799/1"}}},"commit":{"parents":[{"commit":"eddaf86b36761f0a3a0ebad4470bd8b221b18931","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eddaf86b36761f0a3a0ebad4470bd8b221b18931"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:53:18.000000000","tz":0},"committer":{"name":"yatin","email":"ykarel@redhat.com","date":"2025-05-05 12:50:34.000000000","tz":0},"subject":"Add wsgi module to Designate and remove wsgi script","message":"Add wsgi module to Designate and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Designate from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Designate to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nIn order to make the CI work, this patch was merged with:\n\nzuul: Drop centos9 and make grenade non-voting\n\nThe requirements repo has dropped upper-constraints for python3.9, so\nthe centos9-based jobs are no longer working on master, let\u0027s drop them.\n\nAlso the grenade jobs are broken due to the referenced bug, make them\ntemporarily non-voting, so that we can merge the fix[0] on master first,\nbackport it, and then re-enable them.\n\n[0] https://review.opendev.org/c/openstack/designate/+/902846\n\nRelated-Bug: 2109577\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: If90056ae1de4a6a3557fc870497e6233088bfb02\n(cherry picked from commit e6d83dfcf729fff333cfc5b700c62efb8aa4aebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9833f7660a32b0426c8fccb693bf8c9f41a13918"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9833f7660a32b0426c8fccb693bf8c9f41a13918"}]},"branch":"refs/heads/stable/2025.1"},"741f1ecb7da6057db815bc8e127d425cfacdd2d6":{"kind":"REWORK","_number":2,"created":"2025-05-21 16:11:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/948799/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/948799/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/948799/2"}}},"commit":{"parents":[{"commit":"a891d81974a4c3aa56a143252b26df9dcbd5fba5","subject":"Merge \"Fix neutron job for stable/2025.1\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a891d81974a4c3aa56a143252b26df9dcbd5fba5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:53:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 15:59:45.000000000","tz":60},"subject":"[Stable-Only] Add wsgi module to Designate","message":"[Stable-Only] Add wsgi module to Designate\n\nThis is modified change If90056ae1de4a6a3557fc870497e6233088bfb02 that\ndoes not drop \u0027wsgi_scripts\u0027 since we still rely on this. We will remove\nthis separately.\n\nRelated-Bug: 2109577\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: If90056ae1de4a6a3557fc870497e6233088bfb02\n(cherry picked from commit e6d83dfcf729fff333cfc5b700c62efb8aa4aebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/741f1ecb7da6057db815bc8e127d425cfacdd2d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/741f1ecb7da6057db815bc8e127d425cfacdd2d6"}]},"branch":"refs/heads/stable/2025.1"},"0f783cfb897a1ad0aff2dfadc41739f7e04a656f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-05-21 16:32:37.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/948799/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/948799/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/948799/3"}}},"commit":{"parents":[{"commit":"a891d81974a4c3aa56a143252b26df9dcbd5fba5","subject":"Merge \"Fix neutron job for stable/2025.1\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a891d81974a4c3aa56a143252b26df9dcbd5fba5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:53:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 16:32:37.000000000","tz":0},"subject":"[Stable-only] Add wsgi module to Designate","message":"[Stable-only] Add wsgi module to Designate\n\nThis is modified change If90056ae1de4a6a3557fc870497e6233088bfb02 that\ndoes not drop \u0027wsgi_scripts\u0027 since we still rely on this. We will remove\nthis separately.\n\nRelated-Bug: 2109577\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: If90056ae1de4a6a3557fc870497e6233088bfb02\n(cherry picked from commit e6d83dfcf729fff333cfc5b700c62efb8aa4aebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f783cfb897a1ad0aff2dfadc41739f7e04a656f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0f783cfb897a1ad0aff2dfadc41739f7e04a656f"}]},"branch":"refs/heads/stable/2025.1","description":"Edit commit message"},"4f0185901f90630ec980ade00d8001da715d68da":{"kind":"REWORK","_number":4,"created":"2025-05-22 09:51:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/99/948799/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/99/948799/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/99/948799/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/99/948799/4"}}},"commit":{"parents":[{"commit":"a891d81974a4c3aa56a143252b26df9dcbd5fba5","subject":"Merge \"Fix neutron job for stable/2025.1\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a891d81974a4c3aa56a143252b26df9dcbd5fba5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:53:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-22 09:50:23.000000000","tz":60},"subject":"[Stable-only] Add wsgi module to Designate","message":"[Stable-only] Add wsgi module to Designate\n\nThis is modified change If90056ae1de4a6a3557fc870497e6233088bfb02 that\ndoes not drop \u0027wsgi_scripts\u0027 since we still rely on this. We will remove\nthis separately. We also don\u0027t migrate our DevStack plugin since we want\nto test everything is working before doing so. Again, we can do that\nseparately.\n\nRelated-Bug: 2109577\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: If90056ae1de4a6a3557fc870497e6233088bfb02\n(cherry picked from commit e6d83dfcf729fff333cfc5b700c62efb8aa4aebd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f0185901f90630ec980ade00d8001da715d68da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f0185901f90630ec980ade00d8001da715d68da"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
