)]}'
{"id":"openstack%2Fdesignate~948963","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~I19b3307a4097fa6569cf694acdfb4b0f4687b0ff","project":"openstack/designate","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-07 12:14:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19b3307a4097fa6569cf694acdfb4b0f4687b0ff","subject":"Add an designate-api-wsgi compatibility script","status":"MERGED","created":"2025-05-07 01:01:46.000000000","updated":"2025-05-07 12:15:06.000000000","submitted":"2025-05-07 12:14:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"948963","meta_rev_id":"2ca54ca98fce2fcb78bd570fd4ac83f8a587a471","_number":948963,"virtual_id_number":948963,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-07 12:14:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-05-07 11:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2025-05-07 04:43:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-05-07 11:23:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-05-07 11:23:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 02:07:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-07 11:23:11.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"74b5bbdce0c342a8e006b1d9aa2e8394f6ab7e20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-07 01:01:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8aede5105e5f95a83e8b9f2490ee3e04d130020a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 02:07:30.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f032d8611f34eec9d3c3e1afa1597b9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c18c4f7cde8740478dbb999e89dbbd20 : SUCCESS in 1m 30s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/32c5344513494148ad452c7fb83370be : SUCCESS in 42m 11s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/2d114c68b8364971a73b850db0b67131 : FAILURE in 19m 57s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dad353c3db484a589f065b44b1c3427d : FAILURE in 18m 34s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/fd23e3f36a3f491ebe1ae5669b0db84e : SUCCESS in 33m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e8303da71c9d4f7f91c7a62e0ddddd4f : SUCCESS in 32m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5476075c0145494cb30e8e8840438b3c : SUCCESS in 50m 30s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4facfbc05b4f42e6a04fb0fab1d3cbfd : SUCCESS in 50m 03s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4522037abc104279b2639533a3f75f6b : SUCCESS in 46m 19s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9ab9d8a4a78f4b3582d52880368188f4 : SUCCESS in 51m 09s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cdb744a4dfed407b851773ac6926413f : SUCCESS in 45m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/856cd79228924b629ac1802169527914 : SUCCESS in 42m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7e43154be0a746168d9e730c264305ef : SUCCESS in 6m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b030d34a8bfb4a8781da49ba17572343 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bdf7897dc41c4fdea3ee4c35608a417a : SUCCESS in 4m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9d771fea6bf241fd93e54d72eb11360b : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3fd12f87b144958a96107d4fee9ac1a : SUCCESS in 3m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a98f51b8e11045e1b4022f38231e2232 : SUCCESS in 2m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0b20d1d3c419489c9c082b9ff940a780 : FAILURE in 19m 51s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/48e244d35b3343b9b43b22f9f63a1bda : SUCCESS in 2m 46s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0251aa65996c48f78575a1b4b97e8ed1 : SUCCESS in 38m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3985ec38cef6d94f28c3f5e8dfacdcdbc73bdb02","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-07 02:58:33.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf2c9f5721bd729bd463d40b7e66b3ec89d559b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 04:04:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d5b3c3391644874920c721a24efbe96\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e3b41def692d49e7bbd36f327d05cf1a : SUCCESS in 1m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f467f220120c4654b81e0fa295d6d773 : SUCCESS in 38m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/c2fed6b2b4564157aaba048cc11ff887 : FAILURE in 17m 47s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/e1721f47dee344cebdc6449c2638e22d : FAILURE in 11m 52s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f1ed26f955674acba77990dcdb2e1e20 : SUCCESS in 35m 56s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5751053c50944e979b2b24e3233919da : SUCCESS in 48m 31s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3ca05449840a4768bc8e80f1cfe28945 : SUCCESS in 58m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/16133928724446d389732132f57a4cd6 : SUCCESS in 1h 02m 18s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3ba420b19f804f449983066460b37177 : SUCCESS in 56m 24s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a8a82b0e00b74a768f91acb507b258ff : SUCCESS in 47m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/20f1915d659847cba0ba3babf01b7182 : SUCCESS in 47m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6c58f4cbeb7a4a82813daaf7921b00e8 : SUCCESS in 43m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f17f93ef1f7b443f8271812b4269e1a9 : SUCCESS in 4m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db07929b8b9435199ef6876c9ec7ddb : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ccb8c5520681417881d7f02af40485f6 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e6356e84e0fc4221a31dcc41772b4044 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc1ec5f5f6b340e3ae71569deb9fbceb : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a5263bffe38410796fd43eeeaf25ca4 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e67ed57f95ee47ee83525944b3d09cbc : SUCCESS in 33m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/27817bf3a8c24f1fb63f372c07fb31cd : SUCCESS in 5m 03s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/aab87cef48e5403ba986bdaaefc130a4 : SUCCESS in 42m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3684454a387a2d7985084754c88d2915840a8105","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-07 04:43:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"99fa62be1355cb258d5b6cf07b71625511f909c4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-07 11:23:11.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c74bc5d64f2cb8bbfc0132ec6ca9abf783283f1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 11:23:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c32cc811938788e76932a7aa0504d0251efaf2f6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:14:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed486378a3a4de29afcb621b685f449\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/683bb8727ee8415ea6e02a5f23ba8837 : SUCCESS in 2m 15s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/d3f5be4daf1b442b80d8ee0ce3da68e9 : SUCCESS in 46m 36s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/8eac1e91516f43b6a4030111dc14c2e3 : SUCCESS in 37m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e4380af27b164a1d9891d15f32af7dc2 : SUCCESS in 44m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d23525c826d4450e85730f4bdae6952b : SUCCESS in 49m 31s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a42b0b6fdef14d80875c021219a783ec : SUCCESS in 47m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b506b30c83304767bf0d9c74a881f879 : SUCCESS in 36m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/efd232c7be5a45b2a9fd791475acf37d : SUCCESS in 37m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e011bc15a8834f74ada5c51d5dfe9103 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5a7473ab86014cd893372fa304b71d47 : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c01fbe90cc14f5cbd8b204459457cc6 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bcab5eaed0242eea393249539b75ff8 : SUCCESS in 9m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c073a047429476c8c5e94a4113452ba : SUCCESS in 2m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b34fe23425f248408fa2f45969faa97c : SUCCESS in 40m 22s","accounts_in_message":[],"_revision_number":1},{"id":"5724744ecdfc5a05faa7a593032399db4ba84a27","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:14:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2ca54ca98fce2fcb78bd570fd4ac83f8a587a471","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 12:15:06.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d18586b4962548df8cde9474516b7174\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1c1e9c9b207430db8aacf98f4e2a629 : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ece33cad084c42588ea3d13e5055cba5 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5f4d6c4c6b68bea0444b516ad000ef1fb6b67673","revisions":{"5f4d6c4c6b68bea0444b516ad000ef1fb6b67673":{"kind":"REWORK","_number":1,"created":"2025-05-07 01:01:46.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/948963/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/63/948963/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/948963/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/948963/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/63/948963/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/63/948963/1"}}},"commit":{"parents":[{"commit":"eddaf86b36761f0a3a0ebad4470bd8b221b18931","subject":"Update TOX_CONSTRAINTS_FILE for stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eddaf86b36761f0a3a0ebad4470bd8b221b18931"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-07 01:00:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-07 01:01:45.000000000","tz":0},"subject":"Add an designate-api-wsgi compatibility script","message":"Add an designate-api-wsgi compatibility script\n\nThis patch adds an designate-api-wsgi script that gets installed with\nDesignate for backward compatibility on the stable branches.\nThe pbr wsgi_scripts function no longer generates this script\nautomatically due to recent changes in setuptools.\n\nThis is a stable branch only patch.\n\nChange-Id: I19b3307a4097fa6569cf694acdfb4b0f4687b0ff\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f4d6c4c6b68bea0444b516ad000ef1fb6b67673"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f4d6c4c6b68bea0444b516ad000ef1fb6b67673"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
