)]}'
{"id":"openstack%2Fdesignate~949094","triplet_id":"openstack%2Fdesignate~stable%2F2024.2~I19b3307a4097fa6569cf694acdfb4b0f4687b0ff","project":"openstack/designate","branch":"stable/2024.2","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-17 16:32:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19b3307a4097fa6569cf694acdfb4b0f4687b0ff","subject":"Add an designate-api-wsgi compatibility script","status":"MERGED","created":"2025-05-07 22:02:29.000000000","updated":"2025-05-19 11:07:22.000000000","submitted":"2025-05-17 16:32:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949094","cherry_pick_of_change":948963,"cherry_pick_of_patch_set":1,"meta_rev_id":"92ccf9fb9a9b6f3a2aad4d67be755462017c4956","_number":949094,"virtual_id_number":949094,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-17 16:32:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-05-16 16:01:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-17 15:41:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-17 15:41:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-05-16 10:28:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 23:01:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 10:28:42.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-05-17 15:41:37.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-05-19 11:07:22.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"CC"}],"messages":[{"id":"b2c9c534549e964ce7188d660b247738a5368d2e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-07 22:02:29.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"f2c93eedf2e58af1b7072d3735879ffe6cbd9d36","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 23:01:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46d11f02d7534628a29bd931431d9d1c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4da575899c6b4e559d1ddbb2a9b8575b : SUCCESS in 1m 29s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c6561bfcce57458bbd4dfda205164897 : SUCCESS in 45m 20s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bdb817da1405423d85db77420b8c3b82 : FAILURE in 14m 58s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/a0230912b16d45ebac0e58a9b87e94f5 : FAILURE in 18m 37s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4ced4690ce4a418f8156d8328c276fe9 : SUCCESS in 39m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/41bab228370142868737f1dde9676556 : SUCCESS in 33m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/01d07ca07f1a452690086c917fd268b7 : SUCCESS in 49m 55s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6790eb349e70497baaad6bffc5d0829c : SUCCESS in 57m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/972f8385aef5429786e204466b08f482 : SUCCESS in 34m 39s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/10fe1b13d97648cd81b58a944f101f44 : SUCCESS in 38m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbea5283cae44ef380b5012bbd3323db : SUCCESS in 4m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27280b605a154f478c5951163b27d25d : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3f4ff751605f4e2484f319a93c535c7d : SUCCESS in 3m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0a4757c72902441d8bd624165dd84e3f : SUCCESS in 4m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3b530ac3769849b68bb03273a2d071ef : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a1de228d4df54f729b75617c4bce0857 : SUCCESS in 4m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/90426dd36be4410db0d3f8e74a43dbd4 : SUCCESS in 4m 00s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/66eecc43f0384c8e8ede3f8b3db22bc0 : SUCCESS in 20m 56s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b61248b5e8484befa2cdbe9c37517a1b : SUCCESS in 4m 25s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b2cf6f146fa647eea57e6276ed70ee4f : SUCCESS in 47m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f4cae4fdba51b3ceecb93df21979d2db31fff449","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 10:28:42.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0e79dc5f81a041224a477832c453415203eec06","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 10:36:28.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d039b80e3ce2d04ebab651a23d1cee2a46a06da","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 16:01:58.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6a9315aa9c70245e324f88388e188320577b5815","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-05-17 15:41:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8a42a7ddd48b9fb9d1d887253da6f6bb64f82022","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 15:42:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0d938f042781004e1e3b77add29a416fe1d0f11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 16:32:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d3824196b274e209bb9635e6ab458c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/10b92373dfc5432489372726dad59b1a : SUCCESS in 2m 02s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f86ed4375d2149c9bd2c213e0aad14e6 : SUCCESS in 33m 16s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5dd6870c61c6464f9cd1df2cfa1281c3 : SUCCESS in 31m 35s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a658825d64c648cb8e39069b6441f287 : SUCCESS in 30m 02s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7330b459d7f04c99b05107814eade115 : SUCCESS in 48m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d19726f27224414e93f3b68619ad73da : SUCCESS in 29m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c91c4b4b1f194f6c9c65de7c29caa193 : SUCCESS in 36m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1824e634d94afca247d16a766809d0 : SUCCESS in 2m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a90dc5c0fb484219a6822cdaea9d5459 : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a47bc5c608541f1a90b7c9251e793cb : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05276d1ddd1c42fe8cdf12952c2148cc : SUCCESS in 3m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4bbf42c075b8487886969d0d4afc95cc : SUCCESS in 3m 19s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/d77f417b23504f96ab629d01205558ed : SUCCESS in 26m 59s","accounts_in_message":[],"_revision_number":1},{"id":"f8a6ff4fd9956e0f99dc5acaf25efda4ccd3a948","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 16:32:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9bbd8c21721dad873b42004448588c39912481b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-17 16:33:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/301d28f32a1b4e98898218e1465c1b9f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbd187d68b6a4138aec56ab771e99a38 : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9912220626b3461284fdaf87ac9df5b1 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1},{"id":"92ccf9fb9a9b6f3a2aad4d67be755462017c4956","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-19 11:07:22.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5bfc35864c28a97ded6104dea05fa9691a7f14d8","revisions":{"5bfc35864c28a97ded6104dea05fa9691a7f14d8":{"kind":"REWORK","_number":1,"created":"2025-05-07 22:02:29.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/94/949094/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/949094/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/949094/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/949094/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/949094/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/949094/1"}}},"commit":{"parents":[{"commit":"5490df4e6b60f00e31e0ff57f863e406728e4eef","subject":"Support \"-\" in region name for FIP\u0027s","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5490df4e6b60f00e31e0ff57f863e406728e4eef"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-07 01:00:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-07 22:02:29.000000000","tz":0},"subject":"Add an designate-api-wsgi compatibility script","message":"Add an designate-api-wsgi compatibility script\n\nThis patch adds an designate-api-wsgi script that gets installed with\nDesignate for backward compatibility on the stable branches.\nThe pbr wsgi_scripts function no longer generates this script\nautomatically due to recent changes in setuptools.\n\nThis is a stable branch only patch.\n\nChange-Id: I19b3307a4097fa6569cf694acdfb4b0f4687b0ff\n(cherry picked from commit 5f4d6c4c6b68bea0444b516ad000ef1fb6b67673)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5bfc35864c28a97ded6104dea05fa9691a7f14d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5bfc35864c28a97ded6104dea05fa9691a7f14d8"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
