)]}'
{"id":"openstack%2Fdesignate~949095","triplet_id":"openstack%2Fdesignate~stable%2F2024.1~I19b3307a4097fa6569cf694acdfb4b0f4687b0ff","project":"openstack/designate","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-18 09:59:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I19b3307a4097fa6569cf694acdfb4b0f4687b0ff","subject":"Add an designate-api-wsgi compatibility script","status":"MERGED","created":"2025-05-07 22:03:13.000000000","updated":"2025-05-18 10:00:36.000000000","submitted":"2025-05-18 09:59:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949095","cherry_pick_of_change":949094,"cherry_pick_of_patch_set":1,"meta_rev_id":"7544c93596b976ef5ed474fc41d32fcbf6de385e","_number":949095,"virtual_id_number":949095,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-18 09:59:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-05-16 16:02:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-18 09:19:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-05-18 09:19:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-05-16 10:29:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-07 23:24:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 10:29:04.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-05-18 09:19:04.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"3b6aa526a5f8f2ad2aa536246356743f476c9809","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-07 22:03:13.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.2.","accounts_in_message":[],"_revision_number":1},{"id":"2c6bdc67d55764eafde131ce73346d7f23ce0860","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-07 23:24:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ca71eb7bc844b22897b5d545817aef9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f55a8e3e08445e68be92660501316c2 : SUCCESS in 2m 25s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0c7c62d167e54b349b1dcedb2e076203 : SUCCESS in 35m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/1072a85624c245489d23dea320f992be : FAILURE in 18m 53s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/b1850d6071a249069e4c35572bae7225 : FAILURE in 26m 56s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/aca1fc10a27649edb38aa1077f4004c2 : SUCCESS in 40m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5458477c26d94717b998d17e141e71b8 : SUCCESS in 37m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b6e26b2f67ce4155b7069dab03c0c1ee : SUCCESS in 1h 04m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/183b1845590a46c29506b534a99e3acb : SUCCESS in 1h 01m 05s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d0275028194f4ba18169de4b9642420b : SUCCESS in 35m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c0ddc3a6e7ba439b9228c1a0bde2b71a : SUCCESS in 29m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/43177e59b544469fa6dceb26cfd542f6 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c824801f97124d7cb5c78a8b5f6d4880 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b486caec1e14ff392a1b41115280de0 : SUCCESS in 4m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c842c5e7e014675aa8c211340b22502 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60589b0682e24f7eacf944c2ef9d6d00 : SUCCESS in 6m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c7034200d7b74e7db27bfde6b9bbc4ef : SUCCESS in 2m 54s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/73c255282c6d4253beddf91a973f32e0 : SUCCESS in 17m 00s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5e6d315de61b4c1896e93da34a2a00c5 : SUCCESS in 3m 58s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8c21bd36ead64c568e37292915830bbf : SUCCESS in 34m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d96b9bcc0bb756e8c918833f5d5c745da62ec9c4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 10:29:04.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5f485c360a6b961f3ea72813b23b642f2624d212","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 10:36:49.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f5d61f8b5e7fc8c4ebf063fa99701d1e2fe450f6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 16:02:08.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fbc42c6a2e0d02ea10bb2dd89027a364f75816d2","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-05-18 09:19:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3a3e28e5b6761ff1d2758ae8a4192fd93213f87d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:19:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"39c873b8d0f3e4f8bed9bc4dc8089518f55e892b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:59:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c839b00171a4591a4589099cc6bc902\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5414ac7b82c241a795ff4cea79a5ad6f : SUCCESS in 1m 43s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/692173aa5e2d42cfa33f9f48fedc60d2 : SUCCESS in 33m 55s\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/75595cacfb9345128dfbe8af57794907 : SUCCESS in 38m 44s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cd4c824d7bad490bab5b8f398e9eda69 : SUCCESS in 33m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/02a060b8df5d41229e278fbfaa1d18b6 : SUCCESS in 34m 24s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d078de480d7e45f1ba89dc0e4e9b820c : SUCCESS in 39m 56s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/15b701630b064ceaa530209813ef124f : SUCCESS in 35m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94ebd7834ae24a689d524f687764086f : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09e0be73a92c4fc885c516d15db67890 : SUCCESS in 3m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/52305be1be3745359eeeff5afa799dfa : SUCCESS in 4m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5181313fd0ba4be18c098898777e307b : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/edbef2585d70422d9948459e52dcdaf8 : SUCCESS in 3m 32s\n- neutron-tempest-plugin-designate-scenario-2024-1 https://zuul.opendev.org/t/openstack/build/2e5a013c99b7483dafbdbac336c91948 : SUCCESS in 27m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4059a2e7adb44cbac1b36f676d25ff98dab58bf9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 09:59:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7544c93596b976ef5ed474fc41d32fcbf6de385e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-18 10:00:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0df3b386c1d415787a91413e0cdbb46\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ad06bd27a7f4f88845bca17a6c39730 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/701e3245049d425e9159e19407db13fb : SUCCESS in 59s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"af7445fa0add6950f564cee6d9bbc052ba90d90c","revisions":{"af7445fa0add6950f564cee6d9bbc052ba90d90c":{"kind":"REWORK","_number":1,"created":"2025-05-07 22:03:13.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/95/949095/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/95/949095/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/949095/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/949095/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/95/949095/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/95/949095/1"}}},"commit":{"parents":[{"commit":"a33f7376407c57d86963b7f763c4ded7b4e16d3a","subject":"Support \"-\" in region name for FIP\u0027s","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a33f7376407c57d86963b7f763c4ded7b4e16d3a"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-07 01:00:39.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-07 22:03:13.000000000","tz":0},"subject":"Add an designate-api-wsgi compatibility script","message":"Add an designate-api-wsgi compatibility script\n\nThis patch adds an designate-api-wsgi script that gets installed with\nDesignate for backward compatibility on the stable branches.\nThe pbr wsgi_scripts function no longer generates this script\nautomatically due to recent changes in setuptools.\n\nThis is a stable branch only patch.\n\nChange-Id: I19b3307a4097fa6569cf694acdfb4b0f4687b0ff\n(cherry picked from commit 5f4d6c4c6b68bea0444b516ad000ef1fb6b67673)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af7445fa0add6950f564cee6d9bbc052ba90d90c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/af7445fa0add6950f564cee6d9bbc052ba90d90c"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
