)]}'
{"id":"openstack%2Fdesignate~949126","triplet_id":"openstack%2Fdesignate~master~I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd","project":"openstack/designate","branch":"master","topic":"add-service-clean-cmd","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-07-03 16:33:06.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"12404":{"account":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"last_update":"2025-07-03 17:35:36.000000000","reason":"Change was submitted"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2025-07-03 17:35:36.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-07-03 16:26:32.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-07-03 17:35:36.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd","subject":"Add `designate-manage service clean` cmd","status":"MERGED","created":"2025-05-08 08:06:21.000000000","updated":"2025-07-03 17:37:44.000000000","submitted":"2025-07-03 17:35:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":22,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949126-add-service-clean-cmd","meta_rev_id":"7a79f6370c0e89b403d6f26a097c1888f6bceebc","_number":949126,"virtual_id_number":949126,"owner":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-03 17:35:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-07-03 16:33:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2025-07-03 16:26:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-07-03 16:33:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}],"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-08 09:20:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-09 06:44:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2025-05-09 06:44:50.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-05-09 06:45:02.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-05-09 09:06:45.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2025-05-09 13:58:13.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"},{"updated":"2025-05-16 12:35:24.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"0bb15214b4533c38dcfb5bce42613c68625bf2b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 08:06:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4e9604346c6bac526ef173a4af7254ce67415679","tag":"autogenerated:gerrit:setWorkInProgress","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 08:06:42.000000000","message":"Set Work In Progress","accounts_in_message":[],"_revision_number":1},{"id":"46da9cbf44578d57caef5d95ff2715085652a331","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 08:07:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"41ffb8314eb00a57091d627492912d2e0d7fefe9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 09:20:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/60403aaaa5a0413baf635c28fc4612c1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e480540b2e2f4a649af637f8b7c20ac5 : SUCCESS in 2m 22s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f3a70ee83b7743d58da5a24ca4dc5e1a : SUCCESS in 39m 26s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2133dfe2d3a84b79a6306aa3d18ac661 : SUCCESS in 36m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/84e24a164dbd4aa08f09b3df91017191 : SUCCESS in 46m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/49d1b9d46c9741859fb6ac598bb5fc3a : FAILURE in 41m 48s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c7afde54336c43118d6bd2d28a5e26f5 : FAILURE in 19m 50s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a5ab64fb4b9d41708b442965708be9c8 : SUCCESS in 50m 56s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4c285e0f14044046b6367421c01a9815 : SUCCESS in 55m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/09d6d5d6a02148539ee205f491d494e9 : SUCCESS in 49m 00s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d7df55a07eb34263ba696d793bd81383 : SUCCESS in 40m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ec817d284094a28b0b8555549a527dc : FAILURE in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb37c47f7d824dc7b9b2919055d76d63 : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5044f256fb4440ea839c9c4b78d6455f : FAILURE in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/25a947621a30482b82ef7f7c9e197786 : FAILURE in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/824c3ec75fef44d2b08d734c3440b2d7 : FAILURE in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75f2ced1cc9b481cb20f19a32416b517 : SUCCESS in 9m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/324be2783d0a479eb2cdc7ea59913322 : SUCCESS in 3m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5cfd8562b4e3436ba2edaf9320fd35ea : SUCCESS in 43m 03s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e8569a740ea04e5eb55b545c20b4f023 : FAILURE in 3m 51s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d7c8a101864242bf9ea5fc2a65cfb52f : SUCCESS in 39m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16ed6eb42be2e01a34b226ade204c7613cdc30bf","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 14:04:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"eb2f18da530b39b6c6cebad0152a8cb67e610606","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 15:09:49.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e17ddbaa10ee4ad78ba5139660d841a4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e762f703ff00422484a72ea3117565e2 : SUCCESS in 2m 20s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e4508447ba434cf48c95f6e375bdab25 : SUCCESS in 49m 39s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/6e03a6c8dab3480083bb0de93433d22d : SUCCESS in 33m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/3158a0dd630b40518ac1e3c6bbda29d3 : SUCCESS in 48m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/87e5ba89d31b425a916648a05eacd962 : FAILURE in 40m 20s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6ee01442aa3547be84db5688700a90df : FAILURE in 44m 00s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8d69c3d2cfc34769b33525b3b57ef170 : SUCCESS in 54m 53s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/11b5292555b64ceebed9ac95854cd130 : SUCCESS in 1h 00m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6f9444febcbb45b4b0ceb39f649f2108 : SUCCESS in 49m 45s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ffa3445e275049bca82c99bc8560d709 : SUCCESS in 46m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52e4bf37a78246ee8484b89333d61a15 : SUCCESS in 6m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0be281c1853b4364843076d579eaca49 : FAILURE in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/50aaeca841104e54be2723e554cbf447 : SUCCESS in 5m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/80a61fcecdf64700ba501c76f8ec795c : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/89774d0f7d44469abab180def5490996 : SUCCESS in 8m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5b9079d087440e5a1b0099b6e656217 : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e82d6ae40804d8ead6da45d0d01a552 : SUCCESS in 3m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0ca77904b0e94845873a9f031a2d891c : SUCCESS in 35m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/14902341c7fb44adab99255db0997007 : SUCCESS in 6m 50s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/d511aaa4b40943ba9290c68bb0424bfd : SUCCESS in 44m 21s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b5c825d80024313ac355e1fe1df3a9d49fd1bca","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 15:45:30.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"918bb4918ad435ac984b1f8f268eb73153ebc0eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 15:55:42.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"871446ab105996e288510db1c98e11dd337eeafc","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 15:56:11.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":4},{"id":"056a5d036870a1702e839b8772de522909e64362","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-08 16:12:45.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"ab4596255804fdc6cca60e4653a2f5b77991b1fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-08 17:21:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/22eaf367f3234af09ace3fb4dfacef1d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/35b8809290254297a5c50e39ac05c9b3 : SUCCESS in 1m 47s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/348f0647117441b6bb15eb048cd5f644 : SUCCESS in 47m 23s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/fc23dcff38b6427bb7e669e0a06c6490 : SUCCESS in 43m 19s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/09d861a9724149df93bb97e4efdb583e : SUCCESS in 37m 47s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e925cd1e729e4766a64b4e66f884c222 : FAILURE in 20m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ef0dc094f504acc98ceeecaff9c3510 : FAILURE in 20m 36s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3a0bb3f009514cb2a9c4520696567623 : SUCCESS in 1h 04m 14s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e2472bc23dff4543af8e7b4eda66ae68 : SUCCESS in 1h 00m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2d67d2dd5afa449a9f9ff3ce850ec8d3 : SUCCESS in 47m 27s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/06f9314bc83a4d3baf9655b1e558e31c : SUCCESS in 45m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/829a93bed0c24b0286e40fdfab705473 : SUCCESS in 6m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8ba0df6ed204c02bb78559f81c7975b : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fa5d9c9748b4fb98186877c590bec9b : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e2f7a6340fdf405b941a38246be8df3c : SUCCESS in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efc2c00bd30946f3ab27cf61562fba0f : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f470a6b3e2441209b537f9ad63466ac : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2360d38135134deb937c5dca495432a8 : SUCCESS in 3m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3ac3c6ac439649cfbaee00e98d2439ad : SUCCESS in 33m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/373b63e0ce4e4ef29b2fff165f96941a : SUCCESS in 4m 18s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/11351828b76d4f8f9c0fcf18a2abe052 : FAILURE in 39m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"55d64c614ba6dbf85320730e9888e721b60d555d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-09 00:23:21.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"f0617a3a17d56bc5bf1dc17153480a601bd7fb85","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-09 01:31:57.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3b8be703fbe847d989cf7f3216d019f8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f864a18c33de4764adbc859e00683114 : SUCCESS in 1m 26s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a0bdf10620f347c1b5bff364bc65da58 : SUCCESS in 38m 48s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/775ab07a3a01467ba9eb20aae7a3d8c5 : SUCCESS in 40m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/e6692a71b87b43989fe67d61ff9b221c : SUCCESS in 38m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3db2b31615b44393b8c9368a85cbfae4 : FAILURE in 20m 09s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/118b09da13584e3186925a20cb70b40f : FAILURE in 20m 23s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/dcf468b557a14e39af93dcbd4144db6c : SUCCESS in 52m 15s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c4a8ba8778d245efbc7cffc681c8f316 : SUCCESS in 1h 03m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f462d7c7f71544a4b6877bb331229a35 : SUCCESS in 39m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/523a64e45ebe4a5dab39d6cdf26e9ef0 : SUCCESS in 42m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ac657686360441b59a01dab60960d867 : SUCCESS in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b9a6df5f3d49d686188557189f9373 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/63b27f68b0b74e3aa4d4340e834142f6 : SUCCESS in 2m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0813b47606f848aa881e6454cf456074 : SUCCESS in 4m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/78e1a5c66b354f218d9d76120ddd804d : SUCCESS in 5m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab193a27608b41c29d4c9067bc9003d5 : SUCCESS in 7m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13e3b572a7b149109351b9f90ce9b31d : SUCCESS in 4m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/125a62ffc849451eb00b1110e9d5c34a : SUCCESS in 32m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bb958ec42ef5443b9fd71cbefdbbf527 : SUCCESS in 4m 04s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/4d8b09df42974f08a8493beba6c10ec3 : SUCCESS in 37m 04s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"98f4b1fb1ddd90f85348b71026b23e442f0b7ee0","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-05-09 09:06:45.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4e6b8f57107cf0b1b8c3e40e4bcdf3814e0c4246","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2025-05-09 13:58:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"21c1a3112bd03b94a29f54ea1e9db35d94383a38","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-09 19:27:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0e6bfdc1e4fcf8fecd1b3b89bf5c72453c1ed100","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-12 14:51:28.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"d3b8927dadd1ef2253dea0c086e1e86e00f42db9","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-12 14:52:51.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"f1d13346fde81e4e4210ff94365e984c5883bdf1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-12 16:00:41.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fd6a205cf66e4a828e6e40564303d4b9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1f09e6e5c6c3497cbc043826ac651b4b : SUCCESS in 2m 03s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6f8c364943c547a88ed4452dc96e0d27 : SUCCESS in 42m 33s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2375b951ee4f4e9090c7cacf64bf8b6a : SUCCESS in 35m 52s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d1f6c7ceba8248aaba296c5646b86ee6 : SUCCESS in 43m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d6539def3835417fb85d8ade2a274ce9 : FAILURE in 42m 37s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1e11effa1cda454cb5122584187a6266 : FAILURE in 45m 06s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9032de47c4764ac5b88cdef3d1416524 : SUCCESS in 1h 03m 01s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/75395cd1770e4f34a0cad70464653953 : SUCCESS in 1h 08m 35s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1aaa7324d9124239af9d32289ba4751b : SUCCESS in 41m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a3867757cfe4312941d62c5eb848807 : SUCCESS in 41m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e49768cbea434a5b9c4c8d46717debc6 : FAILURE in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4725e6f8d69c41e19c9b877f46501773 : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90588284d38141898e0e1f03be354526 : FAILURE in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed420b51e2424fa5a0a696fba4f2e3bb : FAILURE in 4m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c7f63f9848ee4268b9c08b93d9d82fc1 : FAILURE in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d910134cf6745c490f2a5f1726cd947 : SUCCESS in 7m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b3f49e8db8784237982b446cd56410fe : SUCCESS in 3m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/58871aa6eddb406b83943805bdf14533 : SUCCESS in 41m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a7b36945f9f74d4082db08b04461c530 : FAILURE in 4m 10s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/e2a33705be8a486991d9eea3726a511a : FAILURE in 48m 55s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"e74dbd86fa7fafe8c85c4ce7bba8d4ff35bed5ed","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-13 00:28:03.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"6b1170bb1fce789ad3aae662901f40f83323e5bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-05-13 00:33:58.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b9643f0b1ed0bcf338760341b14276ce2a82f484","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 01:26:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/079fba0feb344324b9c69a28ccc531eb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56e583d5d9144a51977951fbf8f0abf2 : SUCCESS in 1m 17s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad8e308d54494216ac12bce1bbad2f48 : SUCCESS in 44m 34s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f5c120d4de864315b163890f5673cb28 : SUCCESS in 39m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/260262d7e21f46a797b09deb0b7ee273 : SUCCESS in 41m 45s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9298bfd28b0e46ab82b8db80c4fd9cc1 : FAILURE in 41m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b7fb552b2fec4834b4d9c6ae9b271c04 : FAILURE in 29m 50s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e6e1c724e9d344ecb2d9ad94e97fe7ed : SUCCESS in 52m 04s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/22c15740d0274aefb24ce1ec4c9be1ae : SUCCESS in 44m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d61931b5dc9f44f8bc67a576af2216bf : SUCCESS in 41m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d31bed4f4a134aad975399bdeed7379f : SUCCESS in 44m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/327d149875644fd280f4553a231d95c5 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4afdc4b8944f49109a9e584587c15872 : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/86f24cb672024161b65dab2658595d8c : SUCCESS in 3m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/05b0e9f421244b349859e4513adca6ca : SUCCESS in 3m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/479311cfbb32488a9f15dfe80b0c7a20 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed455e66692147f1ae93b200c2a4c0b0 : SUCCESS in 8m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b83c84b8103546fd84f015d5fa3a0c6d : SUCCESS in 4m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b9cb7aef180b41f6b508a2b3bd9b2409 : SUCCESS in 33m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c57d9b0dcfb54642988a2d95162d7551 : SUCCESS in 4m 57s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8992139bb41044dca50146fa86aa629d : SUCCESS in 45m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"c39587cb6061bd2cc69195a07acb05a2411bc9dc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-05-16 12:35:24.000000000","message":"Patch Set 9: Code-Review+2 Review-Priority+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3cff7c0e51d1a916adfb08044c55f8931997de0f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-28 15:05:18.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"63c91f05aa557fd6c9a90f0327e93b8977b7ca21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-06-04 05:47:34.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR is:MAX OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"4484f019c5922a0aef53dbaebdd0911dfeeabd56","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-06-04 05:47:42.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"dc8e55360f1a0d35c300ddda90398d0cf6f60e6e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 07:26:51.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/67705f7845a04ec3827003710cab7248\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/ac42d693b05a47c193f0e1a316d97bbe : SUCCESS in 5m 04s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/8c53e6a8a2fa44a5a0bc2f44d3d33479 : SUCCESS in 45m 03s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e7ef923baa31475c9057cbd2b1b657d4 : SUCCESS in 43m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7c4d2a28b94d449bafd8dfda03545eac : SUCCESS in 45m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ed98b28f4a934d16b7c6021617465e42 : SUCCESS in 1h 07m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8e8b0e6aadf9469b9ffe26cc80b54c0b : SUCCESS in 1h 07m 54s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/f0305145308d4dcf814bb4f330f82af6 : SUCCESS in 1h 01m 36s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bd07690615014b5387cab3a8322e3a49 : SUCCESS in 1h 07m 17s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c801629254bc4688aad62f00736262c0 : SUCCESS in 45m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/46d26d0e37ad41cd9a2624c2d1f926e1 : SUCCESS in 54m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cc714fde0c194dd0b7e025999cda42a9 : SUCCESS in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc814db0bfb846b79e0eb67e173f2be2 : SUCCESS in 2m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e1f4cf3318a2496e9264e0bef2d21d67 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/812e8fc2ca0b4235a437585f0dbea2c7 : SUCCESS in 6m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5f663e62657f48f8804cc02b6027bec4 : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/873ee9ff5eba4e669e04bc00598fb1ef : SUCCESS in 9m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9d406187b0464330b360386dd127d543 : SUCCESS in 3m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8db6a7aa8c684f45802c3a57464e2798 : SUCCESS in 34m 01s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b07d80e8276f4bd8995eb77db31b6220 : SUCCESS in 5m 23s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/43bfcd63d785415c952e2c861823eedd : SUCCESS in 49m 17s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"7f5a8b1f530002b938e33d24ef93b2d35a1df77c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-04 14:41:44.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"9a9fbc2c5507000ce0acd6f763a1d360896bb0c9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-07-01 15:03:37.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"4d49da364aecac293c362f63321dc2fdc47eda14","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-01 19:34:00.000000000","message":"Patch Set 10: Code-Review-1\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f3f1c95e8ca18e35c88889c731d59756c7acb167","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-07-02 05:54:17.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"3a17a3a8a56f69c08d1fd3db3005515adfd166ca","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-07-02 05:54:41.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f4407e893b7f4d09847204068455fad85bc11af4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-02 07:11:15.000000000","message":"Patch Set 11: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d3ad6e52c31e4efbaa89256f3a7ed8e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d2f74a764a754d31a15c5cf78142e09c : SUCCESS in 2m 08s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/25d2a1ecd1b840e0b3c4599b8df775ca : SUCCESS in 39m 31s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2962bca61ba4400c971b1e250aebceb1 : SUCCESS in 48m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/933fdd904b3d4b10be8f20494b908805 : SUCCESS in 43m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c172d094f92e42c993620f5abec3d9d8 : SUCCESS in 1h 04m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d7894649dc0942d7bef02a7a2540b130 : SUCCESS in 1h 10m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b484399d77344e6ba361b3b5111ce760 : SUCCESS in 1h 04m 25s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/68f84347e5dd4040a70e30e42963d28f : SUCCESS in 1h 00m 10s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f69ea8b6b03446369d08641df7177950 : SUCCESS in 53m 07s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ed7101e849fd41a6937841751f5c6789 : SUCCESS in 49m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90dfcee12bbc456cae1265797f992bf8 : SUCCESS in 7m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b9dbddbac99478b979bf058a1a1efa4 : FAILURE in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bf8c3ffa93fc4bbcb4641150e6811f51 : SUCCESS in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0da783afeeb64b5c89cd81bd96f4fd34 : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/44b8061d665d4ff2a578874f15d73e7c : SUCCESS in 9m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf413008800a4936a23a9c4496849331 : SUCCESS in 11m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/710cd4b0e926420bae54a3d97adcca8a : SUCCESS in 4m 08s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/365967321acb40038ab4ab9c5cc47ee5 : SUCCESS in 31m 58s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6fc211aea8a74375bbc2c54fdc85568e : SUCCESS in 6m 07s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b4c93fb048d54ae5bcac306894a17da3 : SUCCESS in 37m 14s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"3c5361618c27bc41171358cb470b87ed2189cd05","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"date":"2025-07-03 14:36:24.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"2717a6de5fcbb9e7d001833128eb2ce69b857c93","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 16:00:00.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce6dc46792ab45af8133a12f5f657753\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/64d76be849774e0186c2abca3827d420 : SUCCESS in 2m 21s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e6b71625c34f471bb4186c68112d8139 : SUCCESS in 47m 18s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/9fd425d6a4b7465bac52ff01e78eaea6 : SUCCESS in 47m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/b1c69beddc5c4ee2842c6051cac5c44a : SUCCESS in 49m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e070e83df5904096b5aeb21050e54e00 : SUCCESS in 1h 16m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a1a671f3ab834c4b8f6d0e0c390f9172 : SUCCESS in 1h 06m 46s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ecbe109807994104aa86887636e06f6b : SUCCESS in 1h 07m 37s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5a94a345e60946ce98b1e17e44ffd050 : SUCCESS in 1h 08m 59s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f90579777e2444e7a1cc0d8b5f3146f1 : SUCCESS in 48m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4e27d6bf497140fb8bd6903f10b2e06f : SUCCESS in 52m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2899a758b137439ca88fc10dcb8542b0 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d707934589434fa8b9af831cbb154eb9 : SUCCESS in 1m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b5df9da852534d58a54e5de00fe26179 : SUCCESS in 4m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6af0139355654a8781e3592c5b271792 : SUCCESS in 7m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d015bb51bec7475ba8c893caa19fa615 : SUCCESS in 9m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95b1ca8bd3bf405ba09c9e5a940ba72d : SUCCESS in 10m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a81ce3b5f1684bc8867b29bc30567a4c : SUCCESS in 6m 13s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b93760bc2adf4839a7e3f824e8f9ca95 : SUCCESS in 47m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cd2fb01e24914b76a211552ce222a0b2 : SUCCESS in 6m 38s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/60c595f42c474061b4a28707e2834135 : SUCCESS in 51m 44s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"6a52af8517d5437b3cb3de73c06ab2d9b39421e9","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-03 16:26:32.000000000","message":"Patch Set 12: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"5b3816d187fea3b423f612f7ad3d4b5671dd6b87","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-07-03 16:33:06.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"bbe65b2b8465c3abf3a9d552db9dedeb40c60dab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 16:33:32.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"6fa54f33c1ce99927bb735ab7dc03233127877ae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 17:35:35.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a88569cb218d4ee0b6dce8a8416df420\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/5730dc24a190452ebf907693ed13f708 : SUCCESS in 2m 17s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/677d0ab4c7e44a7fb915ddc9602a2d80 : SUCCESS in 45m 02s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/be489edc89dd41e7ad4cf2db9840dee4 : SUCCESS in 44m 29s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/be571ea5c5234be087d7925f376575db : SUCCESS in 40m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5d861d13ae0c4ad68b1687fc2f65b7c9 : SUCCESS in 1h 00m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/94b6ccb69f414ccc83f3ee421055e85c : SUCCESS in 51m 53s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5b6543f9ffc548609ad1c3d94c8ab126 : SUCCESS in 45m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d09e4d72f21347e280ada465c4303307 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4601d8a6e0344f7b4542e6ab88c8915 : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4c00cf00bee843a6a3572338ce5481c4 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03fcca6554804f11aea3160cec36136d : SUCCESS in 10m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f427306bf523444e998bd6050e13f642 : SUCCESS in 2m 15s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4f327dec8ec04192871be325924c9179 : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":12},{"id":"7bb0be0289c4cd3722f73cd593952e9ed86dbf66","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 17:35:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"7a79f6370c0e89b403d6f26a097c1888f6bceebc","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-03 17:37:44.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7386f4c77bab4054a1ae8778d3906f37\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98cc8cba1ee044308164973b7556cc1e : SUCCESS in 1m 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5768d73c9034cf7b86b3690c6e0afed : SUCCESS in 1m 54s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"6c610e23ce3937087a86e2599c9ca18b4bf988c9","revisions":{"87068258826d1a04970007437f5def7d2fa02b57":{"kind":"REWORK","_number":1,"created":"2025-05-08 08:06:21.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/1"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 07:23:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 08:06:05.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in crazy rpc\ntimeout), it can alway update and create it\u0027s service status back.\nSo we\u0027re find to delete the service status in this case too.\n\nAdditionally, central rpc api version is now bump to 6.11 to add\n`delete_service_status` method.\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87068258826d1a04970007437f5def7d2fa02b57"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/87068258826d1a04970007437f5def7d2fa02b57"}]},"branch":"refs/heads/master"},"a677f6d316d33d19744251c8943fa6344bacd07e":{"kind":"REWORK","_number":2,"created":"2025-05-08 14:04:13.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/2"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 07:23:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 14:03:57.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in crazy rpc\ntimeout), it can alway update and create it\u0027s service status back.\nSo we\u0027re find to delete the service status in this case too.\n\nAdditionally, central rpc api version is now bump to 6.11 to add\n`delete_service_status` method.\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a677f6d316d33d19744251c8943fa6344bacd07e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a677f6d316d33d19744251c8943fa6344bacd07e"}]},"branch":"refs/heads/master"},"46d5c863456792ad8e985ff6a172beb8adf7b50c":{"kind":"REWORK","_number":3,"created":"2025-05-08 15:45:30.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/3"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 07:23:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 15:45:19.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in crazy rpc\ntimeout), it can alway update and create it\u0027s service status back.\nSo we\u0027re find to delete the service status in this case too.\n\nAdditionally, central rpc api version is now bump to 6.11 to add\n`delete_service_status` method.\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46d5c863456792ad8e985ff6a172beb8adf7b50c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/46d5c863456792ad8e985ff6a172beb8adf7b50c"}]},"branch":"refs/heads/master"},"ff5a26baba4e31b1a277a7e80fa48b1bf078c2bb":{"kind":"REWORK","_number":4,"created":"2025-05-08 15:55:42.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/4"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 07:23:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 15:55:32.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in crazy rpc\ntimeout), it can alway update and create it\u0027s service status back.\nSo we\u0027re find to delete the service status in this case too.\n\nAdditionally, central rpc api version is now bump to 6.11 to add\n`delete_service_status` method.\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff5a26baba4e31b1a277a7e80fa48b1bf078c2bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ff5a26baba4e31b1a277a7e80fa48b1bf078c2bb"}]},"branch":"refs/heads/master"},"6665d15e815bf85be9dd5d01f82f374005ee4261":{"kind":"REWORK","_number":5,"created":"2025-05-08 16:12:45.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/5"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 07:23:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 16:12:34.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in crazy rpc\ntimeout), it can alway update and create it\u0027s service status back.\nSo we\u0027re find to delete the service status in this case too.\n\nAdditionally, central rpc api version is now bump to 6.11 to add\n`delete_service_status` method.\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6665d15e815bf85be9dd5d01f82f374005ee4261"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6665d15e815bf85be9dd5d01f82f374005ee4261"}]},"branch":"refs/heads/master"},"1f949fdadad3119a397aea39f428642885a5cc20":{"kind":"REWORK","_number":6,"created":"2025-05-09 00:23:21.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/6"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-08 07:23:41.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-09 00:23:10.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in crazy rpc\ntimeout), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nAdditionally, central rpc api version is now bump to 6.11 to add\n`delete_service_status` method.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f949fdadad3119a397aea39f428642885a5cc20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1f949fdadad3119a397aea39f428642885a5cc20"}]},"branch":"refs/heads/master"},"7a210870097c182bace9d295d84cee5ca85fe92e":{"kind":"REWORK","_number":7,"created":"2025-05-12 14:51:28.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/7"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a210870097c182bace9d295d84cee5ca85fe92e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a210870097c182bace9d295d84cee5ca85fe92e"}]},"branch":"refs/heads/master"},"15558b0de91375462d222bd6608309baa07661b8":{"kind":"REWORK","_number":8,"created":"2025-05-13 00:28:03.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/8"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-13 00:27:53.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15558b0de91375462d222bd6608309baa07661b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15558b0de91375462d222bd6608309baa07661b8"}]},"branch":"refs/heads/master"},"d9c7c3561943fef861c1af2c195e14b335c30afb":{"kind":"REWORK","_number":9,"created":"2025-05-13 00:33:58.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/9"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-13 00:33:49.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9c7c3561943fef861c1af2c195e14b335c30afb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d9c7c3561943fef861c1af2c195e14b335c30afb"}]},"branch":"refs/heads/master"},"c0a13277699f53ed0b819487b038e83e4036a108":{"kind":"REWORK","_number":10,"created":"2025-06-04 05:47:34.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/10"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-06-04 05:47:24.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0a13277699f53ed0b819487b038e83e4036a108"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0a13277699f53ed0b819487b038e83e4036a108"}]},"branch":"refs/heads/master"},"33b578977b92821c96aadb014cd562760da81c89":{"kind":"REWORK","_number":11,"created":"2025-07-02 05:54:17.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/11"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-07-02 05:54:04.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33b578977b92821c96aadb014cd562760da81c89"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/33b578977b92821c96aadb014cd562760da81c89"}]},"branch":"refs/heads/master"},"6c610e23ce3937087a86e2599c9ca18b4bf988c9":{"kind":"REWORK","_number":12,"created":"2025-07-03 14:36:24.000000000","uploader":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"ref":"refs/changes/26/949126/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/26/949126/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/26/949126/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/26/949126/12"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-07-03 14:36:14.000000000","tz":480},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c610e23ce3937087a86e2599c9ca18b4bf988c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6c610e23ce3937087a86e2599c9ca18b4bf988c9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
