)]}'
{"id":"openstack%2Fdesignate~949605","triplet_id":"openstack%2Fdesignate~master~I40f9b914af6a714fe2e9b4180c4a9727bbafad48","project":"openstack/designate","branch":"master","topic":"remove-wsgi_scripts","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-05-30 11:19:45.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-08-20 17:16:03.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-08-20 17:16:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I40f9b914af6a714fe2e9b4180c4a9727bbafad48","subject":"Add pyproject.toml to support pip 23.1","status":"MERGED","created":"2025-05-13 11:33:54.000000000","updated":"2025-08-20 17:17:12.000000000","submitted":"2025-08-20 17:16:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"949605-remove-wsgi_scripts","meta_rev_id":"66430aba6a457ddefbe5a15ea7856d1335388e74","_number":949605,"virtual_id_number":949605,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-08-20 17:16:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-05-30 11:50:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-01 15:31:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-08-20 16:04:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 12:51:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 16:01:01.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-05-30 11:50:55.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-01 15:31:01.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"c4f8faf4b0b5af3791bd7534a01122b64cdbddac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:33:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8c87b6a5eba984855310b884a5de251da1284768","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 12:51:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab2240d73da14859afe51fd66dcaf3ff\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e93f1b5380db486e9b12be85058b9caf : SUCCESS in 45m 55s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/2505cba9175c475883a5624bc92a35a4 : SUCCESS in 48m 42s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/21aca8ec2ba54f5b9eef33a8726890a3 : SUCCESS in 38m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/80d4f56ef299466580e34030538b3f7d : FAILURE in 43m 28s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/203688eeb2dc4354a65e2b83ded8c4d0 : FAILURE in 49m 58s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3d80ed7bce514c3e858917d84802815f : SUCCESS in 59m 36s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/afca5f81f50d400e96cf554b509997b1 : SUCCESS in 1h 02m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2c57873165074e928fda64cbb1ff2fd0 : SUCCESS in 45m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/c7e80189ecad41ebbb13556bbfce836b : SUCCESS in 49m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96141843b0d947d786df0ab1a5655bfe : SUCCESS in 6m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c74b6635133e4296ac6eb86b635428c5 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e3ba2c4213b46a9bfc82aca7515cfc6 : SUCCESS in 4m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d84da5126d744693b98183f7d9690174 : SUCCESS in 5m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d6db8bcf2a504b2f99c2e6687f4c016e : SUCCESS in 7m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9bb9b9b98c04abcb12784b88d48ae58 : SUCCESS in 8m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e325e629a7f34ea7bc59f7c6345b3d5a : SUCCESS in 38m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1c1f53808dcc4f1d8bd735cb14a1a6cd : SUCCESS in 5m 06s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/132312483fd4422db5302f8845bd079c : SUCCESS in 45m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"540feeb24c6fac6551d742f0ac0f81413bcb4272","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 16:01:01.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"704bd2952ca94ee09eafd5874ee2c810fa3c68ed","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 11:43:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"428138d192b878fc846b2214555da70c312062d8","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 11:45:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2b6c350bfeba788864719f7cf01c222536418198","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 11:19:45.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2e9a6c752805cba6759efdb60f71bb3cf612c09c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 11:50:34.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c539aa654764d0bdb97b8659bfb9ec81fc8b3442","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-01 15:31:01.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"676c55853257a1ce2367a96cdff4d26aed06a926","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-20 16:04:42.000000000","message":"Patch Set 1: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"05f332b3b373e0ecdab940ba9ea070ed8fa01319","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 16:05:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d4585f5e160767c8678683ecdaec4ca9ab9a634c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 17:16:02.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6dc0b85f27f7416db430d0b8e65ac514\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/40122fc5c7c04d3d89e0bb49df405512 : SUCCESS in 35m 15s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/06be5c38fe004a889474f2f41febe95f : SUCCESS in 45m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/39f008d9ee284ba0a302f04631b4ca8e : SUCCESS in 41m 27s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9a9ca47a37c34118952a38adb4c69c95 : SUCCESS in 1h 06m 21s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c9e8704e254b4d0eaf057b10b12b37b1 : SUCCESS in 44m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/1b342ad6c2ce408798067de79780f559 : SUCCESS in 39m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6859c97d68b4e5a9799e3f5d1fc69ec : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/809bd7d12aac40a5bf0236f86344c566 : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/597d7fd427e44cbda2e0467c89a299ef : SUCCESS in 7m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a9c17dfb8f24d0791037b5ea1023ae0 : SUCCESS in 7m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/212ef7a3eb87462abfbdc62fc74dec1a : SUCCESS in 30m 54s","accounts_in_message":[],"_revision_number":1},{"id":"118a2e2a1279ce0dcede8fabfbea5c378362ed8e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 17:16:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"66430aba6a457ddefbe5a15ea7856d1335388e74","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-20 17:17:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da982d8b05fb4440a4fe163fd931b1fa\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c30da57fe53648b29726215da8aa06e3 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a57e02ea57caaf0b22d212fbf5eea14bd031ff3","revisions":{"7a57e02ea57caaf0b22d212fbf5eea14bd031ff3":{"kind":"REWORK","_number":1,"created":"2025-05-13 11:33:54.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/05/949605/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/949605/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/949605/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/949605/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/949605/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/949605/1"}}},"commit":{"parents":[{"commit":"b448b52f8f0de50dd7bc7943299abe6cf4023815","subject":"Merge \"Add wsgi module to Designate and remove wsgi script\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b448b52f8f0de50dd7bc7943299abe6cf4023815"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:33:42.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:33:42.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip. pip 25.3 will go a step further and enforce presence of this\nfile.\n\nTo address that, this change adds the minimal pyproject.toml to enable\npbr to be properly used to build editable wheels.\n\nSee [1] and [2] for more details on the removal of the fallback support.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/836\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I40f9b914af6a714fe2e9b4180c4a9727bbafad48\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a57e02ea57caaf0b22d212fbf5eea14bd031ff3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7a57e02ea57caaf0b22d212fbf5eea14bd031ff3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
