)]}'
{"id":"openstack%2Fdesignate~949606","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~I40f9b914af6a714fe2e9b4180c4a9727bbafad48","project":"openstack/designate","branch":"stable/2025.1","topic":"remove-wsgi_scripts","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-07 13:52:38.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I40f9b914af6a714fe2e9b4180c4a9727bbafad48","subject":"Add pyproject.toml to support pip 23.1","status":"ABANDONED","created":"2025-05-13 11:34:13.000000000","updated":"2025-11-07 13:52:38.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1ae8cd825516ffccfd95fbabc6266e4b9b9b5dbc","_number":949606,"virtual_id_number":949606,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-06 12:24:06.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"date":"2025-11-06 11:25:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 13:18:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 16:01:10.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"56a9741384df0b424a77bb9b1035ef0faae44866","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fca3c100b8b8828609085e7465c6868e1fc3a5ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:18:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8b04cf3859e44ea7a7741d83feaf2f8e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7e273d68594d453db8403a2ff9cc33b8 : SUCCESS in 42m 50s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/67d8bce85f1848d48b2ea85480369ce9 : SUCCESS in 52m 31s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/18e7f0ab7dab4874a5434ea13af558a2 : FAILURE in 44m 49s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/0647463511dc445fae55422dfed9107c : SUCCESS in 44m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/83b68dc6a80c4d2b9a647ae4f4576c1b : SUCCESS in 52m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/67ba6ca8c7564e17843c7ec4e1876cd4 : FAILURE in 1h 04m 35s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2727213769fc4c76b70c8ae1c981b6b2 : SUCCESS in 1h 04m 52s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5aeb63eed06e4b7fac9aa3ace4e39020 : SUCCESS in 1h 05m 18s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d6f1bb06fad54759bece62a36d75b06e : SUCCESS in 53m 56s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/cae580c6f38f42ec868d35d69e266a61 : SUCCESS in 44m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9c18433235024d90857883d4a5b01c00 : SUCCESS in 40m 24s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2d146436c3d64f8caf9e44ff36fd04cc : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/914f6a208508455cb0009bc501ad1dcb : SUCCESS in 3m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc77ca5d9ac6431da64897d3d789eef8 : SUCCESS in 3m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b7f40b0234e421ea011f0fff05b3c0e : SUCCESS in 4m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/289a8dc3a7ff4db0b8a38f9c05d5b173 : SUCCESS in 5m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4d16b7ea0a26434b88778b2b6ef60984 : SUCCESS in 33m 14s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/90e4cdffa80f46acb8b1a3f53159ef2d : SUCCESS in 4m 10s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2908501b3a8443628afef1264f500626 : SUCCESS in 38m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dac8431edcddacdba87408741ca8f9b4615ee1ee","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-16 16:01:10.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"73a7010d50d1d25e136e2dfda1273f4de2c4ac14","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 11:42:19.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"3e849e74d612b31fb0796b687f4c89eb4f3d68cb","tag":"autogenerated:gerrit:restore","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-11-06 11:24:33.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"dbdaebcd293fdfa64c746427f4aa8d5b944cba06","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-11-06 11:25:02.000000000","message":"Patch Set 1: -Code-Review\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90c42838eadcf1fb6dd0be2ae953e9e04797519a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-06 12:24:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3a551c9af244439599e75bbb0b042a58\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/83c3ca8f55ba45ba924d3f4d2d1c43ab : SUCCESS in 1m 56s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/76b662d7e6fa44a8995d6a23085a63cd : FAILURE in 34m 14s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/b4694eb4d4a7441e9f50c7980214fac2 : FAILURE in 14m 00s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/17d880535aa34e80b6d4c469dfc14cfa : FAILURE in 12m 29s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/77b43d44c05b4d55890c4aa98c777e0c : FAILURE in 13m 38s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a489f446f5764ad895fd6ca6d2d97865 : FAILURE in 16m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/76e7e5c82abf45d08ac5073b66670766 : FAILURE in 35m 05s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8982756a1492419a965ff9379473c5ad : FAILURE in 46m 31s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e6be818033d64ff38e534cdf5744ef9c : FAILURE in 20m 45s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/63cf3cead3394f98978776865c0a0dd2 : FAILURE in 18m 40s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/37f8140be43d45228fc75384dfeedc63 : FAILURE in 24m 13s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/70ecaf103f4a408799f59f76c8d5c542 : FAILURE in 25m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c398edf075ae496ebd2eaa592024d98d : SUCCESS in 6m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/166bc33e9b334d4e9e0958c41675d13f : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5137d2435175463aa349cbf87ac50566 : SUCCESS in 4m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a1a0785d09d491fb3839ffd7f51b791 : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbf76a25fc75490b99271a4485ac0fb6 : SUCCESS in 6m 27s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/100e205692184eb09886fa92e0d275f8 : FAILURE in 29m 50s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f715d3a9dadd44d9a4e478c886d05aa6 : SUCCESS in 6m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/8a377d2a1cfc4cca8ba978b1f153d82c : FAILURE in 19m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1ae8cd825516ffccfd95fbabc6266e4b9b9b5dbc","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-11-07 13:52:38.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"248cd680c1b10b483ce95f257b163818a8807dc5","revisions":{"248cd680c1b10b483ce95f257b163818a8807dc5":{"kind":"REWORK","_number":1,"created":"2025-05-13 11:34:13.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/06/949606/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/06/949606/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/949606/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/949606/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/06/949606/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/06/949606/1"}}},"commit":{"parents":[{"commit":"5f4d6c4c6b68bea0444b516ad000ef1fb6b67673","subject":"Add an designate-api-wsgi compatibility script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f4d6c4c6b68bea0444b516ad000ef1fb6b67673"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:33:42.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:34:08.000000000","tz":60},"subject":"Add pyproject.toml to support pip 23.1","message":"Add pyproject.toml to support pip 23.1\n\npip 23.1 removed the \"setup.py install\" fallback for projects that do\nnot have pyproject.toml and now uses a pyproject.toml which is vendored\nin pip. pip 25.3 will go a step further and enforce presence of this\nfile.\n\nTo address that, this change adds the minimal pyproject.toml to enable\npbr to be properly used to build editable wheels.\n\nSee [1] and [2] for more details on the removal of the fallback support.\n\n[1] https://pip.pypa.io/en/stable/news/#v23-1\n[2] https://github.com/pypa/pip/issues/836\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I40f9b914af6a714fe2e9b4180c4a9727bbafad48\n(cherry picked from commit 7a57e02ea57caaf0b22d212fbf5eea14bd031ff3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/248cd680c1b10b483ce95f257b163818a8807dc5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/248cd680c1b10b483ce95f257b163818a8807dc5"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
