)]}'
{"id":"openstack%2Fdesignate~949610","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~I020bda6f05f1c79ff1576925fe83f66796d06371","project":"openstack/designate","branch":"stable/2025.1","topic":"remove-wsgi_scripts","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-19 11:42:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I020bda6f05f1c79ff1576925fe83f66796d06371","subject":"Revert \"Add an designate-api-wsgi compatibility script\"","status":"ABANDONED","created":"2025-05-13 11:38:57.000000000","updated":"2025-05-19 11:42:25.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"38c9eebdf09fdc56c4c268df5723c50ee201f761","_number":949610,"virtual_id_number":949610,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-13 13:28:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2025-05-16 14:50:42.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-13 13:28:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-16 14:50:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d69aaf57f2b722735ded9da976afe2a35d29ab92","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-13 11:38:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"711a26c325d63f6bbcf6bcff82ed6b36e07e873e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-13 13:28:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ff71a5bcceb444498a13dfdbfda2759c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/98a5942d3dd542d48ecc649a3bb994a4 : SUCCESS in 3m 25s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/63ac36e7b8764a04b9a3e4d8f3e1ec97 : FAILURE in 33m 56s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/bf563f3f26b04b51988b1a63c4cd9804 : FAILURE in 36m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/74186fb8e7e34baf9ce3a8be5b602ac1 : FAILURE in 33m 42s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/0719633752a6495ba1a0774b84179097 : FAILURE in 26m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c7a87566d7b94c638bea674792f5aba6 : FAILURE in 27m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/57001907046341adb3dc62446f53bec9 : SUCCESS in 1h 08m 39s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/04feb66aba82412780325cfb633a3c18 : SUCCESS in 59m 17s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d012d87255b94139a1acaba691875cea : SUCCESS in 1h 02m 42s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8963d724099c4da2979a715d204815f8 : SUCCESS in 55m 29s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/d7f458c79b3d49949234f70b5199c036 : FAILURE in 27m 08s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/56af151a504f496eadbcd2bfa779348b : FAILURE in 24m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be9941ff8e254a3eb272a069e6a2feb5 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/026d740c6c234dbd803fdeeb2cc0b771 : SUCCESS in 3m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d09929f7492c4d5693ae7ec83c072255 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef346568a4ee4733855a82eb7fea54ca : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dacebad143874aa3abd254d34590ae1c : SUCCESS in 8m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c42f1daf8a714157966d696f141b4fff : SUCCESS in 5m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1cb6f96412a842d5b272c35183cd55a3 : FAILURE in 43m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2c265208079348cb8f50f30bd8113b12 : SUCCESS in 6m 10s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/1bd3022ccdcd4cafb4ae145715b97934 : FAILURE in 32m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b8c67f60af429dcf83be698f67735002aa9a8056","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-16 14:50:42.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"38c9eebdf09fdc56c4c268df5723c50ee201f761","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-05-19 11:42:25.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"50c5017c0b25d8ee3238ea25b6f25034f13bbf4e","revisions":{"50c5017c0b25d8ee3238ea25b6f25034f13bbf4e":{"kind":"REWORK","_number":1,"created":"2025-05-13 11:38:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/10/949610/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/10/949610/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/949610/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/949610/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/10/949610/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/10/949610/1"}}},"commit":{"parents":[{"commit":"248cd680c1b10b483ce95f257b163818a8807dc5","subject":"Add pyproject.toml to support pip 23.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/248cd680c1b10b483ce95f257b163818a8807dc5"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:38:31.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-13 11:38:45.000000000","tz":60},"subject":"Revert \"Add an designate-api-wsgi compatibility script\"","message":"Revert \"Add an designate-api-wsgi compatibility script\"\n\nThis reverts commit 5f4d6c4c6b68bea0444b516ad000ef1fb6b67673. This\nshould not be necessary once a pyproject.toml file is present to ensure\npbr is properly initialised.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I020bda6f05f1c79ff1576925fe83f66796d06371\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50c5017c0b25d8ee3238ea25b6f25034f13bbf4e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/50c5017c0b25d8ee3238ea25b6f25034f13bbf4e"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
