)]}'
{"id":"openstack%2Fdesignate~950465","triplet_id":"openstack%2Fdesignate~master~Ic2a52c57561e778cce8fb696670fd624f547059c","project":"openstack/designate","branch":"master","attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-20 22:49:28.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ic2a52c57561e778cce8fb696670fd624f547059c","subject":"Fix zone abandon on catalog zones","status":"NEW","created":"2025-05-20 21:45:33.000000000","updated":"2025-05-20 22:49:28.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"14c8f46da0947e9795688095aff4ac9edb02e7d9","_number":950465,"virtual_id_number":950465,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-05-20 22:49:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2025-05-20 21:45:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-20 22:49:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c2f1ecb0d1eb8a75e17a5e5bee16e3c69a3bc1d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-20 21:45:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c09f5c744972174bf77d42705b8869edbe689d59","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-20 21:45:57.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"14c8f46da0947e9795688095aff4ac9edb02e7d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-20 22:49:28.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0287a89cbacd49cb9258b8c925a346d8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/aded46b5fae84df983f86b449ced430a : SUCCESS in 42m 18s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/46bcea069df24a18b69b2844100d0308 : SUCCESS in 35m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/0f67afccdc864ed2b1118bb933205738 : FAILURE in 10m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/de79aeda94fb40da8a72f992270f2654 : FAILURE in 40m 42s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/77614083f60446cdbffa56fdf64204ab : FAILURE in 44m 15s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9ff19a981a714e03bfb0d45f3facd2cf : FAILURE in 17m 57s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/803449aed17f4dcb820a5281caddc889 : FAILURE in 21m 47s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/2d7a7f46fc084b5094951381c6576ef4 : SUCCESS in 38m 30s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5e35ffdab9df458c9d1ac8a35704982a : SUCCESS in 48m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6a73aa99b694aeda88d7fdee340fb39 : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d11dbcd26c704e4f8339b0110cb78b8d : FAILURE in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd70ea11e5554701a65324543d96adb0 : FAILURE in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/235f151928f24403888951ebd493e7cd : FAILURE in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/38d85abd59464767b5051ed66c5a44a2 : FAILURE in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0ea6a0e8e149b097f9265007af13c8 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf7220a9abf24ac79009d9cedea83b6d : SUCCESS in 18m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/607cf8cf75c447d9ab2e8468b410d3a7 : FAILURE in 2m 57s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0ec067d2c21343cda5f3ab8c5f2ae52d : FAILURE in 40m 55s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cb386edf4a88f4e5b707ddbcecf335866ffde8aa","revisions":{"cb386edf4a88f4e5b707ddbcecf335866ffde8aa":{"kind":"REWORK","_number":1,"created":"2025-05-20 21:45:33.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/65/950465/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/65/950465/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/950465/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/950465/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/65/950465/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/65/950465/1"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-20 21:40:00.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-05-20 21:45:32.000000000","tz":0},"subject":"Fix zone abandon on catalog zones","message":"Fix zone abandon on catalog zones\n\nWhen a zone is in a pool that has catalog zone management enabled, you can call\n\"openstack zone abandon\" on that zone. This will cause unintended consequences\nas zone abandon is intended to leave the zone intact on the backend server, but\nsince the zone will be removed from the catalog the backend will remove the\nzone.\n\nThis patch blocks users from abandoning zones that are part of a pool\nusing catalog zones.\n\nCloses-Bug: #2111370\nChange-Id: Ic2a52c57561e778cce8fb696670fd624f547059c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb386edf4a88f4e5b707ddbcecf335866ffde8aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/cb386edf4a88f4e5b707ddbcecf335866ffde8aa"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMIN"],"failing_atoms":["label:Workflow\u003dMAX"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
