)]}'
{"id":"openstack%2Fdesignate~950932","triplet_id":"openstack%2Fdesignate~master~Ib53d53d9ce74453f568bda785b08df656b8ab109","project":"openstack/designate","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-06-03 15:48:27.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib53d53d9ce74453f568bda785b08df656b8ab109","subject":"Drop explicit executor argument","status":"MERGED","created":"2025-05-27 05:25:04.000000000","updated":"2025-06-03 15:49:29.000000000","submitted":"2025-06-03 15:48:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"950932-eventlet-removal","meta_rev_id":"19403b2ed2ab4b53a1a6a93e119bbd973570f32a","_number":950932,"virtual_id_number":950932,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 15:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-06-03 14:31:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-03 13:58:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-06-03 14:31:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-27 06:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-02 20:23:28.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2025-06-03 13:58:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-06-03 14:31:34.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"0ec059d9153c03af2ebbcc94b309edd176ede3d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-27 05:25:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cbdb7cac26477377a0ea0b4b0cc758203ebc481b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-27 05:26:02.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"8a9ec28ea872a2293452dd0afa9a9152871a25eb","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-27 05:38:14.000000000","message":"Topic set to eventlet-removal","accounts_in_message":[],"_revision_number":2},{"id":"9f552933d83e3354888380026b4c55b360e078b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 06:21:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fc51634deb934825aa2728dbd8ea05db\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1112c16913234e6ba1c108207274fd2d : SUCCESS in 41m 17s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e69537d8dbee47e99a794adb075a2178 : SUCCESS in 41m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ffeaea85d6b743329221bae6a185b808 : SUCCESS in 46m 55s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c658bede220b46a69c865734b335e26b : FAILURE in 36m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/28ddca244ef24e1a88a193c5c2098217 : FAILURE in 42m 06s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/7de7edcd89f34739aa82d2681e7e891b : FAILURE in 16m 14s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/24bb2d1a14df4e908aed5cecd6bf7a6a : FAILURE in 21m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0244088d6ec84731bffea7afa276da55 : SUCCESS in 33m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b56ad856ad994cb38c4cced8a136d0a0 : SUCCESS in 51m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bfc3c6b2550146a990f14cde2a1b12f7 : FAILURE in 4m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d1409facdd4e41a2c4837b588f0248 : SUCCESS in 3m 18s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8da51f3e9e1a4c8da434615d894ed6dd : FAILURE in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e67773fc74b0470c8a4d7e301e490197 : FAILURE in 4m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbe69927ab97404e81d8da995d8f9dfa : FAILURE in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a33bfc84511441bab06b31fde71450d0 : SUCCESS in 9m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a8360225d0d74a44bac7f97149bec220 : SUCCESS in 34m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/bd17a6053f8449a3a2460bd64ad8ecb0 : FAILURE in 4m 41s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2e32e1469b534e08957dd63c7dce4cbc : FAILURE in 49m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a067d5fc894e0c4935768251be09d576438dd60f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-05-27 06:37:51.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"35bd455e9c83bb0380a75b9c01ca98cb0fca7646","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-27 07:34:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fb41189c2e44b5ba35b3b4a042efd12\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/028973292aa0484a9ea014323f2426d7 : SUCCESS in 33m 51s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/74db77883aa442da8a26fb23899c94d6 : SUCCESS in 49m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ea9c321fd23e46acac38f592278e3d2a : SUCCESS in 35m 23s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d1576f15daf142b59c8dfdd3c3d6b7a0 : FAILURE in 40m 33s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2cfc6bacefa04bf0a3b8e4b02796505a : FAILURE in 28m 45s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9ee483b99cb34cdcbb5cd90317cbc16a : FAILURE in 23m 42s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0dabcd30087f4672bfc00d2ad6ed20c3 : FAILURE in 23m 52s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b1b6f0c690414224acea0687650135d8 : SUCCESS in 38m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/bc79c59ccf744cc0ae116adaca5bf8a3 : SUCCESS in 44m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c87eafee91b44925b5b7f3909fab55c6 : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf9e3a79fb774bbd8308196b8c85ec3f : SUCCESS in 4m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e965c0f8b4514bf6a925add6525385fa : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b9c2f2999e74c41a1691fe151134245 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/21acc0f6fad84b7f8aac46ced1ad9164 : SUCCESS in 7m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc05404f64bd44bfa1c6e8283c0efe7e : SUCCESS in 8m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f708a7b38da34592b93d121c99da7c7e : SUCCESS in 31m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eb1bbf5d51a840bebc860ce1b06f4982 : SUCCESS in 4m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/7c6f82d5b4ae4013bebae8456e974dbc : SUCCESS in 44m 01s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"857595d4b9b3b2792a55f4ee0979f422a838887a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 20:23:28.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a31d9400e3a65dea7c7a6b2cca5f3a793e8708dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:48:21.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d6230ec016b48809e29e1b865c59eeb\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bf7b8648ceee451f96ddb3c4ff761f7d : SUCCESS in 49m 01s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/5393d9fe6aa347a78a5213ebe78bf0d5 : SUCCESS in 42m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/af918b0778e84ae3a3f17de101f4efe4 : SUCCESS in 47m 32s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/114678c24ca3438fb7fe20530727e413 : SUCCESS in 58m 46s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/8be514ff135f46c5b929c266d8ba640e : SUCCESS in 1h 07m 12s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6693e35ebd774950bc757d79a0917ec3 : SUCCESS in 1h 06m 06s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/68000babb87843638bf666a28c619f2e : SUCCESS in 1h 07m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/6e2eff2f3f534c70af3a8c824ec846ce : SUCCESS in 43m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/926272c471824558901450be96beacdb : SUCCESS in 51m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f59b15f664814e22ae613bfbeb184206 : SUCCESS in 6m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc417642209f4ea590f111b5eb9c0ee1 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d04297d580bc44bbbac65057787e2491 : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ada9d8204daf4c1582d68e09ad17332f : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2a57021abbd646e3b09071d3e6762412 : SUCCESS in 7m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4d93b84c37644d3b1d02071b8dc4231 : SUCCESS in 8m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/92ddb7ed27c94de19ff3cea03018964a : FAILURE in 21m 28s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d32c21bb2ac8447eaba4ac9f185e47d4 : SUCCESS in 4m 23s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/712c6f52749f4d8eb745b9929a6d7c0f : SUCCESS in 44m 18s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"a20334866bfe62d18eb3a18fd38fd8e466d97085","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 09:49:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"db26ea9de5e007eb62a72127c090bfcbf9427678","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 10:58:46.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc541c6293254ad79de411afa01f6920\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/065b26f5c36440868c5c96207eaff13b : SUCCESS in 39m 46s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/844c8baa13ab4847830c50486f3a16bf : SUCCESS in 40m 03s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/22065379bfa644eb946c4b991b2e1327 : SUCCESS in 37m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/b08450952997479680914a26603a838c : SUCCESS in 1h 06m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9993f8e5e48147f1bb33de3d912cf8c4 : SUCCESS in 57m 42s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d117c8f412254e028e22f00a6b54ddbf : SUCCESS in 59m 25s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/857c52e546ff4cc0b7cd424525dfaeee : SUCCESS in 1h 07m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1f0cfb93dcaf472d957aae3f89477e71 : SUCCESS in 43m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3c36a9b356d84dbba8cd6ebdf89c0274 : SUCCESS in 45m 26s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d3d8b942759b4258864a18a7292a966f : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/159b0ce18af24483835891ce850ca8e9 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa66946823674013863c2ea42a9e7767 : SUCCESS in 4m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0aa7ba2322a947fcb90c105d62c1f61e : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/23a2ddafe26b4afb95afb612285947e3 : SUCCESS in 7m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/271d52ed73d84c05b2447c626fd242c2 : SUCCESS in 7m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/adc01318277b4d91b6f25ee75f8d427b : SUCCESS in 32m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2980bb7e6a6e4c828edb0db9096a1c9d : SUCCESS in 4m 34s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ed31d94e854b43718be0279db4381023 : SUCCESS in 44m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"7d4047004bff3224950dd97a71198f95d4756c3c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-03 13:58:20.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"0e2a1c2959bcae6d52e5211169569bc986a2a7dd","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 14:31:34.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"fb2d0a581ffe713ccb1160d16bee4c7416e99516","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 14:32:13.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"86948f9349dc4f877a34af79ee6803fe1c082a42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:48:27.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7420a6ded7a4f27aff2be86a898c8cc\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/081f068bae9040fcac751e0a1850cef2 : SUCCESS in 47m 03s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/733f1da49db84789811dc62b0446d784 : SUCCESS in 43m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/33a8b4a699404fc1845d6caeb0be0fe9 : SUCCESS in 46m 00s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9ca1458100ee48419e11baf7998be67c : SUCCESS in 59m 04s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c2ccd9409c93463c84d32c98dafcef7b : SUCCESS in 38m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/baa81117543345bc82b062ba29ea09f8 : SUCCESS in 52m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc8f546a44f248939d772fe874af0e50 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e991c64c5864f4b832e18398876e5cd : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/369609a58c0b4307961fb996e500e73c : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c20f0b85bc7d4143ac01b78f16e06e0f : SUCCESS in 10m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/89849eefbc234c119c3a5acf9fcd16fe : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":3},{"id":"7644741a200382e3fc802b43416081786ba9ea1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:48:27.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"19403b2ed2ab4b53a1a6a93e119bbd973570f32a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:49:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6166697bfc14a1db5ab4c34e34399ca\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26962032c7074831bc3cbc2320a8dcec : SUCCESS in 48s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b211ff4e06286601ffc6ae38df17ed847dda90e8","revisions":{"5615542b784d8fedab52bfa313c62e3cdb38074b":{"kind":"REWORK","_number":1,"created":"2025-05-27 05:25:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/950932/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/950932/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/950932/1"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 05:24:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 05:24:52.000000000","tz":540},"subject":"Drop explicit executor argument","message":"Drop explicit executor argument\n\nThe argument was deprecated as part of huge deprecation of eventlet\nsupport. If the argument is not set, the appropriate executor is\nselected according to whether eventlet monkey patching is done or not.\n\nChange-Id: Ib53d53d9ce74453f568bda785b08df656b8ab109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5615542b784d8fedab52bfa313c62e3cdb38074b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5615542b784d8fedab52bfa313c62e3cdb38074b"}]},"branch":"refs/heads/master"},"4f4a1616f77390556084f48a1eca7cf0c812cd5e":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2025-05-27 05:26:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/950932/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/950932/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/950932/2"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 05:24:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 05:25:36.000000000","tz":540},"subject":"Drop explicit executor argument","message":"Drop explicit executor argument\n\nThe argument was deprecated in oslo.messaging 16.0.0, as part of huge\ndeprecation of eventlet support.\n\nIf the argument is not set, the appropriate executor is selected\naccording to whether eventlet monkey patching is done or not.\n\nChange-Id: Ib53d53d9ce74453f568bda785b08df656b8ab109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f4a1616f77390556084f48a1eca7cf0c812cd5e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4f4a1616f77390556084f48a1eca7cf0c812cd5e"}]},"branch":"refs/heads/master"},"b211ff4e06286601ffc6ae38df17ed847dda90e8":{"kind":"REWORK","_number":3,"created":"2025-05-27 06:37:51.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/32/950932/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/32/950932/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/32/950932/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/32/950932/3"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 05:24:16.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-05-27 06:37:42.000000000","tz":540},"subject":"Drop explicit executor argument","message":"Drop explicit executor argument\n\nThe argument was deprecated in oslo.messaging 16.0.0, as part of huge\ndeprecation of eventlet support.\n\nIf the argument is not set, the appropriate executor is selected\naccording to whether eventlet monkey patching is done or not.\n\nChange-Id: Ib53d53d9ce74453f568bda785b08df656b8ab109\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b211ff4e06286601ffc6ae38df17ed847dda90e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b211ff4e06286601ffc6ae38df17ed847dda90e8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
