)]}'
{"id":"openstack%2Fdesignate~951260","triplet_id":"openstack%2Fdesignate~master~I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74","project":"openstack/designate","branch":"master","topic":"fix_grenade_jobs","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-06-02 13:50:52.000000000","reason":"Change was submitted"},"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2025-05-30 13:30:09.000000000","reason":"\u003cGERRIT_ACCOUNT_16688\u003e replied on the change","reason_account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-06-02 13:50:52.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-06-01 18:54:45.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74","subject":"Move grenade skip-level jobs to start from 2024.2","status":"MERGED","created":"2025-05-29 10:22:36.000000000","updated":"2025-06-02 13:52:17.000000000","submitted":"2025-06-02 13:50:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"951260-fix_grenade_jobs","meta_rev_id":"06479d1d48c9b158f39407e12f8e133c303060aa","_number":951260,"virtual_id_number":951260,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-02 13:50:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"recommended":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"all":[{"value":1,"date":"2025-05-30 13:30:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-06-02 09:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-01 17:01:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-06-01 18:54:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":1,"date":"2025-06-02 09:29:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":2,"date":"2025-06-02 09:29:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-29 12:34:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-30 11:20:39.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2025-06-01 17:01:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-06-01 17:01:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-06-01 17:01:52.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"eda4bf0daf1c45334b13a9b85003a1f88c8abc41","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-29 10:22:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5d75f83f1403d63c4ed3887cde6f60670f3a7330","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-29 10:37:38.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e43380252f034cfeac451d29367e970e9552f2ea","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-29 11:36:04.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"4487224e5f242e03ea8ab600f4a9e904845bc851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-29 11:39:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ba0932b94ca6a54f86564799b517160421390889","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 12:34:34.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9941d0f5005c4ce8848cd37e61e94d04\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dc14969d913245cd9231f639067b4179 : SUCCESS in 47m 21s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/81302be65acf4b418e346433bdae2b7f : SUCCESS in 39m 47s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9f1b31b3d3d84071be119b16bfd5d4b2 : SUCCESS in 40m 16s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/75f0c1f830024550b0c240e4f3d23b18 : FAILURE in 40m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f157b82da9ec45319a4aa168997c97d0 : FAILURE in 40m 01s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/fa8922e7ac47499a8a788fa4576da7ef : FAILURE in 42m 09s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cdab8e09e5244a4590ccb118104730ee : FAILURE in 46m 37s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/a06fdd5a304c47418f6e14b328d1d156 : SUCCESS in 43m 34s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b9d79a8dee8a4999b51ee8a5e3560947 : SUCCESS in 51m 20s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f799ddda48741efb6a125a07cb25101 : SUCCESS in 6m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6e6a30edc37b4e9eb0aa3f936410d682 : SUCCESS in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae0a68db2ef449e88b6181d11db1efbe : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fc34357ad694f01a60ac5413ff77fdb : SUCCESS in 7m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7e6f6a72a7ff4a6da9847a94eeb46672 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e20d61e94c324ec193472c55061cba46 : SUCCESS in 8m 49s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d66a9c22e2c34513aa7ab61461299c50 : SUCCESS in 51m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2930f6bcd26d4bb8967074e52634c343 : SUCCESS in 4m 46s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2ac4dd4d49e148b2a7355c10ea086525 : SUCCESS in 45m 35s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"03372dd1a251935e946161cc2516028d915b2ee2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-29 12:47:13.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"fcde544bca43fc094a4f5cdf0748895963fa557d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-29 12:49:23.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d70b61a243f65515d9a2a2430dc385d9ac7a888b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-29 14:01:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"f21a25503caa6ec395ee3b95bdc07a91d9ef166f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-29 15:48:03.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/76b676c188624f718a9d6cdb0d5b0077\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/256544099a8840c4806cceadba2303c2 : FAILURE in 44m 26s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/8362a56cc57c47bea3aecb6c01831518 : SUCCESS in 42m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cbc5572d55c34262a53cb303cf44a39d : SUCCESS in 49m 46s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6f0ef9f4118c425d9e5e7101a6cd7f5b : FAILURE in 38m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4b6445536a6148faa805fdbfe80109b0 : FAILURE in 41m 13s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/253bd281c2de4eb9a062b85a4a4c2ce9 : FAILURE in 42m 45s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7f4e7630f3e04ab58478c501661469eb : FAILURE in 48m 13s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/abc0d7982e0248dc98473f5126a3bc99 : SUCCESS in 46m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/100fc385292749f6b37d427d21a8343c : SUCCESS in 44m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/36d31eec779a419e915ed8c31cf888a7 : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c4a5ccf660d48ef954e5f2c99152a8c : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/942aa95b6ac54fa1aed3376490fc3888 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a2b5bcc7c744862b1b72eca4868f8d5 : SUCCESS in 6m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4088cb6b16314292a52c54c855d699a2 : SUCCESS in 7m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6439d47b1bbc4da296d427dcdbc616d3 : SUCCESS in 7m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eabc498bff434177a6852dde4c830142 : SUCCESS in 54m 29s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/89541dc5bbdf4965b495cbbb1e5f5f4c : SUCCESS in 4m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c431a9d314e94c77bc555aee3d8df687 : SUCCESS in 47m 15s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"d971baff64bd2e84f37c6d97be4aa9b59db48265","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-30 05:47:50.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"9a63e07a674e9564da708e508a70f2b4ffa87a3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 06:42:24.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/87afcb88e1214159beaa58f625f3dfd8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1e6b60e683414eac8473b9aabacf2956 : SUCCESS in 44m 33s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e8de90f29007440b981b7e0ff473d70c : SUCCESS in 45m 13s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/eebef149224349b58b6c1f034437d687 : SUCCESS in 48m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/d423c96c6b184020b4b8e5290b3bd016 : FAILURE in 44m 10s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c86ed8e2a34e4d4785c616d265eac1cc : FAILURE in 29m 15s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8b9f08bf744e409da0c4ff93789e552a : FAILURE in 42m 41s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2575e1b312214a0e9e2e847cd0ad53b9 : FAILURE in 28m 34s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0a3705f9223a49bcbe85433fd4316e3a : SUCCESS in 35m 17s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8e9aa09f784049a2ba38a1d728bebee7 : SUCCESS in 46m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d80cb26399f41a7a00782dcaba223ba : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ef06ffe9ec44a9d9d0c48a8e2ba10e4 : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fb3a628246ec4c06b11c9ab39bc14a7e : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/179472a0cfac4b13921dde898febf49b : SUCCESS in 6m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ca813620fca4c9cb859eca3b7f0fedc : SUCCESS in 8m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6902fb4bbb4846c5acb3b8fdeaf77f67 : SUCCESS in 8m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/46cc89d036ae4203b5eefd9d3a80d966 : SUCCESS in 33m 36s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a47bd643419542de97d2af80b8063004 : SUCCESS in 6m 03s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/496818a666eb4774bbf6a357af500535 : FAILURE in 39m 28s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"2f107299228d8e26d1baa974fdd2c330565dfc16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 09:19:04.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"0376bc837e993c62be599a54d4eb3cd1e61a3a70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 10:44:15.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03032c255af949ff8505c7521b87fd6f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/40228080056340db835e15837b5685d2 : SUCCESS in 39m 32s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/918433230b794a8d8a7230139ca290c6 : SUCCESS in 41m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9cd5318614fb46298cefc07a1fc74e71 : SUCCESS in 37m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/36210e1ce0174b0b94ceb5630dcfbe51 : FAILURE in 48m 26s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dc82659e4b7244b9859f104d86aeb6b3 : FAILURE in 38m 40s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6c866cf96bf642f188db53ed4998b6b7 : SUCCESS in 1h 01m 20s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e07da8a08d3242598088fc885fd287d7 : SUCCESS in 58m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/e2246ff7f70744f78e364492cbe44f73 : SUCCESS in 44m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/da9161eff475422ca58bcac7716edd01 : SUCCESS in 44m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/19d5e3c0356540d9aab0d27fa6a625c4 : SUCCESS in 7m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01a70d245b3041af9663a03de145ad0a : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/336ae4023afd4d44a993e2f1af778f2a : SUCCESS in 4m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/251e3d60adc0436e898736d47af9b1f7 : SUCCESS in 5m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bdf53859d6df4340aba6c01769b61272 : SUCCESS in 7m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a43ed774575d423dab59a9d0684ae4d9 : SUCCESS in 8m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/65b6149bc82c4668b04404db8630bff6 : SUCCESS in 46m 25s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e43f01ef48cd4cb4992fde1fd39a8587 : SUCCESS in 4m 24s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/a9e984e37a8f4764a47c1335a37da160 : SUCCESS in 36m 21s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"14a1469882afbf1eef463aad3a10b03d4ae46f2f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 10:46:28.000000000","message":"Patch Set 9: Review-Priority+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"527415c60d7aaa13f1c886b0ad9eca143906d720","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-30 11:20:39.000000000","message":"Patch Set 9: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"f7d9581dd74d8483fb8d645f981ed8545355fde9","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-30 11:23:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1aec1908684ddc49e68edc7a6a5a7be411235209","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 11:28:25.000000000","message":"Uploaded patch set 10.\n\nCopied Votes:\n* Review-Priority+2 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE OR **is:MAX** OR is:MIN\")\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"6569f42913e3329dbb9f04567316e81edba90d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 13:06:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e4e6c47b92404949af99ae5c0b954b4f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1d32dc19688b4ac6a86edd43d0d5dd69 : SUCCESS in 48m 35s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/1438601159ff44db905c8e8749d293d4 : SUCCESS in 44m 11s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c4cbe9d1e1724bf59e51fb1e7897b6d4 : SUCCESS in 45m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0bf96f5c14944d228087492f6da0848f : FAILURE in 43m 50s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/734d40c2fc394f319e92d5940c7ff224 : FAILURE in 40m 50s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/27ce841835474c1a991b43d744700237 : SUCCESS in 58m 52s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cc5d7defaca5459b839fa68cf8edf847 : SUCCESS in 1h 07m 20s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/00764edb70c44754a7b6a8c06285f3d1 : SUCCESS in 40m 22s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/57939d211a04413c8abf1c9d8fb8b3c1 : SUCCESS in 42m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/edbebade278045dfa0f556e4bb01f05a : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b178c1a71494e0c9b94f27763d83c39 : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/03e144f10bc143df8bf7a23de07a543d : SUCCESS in 5m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a4c6e75e4d7478d81842172a7add722 : SUCCESS in 7m 27s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6a74273c70284fa78ee7bc67ac05fa4d : SUCCESS in 7m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44f025fae2149bbb718d270296b346f : SUCCESS in 8m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7eeac49502414bd8ac30e60230c2ae89 : SUCCESS in 45m 38s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/137b980f663e496cad04b922c12f0335 : SUCCESS in 4m 57s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0038d8527d5f417e966407c91c53afdd : FAILURE in 41m 31s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"49e25b171855776d57b246741d96290e03f4a895","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-05-30 13:30:09.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"cdbd5713b16801925743e8787014e94777f7c18c","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-01 17:01:52.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"f3003b0eab68c6634ad57783013fe185c4c25cbb","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-01 18:54:45.000000000","message":"Patch Set 10: Code-Review+2","accounts_in_message":[],"_revision_number":10},{"id":"bc82611149ece598491c34d8987b560d436e7870","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 09:29:56.000000000","message":"Patch Set 10: Code-Review+2 Review-Priority+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8c058387c5ca3f3982eeb43eba248369450b36d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 09:30:26.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"2edde6b651bb0d6cd808332fac1e0646833e070d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 09:43:44.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a2638f1f047456e8d3d3ff9f6851a47\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/35f5a14a7e264cbdad6f4ad6b858d701 : CANCELED\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/595b3001e19149499e801f00c6057f91 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/22fb3f1c55a14291bb9de2cf0b4fe198 : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4865a35bcd9b417cbbc0c00a586455b7 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1f1a17fd10484022a82fd15c0934d988 : FAILURE in 12m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/cb158339a1724fdc8919b6d76c02dc1c : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba50adfdda14443ca000701baf60557a : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/67dce5bb68df466eb9bc55854cd96082 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5e7e9deffcd4422cb513e6f649a06ea7 : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cad43e0db70549dc82cdf716ecdd3b55 : CANCELED\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9227e30ed7834c289cd4b938b86a59cf : CANCELED","accounts_in_message":[],"_revision_number":10},{"id":"8689f5731582016228e99c81b4a7590fb312f8b7","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-02 09:44:10.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"e1360ce837b3b7d198698a2f415e300ceb1e9799","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:01:11.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ce12cbc69e9426c97f36740675efb74\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/86897086feb24ae19ef51cebed37d63b : SUCCESS in 47m 00s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f94cc2790e8d4822ab478417f3a9f9fc : SUCCESS in 46m 31s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9afd33ec70384ca9a090188b8ae30ccf : SUCCESS in 49m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/15287c37f38a459abc69e824b5846d52 : FAILURE in 44m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7c159e30608b4dcd8c7e205d3aabb646 : FAILURE in 35m 48s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/dbf267c826c24b6bba04b6241ddd66d0 : SUCCESS in 1h 01m 34s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e8fae11afe73463c8837d2d516ef90bc : SUCCESS in 1h 11m 25s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9cc02c01207b4741a48f3ce91671b475 : SUCCESS in 45m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/13552e2b16a745f6bad124b1d37dea32 : SUCCESS in 48m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f53861865264f98ad10c629660b21d7 : SUCCESS in 5m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7ba704ef28471d8bed45998b345da6 : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0aaa4a4fc3604c21ab73865f4652622f : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a467c6fd037d4c73b7cea695680aadc2 : SUCCESS in 5m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/804a6b74952b431eb6299676db721691 : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68778cfbbc1b4c0bb36a8560bdadec21 : SUCCESS in 8m 36s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8e76e012663944bf8ae6269fbbaae6dc : SUCCESS in 31m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6e283c598edd439ca1919420bffb03df : SUCCESS in 5m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/58f87cccaa1c461a9031d8da0ebc5d2d : SUCCESS in 41m 55s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"555229210d292aa3b16291ad6f1ef39175087837","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:01:42.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"e01a3bcf3d59f2f3b48c349bd8a2c0464aaf2055","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:13:09.000000000","message":"Patch Set 10: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d5210a1aa17344bd8e64fc16217101a7\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bdc982d7b5b94191ad3415d6f6195b38 : CANCELED\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/a1df78612cf949f88a6622afd6193215 : CANCELED\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4dcfa9843cd443bb8f7505a26e18cd98 : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/97329d5c228d47508ba00f84025a09d7 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/06abe449ffe745a594ba5efe99c1463c : FAILURE in 10m 55s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/639b7fdcd5dd4de58f0012fb1a295446 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b26c2c0613df402eabb2093a4a273a0d : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7740331baec4545b2ee5842e542d160 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c1e2343143a64e60920dc825e37c3844 : SUCCESS in 4m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a35d4662846a4062ab12cf4e3b5fe665 : SUCCESS in 9m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/48a3205b3ceb482f8ae4f50a0131734e : CANCELED","accounts_in_message":[],"_revision_number":10},{"id":"be3058c4d02d899f4b86cb5423cab18a92079e48","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-06-02 11:22:22.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"35607ade1c5a7b6adc909ad13b8b4fa6f8372c3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 12:51:49.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3dfc05e6819047699e68a30ee2997224\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dee30b8d3d924e28a490537989857765 : SUCCESS in 47m 36s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/519827d8afca4acfb5569474bafecbf0 : SUCCESS in 42m 32s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1a6868283d1f450b8c9cc8bc64142c02 : SUCCESS in 46m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5da4ced7ec57403483499f1939fb9f0d : FAILURE in 52m 13s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3f78c56888d647db91937d6f5aae9236 : FAILURE in 35m 51s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ea82e2575feb4e5e8a67690d8896ee39 : SUCCESS in 1h 09m 05s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f5211f5a3c2e4597a4adc6a293ffadd5 : SUCCESS in 58m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/1dba53aac016433e8314b0f4192a3324 : SUCCESS in 41m 49s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7e6014fc950c4bd9ba2c68d2dbe474af : SUCCESS in 49m 10s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/373de63518e741acb43aa4d707336e98 : SUCCESS in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0836e17004347e188790d11491e9ddc : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4acf287c5f3478d960769aefa36f03e : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/45fc275de56442e9b8efe2d8a4429cd0 : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8acecc264de14f1987c405476173ab1c : SUCCESS in 7m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f91edd9e50234528ad8cb1bf1da8f981 : SUCCESS in 7m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7c9ca1723e044eea9713173e981f50ac : SUCCESS in 33m 20s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2b8e328cf03e4202897a81f583a608e0 : SUCCESS in 5m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/671a968ad82643e6bdb782d2b0679c80 : FAILURE in 42m 44s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"84e423bf0238626d238d2f909096ee9ac81e26dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 12:52:21.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"eae3abc65252a63194032474c6b431c1f0b9b4a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 13:50:52.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f014e5022e3342fa8e86c5a4aef9ada9\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/e2c6afc5f4cb4079bdf5f9079953b80a : SUCCESS in 39m 47s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/23ce4b0536a14b17b65863d398992f25 : SUCCESS in 41m 54s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1f0c2f89aeea4a02adef5c93ba6839b2 : SUCCESS in 39m 59s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fe356602a46f4877bb708e26aeb0c488 : SUCCESS in 54m 12s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9b846da97e744fc5b0ea6094057ad0aa : SUCCESS in 38m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e0da1d5b6c2946319dddb3bd8bada3f4 : SUCCESS in 40m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94fa225d18234704aabaaaeef6ad6436 : SUCCESS in 4m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/afd54622a6fc48d685b72e0099e2c941 : SUCCESS in 4m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c9b2627bd46e4269aa77779e4ba3ecae : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/008c41feaba448b3a85264d0d130b91a : SUCCESS in 8m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/137d50b9fc364c909d9f7a3d1c86bfa0 : SUCCESS in 54m 48s","accounts_in_message":[],"_revision_number":10},{"id":"fc2dcbcd18af4282910b7e6a27eddf0524fa9642","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 13:50:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"06479d1d48c9b158f39407e12f8e133c303060aa","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 13:52:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/05f679643165487aacd31dbf838230af\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a2f08e52354493b8cb31b05b0ec9187 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"c2f1473909069cf7513a62392222a3d76d3247db","revisions":{"c0737ffee4b1fbe7e918e4560551a8408d697d70":{"kind":"REWORK","_number":1,"created":"2025-05-29 10:22:36.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/1"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0737ffee4b1fbe7e918e4560551a8408d697d70"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0737ffee4b1fbe7e918e4560551a8408d697d70"}]},"branch":"refs/heads/master"},"f9bd463c0c1784b6cf059aa73b3e75e5b96703b9":{"kind":"REWORK","_number":2,"created":"2025-05-29 10:37:38.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/2"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:35:06.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9bd463c0c1784b6cf059aa73b3e75e5b96703b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9bd463c0c1784b6cf059aa73b3e75e5b96703b9"}]},"branch":"refs/heads/master"},"e2f8fd724ffc4b529b07a0d975d487c9755637aa":{"kind":"REWORK","_number":3,"created":"2025-05-29 11:36:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/3"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 11:35:51.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2f8fd724ffc4b529b07a0d975d487c9755637aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e2f8fd724ffc4b529b07a0d975d487c9755637aa"}]},"branch":"refs/heads/master"},"f6cd337fd4c3529c365aaa2f75edd32119bcc326":{"kind":"REWORK","_number":4,"created":"2025-05-29 11:39:48.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/4"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 11:39:05.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6cd337fd4c3529c365aaa2f75edd32119bcc326"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f6cd337fd4c3529c365aaa2f75edd32119bcc326"}]},"branch":"refs/heads/master"},"5d5dd17a062f6184c0c2c560f894cc41bef55e99":{"kind":"REWORK","_number":5,"created":"2025-05-29 12:47:13.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/5"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 12:46:43.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d5dd17a062f6184c0c2c560f894cc41bef55e99"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5d5dd17a062f6184c0c2c560f894cc41bef55e99"}]},"branch":"refs/heads/master"},"0acc10e97b87d01ed8c236908270c906ce286575":{"kind":"REWORK","_number":6,"created":"2025-05-29 12:49:23.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/6"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 12:49:11.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0acc10e97b87d01ed8c236908270c906ce286575"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0acc10e97b87d01ed8c236908270c906ce286575"}]},"branch":"refs/heads/master"},"11aa08bde9c50c8dac7c12b0fd6b0e2baad8a9a2":{"kind":"REWORK","_number":7,"created":"2025-05-29 14:01:36.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/7"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 14:01:29.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11aa08bde9c50c8dac7c12b0fd6b0e2baad8a9a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/11aa08bde9c50c8dac7c12b0fd6b0e2baad8a9a2"}]},"branch":"refs/heads/master"},"3eb53de531b01e28c43be9dd08b6b2855145fd4f":{"kind":"REWORK","_number":8,"created":"2025-05-30 05:47:50.000000000","uploader":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"ref":"refs/changes/60/951260/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/8"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Rodolfo Alonso Hernandez","email":"ralonsoh@redhat.com","date":"2025-05-30 05:47:48.000000000","tz":0},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3eb53de531b01e28c43be9dd08b6b2855145fd4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3eb53de531b01e28c43be9dd08b6b2855145fd4f"}]},"branch":"refs/heads/master"},"d1155ed3e7afbed0e4699f27d25398be1ee1ea22":{"kind":"REWORK","_number":9,"created":"2025-05-30 09:19:04.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/9"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-05-30 09:19:03.000000000","tz":0},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1155ed3e7afbed0e4699f27d25398be1ee1ea22"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d1155ed3e7afbed0e4699f27d25398be1ee1ea22"}]},"branch":"refs/heads/master"},"c2f1473909069cf7513a62392222a3d76d3247db":{"kind":"REWORK","_number":10,"created":"2025-05-30 11:28:25.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/60/951260/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/60/951260/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/60/951260/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/60/951260/10"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-29 10:17:57.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 11:27:33.000000000","tz":120},"subject":"Move grenade skip-level jobs to start from 2024.2","message":"Move grenade skip-level jobs to start from 2024.2\n\nAs our master branch is 2025.2 (Flamingo), we should run grenade\nskip-level from stable/2024.2 instead of 2024.1.\n\nI also reverted designate-grenade-bind9-skip-level to be a voting job\nagain.\n\nChange-Id: I53d1315d373bb08ad9f295c5aa7bb78a25ae6a74\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2f1473909069cf7513a62392222a3d76d3247db"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c2f1473909069cf7513a62392222a3d76d3247db"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
