)]}'
{"id":"openstack%2Fdesignate~951391","triplet_id":"openstack%2Fdesignate~master~I31ab824b0ed32580de93f94c5b25ac65f0eb6897","project":"openstack/designate","branch":"master","topic":"fix_grenade_jobs","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-06-03 15:28:39.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I31ab824b0ed32580de93f94c5b25ac65f0eb6897","subject":"Set grenade and tox-dnspython jobs as voting","status":"MERGED","created":"2025-05-30 10:45:50.000000000","updated":"2025-06-03 15:30:21.000000000","submitted":"2025-06-03 15:28:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951391-fix_grenade_jobs","meta_rev_id":"5a4caf17dc655a4b4ec6d91da267853bea5e0dc7","_number":951391,"virtual_id_number":951391,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 15:28:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-03 03:26:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-06-03 13:57:08.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-06-03 13:57:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 11:34:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-05-31 01:49:25.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-06-03 03:26:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-03 13:57:08.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"aa1cda9fe4e2ced376c2cfd8e3b65515233bdb51","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 10:45:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cf89766eeed3723aba9840cad06dfe971dda254c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 11:28:25.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"bb2ac89a8d21aadeb3eac6ec4e7d10871b8ed36e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 11:34:16.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aaf142eda1a2450650e2c9bb7b73e2ac0bf10f3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 11:34:42.000000000","message":"Patch Set 3: Verified-1\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/designate on branch master.  The\nproblem was:\n\n  Configuration item has more than one key.  Each zuul.yaml\n  configuration file must be a list of dictionaries with a\n  single key, for example:\n  \n  - job:\n      name: foo\n  \n  - project:\n      name: bar\n  \n  Ensure that every item in the list is a dictionary with only\n  one key (in this example, \u0027job\u0027 and \u0027project\u0027).  This error\n  may be caused by insufficient indentation of the keys under\n  the configuration item (\u0027name\u0027 in this example).\n  \n  The incorrect values are around:\n  \n  job: null\n  name: designate-grenade-skip-level\n  parent: designate-grenade-common\n  vars:\n  ...","accounts_in_message":[],"_revision_number":3},{"id":"fed13367cf7c31b2a57a2c1a832573418576d2d9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 11:37:23.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"62ac85755c14af44856e995beb8cf46641c2930c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 11:53:01.000000000","message":"Patch Set 5: Published edit on patch set 4.","accounts_in_message":[],"_revision_number":5},{"id":"09c8a777c3dafbcda6b1f2ebae93c80b29e5175f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 13:10:15.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1f0ff73f28a5499aa53ab7735020859b\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/07dedf524a0e45839c2dce112a12a24e : SUCCESS in 47m 50s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/b5e761362a8a46da889491f221be3827 : SUCCESS in 47m 24s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/77e8b771ee2648dfa742df429593f190 : SUCCESS in 44m 10s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9e069a1aa4184c89bb11a7abae7747f5 : SUCCESS in 58m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f6d97755d3124dc6b41e66991c525422 : SUCCESS in 1h 14m 39s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/03d4b38605b34643ac43014f86d56a7a : SUCCESS in 58m 11s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6adb6ad8ad1a4e83b316a66d6b2c3dca : SUCCESS in 1h 08m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/609aa9fc7024424cabb06e0e230b164e : SUCCESS in 41m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/f66634b0a9a94c609cd2f69976c39d4b : SUCCESS in 46m 05s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f8a5b95d2b647a2bb3aaa2179b27c8e : SUCCESS in 5m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd17ffd34f904adca7945fd7dafcca4f : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de54f09b591a435f9337e44832ed8f74 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6f2f634d72a747feae85a7681d9f5a82 : SUCCESS in 6m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/30eae6e8867c4f42b2da8dca448fbe87 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25525ce48bd546d3ae2ad6a1290cfe01 : SUCCESS in 8m 55s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b89eaa3b087d404781ec940e304e7958 : FAILURE in 23m 09s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ed0df7b124c945af957734c642de298b : SUCCESS in 5m 41s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/004c6b9004934f6f96b5978757006c4d : SUCCESS in 46m 39s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"67040fba680c13b3c5c6ad8d85b811594f6066c2","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 13:12:07.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ff73f5469c82b27511b8b8aa8bd7c80cccecdd0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 14:58:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a72e3dda15348698ce421c71d192f4d\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/1a4a22db54a24935ae68e068a87c5133 : SUCCESS in 48m 50s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/98d254c3d636409495806001906071fb : SUCCESS in 44m 06s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/04eabe568fa840e5a99a9faa10d426d3 : SUCCESS in 39m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/0b13776ee2ae4f0a8d63604225c31afb : SUCCESS in 58m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c2e33953b6e249beb9f7c6b6da6a0b63 : SUCCESS in 52m 08s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cb09b09d4e894aefa1c2d47bf9f4a0e5 : SUCCESS in 1h 02m 02s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/abd6d2c18efd4420b5265278acc4c27d : SUCCESS in 1h 07m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/88bf0429d1db410683bb7fa3b71adc2f : SUCCESS in 52m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/aaa056d856284fe99434b6b673a8f6e7 : SUCCESS in 49m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/242620c126604436925a4f175cca5384 : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1511b788a4484d92994c62f7d81faace : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/654da7faedd64bb5877681b48e17275d : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/68abe1166c044734ac2ea9ddad3e79bc : SUCCESS in 5m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59bf1bd4ecb04537bb8b922db2a600bd : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0e76f856b8a43f7b24f2a645b88b8cf : SUCCESS in 8m 56s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/eed7c64e392146a7a0548528b3cdec34 : FAILURE in 20m 17s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9ba22097632c42c4b8451486a966e109 : SUCCESS in 5m 31s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2a5c49465e134510b4e273ebc07d80bc : FAILURE in 40m 49s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"62421c1c84418e67271d5a58459b305f2a20e978","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 15:28:44.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4d0eccf462caefd7c862b4b3c8afd2d588ea3396","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 16:43:28.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/381150efc4d2444986412cb9d238670e\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/0dfb91c4b2e948028b106c7f6a36fcfc : SUCCESS in 41m 50s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/23e2e88cf42949989193714e08ea5b9c : SUCCESS in 43m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2539cb974e914953bb1e9122ad7d097f : FAILURE in 42m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9a2caeabd1ed4b23b585cdf9b34fdcb7 : SUCCESS in 59m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f0641c503dd94502bf49fd071cc5769b : SUCCESS in 1h 04m 19s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/49caa24e765e4b7a8c1dbe39369401d6 : SUCCESS in 1h 04m 44s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/459561c40cf04ecf90b80c76ffe8ed26 : SUCCESS in 1h 08m 02s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/536422a5784846caab239690d4e09760 : SUCCESS in 50m 41s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/118784452a62454c87b20d1b127bd83a : SUCCESS in 46m 42s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d05cbe118d504de3913307707a710dcf : SUCCESS in 7m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a121edbf86f41f1a296d9177d8a06a1 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b8ea2aa16dd94ccf9881b667cefc44a3 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4d5bcbebb46849019cef498ec8a56e22 : SUCCESS in 6m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/185254312479455b8ef5d993b04c0456 : SUCCESS in 8m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cefc6dc4299644f48df1f9fb19c2f1c2 : SUCCESS in 10m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6d8d127301074572a9f0e7115164abc4 : SUCCESS in 49m 30s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fc95dcbe10b147a7b587e27d9326e906 : SUCCESS in 6m 51s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/bbf4e59294794bd29a737f6a1ffd49c6 : SUCCESS in 49m 42s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"dff2fc9fac09b66202bc7827d3ef95bd694c3ac5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-31 01:49:25.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"15ba39697fc72ba46d59769afdb669090bced3a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-31 03:34:51.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6846fc7a86e249af9ac07ecb9a83529f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/6a0ce0b51d684d6aa29419961b7e7a0c : SUCCESS in 44m 46s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c73f9fc3cca64fa7aa81bf59d5e2e8db : SUCCESS in 56m 28s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5634c2678e794da688e060467c3bcab1 : SUCCESS in 45m 05s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8f9dc64ca18349f395f53260d6b086ca : SUCCESS in 1h 05m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b649c231291a4cf0a113f203b41df1c1 : SUCCESS in 1h 17m 43s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/7886ecf232464d578c601fc05714bd02 : FAILURE in 1h 18m 02s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0312c2dae4e449539acb19564b53b846 : SUCCESS in 1h 20m 26s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/168436bf6821452ebdec25007c61fdc8 : SUCCESS in 56m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/91502bc0a718491ab6cad237ba5dc7ef : FAILURE in 27m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9d4bd289a9984edaa9d450a96b95a1fb : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cf8d63e0af4941d7a0e4cfcbc8b63e3b : SUCCESS in 3m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/762f0e718a124863a877e42c4ecf5bc9 : SUCCESS in 3m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e52c5e02563d474a891e6d9edaf80fae : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8a4e58466e6c4fffa583396b3ed6a3b3 : SUCCESS in 5m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc6e2e45c0914191bcec596e72da5667 : SUCCESS in 7m 17s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9d321f214e324766991a9e3cd7bdb94a : SUCCESS in 43m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dbde4d9eed3044b1b8958e0f3aae9f78 : SUCCESS in 4m 25s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ecb376a5b9854fbc8544a88ed75408b7 : FAILURE in 40m 46s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"a5a478fc09c50d5b33b6cca4c5b08037a65a7f80","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 09:33:29.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ce668dfa668f15a01e57f4b2a87705f4b083b310","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 11:29:20.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a881c2c27a1244a0815970e973b27774\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c3a46fedaa544ba3b13f8dd6fc8bd97a : SUCCESS in 41m 15s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/217f89e77ebc4922a8f0921572d45333 : SUCCESS in 39m 01s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9d143f05813f4ccf87e55e3c3d485a72 : SUCCESS in 44m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/be7807814bc244be818e1a8060b6c16c : SUCCESS in 1h 12m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/622db3a0a3814624a713b613360608d8 : SUCCESS in 1h 03m 29s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a5f093b13c2f4f6e967c2a93250d47b7 : SUCCESS in 1h 07m 36s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7801594aaa564d7c94f6b853d215a34d : SUCCESS in 1h 00m 07s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/17ef7c29a71747d38fdf9b44a25b773c : SUCCESS in 41m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/58196d30e814456ab619163675037812 : SUCCESS in 41m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32ab9aa401274129a51755a272aefb16 : SUCCESS in 7m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66ab60e1a8c549c4b3fef08a43aab43c : SUCCESS in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06048ece90f14c8e947fde14a277c6e3 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89c6efea8e7647568c04ab113bf5d874 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e6dd92f926a144d0b82cfc028cd3cbd3 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79ce55495ba744cb9279636c3a08baf8 : SUCCESS in 8m 30s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/578d20bc1eb74452b921807cbddf81e0 : SUCCESS in 45m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f50cc7ff78644dd79b5028a5a5413846 : SUCCESS in 4m 41s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/deb523e28a2c490285839b7d58e658e1 : SUCCESS in 45m 52s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"135c33e00322be71f626a61cbfe3dbfdd1656aeb","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-03 03:26:16.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e852c6ddd3cfde3a7d052a9bbb809c26a506bb40","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-03 13:57:08.000000000","message":"Patch Set 5: Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"9ebc4966139e66559533cf496dbac41b5a035efa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 13:57:41.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"8d513d7d212c195510089eafc9b7a8fccd44b9c7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:28:39.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a7f48fb737940eba4a4dc7e01be53e1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/dfe1c2f38a024103ad123860c663f31a : SUCCESS in 42m 58s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/532ba1c334384bf8b86a34f47738035b : SUCCESS in 45m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/a7157f0b607f4493a2e0dd0dec55b457 : SUCCESS in 49m 02s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d4e2e6a2df5e4ec0a4ccbb3d34f49fd1 : SUCCESS in 58m 30s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7bf0556ed900456c9d6bb94e176a0c62 : SUCCESS in 39m 38s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/51ef44c895b9445a9c625529c4aa32a1 : SUCCESS in 39m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb2f407857e149faa411dbf7bcbc732c : SUCCESS in 2m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c1d4a021cc9346fc9e9a550a340bb88d : SUCCESS in 6m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce0a042a53de421b82f25d52463ec7ce : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/602e53961d774f118c5fc11673e6d162 : SUCCESS in 10m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a20fa6fc35ff40e39c36525813d305ee : SUCCESS in 33m 34s","accounts_in_message":[],"_revision_number":5},{"id":"7c75cd96f73c20b7a59e0e75ee029de417f58dd7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:28:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"5a4caf17dc655a4b4ec6d91da267853bea5e0dc7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:30:21.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cf04e37196a4a67a029c43eaf19f5c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4bb56990bf7747e1ac8ce5b094605abc : SUCCESS in 46s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"e515ddaddca7c5ef5ab75e0809ce25a1bb12fb74","revisions":{"a22a79a99e54d7855dd184014f549e38c7df747d":{"kind":"REWORK","_number":1,"created":"2025-05-30 10:45:50.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/91/951391/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/951391/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/951391/1"}}},"commit":{"parents":[{"commit":"9e2214cf0a07edc6423b88ed7e750f87d277d718","subject":"Fix uwsgi config for grenade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9e2214cf0a07edc6423b88ed7e750f87d277d718"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 10:43:04.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 10:43:04.000000000","tz":120},"subject":"Set grenade and tox-dnspython jobs as voting","message":"Set grenade and tox-dnspython jobs as voting\n\nWe recently fixed grenade jobs, and tox-dnspython was added 3 years ago\nand looks stable enough.\n\nThis patch sets those jobs as voting.\n\nChange-Id: I31ab824b0ed32580de93f94c5b25ac65f0eb6897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a22a79a99e54d7855dd184014f549e38c7df747d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a22a79a99e54d7855dd184014f549e38c7df747d"}]},"branch":"refs/heads/master"},"72210b5c9fdea91a96fab6fecd162891dd58e25a":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-05-30 11:28:25.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/91/951391/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/951391/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/951391/2"}}},"commit":{"parents":[{"commit":"8222a367424053572f9bd962ddac289d7f0908ef","subject":"Fix uwsgi config for grenade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8222a367424053572f9bd962ddac289d7f0908ef"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 10:43:04.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 11:27:34.000000000","tz":120},"subject":"Set grenade and tox-dnspython jobs as voting","message":"Set grenade and tox-dnspython jobs as voting\n\nWe recently fixed grenade jobs, and tox-dnspython was added 3 years ago\nand looks stable enough.\n\nThis patch sets those jobs as voting.\n\nChange-Id: I31ab824b0ed32580de93f94c5b25ac65f0eb6897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72210b5c9fdea91a96fab6fecd162891dd58e25a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/72210b5c9fdea91a96fab6fecd162891dd58e25a"}]},"branch":"refs/heads/master"},"b719c3f932b25ed7bd91c0a34d83f4f92da975f2":{"kind":"REWORK","_number":3,"created":"2025-05-30 11:34:16.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/91/951391/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/951391/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/951391/3"}}},"commit":{"parents":[{"commit":"8222a367424053572f9bd962ddac289d7f0908ef","subject":"Fix uwsgi config for grenade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8222a367424053572f9bd962ddac289d7f0908ef"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 10:43:04.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 11:33:56.000000000","tz":120},"subject":"Set grenade and tox-dnspython jobs as voting","message":"Set grenade and tox-dnspython jobs as voting\n\nWe recently fixed grenade jobs, and tox-dnspython was added 3 years ago\nand looks stable enough.\n\nThis patch sets those jobs as voting.\n\nChange-Id: I31ab824b0ed32580de93f94c5b25ac65f0eb6897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b719c3f932b25ed7bd91c0a34d83f4f92da975f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/b719c3f932b25ed7bd91c0a34d83f4f92da975f2"}]},"branch":"refs/heads/master"},"58453f805743a7e9aaad740e3f035246052b0024":{"kind":"REWORK","_number":4,"created":"2025-05-30 11:37:23.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/91/951391/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/951391/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/951391/4"}}},"commit":{"parents":[{"commit":"8222a367424053572f9bd962ddac289d7f0908ef","subject":"Fix uwsgi config for grenade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8222a367424053572f9bd962ddac289d7f0908ef"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 10:43:04.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 11:36:46.000000000","tz":120},"subject":"Set grenade and tox-dnspython jobs as voting","message":"Set grenade and tox-dnspython jobs as voting\n\nWe recently fixed grenade jobs, and tox-dnspython was added 3 years ago\nand looks stable enough.\n\nThis patch sets those jobs as voting.\n\nChange-Id: I31ab824b0ed32580de93f94c5b25ac65f0eb6897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58453f805743a7e9aaad740e3f035246052b0024"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/58453f805743a7e9aaad740e3f035246052b0024"}]},"branch":"refs/heads/master"},"e515ddaddca7c5ef5ab75e0809ce25a1bb12fb74":{"kind":"REWORK","_number":5,"created":"2025-05-30 11:53:01.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/91/951391/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/91/951391/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/91/951391/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/91/951391/5"}}},"commit":{"parents":[{"commit":"8222a367424053572f9bd962ddac289d7f0908ef","subject":"Fix uwsgi config for grenade jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8222a367424053572f9bd962ddac289d7f0908ef"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 10:43:04.000000000","tz":120},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-05-30 11:53:01.000000000","tz":0},"subject":"Set grenade and tox-dnspython jobs as voting","message":"Set grenade and tox-dnspython jobs as voting\n\nWe recently fixed grenade jobs, and tox-dnspython was added 3 years ago\nand looks stable enough.\n\nThis patch sets those jobs as voting.\n\nChange-Id: I31ab824b0ed32580de93f94c5b25ac65f0eb6897\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e515ddaddca7c5ef5ab75e0809ce25a1bb12fb74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e515ddaddca7c5ef5ab75e0809ce25a1bb12fb74"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
