)]}'
{"id":"openstack%2Fdesignate~951398","triplet_id":"openstack%2Fdesignate~master~I1d1e3e0806eaff5b55ba01e25a015f5cac3e8131","project":"openstack/designate","branch":"master","topic":"remove_py39","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-06-03 15:42:07.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}},"hashtags":[],"change_id":"I1d1e3e0806eaff5b55ba01e25a015f5cac3e8131","subject":"Remove python3.9 support","status":"MERGED","created":"2025-05-30 12:19:06.000000000","updated":"2025-06-03 18:36:03.000000000","submitted":"2025-06-03 18:35:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"951398-remove_py39","meta_rev_id":"6821867eac5c6b60a7aebb78f8f3600c8f86e350","_number":951398,"virtual_id_number":951398,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-06-03 18:35:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-06-03 03:20:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-06-03 13:57:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-06-03 13:57:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-05-30 13:39:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-03 03:20:49.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-06-03 13:57:20.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"69b3df66aa9057d79e1b301611388c7d06c842c2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-05-30 12:19:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db76eea7db7ca6403d15ae09e3200617f8e61f3d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-30 13:39:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1d303c4f08d4051a0286de64f6f1b5c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c7b1259de8b749669708a7299171b93d : SUCCESS in 2m 14s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/cb9338f2b9b54a9da6d3913cdaa7d557 : SUCCESS in 45m 02s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/a23e521602b24c088046bec87502ff41 : FAILURE in 13m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/2abe793ff64145ff83be5bcf207c90b2 : SUCCESS in 41m 02s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c85b0eb0775a43bc8bae65d18adcb931 : FAILURE in 39m 12s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ac51754887234251baed41978cb7283d : FAILURE in 47m 39s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/66c3e88124f74ae88513600d59948685 : FAILURE in 15m 46s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e5097f0344cd4dee95c499085dc343f4 : FAILURE in 22m 06s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ca6f42b6a0a24cde9c013ca0f6089c62 : SUCCESS in 39m 46s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/5625979d2aa74757b27dad6a5047f6cd : SUCCESS in 42m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fb4afdc1ae4427fbb4eed1d518a3578 : SUCCESS in 6m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c82137c03f124972b33b55673f1f45db : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28e9cf65d2d94e9fa9ceeccb53468dd6 : SUCCESS in 3m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/17ef026ac1ff48e5b2753e13233e548e : SUCCESS in 4m 21s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/22ebc105c4494ac9aca16430e29a20f4 : SUCCESS in 7m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/136a00204a8c4ef485bf5597bd8ab93b : SUCCESS in 8m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e1f573e968f14d948eb57cc8ee7e3cc7 : SUCCESS in 4m 32s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/5a81171ed73a460cbe3ad5c8a388b2ac : SUCCESS in 49m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ecebcd1356c4f99e5312af0608ba011e14282a26","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-02 20:21:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3cdf1fdfef5f18034c9a7fb9ddab91b6c463dabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-02 21:37:54.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0d4677f8ad64ed490ee8abe7e84f1e1\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0f43f0d9f5347e78263a10ec0e5b9ad : SUCCESS in 2m 48s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9f91457e578d44368cc0207ed1205419 : SUCCESS in 47m 20s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f6ac9ab0bc2446fc8bf95e35941f2d4c : SUCCESS in 47m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/96960cb6a2cb4deda579d3750722c21c : SUCCESS in 38m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8131ef76b39e4b6294bedbaaacf80327 : SUCCESS in 1h 09m 45s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/da026bd53b4f43bdb567c218c7c3702f : SUCCESS in 1h 06m 58s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a78ab076b7ca462fa8958a5812351cf9 : SUCCESS in 58m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2e3373871b9745e7bfc30d2708569f94 : SUCCESS in 58m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ee094817464474ea570a49b6f42bbf2 : SUCCESS in 39m 36s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/90b78720f3c143ccb68670fda84ce0af : SUCCESS in 41m 00s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2051570818045a0b01f1e6f64d31a6f : SUCCESS in 6m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/534c533e5dbf44cea61d2e30a362d4d4 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1772a58210cf4eb8b840e83a1c5cdf58 : SUCCESS in 4m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cafc612e9c446b88c6e9e5b1d63051c : SUCCESS in 6m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/84c1181d23c544afa7b9234f5f9acc4d : SUCCESS in 8m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43cdb865421341f18350c745492ea724 : SUCCESS in 7m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/927516101a9640858885134939455bfa : SUCCESS in 4m 48s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/fbfeb696c0d64c0ebbcdc6daa24b33f4 : SUCCESS in 46m 57s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b3841d677646c54691f4467840038953dc92a2d1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-03 03:20:49.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a987385e7005e44e77670594c4f2e9682eeb77d8","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-03 13:57:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"95669b31b28a9a991d62b6d00d5a9a67a2c94c9d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 13:58:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"5e68b55da32d2eb33a73930eff5104d023fbea98","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 15:28:41.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/140ab9d3e42c48689b4997533021cdbb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28e0d7d3a55b46e892fcb26ed23ad497 : SUCCESS in 2m 06s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/610398ade38d48d88ddac8a309e98947 : SUCCESS in 44m 24s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f45dba5aebef4e13a43aca0b0dbd292b : SUCCESS in 42m 48s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d3ffb05b0b024185a225f309ed618e90 : SUCCESS in 47m 37s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a596b67ec37a432d8e4c699254908467 : CANCELED\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/3a3b4062d2174683b1ff0238578eb7fa : FAILURE in 55m 25s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/816e6bcfa64e49c094ef34a179afa934 : SUCCESS in 42m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2904446d8dcc41c2b4685a26b113c010 : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba996bf6a25c434bab33d874d30aab17 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c13af4a6cb5e4e1884a508928c4f3772 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/057e0be1ceb64c698e21dce64dd2bf1b : SUCCESS in 8m 48s","accounts_in_message":[],"_revision_number":1},{"id":"05fc36169aea4767063bf6f887bbe0bf6fcf0c1c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-06-03 15:42:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1362e20698546c555c98306f03ef629e1de8ad3e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 17:01:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dce8f766ad6a4bd6a64be3373b26a03b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2df96a1951fd46e9bb9700503a7e909b : SUCCESS in 2m 36s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/3598fc8d92a84bf1bb7d08f611550e3f : SUCCESS in 48m 24s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/3dae17f777a440bca41a6adfa78598a4 : SUCCESS in 44m 50s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/913007af856340a094b132fd1c4b5826 : SUCCESS in 41m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/938619d8f9a34bddab621672738075f8 : SUCCESS in 1h 03m 55s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/2493bf37e42346488443ab7796514dce : SUCCESS in 1h 03m 56s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/34f949bea0e547c2a40be3dd1d670eb0 : SUCCESS in 1h 07m 05s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b882e9fb78da4f86bdb4fc422cba56f3 : SUCCESS in 1h 00m 39s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ebe300912fd94f399b02aabf0e54be40 : SUCCESS in 48m 48s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ab4ae0ed4cd349c89b91f8aa408db8b5 : SUCCESS in 54m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0604d8f8e754f52b7d456868086e83b : SUCCESS in 7m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b28a2cfc3a9547a78dac21423a4172ce : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8336e1431fac47488b995d720f803b5d : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6b0d36e1821403f85a2997684a7091d : SUCCESS in 5m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d580aa3e6b3e4698ad63f2ead83f7b54 : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7448a04dc7474f32ad656247b5a37d6a : SUCCESS in 9m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/fd896bef127f4f6f965345ae5f90ae2b : SUCCESS in 5m 17s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ca306e1a62954b97b587f784a53d81fe : SUCCESS in 42m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ddeb4758200032276368e447cd55c6f9df4ec1ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 17:02:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f031e9743e920ffe1d16d9c0d5404d251883a791","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 18:35:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3d7c5b53be54f508eff5ea35e904980\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1558a9d757e6445797142144c8fd51f1 : SUCCESS in 2m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/993234805f46446997536b0a7ebe9d61 : SUCCESS in 46m 31s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/0b32c941641f4f7b9101f297058fd325 : SUCCESS in 49m 27s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8090a8b0ed8c4df2aca89e9ce1e6c493 : SUCCESS in 47m 36s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2019d68605574e6a995227e5411f3e5e : SUCCESS in 1h 05m 31s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/86441a7d68ec4e3f90e545875c3b3a58 : SUCCESS in 41m 43s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/3a56496114c040fe86ac386724430376 : SUCCESS in 1h 01m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d14650b535d241349590fba90bc474d9 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/99cb2455715041d58ea38aa6a985fdb2 : SUCCESS in 5m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bcfc9f5e127d428ab1290b8c9800324b : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3b6da8d343974c98bdac1ab36f0ed0fe : SUCCESS in 8m 16s","accounts_in_message":[],"_revision_number":1},{"id":"099046b53861d5cdfba31efa03902b86f5a1b905","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 18:35:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6821867eac5c6b60a7aebb78f8f3600c8f86e350","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-03 18:36:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6df18ac473be4f6a9fbe85830cb3a467\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1efe594f0864425d9f3f04b7479bd0d5 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d8a980fa0e82bde95dfa170e5d816b4671d74d56","revisions":{"d8a980fa0e82bde95dfa170e5d816b4671d74d56":{"kind":"REWORK","_number":1,"created":"2025-05-30 12:19:06.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/98/951398/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/98/951398/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/951398/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/951398/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/98/951398/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/98/951398/1"}}},"commit":{"parents":[{"commit":"6ebeb372617a236979cd583a656af82d8a136051","subject":"Remove installation guide for openSUSE/SLES","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6ebeb372617a236979cd583a656af82d8a136051"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 12:09:19.000000000","tz":120},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-05-30 12:09:19.000000000","tz":120},"subject":"Remove python3.9 support","message":"Remove python3.9 support\n\nBased on [0] python 3.9 will go EOL before 2025.2 release.\n\n[0] https://review.opendev.org/c/openstack/governance/+/941141\n\nChange-Id: I1d1e3e0806eaff5b55ba01e25a015f5cac3e8131\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8a980fa0e82bde95dfa170e5d816b4671d74d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d8a980fa0e82bde95dfa170e5d816b4671d74d56"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
