)]}'
{"id":"openstack%2Fdesignate~951781","triplet_id":"openstack%2Fdesignate~master~I2191731223993a13e2d63db7441d4f4704161742","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"13861":{"account":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"last_update":"2026-03-16 16:46:20.000000000","reason":"Change was abandoned"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-03-16 16:46:20.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I2191731223993a13e2d63db7441d4f4704161742","subject":"DNM: Testing parent grenade job for neutron","status":"ABANDONED","created":"2025-06-04 14:57:00.000000000","updated":"2026-03-16 16:46:20.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1a218aa155e215914fcf1aef87d3b142c5974e9e","_number":951781,"virtual_id_number":951781,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-06-05 17:02:37.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-04 16:48:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"db92c8bd1d985fb9db34c775f7f41d03ac71ddbf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-06-04 14:57:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"69e7aea9f4826ef6318e4b0ef4be47183a6f65c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-04 16:48:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/24c556e82afb45469f5d16ee990599b0\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/5f2ececbf6a142e59467f09a0f576952 : SUCCESS in 44m 29s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/1c42e9876d194a7785a2ad558e8ad5bd : SUCCESS in 45m 33s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/12253b0016554a17951f625eb6d3b5dc : SUCCESS in 43m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c40bc46d6fbd4c749e3070491c172787 : FAILURE in 37m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/08ca651eed0548419919ddd0106031e9 : FAILURE in 42m 25s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6d5cafc396e04538a6acb777f958ef4e : FAILURE in 39m 06s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/00eaffa998654c70ba86b5ff805e97e0 : FAILURE in 41m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0eb66804091c4ed9b4ed68e0b9ae5250 : SUCCESS in 40m 26s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ca4d1db23d2d42beaeb17480a771b073 : SUCCESS in 48m 13s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eb5bb20792454c6baf2b6215fd4bf7fa : SUCCESS in 6m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752e6abde0c9432d97ae4b779e7063a6 : SUCCESS in 3m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0f6160beffac44af8d01a4d49634650b : SUCCESS in 4m 45s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8e85c77fe99540d9bc07d6510f912a91 : SUCCESS in 4m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0c502d897f334f589f397d780bd051a6 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc52b12f12db4a39a8badc4e2271747b : SUCCESS in 8m 16s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/def18e9b9ade4879983672ef6110c708 : SUCCESS in 34m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0176fef070d740c3b0a13d68a2844547 : SUCCESS in 5m 32s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/e86a79282fbd41c18a1fea64caca1ddf : SUCCESS in 39m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3c864b260d721e5ab4d4d7ab2229c593a5c1c339","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-06-05 15:55:45.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"773ad01d03e51d988a2fd4585dc76b4b6a0543ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-05 17:02:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f346978b7524d04b753a43b30254e53\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/16f09243cce842d39dddf883465d00ad : SUCCESS in 50m 07s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/454d7fe503f345799f867b4f93ac37e5 : SUCCESS in 41m 46s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/8dd589a78ac8413d9d28acf0a0f4381c : SUCCESS in 42m 09s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/04d63d3e1a30487e9618e184f5d0b6b3 : SUCCESS in 55m 24s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/80c5545bd6e44500ba7b1f68aa5dc37c : SUCCESS in 57m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8684b955607a420887ab2114f7978697 : SUCCESS in 1h 01m 13s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7a30d597abf842e58d3ce1e6d85e4b8e : SUCCESS in 54m 41s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/fedea75fe4c34b718948f413cfd44242 : SUCCESS in 39m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9f31c70574904357adca0081ca000d9e : FAILURE in 8m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09762a7ccf9243d3b05f3f9f71493d1d : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/673a3c33073b4aa2a7ad29488c99ec29 : SUCCESS in 2m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6327fb988bd443bc8f14ac632aaf5a84 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4b3c6452bdf7453593cc56db7046ed97 : SUCCESS in 3m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4d557e4c17d249f39a7eff06f4db5f64 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/746f844193c24e5584ddedf6753d7fe7 : SUCCESS in 10m 04s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/e3ecd72b70ce4f3f953b367e6bf9bec3 : SUCCESS in 22m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9ba76b0317eb4b8e9f628625127b5b21 : SUCCESS in 3m 21s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/af6d1d22fa264666944c84a30cbe2a5f : SUCCESS in 52m 10s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1a218aa155e215914fcf1aef87d3b142c5974e9e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-03-16 16:46:20.000000000","message":"Abandoned\n\nNot relevant anymore","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"f9486c9607e4f656e0630a15375fd7228bae2321","revisions":{"1e5bf09fd9b09c17e02a870ff945d558cdbfe989":{"kind":"REWORK","_number":1,"created":"2025-06-04 14:57:00.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/81/951781/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/951781/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/951781/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/951781/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/951781/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/951781/1"}}},"commit":{"parents":[{"commit":"c1c89d35a15eca75ad70720f253d1e324703d221","subject":"Merge \"Remove python3.9 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1c89d35a15eca75ad70720f253d1e324703d221"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-04 14:56:38.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-04 14:56:59.000000000","tz":0},"subject":"DNM: Testing parent grenade job for neutron","message":"DNM: Testing parent grenade job for neutron\n\nChange-Id: I2191731223993a13e2d63db7441d4f4704161742\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e5bf09fd9b09c17e02a870ff945d558cdbfe989"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1e5bf09fd9b09c17e02a870ff945d558cdbfe989"}]},"branch":"refs/heads/master"},"f9486c9607e4f656e0630a15375fd7228bae2321":{"kind":"REWORK","_number":2,"created":"2025-06-05 15:55:45.000000000","uploader":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"ref":"refs/changes/81/951781/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/81/951781/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/951781/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/951781/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/81/951781/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/81/951781/2"}}},"commit":{"parents":[{"commit":"c1c89d35a15eca75ad70720f253d1e324703d221","subject":"Merge \"Remove python3.9 support\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1c89d35a15eca75ad70720f253d1e324703d221"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-06-04 14:56:38.000000000","tz":0},"committer":{"name":"yatinkarel","email":"ykarel@redhat.com","date":"2025-06-05 15:55:42.000000000","tz":330},"subject":"DNM: Testing parent grenade job for neutron","message":"DNM: Testing parent grenade job for neutron\n\nWith https://review.opendev.org/c/openstack/devstack/+/932203\nq-svc \u003d neutron-api and there is no q-svc service.\n\nChange-Id: I2191731223993a13e2d63db7441d4f4704161742\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9486c9607e4f656e0630a15375fd7228bae2321"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f9486c9607e4f656e0630a15375fd7228bae2321"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
