)]}'
{"id":"openstack%2Fdesignate~952388","triplet_id":"openstack%2Fdesignate~master~I3df892e02a35d2ce70c51a38369637996d955b2a","project":"openstack/designate","branch":"master","topic":"devstack","attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-01-28 15:29:55.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-25 14:59:39.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-09-21 12:18:04.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I3df892e02a35d2ce70c51a38369637996d955b2a","subject":"devstack: Enable proxy header parsing","status":"NEW","created":"2025-06-11 17:22:46.000000000","updated":"2026-01-28 15:29:55.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"0980dc7358104a459c6426604bd540b9dc9c2e41","_number":952388,"virtual_id_number":952388,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-11-18 17:05:27.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":1,"date":"2026-01-25 13:34:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":2,"date":"2025-09-23 15:02:57.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-12-12 21:52:43.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2025-09-21 12:18:04.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":-1,"date":"2026-01-26 09:52:57.000000000","permitted_voting_range":{"min":-1,"max":0},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":1,"date":"2025-09-23 15:02:57.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 18:40:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-18 14:50:04.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-09-23 15:02:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2025-11-26 02:36:43.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2025-12-12 21:52:43.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2026-01-25 13:34:47.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"5634ff51ceff94979046ba700c276963e0909916","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-11 17:22:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34d127c0a677c320bbe7baa263ea424ee546cd47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 18:40:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/427a6a72e463484690fa420e0839f25f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/01f495ad8aa440e7b6c673ec9981dba6 : SUCCESS in 44m 12s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/49fee7cac6ec4acf95ac3c3ef47377da : SUCCESS in 40m 23s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/5d7f3e5e4265499ea1527c976165f011 : SUCCESS in 37m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/c0631c106e4f4ba1b4f2b1b286affa2e : SUCCESS in 1h 10m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6e68321e7c184a258628d0f26e6a57fe : SUCCESS in 1h 10m 46s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/374b4b1e6a874110a7e6db512a10c01f : SUCCESS in 58m 25s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/5620057e12eb41738ab78ffcc16146f6 : SUCCESS in 1h 10m 50s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9780a6fa69d44bdb89867ab9e32f09b3 : FAILURE in 24m 24s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/97eafbf75b674009b3c0822aa969ec79 : FAILURE in 30m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc11004e3f434568a4d35961228e5661 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69d4defba3c4658a19ceea42a4950bf : SUCCESS in 2m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/06a6c175d4554c4b94c73e75326a6666 : SUCCESS in 4m 20s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/465056d030004d1297803aeeba7d545b : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/713a39c32c8b4ddf81ad00338d47cced : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62a3d17ba10b40d980db2f7d39dcbfa0 : SUCCESS in 7m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de9c8fce0c7242e085254705c3c6660a : SUCCESS in 24m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d266eb27d52a4883ab509bdbef30dd03 : SUCCESS in 4m 59s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/894ca6e901104f1db3ffbf3ea4457069 : FAILURE in 44m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d44b44063e26a0fca3cd744effd8d306c14243ca","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-16 11:56:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fecac596f2d6d861a19a47263813f240a6f6b4d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-17 12:29:18.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"34830ed1164d9cce1da83d3d6fdd9eeca6ece331","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 13:36:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f7be6d90b78541e7a73e56f0ea6220e2\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/c3072c9601044f40a5c4df07a2fe40d8 : SUCCESS in 49m 37s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e5affca42f9f4134910a05323530e75a : SUCCESS in 39m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/c449aa4451ec45349bddb8cfdd39d245 : SUCCESS in 46m 51s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/bd19571a62904ad7be1560d5478ab734 : SUCCESS in 1h 00m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f66c60ddab0e46c5bf21b913d5678778 : SUCCESS in 58m 59s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/0768ec61903f467b9c409cfea23dddb0 : SUCCESS in 1h 03m 55s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/02aab082a4e54b1c9716d8c5f6f6e7fd : SUCCESS in 59m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/887fa2ed3b8b45489089f08650824687 : SUCCESS in 44m 09s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b1a965dcb458498a932c3e17ca799029 : SUCCESS in 42m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8823c79fb51048688479de96ec65ace2 : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/71f2828e61a245bcaf0c490d1dd68325 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8cb38ed07a2147c082078bc3ea1c7b83 : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4e84f739b40f4591ab77118aeab45955 : SUCCESS in 5m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5adae60be7004b4b8949ec3d6212ae39 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79c925f17d4c45fa8548690f0e108683 : SUCCESS in 8m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/34c56c4f74e34e83bd4b59e9735e5c8e : SUCCESS in 31m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/393f0b8c4b0a4787aaf81cdb0fb5729f : SUCCESS in 4m 44s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/a512a2d9c66643d383379b163f13c5c0 : SUCCESS in 48m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6710570d0337526e94aa262970b0ff271d60f34b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-18 14:50:04.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"18403c509d0bf754ea662d40ae49a5ccb1b03266","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-20 18:25:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"cf38408740e6cf805b8cf857930c7eff32a52978","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 19:44:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bb1d497403f4d9db717fcb703aab0a1\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4624404637fd4ca8b86655c202574c26 : SUCCESS in 1h 09m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0ad3505c518149b3848555eb04046091 : SUCCESS in 1h 07m 30s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4f08f8f1ab3a41d3a5338cec8efeecc8 : SUCCESS in 1h 08m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/697c01f28a8a43eabef807c3144a9380 : SUCCESS in 56m 21s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/972376be905a42fc9f9ca86317609097 : SUCCESS in 6m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa31acfecf63417895fe3d6dc62736a5 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07db5460adcf406cbbbb91dae6245c99 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2e74c46d6597436cb8eb20011fb28c7c : SUCCESS in 4m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fbfbbcf05c9d4530adac67e4044a9dd0 : SUCCESS in 8m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/69c31949029a42c6a6045c7a7dd6d7d5 : SUCCESS in 7m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d7f5c58d5dfb4712ae9db7b2ed734dbf : SUCCESS in 36m 54s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/cc85ca0427ff41118467ddc2480e20ab : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":3},{"id":"6abf5360e3520511246ba650084d5fdd9de8a579","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-09-21 12:18:04.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"cf46004bf6233d33fbb3d8eeff41b28c4b26aab6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-09-23 15:02:57.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"678da150b048fd59254317dab5a8c634e8f8f9a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 15:03:26.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"1feab3d74e4daaeda917d1e9e40ffcc759ef2273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-23 15:34:18.000000000","message":"Patch Set 3: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/acc1d65094d94029bc2007780b595a75\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/cea82b19e3ec4635986e3e52c006798c : CANCELED\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a38e188dedfc408bb1e0d1a7d02aae17 : CANCELED\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/438240cb729e4725953fb2fa71ef2299 : CANCELED\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/28e4acf7cb564b418c5e2b581f6bd108 : SUCCESS in 28m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f85464640a964f06976dd6c394b2ad2c : CANCELED\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f9b932a4cd984e6c9b6b797aa67b359f : CANCELED\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/876b243c39924ee8a3f9dc721ac67798 : CANCELED\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a9b27e64400b4b248ad5d37ebf5d32b8 : CANCELED\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/53f47a09f9c948e4ac30ca14f5275c0b : FAILURE in 28m 33s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c57cf5bea7f349849342d7dabf740813 : CANCELED\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d1aea7aa9a448f6a870423603363215 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/38c93dd8b3f442b98183b8a5c6abb3f5 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/840243c738d149599fd9043da4b18bdc : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8925c3fc517d47e19d41358b19fe8e3f : SUCCESS in 9m 12s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/9f3979271711451996248ad36352a809 : SUCCESS in 29m 59s","accounts_in_message":[],"_revision_number":3},{"id":"59c7b15a539a842b1a8951544c62688bf4d6b642","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-18 15:53:04.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d7f9be460d0085bd54ca9e3791e50e05b5a230dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 17:05:27.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be736cc6ac4d423ca4514324df2dcc1f\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/0fb928a73acd42ceb9084d33f42ff338 : SUCCESS in 52m 19s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1ffb94e620ef48049bea6295d07ad62a : SUCCESS in 42m 38s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/6b92ed030f3f43318fcc5cb55aeff6a8 : SUCCESS in 51m 37s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/e50c5a61125a49658ba775c48c00d00e : SUCCESS in 30m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/28107f53afd744f1a1abc5b94e3457ef : SUCCESS in 55m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fb91a70164fb4b649135c7419eff0386 : SUCCESS in 57m 34s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/45973425681f47f7a38519237fc934f1 : SUCCESS in 1h 07m 23s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/4b4e9ab8df6f41f4acc0e6e58fdf920f : SUCCESS in 1h 10m 00s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/8cdea967674a484aa4644985bac1f0c1 : SUCCESS in 52m 13s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/98acb65ac1544822a7dbcd6eb382ea09 : FAILURE in 16m 50s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/71b95b663b6d4a0e9354f7150d373661 : FAILURE in 32m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/97135e67d3544bb5b04e473f4728c5f5 : SUCCESS in 4m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3cb346d1021f49f287fa97c7bcdaf1db : SUCCESS in 3m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbe3dbc2a947466581e558f733e92422 : SUCCESS in 3m 36s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1eb948d816bf4bdcb877ba8985614ed8 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d651f48d1684edda8f599f7c2cc664d : SUCCESS in 7m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dcac3eb7c4884370b37de12c82abb620 : SUCCESS in 30m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/06fe97a715fa42e78c81cff513c0eec9 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":3},{"id":"3992c327eed0dd62117718027ddd646874f1a1f0","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-25 11:50:11.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5cbc8024b565a102935f219ae355495207cbb1c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 13:16:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dd7e3401c4b8434a92a1dbd031f5cac7\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/9a281386cb9d4d399a4f1f3943441053 : SUCCESS in 45m 46s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/84347a1377994c49b8f4b661efa04ff2 : SUCCESS in 47m 18s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/9899dca868b54517a322aa38cf1ab396 : SUCCESS in 42m 25s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/06fecfc76ea8449c935bc9d57cfe9234 : SUCCESS in 27m 56s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f99fe8eb18a84920b7ea10eb3b9df917 : SUCCESS in 1h 12m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1efd1cc90fa448bc8016da0a9bda6c56 : SUCCESS in 1h 03m 01s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b76db96b8d4f49c49d92a1e296169f37 : SUCCESS in 1h 14m 09s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c7e33e00084d490ea3c363e29a247acc : SUCCESS in 1h 07m 38s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4aacc3620c4e488d85999b3249d6405c : FAILURE in 50m 55s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/4e309f27e2a94fa0ba9d2f9c4b3c91b8 : FAILURE in 27m 15s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/b3e7850344ea408f9b5f63a518e20d49 : FAILURE in 28m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86b3695b46e248da90c85ab173223052 : SUCCESS in 4m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c300de336aa49d2931a216e0addcd0e : SUCCESS in 2m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/54ac639f72334304bac6242a3af40c80 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6f64cb1c86864076ab38d77342166de5 : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0cdb9154cfe42e28127f8929ebab71d : SUCCESS in 4m 34s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/559a6f52d7334c659c7615f677f20cd1 : SUCCESS in 32m 33s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5164d6ead19b48ed934f8bc7eb89e790 : SUCCESS in 5m 18s","accounts_in_message":[],"_revision_number":3},{"id":"6a5cbfc138eed572a95a389b31c0c8d928343e53","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-11-26 02:36:43.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a8d13b32e95a2771363cbadf1009eeae6b2a091b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-26 03:57:53.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ef552c71f25a4e79a105549bb4c00daa\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/ac778303535a4ea6a0a5b3151ee36f7f : SUCCESS in 44m 59s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/34b2c72098ce4d5cb742f8d599c23065 : SUCCESS in 46m 58s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/8f763ba99b814d1aad5812026e9c582d : SUCCESS in 46m 36s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/96a636c221ba4d7c83d9472ce136ce69 : SUCCESS in 31m 21s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5ce84b2cb6cd4c698b7f0f279039b2d7 : SUCCESS in 54m 40s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d2c8f2a183154554a88c351c57da08d7 : SUCCESS in 54m 31s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5389b02352be49709aca8ad1a9ee5a84 : SUCCESS in 1h 13m 55s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/38fb04ccde894d46b56f757cb266b63e : SUCCESS in 1h 07m 02s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/33a7907763614f20acf91bd0531a524f : SUCCESS in 45m 16s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/cc4972757361450887b7fbf3e703b7c7 : FAILURE in 35m 57s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/2d0287f4861a4abd9f92aa25b8ba68e6 : FAILURE in 32m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/38d0739b85ff415f9b25038aa8a432a4 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c64ccb74e72346abbf8da17b1f611ae3 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29611801789b4d9e9884bbec972f2423 : SUCCESS in 5m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c2e04d4fc544ba7b5311b8b9180d3ac : SUCCESS in 6m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c74c1076eb4e5aa4e5839f931af8ee : SUCCESS in 9m 45s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/35d1cf6f9a694b1e8234b4a723fc4f16 : SUCCESS in 34m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1e8692bd609c4d5e9dbff7eab7fdba02 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"8f3bd4b62bc96ca4c1d72ba401201795e73563d3","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 21:52:43.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"6e530a84676f3b0702a1255e3be03f17df32c30d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-12 23:09:24.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a25547bc2de54daa8e995ef9dbd93552\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/9e2afbbab5034a7fb21a6b05bef61905 : SUCCESS in 41m 58s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/85fba2e4b4e64ee0a2ab6b4f5816470a : SUCCESS in 47m 23s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/acf27e85712b455089778950d8daf138 : SUCCESS in 48m 38s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/3807cec9ea3344eca07bf011efad9138 : SUCCESS in 31m 35s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fbca7714fdd54d9c8f767eceefe2e2bc : SUCCESS in 1h 06m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/c3c47465da9a4c3a8cc7550cdbd3782a : SUCCESS in 1h 11m 28s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6094ec3afed749d29204bf5331bfa4d5 : SUCCESS in 50m 00s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6bb933a49c0d434a9eb7b04ca80177cd : SUCCESS in 54m 04s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/6032729af4b242a28c9aa2fc49073260 : SUCCESS in 49m 04s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/c4562d75121e40bcbbf8952b279d4553 : FAILURE in 15m 41s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/82234f8b1d884d39a0cbab17182a6a2e : FAILURE in 33m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e1330fda6b046358dd703765e488e2e : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a9cec142118449b3ae6c5beb5bff6acc : SUCCESS in 3m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/27b4e2dadb0f4937a88bd781b33b589a : SUCCESS in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/396227eb132c439080e86183df252c86 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d53db99eed144369991127f27f38e70d : SUCCESS in 8m 47s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7e067a4afe2e47d595ca7146f9b815a0 : SUCCESS in 32m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dca8923557f7446eb9eea2f687694ed0 : SUCCESS in 4m 54s","accounts_in_message":[],"_revision_number":3},{"id":"83afcc8de7b10f815bc7dea9220a0ec46d720331","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-25 12:15:38.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a899cc895dd052d9f0d99d63adcda37c4555abaa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 13:33:02.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/98daf14ac5fc4966922e7731d2c88bad\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/264e9e51e70a4042b464ebf11a3d8b80 : SUCCESS in 50m 26s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d164266f621641328a1552498a80510a : SUCCESS in 44m 18s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/960d80bd13504b4d92dc820ac24831a7 : SUCCESS in 47m 51s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/0d8081050b2b4e4f8d19d859ac4ad659 : SUCCESS in 31m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/138ec757f400468e9e27be75d8185982 : SUCCESS in 1h 07m 41s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/ab3782f99c394916947b31e823b55d78 : SUCCESS in 1h 08m 01s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c372cb50403f4222bca06c6e617eeff1 : SUCCESS in 52m 36s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/20729d96850e487ca5e631b41ae23384 : SUCCESS in 56m 27s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/cbc7372f093948b5aaab95c37165b298 : FAILURE in 48m 28s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/a63cd6b8c34641bcbfe3ea1f59dcca02 : FAILURE in 15m 46s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/908bff1aea784fd098854437dd52cba0 : FAILURE in 33m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b2b224807f5f48488703bd30d1942fd6 : SUCCESS in 5m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b24b5e9efbf041928ba86414760fb03a : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f642b474453436694b0edb292ef84e5 : SUCCESS in 4m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7a3ed5f2789b49eca8b83a9706383306 : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dccc009d4d3c446eaaeb6d322e0f2e12 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3d906fb019e840588043525fb24f4bc7 : SUCCESS in 33m 06s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7e1892cb6b9946adaddf93d4048bdc30 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"095a214ce37ef4a590d7c229f6fe617b75dddd0c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-25 13:34:47.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f8967291be1b9bc3f24beec2d7d69a237b88f1e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-25 14:59:39.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a03129ff46e4e9a8137cfc9fc7e3a08\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/26379d6729744d4f9d7888d4a6f8d108 : SUCCESS in 43m 35s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/fb255fd65f3a46828c42f12431840e4b : SUCCESS in 34m 53s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/25bccb1eda4f4d8c98c5b9726f8c0295 : SUCCESS in 41m 04s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/06d0d3f6774c44e485d304987065b26a : SUCCESS in 31m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/388e928309bf4c379fce3d293ae0de55 : SUCCESS in 54m 57s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/cf8a49fb008741d4a60a92dccfc039db : SUCCESS in 1h 07m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/06594803b09949c68df5a31bb1be6129 : SUCCESS in 52m 51s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ffcc9217d81c40ffa9892b144e4dcf7b : SUCCESS in 1h 11m 23s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/cf3057f8aee8463aa1ceb0449a8432fa : FAILURE in 36m 39s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/0fcdaa59a60e40f0aeeecd1e5840801c : FAILURE in 31m 40s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/a916aa2a585c4f75b0bcda888e8f325d : FAILURE in 34m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06fb061fb676491aa55cdb15f3eb963a : SUCCESS in 5m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b94e76ae3b164b05bfc9b58bcba72f85 : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b89eaf74cc0f4053a458a439ad65fc7c : SUCCESS in 4m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa874e3aed9f4010a8de1721d46b6b97 : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d95bf4b95ef4f439ad3c79d5581b0eb : SUCCESS in 8m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de92b5d2019f441b9e5b6a8258f2251c : SUCCESS in 32m 57s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/363d7eaf11c14fd084ba8b6eed099adf : SUCCESS in 5m 03s","accounts_in_message":[],"_revision_number":3},{"id":"513eaa5776d24bd0455b9e1e726785c90b310854","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-26 09:52:57.000000000","message":"Patch Set 3: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0980dc7358104a459c6426604bd540b9dc9c2e41","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-01-28 15:29:55.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"075c2b3d315faff398f49cd2562c8267238cc892","revisions":{"161b8418f7b54c2e91f70b5136d36ae9fbf07263":{"kind":"REWORK","_number":1,"created":"2025-06-11 17:22:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/952388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/952388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/952388/1"}}},"commit":{"parents":[{"commit":"698f7b605926ab26eb1577f8728f77c71e2fbda1","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/698f7b605926ab26eb1577f8728f77c71e2fbda1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:07:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:22:14.000000000","tz":60},"subject":"devstack: Enable proxy header parsing","message":"devstack: Enable proxy header parsing\n\nWe are currently using the http transport [1] to work around a bug in\nuWSGI [2]. Enable proxy header parsing to ensure the Host header is\nrewritten correctly. If this isn\u0027t done, the version document (and\nlikely other APIs) returns the IP address and port of the backing\nservice rather than that of the proxy server.\n\n  $ curl -g -i -X GET http://10.0.111.187/dns -H \"X-Auth-Token: {token}\"\n  {\n     \"versions\": [\n       {\n         \"id\": \"v2\",\n         \"links\": [\n           {\n             \"href\": \"http://127.0.0.1:60053/v2\",\n             \"rel\": \"self\"\n           },\n           {\n             \"href\": \"https://docs.openstack.org/api-ref/dns\",\n             \"rel\": \"help\"\n           }\n         ],\n         \"status\": \"SUPPORTED\",\n         \"updated\": \"2022-06-29T00:00:00Z\"\n       },\n       ...\n     ]\n  }\n\n[1] https://uwsgi-docs.readthedocs.io/en/latest/HTTP.html\n[2] https://github.com/unbit/uwsgi/issues/2368\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I3df892e02a35d2ce70c51a38369637996d955b2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/161b8418f7b54c2e91f70b5136d36ae9fbf07263"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/161b8418f7b54c2e91f70b5136d36ae9fbf07263"}]},"branch":"refs/heads/master"},"3729e4ad1c934a1143df0e389398c97f45733868":{"kind":"REWORK","_number":2,"created":"2025-06-17 12:29:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/88/952388/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/952388/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/952388/2"}}},"commit":{"parents":[{"commit":"698f7b605926ab26eb1577f8728f77c71e2fbda1","subject":"Merge \"Update python classifier in setup.cfg\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/698f7b605926ab26eb1577f8728f77c71e2fbda1"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:07:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-17 12:28:34.000000000","tz":60},"subject":"devstack: Enable proxy header parsing","message":"devstack: Enable proxy header parsing\n\nWe are currently using the http transport [1] to work around a bug in\nuWSGI [2]. Enable proxy header parsing to ensure the Host header is\nrewritten correctly. If this isn\u0027t done, the version document (and\nlikely other APIs) returns the IP address and port of the backing\nservice rather than that of the proxy server.\n\n  $ curl -g -i -X GET http://10.0.111.187/dns -H \"X-Auth-Token: {token}\"\n  {\n     \"versions\": [\n       {\n         \"id\": \"v2\",\n         \"links\": [\n           {\n             \"href\": \"http://127.0.0.1:60053/v2\",\n             \"rel\": \"self\"\n           },\n           {\n             \"href\": \"https://docs.openstack.org/api-ref/dns\",\n             \"rel\": \"help\"\n           }\n         ],\n         \"status\": \"SUPPORTED\",\n         \"updated\": \"2022-06-29T00:00:00Z\"\n       },\n       ...\n     ]\n  }\n\n[1] https://uwsgi-docs.readthedocs.io/en/latest/HTTP.html\n[2] https://github.com/unbit/uwsgi/issues/2368\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I3df892e02a35d2ce70c51a38369637996d955b2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3729e4ad1c934a1143df0e389398c97f45733868"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3729e4ad1c934a1143df0e389398c97f45733868"}]},"branch":"refs/heads/master"},"075c2b3d315faff398f49cd2562c8267238cc892":{"kind":"REWORK","_number":3,"created":"2025-09-20 18:25:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/88/952388/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/88/952388/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/88/952388/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/88/952388/3"}}},"commit":{"parents":[{"commit":"035453ffd1ba0f574a1d7f23b458222c0fd8fc68","subject":"Restructure Designate CI jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/035453ffd1ba0f574a1d7f23b458222c0fd8fc68"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:07:52.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-20 18:24:35.000000000","tz":540},"subject":"devstack: Enable proxy header parsing","message":"devstack: Enable proxy header parsing\n\nWe are currently using the http transport [1] to work around a bug in\nuWSGI [2]. Enable proxy header parsing to ensure the Host header is\nrewritten correctly. If this isn\u0027t done, the version document (and\nlikely other APIs) returns the IP address and port of the backing\nservice rather than that of the proxy server.\n\n  $ curl -g -i -X GET http://10.0.111.187/dns -H \"X-Auth-Token: {token}\"\n  {\n     \"versions\": [\n       {\n         \"id\": \"v2\",\n         \"links\": [\n           {\n             \"href\": \"http://127.0.0.1:60053/v2\",\n             \"rel\": \"self\"\n           },\n           {\n             \"href\": \"https://docs.openstack.org/api-ref/dns\",\n             \"rel\": \"help\"\n           }\n         ],\n         \"status\": \"SUPPORTED\",\n         \"updated\": \"2022-06-29T00:00:00Z\"\n       },\n       ...\n     ]\n  }\n\n[1] https://uwsgi-docs.readthedocs.io/en/latest/HTTP.html\n[2] https://github.com/unbit/uwsgi/issues/2368\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: I3df892e02a35d2ce70c51a38369637996d955b2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/075c2b3d315faff398f49cd2562c8267238cc892"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/075c2b3d315faff398f49cd2562c8267238cc892"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"failing_atoms":[],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
