)]}'
{"id":"openstack%2Fdesignate~952389","triplet_id":"openstack%2Fdesignate~master~Ice4c3bf3659460516158bb8be659fbacfc15bc43","project":"openstack/designate","branch":"master","topic":"devstack","attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-12-12 21:58:37.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-06-18 14:49:40.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-01-25 13:34:30.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-09-21 12:17:48.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ice4c3bf3659460516158bb8be659fbacfc15bc43","subject":"devstack: Remove support for deployment under mod_wsgi","status":"NEW","created":"2025-06-11 17:22:46.000000000","updated":"2026-03-17 15:12:24.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e4274874a870726bcc168be89dcd9808b21572c5","_number":952389,"virtual_id_number":952389,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-03-17 15:12:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"recommended":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2026-03-17 13:55:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-03-17 13:55:47.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":2,"date":"2026-03-17 13:55:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-06-11 18:34:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-06-18 14:49:40.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-12-12 21:58:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"},{"updated":"2025-12-12 21:58:37.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2026-01-25 13:34:30.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"5645b8e9418988ddbf43ff35b10d8b9b0285be84","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-11 17:22:46.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0ea45878e8d7bc9c2f2f3884bb9da90a8f177194","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-11 18:34:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5480c9762c248c0ae075dd654de2544\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/f8fb75934aa641edb91a41ecb061782c : SUCCESS in 51m 16s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/414d24145b924973b6948e4bd3863a37 : SUCCESS in 44m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9bc49879ef874eecb1205a5b74d4cfb3 : SUCCESS in 43m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/91cad542217647b19cb31aaa2124e53f : SUCCESS in 58m 27s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/97e42205e35d486db9ffdc82b5b1f363 : SUCCESS in 57m 21s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d8b4776b2dbe4abd873ec23d821a0391 : SUCCESS in 1h 09m 46s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0261ddc2ebb44b259fb2e4fd17395229 : SUCCESS in 1h 05m 53s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f2d99704fc02413abce01062a7fcab8d : FAILURE in 29m 44s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/e53f694a1a364d01a69c8a3f32132f6e : FAILURE in 30m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96bb195c2d94417e944c3b2a09b55d08 : SUCCESS in 5m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/11a099272b4b48028dbb86d06168626d : SUCCESS in 2m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/60538ee6e38141c5bdb65e9f17cc5c02 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f156c71479914ad79f2f44feccb67802 : SUCCESS in 2m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/63aa141fed1a413fbdaa1f5af6e44f23 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/63c184417daf44aeb9a804b7702d554d : SUCCESS in 5m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5f459bc832bf4c3f868387bc1d28694a : SUCCESS in 18m 02s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/daf693ecec4e47fea1e3c30a933f5c6d : SUCCESS in 5m 49s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/0f63d8585318424b9026017e16892ebb : SUCCESS in 51m 49s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9c3dcdcc56568079bb76d78ead4fb8bb41c09faf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-17 12:29:18.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"e76edc71ad8bbc483d19a10ad0c3f08de7b019f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 13:37:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7858ffbb5d114bbfbce4eedcca87de31\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/25face3a64e1495c92e91df6d7593d4c : SUCCESS in 39m 07s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/3360557773a0440c8ce3374e6f99bd9b : FAILURE in 34m 57s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/60f9223078c74e2b9b450cbcc0000a1b : SUCCESS in 41m 25s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/49d901b76dba4e65b4da48321266c700 : SUCCESS in 57m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b0cb843f355b4655b2f8d6872cc47ceb : SUCCESS in 1h 03m 43s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4f082303b5964763aa8c5c1e9e31a851 : SUCCESS in 1h 00m 48s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/6f247a3430ec46ed9e0ec8ccc3e51013 : SUCCESS in 50m 46s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9ca9fc2eaf1d4f44bfa43572d2548c59 : SUCCESS in 36m 03s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/6f576f6a190045a1bbe89a1ca96102b9 : SUCCESS in 41m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9b88461ebda4363b2445d145bc73c60 : SUCCESS in 6m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00b060bb14c442449a489166e512a9a2 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2c2a76450634262b801a3e12e14bbde : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fac3454cd38442619d3c86156e979ffa : SUCCESS in 4m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cfcf7be9d05d406caae0024bdd2088a6 : SUCCESS in 6m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d15abfeac7b44e30a341e38e1834f2c7 : SUCCESS in 7m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/68c7ebec9794452297bf486d4ab87437 : SUCCESS in 32m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/6abd949639b64ca78fb0e51b7e735728 : SUCCESS in 2m 48s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/baf35936cdac49d5a2c84cbec3ca6a2b : FAILURE in 51m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3350d98281e87df355f23358976ecc0f34870440","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-06-17 15:12:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"91e6aa112f1e0e8f115e70ff997e18b8c259963b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-17 16:32:38.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ff6f94753c142a6a6e17c96250f7b9a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/ad3c730dc9ca4a7a8f05701c21de8344 : SUCCESS in 46m 07s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e98139ee39b74ffab4311e855334c98f : SUCCESS in 41m 34s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/9b4f74b48cd84c31a04afaa9b2b7f80f : SUCCESS in 44m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/26ef76fd254346089912b6f4085be732 : SUCCESS in 1h 12m 23s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/1af9986b32fa436cae7194689084fb56 : SUCCESS in 1h 04m 37s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d869e4a4b4e2410e9e9f435d6713ac6e : SUCCESS in 1h 10m 54s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/60688fc2bd6d4696a20fdcfaeb1db141 : SUCCESS in 1h 12m 55s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/9cad78f1f8f444cfa840a2e2c8a2cd12 : SUCCESS in 43m 23s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/dffd962da51b47db9a3a4abed9b68c6b : SUCCESS in 48m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7f8bcfccccbc420dafbf226553763b5e : SUCCESS in 6m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2b1a0ccd57741d5ac35d119e9dcae65 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d25b97ee89e4a0e9978408751a75da0 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f053bf8752a34af18cf9db69a3f3d027 : SUCCESS in 6m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3251195ea6014ce3b7cc13796891bb9d : SUCCESS in 6m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7edb00478cc246c5bbed9b1cacbf4bd9 : SUCCESS in 8m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0869004eb92e490fb261fe1e52beb340 : SUCCESS in 32m 46s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/859045a93d334930aa16975edaedd077 : SUCCESS in 4m 47s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/36af333b7e4648f3a5d67a4208498da6 : FAILURE in 41m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"5c27730003746f4f031cc03a227ffc20ea581791","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-06-18 14:49:40.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"1692aa915376fe96140eac2e38de6163c7adf723","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-20 18:25:04.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"73463d21a9890bce3163b4333215a2fefe1b7b76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-20 19:44:00.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c5c26e79c4f4880bfb3c47e66dca441\n\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/881286f280d74dae82df226e68597549 : SUCCESS in 1h 12m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0d28228b968c4c9281435aa35f9e6ec2 : SUCCESS in 1h 04m 15s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a81b344db4c84c1890c0a65dcbc35487 : SUCCESS in 50m 57s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ad0c2b31a80d45aea2ebf5571ff06da8 : SUCCESS in 1h 08m 06s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5a24628925a4089b0dcd1a3dba80b35 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c84b4e79cfbd4e838acdecb6807ae1b2 : SUCCESS in 3m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ea9a8e9c1d854c978f25f291990be693 : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e27591f70f044941a35e9bbc9073bc5b : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/8b35f039224f46c8af01813a2d225a18 : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc194c71a8ee426293ecf1ff20cf50ac : SUCCESS in 7m 46s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f39433e3184047b8a2d4fb6cca21e122 : SUCCESS in 30m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2646f0a90657425792e2087f79772087 : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":3},{"id":"3533993063491abfc1df2b30692b0a8430f54401","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-09-21 12:17:48.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"e4f6ab9954d9af07e7dc746e63a6cfff4eed8815","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-12-12 21:58:37.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"7f653eb5c4d0079476d36a3372887ab9cae3f3c2","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-01-25 13:34:30.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"b34a71760ac89bc503da6d96b6c7f6a0a1812168","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-03-17 13:55:47.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"e4274874a870726bcc168be89dcd9808b21572c5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-17 15:12:24.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44108508fb314d57b310e7a05a6f3fa2\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/1cb7bcbc76274e4aaad6e25975f49ca3 : SUCCESS in 48m 18s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4895cb32fed54a2689d4f427d3e94090 : SUCCESS in 40m 21s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/12cecbbff21b4719a326df647ae1fe59 : SUCCESS in 38m 50s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/7608a72c0aee4476afd42ea8032024fe : SUCCESS in 16m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8be31fa40d1b4a3eaa4be1ca85cb1754 : SUCCESS in 1h 00m 32s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/40e44c3307b748da9f734bfaa4430349 : SUCCESS in 54m 54s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/854136390a534447b42ffe4d0134418f : SUCCESS in 1h 09m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7f10d8ec925d4988a1caa2abde31c8a1 : SUCCESS in 1h 08m 08s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/6f30b4eeb9634cde85621ec017e915fb : SUCCESS in 50m 18s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/1f6cd55e327e4b369d9632ed4777101a : SUCCESS in 48m 38s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/a84bd95f178840709a204a2a7c53b58b : SUCCESS in 52m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c778c32def34b179f466d324db6fe63 : SUCCESS in 5m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da5a517f93714570896ffde4bc609b48 : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2bc3fc89f30546f9a82c333ff5c00dc9 : SUCCESS in 5m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e1ad4cca27724873865fcc5e7184e514 : SUCCESS in 2m 57s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/4ea9b092e6d6490695474feef31e726d : SUCCESS in 9m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc682fcc559d45d382e88801a5c654cb : SUCCESS in 8m 25s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/acd4dbab5eff41a0aa6aa23fa1e05a6e : SUCCESS in 32m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4c1180b29ced477d9580e9faf9f82480 : SUCCESS in 4m 09s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"ec321809b74de000a7d618e02a1f5eca58be8254","revisions":{"bfc637a5a68ce1affc1e25b6bfcaeef9c1174b65":{"kind":"REWORK","_number":1,"created":"2025-06-11 17:22:46.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/952389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/89/952389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/89/952389/1"}}},"commit":{"parents":[{"commit":"161b8418f7b54c2e91f70b5136d36ae9fbf07263","subject":"devstack: Enable proxy header parsing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/161b8418f7b54c2e91f70b5136d36ae9fbf07263"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:17:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:22:37.000000000","tz":60},"subject":"devstack: Remove support for deployment under mod_wsgi","message":"devstack: Remove support for deployment under mod_wsgi\n\nThis is no longer supported for other services in DevStack. In\nDesignate\u0027s case, the code paths are not being triggered since no CI\njobs are setting \u0027DESIGNATE_WSGI_MODE\u0027 to a non-default value. Remove\nsupport for the feature wholesale.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ice4c3bf3659460516158bb8be659fbacfc15bc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfc637a5a68ce1affc1e25b6bfcaeef9c1174b65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bfc637a5a68ce1affc1e25b6bfcaeef9c1174b65"}]},"branch":"refs/heads/master"},"730765fbf98b07bf9ef3ec2800817898bdad8dcc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-06-17 12:29:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/89/952389/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/89/952389/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/89/952389/2"}}},"commit":{"parents":[{"commit":"3729e4ad1c934a1143df0e389398c97f45733868","subject":"devstack: Enable proxy header parsing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3729e4ad1c934a1143df0e389398c97f45733868"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:17:40.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-17 12:29:08.000000000","tz":60},"subject":"devstack: Remove support for deployment under mod_wsgi","message":"devstack: Remove support for deployment under mod_wsgi\n\nThis is no longer supported for other services in DevStack. In\nDesignate\u0027s case, the code paths are not being triggered since no CI\njobs are setting \u0027DESIGNATE_WSGI_MODE\u0027 to a non-default value. Remove\nsupport for the feature wholesale.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ice4c3bf3659460516158bb8be659fbacfc15bc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/730765fbf98b07bf9ef3ec2800817898bdad8dcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/730765fbf98b07bf9ef3ec2800817898bdad8dcc"}]},"branch":"refs/heads/master"},"7c2f720fc493a7f15cbb25a4f4b3c0a41a49cb98":{"kind":"REWORK","_number":3,"created":"2025-09-20 18:25:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/952389/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/89/952389/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/89/952389/3"}}},"commit":{"parents":[{"commit":"075c2b3d315faff398f49cd2562c8267238cc892","subject":"devstack: Enable proxy header parsing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/075c2b3d315faff398f49cd2562c8267238cc892"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:17:40.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-09-20 18:24:54.000000000","tz":540},"subject":"devstack: Remove support for deployment under mod_wsgi","message":"devstack: Remove support for deployment under mod_wsgi\n\nThis is no longer supported for other services in DevStack. In\nDesignate\u0027s case, the code paths are not being triggered since no CI\njobs are setting \u0027DESIGNATE_WSGI_MODE\u0027 to a non-default value. Remove\nsupport for the feature wholesale.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ice4c3bf3659460516158bb8be659fbacfc15bc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2f720fc493a7f15cbb25a4f4b3c0a41a49cb98"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7c2f720fc493a7f15cbb25a4f4b3c0a41a49cb98"}]},"branch":"refs/heads/master"},"ec321809b74de000a7d618e02a1f5eca58be8254":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-03-17 13:55:47.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/89/952389/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/89/952389/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/89/952389/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/89/952389/4"}}},"commit":{"parents":[{"commit":"eabf5d15d61624da4ea806cbfe4684f7922e7c6c","subject":"Merge \"[trivial] Fix MySQL migration compatibility with ProxySQL\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/eabf5d15d61624da4ea806cbfe4684f7922e7c6c"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-06-11 17:17:40.000000000","tz":60},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2026-03-17 13:55:47.000000000","tz":0},"subject":"devstack: Remove support for deployment under mod_wsgi","message":"devstack: Remove support for deployment under mod_wsgi\n\nThis is no longer supported for other services in DevStack. In\nDesignate\u0027s case, the code paths are not being triggered since no CI\njobs are setting \u0027DESIGNATE_WSGI_MODE\u0027 to a non-default value. Remove\nsupport for the feature wholesale.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nChange-Id: Ice4c3bf3659460516158bb8be659fbacfc15bc43\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec321809b74de000a7d618e02a1f5eca58be8254"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec321809b74de000a7d618e02a1f5eca58be8254"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
