)]}'
{"id":"openstack%2Fdesignate~955682","triplet_id":"openstack%2Fdesignate~stable%2F2024.2~If90056ae1de4a6a3557fc870497e6233088bfb02","project":"openstack/designate","branch":"stable/2024.2","topic":"remove-wsgi_scripts","hashtags":[],"change_id":"If90056ae1de4a6a3557fc870497e6233088bfb02","subject":"[Stable-only] Add wsgi module to Designate","status":"ABANDONED","created":"2025-07-23 10:43:36.000000000","updated":"2026-04-30 09:16:27.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"f66808c0536797d4ecb11703a0d5d6738201f3be","_number":955682,"virtual_id_number":955682,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-23 11:54:02.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 10:43:36.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2025-07-23 11:54:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 09:16:26.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"c7dd5840e41367840ad4543d930e758417ab5957","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-23 10:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1e6323a6a907f9eff4c7c7f1dba24cecb9770d40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 11:54:02.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bf588b0d1304f96a435f3d491597544\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/19572485164d492fa69244a13debca5b : SUCCESS in 34m 57s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/cd55a7209ef74954af1c052ed9fa2648 : SUCCESS in 54m 17s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/0f019307f2124317a15820a4c74b7659 : SUCCESS in 53m 58s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2b1043d7791540c4ac952f2a363493ec : SUCCESS in 42m 20s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7125c0e0753f4fceb5ff0cbff9c326ca : SUCCESS in 49m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8df4953c90ad42ff9a74d5d5ede3ad9a : SUCCESS in 1h 02m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/80ae78b854e0496a9f388a686dc8d1f9 : SUCCESS in 57m 51s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/21324c68f9d9452ba6ca4cf12b5dfad0 : SUCCESS in 36m 16s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/eefc60e103ba427ebbc35b61ca54f318 : SUCCESS in 39m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96cd500a241444f390a2bb3c12c16113 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56f640b5b58842ff977fd9113e16199b : SUCCESS in 1m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1bc4ef3379194e319e5c92095a19caf9 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2b8d3b1693f34d6bb105f358fbfd48b1 : SUCCESS in 4m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6620bc2e7cc74efcaa451d365eb4ebbe : SUCCESS in 4m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43ecc9d9b12c44f8bcd96e6c8508ec5f : SUCCESS in 6m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/92c0c408b7b8497bae0073defc74cbd8 : SUCCESS in 3m 53s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/8bee82b6055b4fffb674de1260a6f4f1 : SUCCESS in 26m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1f96634234294f4aae1f1fa8d0d8ccc0 : SUCCESS in 2m 53s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/2b68560844474ed38b02b7b776642c6e : SUCCESS in 36m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f66808c0536797d4ecb11703a0d5d6738201f3be","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 09:16:27.000000000","message":"Abandoned\n\nstable/2024.2 branch of openstack/designate transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"07cbf0de74e1f89b3036d757bd5145ce6fabe476","revisions":{"07cbf0de74e1f89b3036d757bd5145ce6fabe476":{"kind":"REWORK","_number":1,"created":"2025-07-23 10:43:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/82/955682/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/82/955682/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/955682/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/955682/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/82/955682/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/82/955682/1"}}},"commit":{"parents":[{"commit":"5bfc35864c28a97ded6104dea05fa9691a7f14d8","subject":"Add an designate-api-wsgi compatibility script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5bfc35864c28a97ded6104dea05fa9691a7f14d8"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:53:18.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-07-23 10:42:51.000000000","tz":60},"subject":"[Stable-only] Add wsgi module to Designate","message":"[Stable-only] Add wsgi module to Designate\n\nThis is modified change If90056ae1de4a6a3557fc870497e6233088bfb02 that\ndoes not drop \u0027wsgi_scripts\u0027 since we still rely on this. We will remove\nthis separately. We also don\u0027t migrate our DevStack plugin since we want\nto test everything is working before doing so. Again, we can do that\nseparately.\n\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\nRelated-Bug: 2109577\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: If90056ae1de4a6a3557fc870497e6233088bfb02\n(cherry picked from commit e6d83dfcf729fff333cfc5b700c62efb8aa4aebd)\n(cherry picked from commit 4f0185901f90630ec980ade00d8001da715d68da)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07cbf0de74e1f89b3036d757bd5145ce6fabe476"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07cbf0de74e1f89b3036d757bd5145ce6fabe476"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
