)]}'
{"id":"openstack%2Fdesignate~955683","triplet_id":"openstack%2Fdesignate~stable%2F2024.2~I0462aec6e6b3bb45ebf5be4118540f758d9ea2bc","project":"openstack/designate","branch":"stable/2024.2","topic":"remove-wsgi_scripts","hashtags":[],"change_id":"I0462aec6e6b3bb45ebf5be4118540f758d9ea2bc","subject":"[Stable-only] Migrate from wsgi_scripts to console_scripts","status":"ABANDONED","created":"2025-07-23 10:43:36.000000000","updated":"2026-04-30 09:16:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ca809b5f21540401687e31029782075739cb059b","_number":955683,"virtual_id_number":955683,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-07-23 11:49:00.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-07-23 11:49:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-30 09:16:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"4724df0c1c0d01ae9769b603c2ce8a1ef8e495c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-07-23 10:43:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8f00da201584c1babf68c5f452389a973f5aafad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-23 11:49:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f4cf5985b01454985b2b2908873bbbe\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c732ab2d540242c0b0a0dca4046c2354 : SUCCESS in 1m 28s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9912fe71e07244d6878629dd11ee2dc3 : SUCCESS in 48m 33s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/10c4fb14d80f4e2fb5cc7be0c24c590f : SUCCESS in 52m 25s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/89a6abb27fb74121a6876e11ba39b106 : SUCCESS in 48m 16s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d544254fa1184869ab9e1de9167578af : SUCCESS in 41m 51s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/f11ecec20fcd4369a99f837cea5ac60b : SUCCESS in 37m 19s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5f443fd8b549458086adfa4745086591 : SUCCESS in 1h 00m 07s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/44a32537b9e24752aaa02d681e9f3455 : SUCCESS in 1h 00m 36s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/edee9a1cf78c482abdd1f74d6e545b8d : SUCCESS in 39m 57s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/94b5bfc1caab4a959ef9b45cc66b79a1 : SUCCESS in 41m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/890f86c14481424fbf24ee379d3fa173 : SUCCESS in 5m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d71ca1e6df4c4629a9ca5d5fb8f945e5 : SUCCESS in 3m 09s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e063b919a5954c41abbba5ed1029c265 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b72eee25dee4275b942267cb70aeba5 : SUCCESS in 4m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e4f4ae1db4fd4635bfd7422e8c37de39 : SUCCESS in 6m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/177662b93b5e470e9a685d32a29e5342 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dcfaf0f20f554601a12bdea383624d5d : SUCCESS in 4m 03s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/56a761a82390496e9b40539bd8f0018c : SUCCESS in 26m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d592f1edcbe54c4d863b6a292f2133bd : SUCCESS in 6m 28s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/b3a6fdeb574048cb96f77190d5d081d5 : SUCCESS in 48m 10s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ca809b5f21540401687e31029782075739cb059b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 09:16:39.000000000","message":"Abandoned\n\nstable/2024.2 branch of openstack/designate transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8eaeedcab25dc0093b9861f204a60e8ca507bcc2","revisions":{"8eaeedcab25dc0093b9861f204a60e8ca507bcc2":{"kind":"REWORK","_number":1,"created":"2025-07-23 10:43:36.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/83/955683/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/83/955683/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/955683/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/955683/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/83/955683/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/83/955683/1"}}},"commit":{"parents":[{"commit":"07cbf0de74e1f89b3036d757bd5145ce6fabe476","subject":"[Stable-only] Add wsgi module to Designate","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/07cbf0de74e1f89b3036d757bd5145ce6fabe476"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-05-21 16:00:15.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2025-07-23 10:43:16.000000000","tz":60},"subject":"[Stable-only] Migrate from wsgi_scripts to console_scripts","message":"[Stable-only] Migrate from wsgi_scripts to console_scripts\n\nThis reverts commit 5f4d6c4c6b68bea0444b516ad000ef1fb6b67673 and makes\nso further modifications to setup.cfg, such that we no long use\nwsgi_scripts (or scripts) and just use console_scripts.\n\nChange-Id: I0462aec6e6b3bb45ebf5be4118540f758d9ea2bc\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n(cherry picked from commit 49082704030f03abeaf27845604368e1293cfb30)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8eaeedcab25dc0093b9861f204a60e8ca507bcc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8eaeedcab25dc0093b9861f204a60e8ca507bcc2"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
