)]}'
{"id":"openstack%2Fdesignate~956696","triplet_id":"openstack%2Fdesignate~master~I67941b5c7525d4b9661b30c4cc03dbd061e86083","project":"openstack/designate","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"16688":{"account":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"last_update":"2026-01-09 19:59:44.000000000","reason":"Change was submitted"},"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-01-09 16:49:12.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-01-09 19:59:44.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-01-08 12:12:11.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"I67941b5c7525d4b9661b30c4cc03dbd061e86083","subject":"Isolate eventlet-dependent services for future migration","status":"MERGED","created":"2025-08-06 15:17:24.000000000","updated":"2026-01-09 20:00:53.000000000","submitted":"2026-01-09 19:59:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":19,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"956696-eventlet-removal","meta_rev_id":"4a66f1c0cf93a85b5e9502f2945ec85491c81970","_number":956696,"virtual_id_number":956696,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-09 19:59:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-01-09 15:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2026-01-09 15:07:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2026-01-09 15:08:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-01-09 15:08:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 16:29:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-07 10:24:55.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-08-25 12:47:42.000000000","updated_by":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"reviewer":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"state":"REVIEWER"},{"updated":"2026-01-06 22:25:27.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ad013cc508930dc060ff9c3fd75c3d66cf3223ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-06 15:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"63967b0f74bedd2ecfb15be0a984bfefbaef4202","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-06 15:21:42.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"779017a4b0c6ed35ccf61142ee0639ec0d5e4135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 16:29:48.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e18b2f5ff8545df8d4d7b2e42aad95a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/857b880c2887479a9bc04510461a4aa5 : SUCCESS in 1m 27s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/84991a45b44c4b09a032fcc5f0b6a486 : SUCCESS in 37m 54s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c61b407ad1734a97992caa2cc5c5f5e7 : SUCCESS in 45m 58s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/77024e666c9c43a494c938baf863cc30 : SUCCESS in 37m 03s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fd885d178efc421e8d0f23d075584229 : SUCCESS in 55m 12s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/dab831d8c27a448dbd3b186bf26991df : SUCCESS in 49m 20s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/3600280a07da47beac40ca3b96a33d76 : SUCCESS in 50m 28s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7c600b89c3854cf3a6534f0682cee5eb : SUCCESS in 1h 01m 54s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0109152538cb4f6f802b7106ddcaae64 : SUCCESS in 41m 47s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/8ee3cc16b9b340fcb57a9543e73cd9ac : SUCCESS in 47m 28s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/308b538b8b6e44e7b843da4cae837d0d : SUCCESS in 6m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/edd44ca667824230a7cf4e14c98978bf : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1b755c9c322f470980f2508784df2bdd : SUCCESS in 6m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/15015210411249609ddc168ce16738cc : SUCCESS in 4m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f80158bb5d3a4551b96d3d7f7c1a906c : SUCCESS in 8m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a9d90bbf0c374cc4be1e0a2678560823 : SUCCESS in 9m 53s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/1c931d9665d24c2c9b1127a30359fa3a : SUCCESS in 30m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a3c9e34382774f469589775c2315797c : SUCCESS in 4m 31s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3b5166748e82496cae4453bd234f4b44 : SUCCESS in 47m 30s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"75048ffefdc34661b025eceb34768c6a94bcac58","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-08-07 10:24:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a11677588eb85be283aa91f642d5da0924db5603","author":{"_account_id":16688,"name":"Rodolfo Alonso","email":"ralonsoh@redhat.com","username":"rodolfo-alonso-hernandez"},"date":"2025-08-25 12:47:42.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a1e48f98fc39e16cfe9faf3a2ab9a1f7fc9bddb5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-30 09:52:32.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d04dcfd323336141964939652b1be9fb466acaee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-30 10:32:16.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent tree, and commit message as Patch Set 3.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n","accounts_in_message":[],"_revision_number":4},{"id":"c0c2230d9eac6491e5b985dc0490e030d7c5c0dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-30 11:48:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad54eb893c554246aaba571d7814806b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d2d62906afc14ba991b6f28f28b8d805 : SUCCESS in 2m 13s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/6eee331443c34619afa4b4f4c1435d05 : SUCCESS in 43m 38s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ba48a6059e8d4f0b92c94d7fef32d8ff : SUCCESS in 47m 15s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/a4f162aed58547048b73402de1eefe9b : SUCCESS in 43m 17s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/62add2f60b614b22b8c7306b5a5ea36d : SUCCESS in 28m 38s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/77282cf43818431baf2e41ac1de802a9 : SUCCESS in 1h 11m 26s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5d6a202071aa495aa65c0de5fd9ba46a : SUCCESS in 53m 02s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/17779bae099449e193e81dff20131da1 : SUCCESS in 1h 11m 18s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/2969a36dba1a44c99b8177df428f11e3 : SUCCESS in 1h 07m 27s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4dd3bacb893245dfa766951d92d8d4d1 : SUCCESS in 46m 29s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/a397f6736a584c329965a39d9413e2db : SUCCESS in 47m 20s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/2e0a35af53c04c5383bdddd6af9e4cc4 : SUCCESS in 45m 38s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8b3ca510a9724dd5b26a3406978b189b : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b13cc8e2049a4519b4dc57cc127fca60 : SUCCESS in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b619e656c9834df0b0ad058b9d3b9484 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4dc4dc86164a42a3bc2c5c962ec2d733 : SUCCESS in 6m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f0d0f401423948028dfdf719a4817263 : SUCCESS in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d88a3000434afea2df229b1e5d8872 : SUCCESS in 5m 50s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/729e4f8b78e04a16a6392784051f0fd8 : SUCCESS in 35m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/195cd53c0a124ca68966640c12aaa4d6 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":4},{"id":"25d85abe659b6a52ddbc068566508238cd5acaa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 12:28:12.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7f3ef315d4dd84d5fa6782b341c2784b98b9b633","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 13:00:16.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"8717f42a04e42e7d7f525f7e8c4041040beba17c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 15:09:07.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/71a9d483c4c849ca96c06255257729e3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/b2989496daa947978e7adb07e079ebc4 : SUCCESS in 2m 32s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/0063af5ddbc04cb99bcdb4ad290113bf : TIMED_OUT in 1h 12m 53s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a0d94428f98942f199483dc2f0f4f1c5 : TIMED_OUT in 1h 12m 59s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/dd768f0db7db443ba4c2cc81a0714bcf : TIMED_OUT in 1h 12m 25s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/fd19c95fcbd94663809abba58ae63b24 : FAILURE in 25m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/82456a49eac940e8809057f0198abbbf : FAILURE in 57m 56s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f91c48996483450b9bf04f393bdb7e73 : FAILURE in 40m 29s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a327f78ae67444baa23ee5e9c85fb10e : FAILURE in 25m 12s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/26ce254e8e254d069cb5a8e69e33c6f1 : FAILURE in 50m 29s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d46de82f0be84933abdddcdf27c7897a : TIMED_OUT in 1h 13m 04s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/dd6fddd8f405402b881371c6d077d8d6 : TIMED_OUT in 2h 02m 51s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/5b0e6f40952940e6b0e11665649feccd : FAILURE in 1h 56m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4deedc5d79b74ea9a59e7a31920bf9a0 : SUCCESS in 5m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b7a0c26dae44d3c898353ec0181d96e : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/65120388fcd243dc9a535a64483b0166 : SUCCESS in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0d3a95342e54445a6c4a4cf10f3af5e : SUCCESS in 7m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d4e920dc9c274e4180201534bec78b2a : SUCCESS in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a187f909cac4581a21837977d807d59 : SUCCESS in 10m 39s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6d4abe3642240be9ef5a5fab1676422 : FAILURE in 44m 21s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/437e24664a814b89810f285c9c0167ab : SUCCESS in 6m 45s","accounts_in_message":[],"_revision_number":6},{"id":"918fbdb135c400c6f9129d21a859d1cbfcae2e4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 20:54:03.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"72b60b873e1747e4d1bebd5e03f349b50ce3733c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 22:14:11.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aa85347973a41c6a10c90d447487039\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11df31cb0ddc4571b2089f612399b48e : SUCCESS in 2m 15s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/ab2416fbdfcb4917a366f0a56f6b7b54 : SUCCESS in 52m 39s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/828d9a65060c400a8d5aeaa71c732f58 : SUCCESS in 44m 59s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/bd4916a5e26e47009fc4aa8abe41d218 : SUCCESS in 46m 19s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/db1fb8dc19ae47878fe24f058af11bbd : SUCCESS in 15m 12s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/468aa9df0e5742eab0aab3bf93cdc0cf : SUCCESS in 1h 09m 43s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/289554d60bc8478fa3d5b3d1b174c4c4 : SUCCESS in 51m 13s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6c6d4eaa073d41d6a4c5a8f558a06cd5 : SUCCESS in 1h 14m 32s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/31e16bb605e8443dbfe249baf0efcffd : SUCCESS in 1h 06m 35s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/12c1e418d0604485a46466c7ab135766 : SUCCESS in 52m 53s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/21117ef49dcb4ab3b2513e0e99c5fea6 : SUCCESS in 48m 25s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/754ca1a42b9040528288d006329fdd12 : SUCCESS in 51m 50s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d3218c595d449758d034c00b56d338f : SUCCESS in 7m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c14f8a798c484f4083a74e5486435339 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa552c8f332e4c47b46322d723f953cb : SUCCESS in 4m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/982e47b6773547ca8d228c43104f4899 : SUCCESS in 8m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8683d080cb3483fb48dabd786326eff : SUCCESS in 8m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/083cf0ef54a84a88af450f5a6a1ab53e : SUCCESS in 35m 37s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/deb3f7884c32460ba869e60fe5f87ee7 : SUCCESS in 5m 10s","accounts_in_message":[],"_revision_number":7},{"id":"2c66338b517af861e99a8610c714a405517862e6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-15 13:11:08.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"00d7fbf7a322c81c26f34d21da803fc7cc801067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 14:32:41.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7d345fec877a4ad298793f047d13a0ae\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/42d403efe7d2453f92f4c8f8f257f23f : SUCCESS in 2m 22s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/41432a12217c4b148f7c8cad54b8e682 : SUCCESS in 40m 56s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/d44d7db8a1954a7f9274d5a03ccef649 : SUCCESS in 46m 22s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/5bbb30b73c7b4be2b7a41cb413310511 : SUCCESS in 45m 47s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f3afac4286274a86a3d64e84bf159188 : SUCCESS in 33m 36s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1dbab53008a54ddcb3e210a9b8b54431 : SUCCESS in 1h 07m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b9483cc756e34f0eae7d0a6fb0a0154f : SUCCESS in 1h 15m 17s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/f95bf9e4ae7e4dcb8ce59d2ac9a679a3 : SUCCESS in 1h 02m 23s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8d4cbb22fdf8467c98666199dec9d82e : SUCCESS in 1h 09m 18s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d78367bc2a714da1b993de0b36b996ae : SUCCESS in 45m 30s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d79254cabebc46b3998219cc8e7b9dc4 : SUCCESS in 39m 21s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/c5d0c455ec954c809bf2e0c02303e84e : SUCCESS in 48m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ef3f45fae4734348ab70b1a85a60d1af : SUCCESS in 8m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7f203142244385a44bdefe3caa1526 : SUCCESS in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a09601f7de7741a5b4036927c26578e9 : SUCCESS in 4m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7682649579ad48cb97a7dcf32af1c3e0 : SUCCESS in 7m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42b478958c8a4702b01a2cf69ae1c96e : SUCCESS in 8m 44s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/5af4d680876d4490b59bf08c670f9326 : SUCCESS in 36m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/acac6fbe889448f0923c5633f343ff5e : SUCCESS in 4m 27s","accounts_in_message":[],"_revision_number":8},{"id":"109430a0b3d3bbb72b114d0dcd69f5effdad0aae","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-11-03 15:52:01.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"89f7c1abdb4543fc2444cbd9791f91d2d94a48f3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-01-06 22:25:27.000000000","message":"Patch Set 8: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"df720920b2176a32abb0374e7adc724cb677b643","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-07 16:50:37.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b170b079e829392176e1ad1e0b5ffd541bc23709","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-07 17:02:42.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"b7472cb83cdc9f7387f04c038bf1ac1e27f80c37","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 18:20:01.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13ebcba114bd44b4a852b7ed6d781be3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/70329e199b444ecaae9b254066ed2574 : SUCCESS in 2m 33s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/a4834658f54e4f0fa42074d7dbea6f37 : FAILURE in 46m 01s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a19562ff13fe4ec09b1025bc408b3c0b : FAILURE in 39m 10s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/75d136d744ea49a787db56f2e23d2228 : FAILURE in 42m 40s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d748538fd84146a1ba425ee194a78b3e : SUCCESS in 34m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/188bbeaa0fd84e3cab00c612625a8104 : FAILURE in 56m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/82eb6a3f7cf04d51a0b81a6121f09443 : FAILURE in 1h 08m 35s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/cc9a4979d0dd4ef6a300735d0b88f181 : FAILURE in 1h 09m 12s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a6f67f487bd643dc8a5d4333963b244d : FAILURE in 55m 02s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/f253b20cff8547e4b09cfaf344c3bcd2 : FAILURE in 43m 06s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/1b1f887773674f23a7d8b9bb1866421e : FAILURE in 48m 57s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f40e68f53e5a4873a7c467503030197a : FAILURE in 51m 34s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fb2e3523fe2463eb72984bb8114eb2f : FAILURE in 5m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32b7d531f2204e19843b3a9632a01ecd : SUCCESS in 2m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de795f3fc56647ee96ed26d8a2e645e8 : FAILURE in 4m 49s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/08d579547528421b8d5e11e55c803a04 : FAILURE in 7m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96a9d2d3543c4f768e7c852f2ccc4b51 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f337b66db7fe4bda91439dd51e898658 : SUCCESS in 33m 34s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/746d88e35423495eac17f6957932c4ce : FAILURE in 5m 58s","accounts_in_message":[],"_revision_number":9},{"id":"483316a9161dc6a627b4fff5097dfa1cb22425d3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-07 21:18:33.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"3a633e18cee508c430746db479e14c95db2633aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-07 21:18:56.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"3f8bad343fd93da960ad2aff39bb6db20657cc88","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-07 22:01:39.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"493a6fcfc82d3ebcb15371603d143b7e6b15477a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-07 23:17:33.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7eff22132b034eb3b70323706d28d548\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9025892e462748b6b6ea8af1bada481a : SUCCESS in 3m 35s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/10fd7453af7b478eab2c2edb27d4ab41 : SUCCESS in 50m 43s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/20a7ad8afa454a70aacbd56004548bce : SUCCESS in 49m 57s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/487617f15c304d50afa407a798303ece : SUCCESS in 48m 37s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/9a83e2431ee74b269e2ebddc973f1e31 : SUCCESS in 32m 18s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/712c1b561b0c4d38ba77eb8065ec0394 : SUCCESS in 57m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/69168bdebcca4cfeaadc2602b5df6a6c : SUCCESS in 1h 13m 20s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/6c23ae83d2cc406abaf4331640114190 : SUCCESS in 1h 00m 47s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a5bbbbf3e84c4dccbc39a798e06acc34 : FAILURE in 1h 02m 27s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4bf42defccd146669dfb6580c778dd3b : SUCCESS in 48m 03s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/edfd2f8c8b34488180f9135db657adb8 : SUCCESS in 43m 55s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/85c0e793dced472db2408f24313cb478 : SUCCESS in 43m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/54e1fd917ccf4fd1be4f7ff18f565b2c : SUCCESS in 7m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b51199dedc5c4fa79088f8b0eae71fc0 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b15bc206c7dc428595d8cc0e9a1e9080 : SUCCESS in 4m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6d9556a25ba949e0aa966ccc01b75296 : SUCCESS in 9m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ebbb7024a8247698579c9ec0a1f656e : SUCCESS in 9m 29s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f3016f6442f34457be0ba92e3ab132b7 : SUCCESS in 33m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c756087e6d0f43008120a692baf27d17 : SUCCESS in 6m 24s","accounts_in_message":[],"_revision_number":11},{"id":"5eaa38bb5fd05869828c1734283649dc9e2fc9b3","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-08 04:36:59.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"7a951ea9dd3f15ea11a3f31c23bf76c1640ce8c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-08 05:45:30.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac36e6d7c4464dd7b0128e706c163e2c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/896ca28895bf4ef3b8ac59b1775e0aed : SUCCESS in 2m 09s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/238ac242ba7b4cd5b0c958e7e33c13a9 : SUCCESS in 47m 43s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3c4fef7be77c47e2a9f951e7bdb2f452 : SUCCESS in 42m 09s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/f74203f7ea72490b8cfce7f6777c01fd : SUCCESS in 48m 52s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/4bb9671ecab442689e4e8b608ec2c654 : SUCCESS in 30m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/47b041e81b02447dbe82e235065d2999 : SUCCESS in 55m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/93bc0b4962e0475ea93382669c370d80 : SUCCESS in 49m 06s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/a7f99b03a44a4962b2dbfa963b557be3 : SUCCESS in 1h 04m 04s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/eab4710bbd6a45b2a94194458026e37c : SUCCESS in 1h 07m 50s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/fba0f877f96447bc870c28dd1fa9973b : SUCCESS in 45m 44s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/51b028a1c6e5486c96faf1a10e695e77 : SUCCESS in 44m 32s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/b1b576c6e68f440393f3bd1657114e57 : SUCCESS in 40m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/282cc2c4a2254065bd50e4613b4a3bfc : SUCCESS in 7m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d6f82b07cfc4b11b2e51462d048326e : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ddb4876d354f496298a92bf3444657a9 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/371d6684a8424ce093e64d4eb273db25 : SUCCESS in 4m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/087f9585921c4a97a4e897a2d3cb8b95 : SUCCESS in 9m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6750936d565d47f48734102c58a50180 : SUCCESS in 31m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/a653d0df02614fa4938d1ce527025ecd : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":11},{"id":"d5404cc6a4029474b97d4ee0cd0c27a270b4e41b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2026-01-08 12:12:11.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"6885bd0b1edc8806ce72ea183a2f021b0544e940","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-08 13:43:50.000000000","message":"Patch Set 11:\n\n(6 comments)","accounts_in_message":[],"_revision_number":11},{"id":"88875cb18588415d3f0b9c95a2c904701a4aa57a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-01-09 00:12:41.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"411731a7e3a265b9d8242574ccc39078a8bdfe21","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-09 15:07:32.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"eee604a4b54656d0bcb4d8ec56a8b6a3a8a81dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-09 15:07:49.000000000","message":"Patch Set 12: Patch Set 11 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":12},{"id":"b82966bc45f64fd638fccca41a1c50a2eca82111","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-09 15:08:51.000000000","message":"Patch Set 12: Review-Priority+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"92ade5333aac76c6bf3d84a70d1091a51fd67f60","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 16:38:24.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cd24abd9724b4d758ffbefd44ec0b4a0\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/94c1f278ac9144c081c75ebeaca9b16a : SUCCESS in 2m 54s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/4332c702320747e7be757251296c51a2 : SUCCESS in 52m 14s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2a2fa89f4767406ebb44d2f726e7f7a0 : SUCCESS in 54m 47s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/17ad1513fb7145f791fe7986ee59adfc : SUCCESS in 50m 03s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/531cb7852f26484ebdefd1b833e976fc : SUCCESS in 34m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9b9b830a62514d55bb89109909d9f5e7 : SUCCESS in 59m 52s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/33ebdea4e3ed4926a7f88e18b32d3f81 : SUCCESS in 1h 07m 38s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1f164dd2b78c45b5beb13439e6f69af3 : FAILURE in 1h 10m 42s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8c0b0faddadc4da7b185464edf6fcea3 : SUCCESS in 1h 08m 15s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/548786432c2447beb936aedc46cecca2 : SUCCESS in 50m 29s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/f9521d2549714140bec977d7f60d68b4 : SUCCESS in 47m 53s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/ce39da1898ba4e7fb8367e11de4cda21 : SUCCESS in 46m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e15f4ec6fdbf4dffa03cb0f2b17fcbbb : SUCCESS in 5m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d078d1d3444314a056cd2205478351 : SUCCESS in 3m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c0814b4aa6745d9bf1e1c1744f63970 : SUCCESS in 5m 09s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2c7e0b4be40f4d7fba7a207435731785 : SUCCESS in 4m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15ef73604aee4a7e93f3f11b33f5d307 : SUCCESS in 9m 40s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ada7281690e41bca25fcf80df529b6a : SUCCESS in 40m 23s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/f39cb98f868e432b9aecf986b70cf2a6 : SUCCESS in 5m 57s","accounts_in_message":[],"_revision_number":12},{"id":"ed13f74c48f2fbefd2167097878633b9755f3ab4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-09 16:49:12.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"4abee29169a23bd2c64654281ced54d81c288397","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 18:35:47.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6576d8425124db9829344d0e8112bf4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/66124689f374450898c159f2274b07ee : SUCCESS in 1m 33s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/465a875e06cf4b3a8a3b1ce583912c7b : SUCCESS in 48m 44s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/2cd1b7c994d046c59ef57a2daacd3b6b : SUCCESS in 55m 31s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/3401c669edf8477aac23101ff85a0f02 : SUCCESS in 50m 18s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b658e7ff43a24e1490bca0608454860f : SUCCESS in 33m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/38f9b0c224384d239e22a43d7f0ab140 : SUCCESS in 1h 10m 42s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7f96fc80c7144738b5bb8617fc0c7ba5 : SUCCESS in 1h 10m 35s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/5565cf5c01434a92afd10f42d2fb4aed : SUCCESS in 56m 13s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a399fca80298471f85b20bb6f722d2db : SUCCESS in 52m 22s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/531b49d6953d493687fcc32e3fe582d2 : FAILURE in 48m 51s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b96c9bf6ccf54140a8fb20e5244feb48 : SUCCESS in 1h 32m 37s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f676e11440754ca3b25b466953fbbb9c : SUCCESS in 48m 44s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ba04247aae24753aefbeb2cbb5cb62e : SUCCESS in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/49ba7a0cbe194d49b9eac31181288c3d : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34c216558b2049e9bd939f0a738fbbb3 : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/45bbc4591df9447fa3e749894fe54881 : SUCCESS in 7m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ee5712a2e1d44b54a9c82f6c333b4ba7 : SUCCESS in 7m 51s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/2b8e46f220064713a588dc7b66106469 : SUCCESS in 32m 49s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c7cd07b8069b413c8102ee949f527d16 : SUCCESS in 5m 04s","accounts_in_message":[],"_revision_number":12},{"id":"46a503b086be433b946a7570349d005be46b061c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 18:43:22.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"258adf5d6ff71ab5ac897cda254ad9b63f180d5f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 19:59:43.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bd829d6e999444f49950987e17b6279c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2f2e01bc233b481f88a50a8e3383a82a : SUCCESS in 2m 43s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/c2fd5411b5124e45b5a296ffeb4efa5f : SUCCESS in 49m 55s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/98fda6b1121a464891ec1fc3aa935109 : SUCCESS in 49m 28s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/8c8a333f517c44f7a9520f02a876f54e : SUCCESS in 50m 28s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/2faf044f968144bbbb6fbb5fad1534e7 : SUCCESS in 18m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/49bc7052e4324b8da283ea331d886010 : SUCCESS in 55m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/017b0b7a3ace4dd397ad1fc4bcc08f89 : SUCCESS in 54m 50s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/47700ecfd4dd4298b6f7e08a72c5607b : SUCCESS in 56m 31s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d6d41dde33084e93b16cc8415720833c : SUCCESS in 1h 09m 51s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/ac19f7a367974df99eb43c9819d2ee43 : SUCCESS in 48m 02s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/5b0e49d2fd0344c3b121d7b936a34d9a : SUCCESS in 48m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/126c32dcc15b43deb62f5976719045b3 : SUCCESS in 2m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37418f8a3121411cb508219665d2fffa : SUCCESS in 5m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b13a5cf7b3544af09b5050b965296d6f : SUCCESS in 7m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52b2929d5108406bbf3dd32b06d9d144 : SUCCESS in 4m 26s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/3503d0e128d44e8f909bbfad0e8b0163 : SUCCESS in 32m 18s","accounts_in_message":[],"_revision_number":12},{"id":"274e70938f4aca9df51115c1b11c29deb9eceff8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 19:59:44.000000000","message":"Change has been successfully merged\n\n11 is the latest approved patch-set.\nNo files were changed between the latest approved patch-set and the submitted one.\n","accounts_in_message":[],"_revision_number":12},{"id":"4a66f1c0cf93a85b5e9502f2945ec85491c81970","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 20:00:53.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a882f8e415dd46d39887fadb85d5c531\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4df5be17199a4a0bb26fec94fe44d6ec : SUCCESS in 52s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"c1a482a673d0c39ca750cc980bda53992358872d","revisions":{"919f131b3c27ad67deb089dc15bcb63f4cdcc29f":{"kind":"REWORK","_number":1,"created":"2025-08-06 15:17:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/1"}}},"commit":{"parents":[{"commit":"19a35e37666dc80f140037a8298cb41f0c09d906","subject":"Merge \"Do AXFR with latest data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19a35e37666dc80f140037a8298cb41f0c09d906"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 13:42:06.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:16:29.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/919f131b3c27ad67deb089dc15bcb63f4cdcc29f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/919f131b3c27ad67deb089dc15bcb63f4cdcc29f"}]},"branch":"refs/heads/master"},"34a36c78b6ba9b6796b354fc4b9e7661b1859dd7":{"kind":"REWORK","_number":2,"created":"2025-08-06 15:21:42.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/2"}}},"commit":{"parents":[{"commit":"19a35e37666dc80f140037a8298cb41f0c09d906","subject":"Merge \"Do AXFR with latest data\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/19a35e37666dc80f140037a8298cb41f0c09d906"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 13:42:06.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2025-08-06 15:21:42.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34a36c78b6ba9b6796b354fc4b9e7661b1859dd7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/34a36c78b6ba9b6796b354fc4b9e7661b1859dd7"}]},"branch":"refs/heads/master"},"64729268691bf0d088c02e8074ad1553bb580933":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-09-30 09:52:32.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/3"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 13:42:06.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-30 09:12:58.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64729268691bf0d088c02e8074ad1553bb580933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/64729268691bf0d088c02e8074ad1553bb580933"}]},"branch":"refs/heads/master"},"52235fd625bd53cc126322a0b03205a2743028f8":{"kind":"NO_CHANGE","_number":4,"created":"2025-09-30 10:32:16.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/4"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 13:42:06.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-09-30 10:31:16.000000000","tz":180},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/52235fd625bd53cc126322a0b03205a2743028f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/52235fd625bd53cc126322a0b03205a2743028f8"}]},"branch":"refs/heads/master"},"c18efd5084701eefc9e591201a7c7b088e354bb7":{"kind":"REWORK","_number":5,"created":"2025-10-14 12:28:12.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/5"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 12:12:45.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c18efd5084701eefc9e591201a7c7b088e354bb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c18efd5084701eefc9e591201a7c7b088e354bb7"}]},"branch":"refs/heads/master"},"6d1a2aa7cc18c36e0beb3ae230237f58f965104c":{"kind":"REWORK","_number":6,"created":"2025-10-14 13:00:16.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/6"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 12:54:09.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d1a2aa7cc18c36e0beb3ae230237f58f965104c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6d1a2aa7cc18c36e0beb3ae230237f58f965104c"}]},"branch":"refs/heads/master"},"81f9ba4c294493ff1db5eb2509f8ef4bb0229e00":{"kind":"REWORK","_number":7,"created":"2025-10-14 20:54:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/7"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 20:49:38.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81f9ba4c294493ff1db5eb2509f8ef4bb0229e00"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/81f9ba4c294493ff1db5eb2509f8ef4bb0229e00"}]},"branch":"refs/heads/master"},"de65267c997e751e5caa862f4d8bcb45eb42de51":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2025-10-15 13:11:08.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/8"}}},"commit":{"parents":[{"commit":"e773de35e72ee46a3eced71e2678342271fc692f","subject":"Merge \"Initialize the client attribute to prevent AttributeError\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e773de35e72ee46a3eced71e2678342271fc692f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-15 13:08:56.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de65267c997e751e5caa862f4d8bcb45eb42de51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/de65267c997e751e5caa862f4d8bcb45eb42de51"}]},"branch":"refs/heads/master"},"552bbd4f3634306c6858fa26ae4a89e1ce1d4f20":{"kind":"REWORK","_number":9,"created":"2026-01-07 17:02:42.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/9"}}},"commit":{"parents":[{"commit":"1db323a3c9c2061174802c75a31df1724d2b063c","subject":"Merge \"Remove old checks for Python 2-3 compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1db323a3c9c2061174802c75a31df1724d2b063c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-07 17:02:29.000000000","tz":-180},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/552bbd4f3634306c6858fa26ae4a89e1ce1d4f20"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/552bbd4f3634306c6858fa26ae4a89e1ce1d4f20"}]},"branch":"refs/heads/master"},"c92163b2da51af95ef2637da189760bd6791a5cb":{"kind":"REWORK","_number":10,"created":"2026-01-07 21:18:56.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/10"}}},"commit":{"parents":[{"commit":"1db323a3c9c2061174802c75a31df1724d2b063c","subject":"Merge \"Remove old checks for Python 2-3 compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1db323a3c9c2061174802c75a31df1724d2b063c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-07 21:17:24.000000000","tz":-180},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c92163b2da51af95ef2637da189760bd6791a5cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c92163b2da51af95ef2637da189760bd6791a5cb"}]},"branch":"refs/heads/master"},"9ed8f9b83d68f9b39215521963f38fa227deda8d":{"kind":"REWORK","_number":11,"created":"2026-01-07 22:01:39.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/11"}}},"commit":{"parents":[{"commit":"1db323a3c9c2061174802c75a31df1724d2b063c","subject":"Merge \"Remove old checks for Python 2-3 compatibility\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1db323a3c9c2061174802c75a31df1724d2b063c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-07 22:01:33.000000000","tz":-180},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ed8f9b83d68f9b39215521963f38fa227deda8d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9ed8f9b83d68f9b39215521963f38fa227deda8d"}]},"branch":"refs/heads/master"},"c1a482a673d0c39ca750cc980bda53992358872d":{"kind":"TRIVIAL_REBASE","_number":12,"created":"2026-01-09 15:07:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/96/956696/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/96/956696/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/96/956696/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/96/956696/12"}}},"commit":{"parents":[{"commit":"d88e46016ca159cae6ae8674ac77663f00b357ec","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d88e46016ca159cae6ae8674ac77663f00b357ec"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 11:21:59.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2026-01-09 15:07:49.000000000","tz":0},"subject":"Isolate eventlet-dependent services for future migration","message":"Isolate eventlet-dependent services for future migration\n\nThis commit creates the structure for migrating Designate\nservices from using eventlet, to using oslo.service threading backend\n(standard/native python threading).\n\nKey changes include:\n* Eventlet isolation: existing eventlet-dependent service entrypoints\n    (central, mdns, producer, sink, worker, manage, status) are moved\n    from designate/cmd to a new designate/cmd/eventlet subdirectory.\n* Updating their correct path in setup.cfg.\n\nThis prepares the codebase for future incremental migrations of\nindividual services to native threading.\n\nChange-Id: I67941b5c7525d4b9661b30c4cc03dbd061e86083\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1a482a673d0c39ca750cc980bda53992358872d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c1a482a673d0c39ca750cc980bda53992358872d"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
