)]}'
{"id":"openstack%2Fdesignate~956697","triplet_id":"openstack%2Fdesignate~master~I3fd123451dea9cbd1615211b41ae1c085537129b","project":"openstack/designate","branch":"master","topic":"eventlet-removal","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2026-02-13 09:41:53.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2026-02-05 21:22:23.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2026-02-16 10:37:11.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I3fd123451dea9cbd1615211b41ae1c085537129b","subject":"Migrate worker service to use native threading","status":"MERGED","created":"2025-08-06 15:17:24.000000000","updated":"2026-02-16 10:38:08.000000000","submitted":"2026-02-16 10:37:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"956697-eventlet-removal","meta_rev_id":"6052f0e8546c34a0de0194148045d9de8f76a817","_number":956697,"virtual_id_number":956697,"owner":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-02-16 10:37:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2026-02-05 21:22:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-05 21:22:23.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2026-02-13 09:41:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-06 17:48:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-11-07 00:52:57.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"},{"updated":"2025-11-14 16:07:03.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2026-01-15 22:25:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"22ed3383cf2a073f148e4088e02b0b216ffbf33e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-06 15:17:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74d607842cd58cf0742b0aa548a191b34f73981d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-06 15:36:11.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"b8b1c006a4dee594d1e15c521bbc79e566cc81de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-06 17:48:37.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b31215c72dfe4399a40c8f17ee591525\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c846e6074000494a9f3e803ca2e7317a : SUCCESS in 3m 45s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/505e0040f4464596822b44af7a8ad7d2 : TIMED_OUT in 1h 13m 04s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/4557c64f70ed43b584d119da385accca : TIMED_OUT in 1h 14m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/10bccbef2f0c464187e8457adcb6c176 : TIMED_OUT in 1h 14m 04s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fe24851044994883827776fdd56a26bb : FAILURE in 1h 04m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/577ac640199a4db48e706ed0f18e4820 : FAILURE in 1h 03m 23s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c5dac98321074f7bab61b4f26b856d29 : FAILURE in 1h 01m 50s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c3241efecc244617bdee6cdfcb61c2ab : FAILURE in 1h 10m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/0d586a9b47444dbdb4cd7b3e17dbd7c4 : TIMED_OUT in 2h 04m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ac66445e01834015b6b9349babd97455 : TIMED_OUT in 2h 02m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6684457bc565481391ae2c95ff0545ca : SUCCESS in 8m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18917d77a75d4137b6ac1687365d7108 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3832185e22cf452c8475e06997306530 : SUCCESS in 4m 07s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a4ff48d6beba48109d6bf0edf49fecdf : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0d9d67dda4584c3c80704e24363d95e4 : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96e6905e89bc42dd9993a69d9ae7cc99 : SUCCESS in 4m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bce7932868634ecf9b1fc96248d315af : FAILURE in 39m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/af309f614ae248639e903223ef1120bd : SUCCESS in 7m 05s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/bc85c7afdc1446edb0596cf7997d50ab : TIMED_OUT in 1h 11m 51s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"a753b88997cba2ad82238d3ccb1e9b133ae6efc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-08-07 09:10:21.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"b680ac44dcb0fd7399c5d04b07826c0ec3b4cda7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-07 11:27:51.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/352d69b6e3bf4a3ba3ffececbfe6cddb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e934a0ae5dca48118534f9524d92d303 : SUCCESS in 2m 27s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/06dff112a15f49f1bde8f4ebc7cea1cd : TIMED_OUT in 1h 13m 36s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f755940f53244200984d4ac8e1464cae : TIMED_OUT in 1h 12m 43s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/096545775b4f46abb1b79bf9b4218373 : TIMED_OUT in 1h 11m 41s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ec1233ae1e3f4a7a913d003b76027443 : FAILURE in 56m 19s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/448ce6965a544315ae0d4b9dc8ee82ec : FAILURE in 55m 08s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d68d63bb36bf406d82f8583d0aa379b0 : FAILURE in 56m 50s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/092cf03544ce4ad4aa7e2fcc44f461e4 : FAILURE in 35m 08s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7ede3ba0f5604fa38141d059bf2d20fc : TIMED_OUT in 2h 02m 52s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/0ae0014854714151ae494e3e64443ceb : TIMED_OUT in 2h 03m 43s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/244cbf901b5b42ceaed98bf2defba89b : SUCCESS in 6m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37a467dd3c174a3a848f8a2a5b8bc780 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e79c42ee13bc4c5b974b4cb95b479736 : SUCCESS in 3m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3333e5207ac743bd9178f36171272b90 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8e49bf10b6448199308e719466763a9 : SUCCESS in 7m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3c760558fece483b85c17dd215d15048 : SUCCESS in 10m 35s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a7a9b9eb39a4437a9c311b9693a4a6bf : FAILURE in 40m 42s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/002af4a7e4184569b8e70433d2f4d5ca : SUCCESS in 3m 19s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/3c25200675a14de7a75444e086824c36 : TIMED_OUT in 1h 12m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"368eedb73f8bff02e8b35ab6c2759a3446a92499","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-07 16:36:39.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fb3df055b76026c86d6ab7b1ad266709a3411b0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 18:03:52.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d418b61f8fdc41038608cf489bcce131\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/de20f7ba719249f8ac9dd512930c34e1 : SUCCESS in 5m 24s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/fadbc00059ad47b6896052b33e511f95 : FAILURE in 7m 25s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3ffb4ccd0ec9429189ff8c31f73d95c2 : SUCCESS in 46m 39s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/1e59a1360233489e8caaa0c719bea1a1 : SUCCESS in 44m 25s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f794081d92134ec8b3fd8fb4ed9ca410 : SUCCESS in 33m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ebc8f68d73a640aaaf47fbacef1d105b : SUCCESS in 1h 08m 01s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/76454f9baa6141fdae24463b98c92098 : SUCCESS in 1h 07m 27s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d75b3a3daef2494184553584bc443814 : SUCCESS in 1h 06m 54s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a16f8f0d7fc64aaf8617d7f37c3e9bb9 : SUCCESS in 55m 53s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/49c5afac4e3449a1b26b84e061ab40bd : SUCCESS in 46m 10s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/bc5ce589499f4a1e9c75523d6dcc7770 : SUCCESS in 37m 57s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/e010c588b6f243968044263bb9d436f5 : SUCCESS in 40m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aee5fbe08e504aec8ac1f0b8551c686c : SUCCESS in 8m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1af2f5683c0246af914e58b42ddbf2ec : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4655b051076e47159235de58a7a7ac35 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ddc8ea0d08c42619822721cbbda3811 : SUCCESS in 4m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c6edc0c3ddb24629b09768fa8e0953c3 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/03135c6df4944c0b80c0d1ff3209fc31 : SUCCESS in 4m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7ba100e7b6854631b3749999fa7515d8 : SUCCESS in 38m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/9f6fee9b21114d5d8810bebd88b68037 : SUCCESS in 5m 32s","accounts_in_message":[],"_revision_number":4},{"id":"fa2b5e8179cc2c16988ccd38c50dbbfddf3fd864","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-07 23:40:49.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7c16704c52a4dabc9541d90b9cf122b0af70484c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 00:55:34.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6bd8fc10d5c4489b359edf4e17a065c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8b6ddaf120414b9f89b72531fa8146c1 : SUCCESS in 1m 29s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/08c0a966e7104c0d9bbb96eb67958001 : SUCCESS in 50m 01s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4926c3dcc82545fcb41533dc3bf075bb : SUCCESS in 37m 46s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/17d91c04906b4d8ba6fe5c42d6988192 : SUCCESS in 43m 48s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f87cdc49fcb24a749014134c2979150b : SUCCESS in 13m 52s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/815dce26ec124da194d81ec00959f977 : SUCCESS in 52m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6eb5219f6b924b5a924f03deac6cf014 : SUCCESS in 47m 02s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/c447643905fd4ea88bd054bb7fb3fd1e : SUCCESS in 52m 30s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/19874299556647afb3c7fc107dff1acd : SUCCESS in 1h 09m 31s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/3deccd612eed474dbd02f95e90d2f15c : FAILURE in 6m 50s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/79838314a01743dabe17ea034ba814b7 : SUCCESS in 45m 29s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/2480f00546764cc1aaa1ca9a150da69a : SUCCESS in 45m 32s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/209bd6126d184fa7bd92f3517f2970d2 : SUCCESS in 5m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/021c9f3367fd40e3be68aa615864e8a5 : SUCCESS in 3m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ab7edf0a68894612beca26cd0eeb6d0c : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/774466816ee04a1e9c1dff9817cd56c7 : SUCCESS in 5m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/428dfdf4c79b405ba11417c5da960f47 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68069326d0614a97ba5f30b9f0aa17dd : SUCCESS in 8m 38s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/d2ae994c2f1444779688071d7a799cd7 : SUCCESS in 31m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/07ff6840d4cd4f1aba5c7f4f6d2a4b5f : SUCCESS in 5m 02s","accounts_in_message":[],"_revision_number":4},{"id":"1fe99131043e92752792f2df4651a066a0e3b2ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-08 14:46:18.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d5059140802ff6c87afe72d85f5942014b961c0f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-08 15:09:42.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"3c4520458ded50a092b1fb303e92663f86485737","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 16:34:21.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0381668633164daeaf4b1b4f82a7adf2\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cafc1e36ca5343d1897bd60e8adaf55d : SUCCESS in 2m 32s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/60df31d90e2a4f59bcb07a23b88c37b8 : FAILURE in 52m 48s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6c8b1fb209ca46349c333be46c1642c0 : FAILURE in 51m 28s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/9e1ce974f5734bf6a44cb0974ae2a87f : FAILURE in 41m 04s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/0101220189d44663aad464c7ccb9359c : SUCCESS in 26m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/2840684fe4e240faabf75ad909fbf0f7 : FAILURE in 1h 14m 07s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d42c15a4bd9a4b1cb31a97827545660e : FAILURE in 54m 01s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/13d0a8faac6f426f984c97a5354c5b69 : FAILURE in 1h 03m 59s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/887fbcfbe1d94f99b6f41b278863f2aa : FAILURE in 54m 44s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/b80b63e748e14128bcdd8faf3c5b82d1 : FAILURE in 46m 33s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/e368ca2e5ed84ac19b30f43dcb9b9315 : FAILURE in 44m 07s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/38a5928b24924cd6bafa38cf424fa750 : FAILURE in 51m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a1aaa055412475b90838f51db7f7ff9 : FAILURE in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04895e99f52943d78138d4f2cad42ed2 : SUCCESS in 4m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e7d975c302d9497e9f933a00efc7f8ee : FAILURE in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0464238421224c5fb82442298aa5cce7 : FAILURE in 7m 50s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0dbc5a6286bc40d1a587b14086ef034e : FAILURE in 8m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fe2dad4f3d8b4222a567aca1cf6171f0 : SUCCESS in 9m 03s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/71b804ee4c0a4ed8907ef032a244c402 : SUCCESS in 32m 11s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c20baf22261a42e7a731f5ad19df8430 : FAILURE in 7m 16s","accounts_in_message":[],"_revision_number":6},{"id":"256710b1a2d11b45159ca03e87160f3f539f8c79","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-09 08:54:56.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"f522921938a1306eb3ddb97b5cc1c95ba6ee8a61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 10:07:05.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb1223f5221f4a97a6a8a79579fc900c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/272c3b011f7d44a0a9c8ff633b250336 : SUCCESS in 2m 53s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/a19b45212168484aa0b7ccbb69701247 : SUCCESS in 40m 28s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ca31a1db9cec4a0dadda8703811c1e46 : SUCCESS in 44m 21s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/c051d8c0890f40e393f86747d82023be : FAILURE in 7m 28s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/79d44593aa1940419ac5cebae835bf31 : SUCCESS in 21m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/310f8bb129fc467ab3c1b97244bc6c41 : SUCCESS in 53m 51s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/0271b5f99aab49c8a4c6fa1bc37e5a17 : SUCCESS in 1h 04m 53s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/7bcd984d0570422082866fd1dfd6226d : SUCCESS in 51m 08s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e8ac3ecb10eb456cb363eb5057a6a4a9 : SUCCESS in 53m 38s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/ae4355eba97740308081d6ec28f49492 : SUCCESS in 48m 34s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/a8dc5bdc9d6d467d8ed40d0a6a81f142 : SUCCESS in 48m 23s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/9b3b2b926a784e729ed0beb3a97c9cdb : SUCCESS in 35m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f3c7a7339f549749b12ecfe51cc4858 : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f53a9ca4d38f421ab1a97b4db790c6eb : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5e7960d378154a8088650e825dd48fea : SUCCESS in 3m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7eec8a413f9845cfbee63a25715f38af : SUCCESS in 6m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/092a6d2736664b92bb608b4ffc519eac : SUCCESS in 7m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3c288bc67964a7999c74b17699cdb16 : SUCCESS in 9m 24s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f120a8e5ae4949b1b0f6e8b6c3f7e63f : SUCCESS in 40m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c396611cd9a24f9e8f659e8c2362e3a1 : SUCCESS in 5m 17s","accounts_in_message":[],"_revision_number":7},{"id":"7147de4d529e7b2cfff537bb3bf584dee92d85eb","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-09 10:52:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"362e28264a2efc60db3453a2006c7bd99f5ec749","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 11:54:52.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/572bb2efef024528994e2b05dca8e92a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fbc07543c5ac4071ba3e90619d77dbad : SUCCESS in 2m 43s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/99cdf0c72c3342f69ea66f1b488c7e9d : SUCCESS in 49m 50s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1f63f11c40274b9d8cad260673df1db9 : SUCCESS in 54m 22s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/a89e6611d7644f67a548f61ae3d4b424 : FAILURE in 9m 59s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/cbb5a7b2aee94b9f8ea4295778a4b772 : SUCCESS in 22m 54s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/37f0b85d4f5b4124882efd38c2bdc1b4 : SUCCESS in 59m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4989540989c448d09f0a6336e1344b0e : SUCCESS in 1h 00m 19s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/fb3daf74fcc74f81b73cf4a23050ac93 : SUCCESS in 55m 03s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/e741c08cb8144a469fbedbf15f9c8883 : SUCCESS in 57m 58s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/185dfb3f87fe44edb82f68a479a0c0e0 : FAILURE in 9m 57s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d16b0656eadc46b1bba6eaf8d86ce21d : SUCCESS in 46m 08s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/a4b0e074b6c3423ab600c535bf26a147 : SUCCESS in 42m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c43fbd068d324c70aed609d021035873 : SUCCESS in 6m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df42c8c84ec94ffc9269cbc3b580a9d0 : SUCCESS in 2m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80a7f6239747425d8e9f1d8b76edca14 : SUCCESS in 3m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3fe6c603275946d8b32cd1d85f811d63 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c31a0434ba5e4cc79df7870df3dee2b3 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/848985af157b4c4f83fa7cf23e99ccb4 : SUCCESS in 8m 11s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c64c1bae658e4f7d8c924f4f42048a70 : SUCCESS in 39m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/54bab97959ab48f1aa5eba5145ad1d27 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":7},{"id":"329cd2fdc49c6986e884742febd27a3207c35c61","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-09 15:10:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"a8ebda6a4dfad598e5c84ee084fd990a37a5ae33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 16:25:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2339db81e09d4e87b69709eb1cd3318a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3e7a9bb471f34026a2ac1e40fea8c548 : SUCCESS in 2m 18s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/166a26ad0dea4fef8689eefe3c229108 : SUCCESS in 51m 10s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f7aa8c17ac034a61af822eeab733b435 : SUCCESS in 49m 03s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/b7a5401dd896492fb1d03dee7df944e0 : SUCCESS in 44m 39s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/5d274304f41046d58f4b746d0818cec3 : SUCCESS in 30m 50s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9070cffc34044ddea4215c2f5c4d2a8f : SUCCESS in 1h 09m 48s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/f2a4ebcb9e49473ab3a8d5c78eda3fe5 : SUCCESS in 56m 40s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8cea2a814e464b5ba0dc321f5f8bfa3c : SUCCESS in 57m 47s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/66832b701ddf401aa667c79060795498 : SUCCESS in 59m 28s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/ce4c725d993b49048654c06023f5ffdc : SUCCESS in 49m 57s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/bfd8a0c5ffc54962829e1b1ca350df2e : SUCCESS in 46m 42s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/44a9b651ca034ad8b29c2673a21da6b2 : SUCCESS in 48m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8c3b2adf9b134362a87fa1dfd606ae4f : SUCCESS in 9m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a6fd80944fb4d8090f787d1f11c71b6 : SUCCESS in 3m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2a123c71aaeb42d7b9899b8e84b8827f : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/669f2978293d4da7a7f5c2e4175b840c : SUCCESS in 5m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3927f23bb35e4991bb49fa7ffbb8a8e0 : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/34adbbdcdd044ef4a374b7fd294ae4e3 : SUCCESS in 11m 54s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/ee79eb46c37d42c9865058bc3f26fa10 : SUCCESS in 38m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/42810edb6c3e461fbf0a55edee29a890 : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":7},{"id":"1b8665a4c79fb5406848477f8a31cd9a97d3bcca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 12:28:12.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"536fa657b2ddcfc9261ac8327e35fefc9f32505e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 13:00:16.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"b40ffb5421f4fc7ef7467107a9ba489dba109423","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 14:11:28.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59b2a058c99c48e78b2d000cdd1d8173\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/19a21a70b096447a90a3eb3fc23d08ba : SUCCESS in 2m 48s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/0c5bf42a3c6b4fb6b5ae25f425a990f8 : FAILURE in 49m 34s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9ef874aa545c4ef48980755781da6236 : FAILURE in 50m 53s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/19a6968a164b46ccbd9237e3c56095c9 : FAILURE in 47m 28s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ff760d720f7c40028b3baa7ab7ef203b : SUCCESS in 31m 33s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/3e2862a7b3574faeb7778fb830b62d95 : SUCCESS in 1h 04m 21s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/217439c01f8b44e3901191bc8993f02e : SUCCESS in 1h 04m 08s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1cc2f61e6ac74e54b44d5e0805ac1ebc : SUCCESS in 1h 07m 54s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/91d02e354de541d0bce3608147208094 : SUCCESS in 59m 07s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/ad838541990c49d188d50160f67d9e1f : FAILURE in 46m 11s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/778e27d5a6774218ad31869732eaa0df : FAILURE in 40m 30s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/e64845f49dce43b787dfde2960b4d300 : FAILURE in 50m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/475dfa061da141c4bdb27bdd41224a5c : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b09a86108766459a8854a7d787926073 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d0a95fbe1d6b45d6960d7fdf9472fd2a : SUCCESS in 5m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16b2867f79234bba8d9ae933072fd81d : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/36aeb9436cb14cd786e270c22b8191c2 : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3e03a1fc750476e8a4c067d68a6eb87 : SUCCESS in 9m 57s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/dc0a49ff61a146c4a564d0c53694bbb5 : SUCCESS in 35m 10s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/3a36bd8e28e5432d8a818e4c2d88b279 : SUCCESS in 5m 36s","accounts_in_message":[],"_revision_number":9},{"id":"73944b391080f799f89f94ffbf3152b71a6b409b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 17:55:24.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":10},{"id":"dd4a0dfec7d450e4d24f439abc7a36456ce54ecd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 19:12:34.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1cc4690eb5904916bbdb2828eec614d4\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/75dffcfb34dd44c78aebacb02cb1769e : SUCCESS in 5m 15s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/938b3a1c0d3f49e1847bf60f1a55b214 : SUCCESS in 49m 19s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/1a9b261447b047378869107a80adb153 : SUCCESS in 48m 13s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/e3aa2e24635a44b4ae3ca9346d03b3f4 : SUCCESS in 44m 31s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/b718cbd203e441cfaf6af11514d33c82 : SUCCESS in 34m 26s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/185d11b63aec4e19a260a605fd64d552 : SUCCESS in 1h 09m 31s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/413f8136355a42faa60608fc570299ff : SUCCESS in 1h 08m 27s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/bc81eada8c024e0fbac3974c454b4603 : SUCCESS in 1h 06m 26s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/17451e7dfd5d44a4ad0d670e57e54613 : SUCCESS in 59m 20s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/6fa9919dc87a4414b5c1784e6b2bade9 : SUCCESS in 48m 57s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/921804fadb8a4e43a506add9581a84c3 : SUCCESS in 52m 27s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f5e1b869dc3c4c7f845c0fcaaf4c811b : SUCCESS in 44m 56s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9ceb9791b5543e9b123f467c6d0d0a8 : FAILURE in 6m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a286025db7b46549789044812100049 : SUCCESS in 3m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d2b5504e503446439bc2671279dc292e : FAILURE in 5m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/16b14bf1a07c41e99cb9d41aa5445ef5 : FAILURE in 5m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c33beb703c2435c9197ecf0f802256f : FAILURE in 7m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2c62a26f30d4f789e6b8a98b4fdc434 : SUCCESS in 9m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bb249e5d85824d388ea66aa2e338ee8d : SUCCESS in 31m 39s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/7503c25036e64257a00d591a760399fc : FAILURE in 5m 03s","accounts_in_message":[],"_revision_number":10},{"id":"0242c1e28ee4dad8ad348e65aa5384f2db561367","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 20:54:03.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"df035109223c61cc6830a711be76a32abc6142da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 22:59:06.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d3b8f958d384363842bc0e8c7d41bf3\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cde3707eb7b84e4281a6682a5f27e6b7 : SUCCESS in 2m 48s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/8fd201f71cc84f2f975544ec0998e8c3 : FAILURE in 48m 06s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ce9461bc28094675bae2e7144a703708 : FAILURE in 45m 50s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/078f7096ff3c4b3c81dc3676c4933a6e : FAILURE in 37m 58s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/46c29d80f7ba4794a5f39bdb32818df1 : FAILURE in 30m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/722089fc767b4aa8817db79ebbd6724d : SUCCESS in 56m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/5f1dfec29af043f1afa019eed10e8c98 : SUCCESS in 55m 09s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1ffff9394bd34c24a9f1a8215e347e29 : SUCCESS in 1h 05m 04s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c5ecf6bff3644f02b8ac2e36d716dfd0 : SUCCESS in 50m 16s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/40002f79085d4affb820e3d414e4aa2f : FAILURE in 37m 32s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/d6b5b87096dc454eb361051b55b93ca0 : FAILURE in 50m 35s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/0c1ea93309f541de94937b7a9bcff9ad : TIMED_OUT in 2h 02m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fe2ac3133c464361a516ca0325610149 : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7653f3eb6be14e298a8108ee9abc96c8 : SUCCESS in 3m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/10887462afa242548f99bf9de19ecf55 : SUCCESS in 5m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/77427ceb782346c2908b7a9994c1ad49 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aaef649776e742da8b686075e679b1c1 : SUCCESS in 9m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/de1af756eee64ee69f8eb9efa143645b : SUCCESS in 42m 47s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0b404e4f3355421bb0cb003de6f82ac4 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":11},{"id":"d6a5af34c3d1a7a6bef2706cb2f4e9db603442c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 23:27:54.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":12},{"id":"363d07dd5196e24a9866dbf4c48b8424d8206805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 00:48:23.000000000","message":"Patch Set 12: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d9cc922c34e43d38db5f3e2df8c76e9\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f53ff48311944749826bf22422aafca8 : SUCCESS in 2m 18s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/4195862825a34138af4a2265c7ffe990 : SUCCESS in 42m 59s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9f0980a140704ee89ced53fe5ac6764f : SUCCESS in 45m 47s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/f9fa13176511487eaf81c321376d32c6 : SUCCESS in 46m 13s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/5e7d3bd8c2e642e3921729c301d1aa20 : SUCCESS in 31m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/03c5c7a637c1458292829db1d022a10b : SUCCESS in 1h 12m 54s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3506db0998124192ba0e24b0280cd0b2 : SUCCESS in 50m 15s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/9aea6655af4b4886be0e957af3f8864b : SUCCESS in 54m 48s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/cb7803a03cc9491ab5e7c4fe16812ca3 : SUCCESS in 1h 06m 19s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/ea022df45a6f4fe1aa0c81f1f725a8ef : SUCCESS in 36m 41s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/233c0f179bc7484e828ac9bd111ee5a6 : SUCCESS in 35m 42s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/9aa828d472224a86bc69e82c8b230f69 : SUCCESS in 52m 39s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0060a7555a304e1b992cb9000ef4795c : SUCCESS in 5m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04d42937d4d5470b8693fd5f421d14ef : SUCCESS in 3m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1fbd9a1501e342b2822d7d27324da2b1 : SUCCESS in 6m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4573a47acc0d426d9c0005bf62ff7ac2 : SUCCESS in 4m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10be8b81877b45d2aa6922d18cfe300d : SUCCESS in 4m 52s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bbfb8f0dc475471f8d3183ebb1de2930 : RETRY_LIMIT in 1m 45s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/75708d72e9ef4a7bad08ba4c7741ce15 : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":12},{"id":"d86534159dc6da131b7c33ab92b1d6786cfaa300","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-15 08:15:51.000000000","message":"Patch Set 12:\n\n(1 comment)","accounts_in_message":[],"_revision_number":12},{"id":"ea105e0c83eea2f9b4e6a2696e02d0c7c02b6fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 09:31:38.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9df7c00dbcd3454b9d4a33ca6a268f7b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4af38b471b4c476ea07319fd1e56a171 : SUCCESS in 2m 15s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/88b7253195a445859956cf6383cf5fff : SUCCESS in 47m 08s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/930351294f98430ca90739d639796d44 : SUCCESS in 52m 06s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/ce46bd375f954f269521789dc540fd92 : SUCCESS in 47m 33s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/fd8118d8ff964470951e1ce20a8972a3 : SUCCESS in 32m 24s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/4d5382f86e634b7c815f68934aea2805 : SUCCESS in 1h 10m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/7e8b9e9bc82c4e92b06e30821efdf458 : SUCCESS in 1h 05m 47s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/204fed04a1f44c5f824897b0934508f7 : SUCCESS in 1h 06m 17s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/d9aec5b0ca74455fac7b494078662331 : SUCCESS in 1h 09m 25s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/1e3de3d406d84e35b165b3815e8270cc : SUCCESS in 52m 16s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/bb80b3dc6e9d40198a8739449675a551 : SUCCESS in 46m 29s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/31e098dc73e440d7a19cdf0515435ba6 : SUCCESS in 39m 30s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7fe5330c7c764b37b217ca9e89baf8f8 : SUCCESS in 7m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bee05dfed2434c3d83b4f9f2b2682be9 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb426391370b47149a4140001186b9ef : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/81bb270366814519937a7f112803a430 : SUCCESS in 7m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14592c2f4a9a48f795e3a01b5e53e5d1 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8ad5fca5610c47c587c8275d4a13081e : SUCCESS in 34m 35s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/0fd2ec3082ca43bbba1326db025d7b78 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":12},{"id":"5a8795ce6e0a79468b86aca9a087fa22651e555e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-15 13:11:08.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"9974f39ed8a6cfae11c5983e5565bd3ab942764f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-15 14:36:37.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e125f991526546bbab26c8cb6de08468\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/93d037a38d974c459f037a961fc1b380 : SUCCESS in 2m 35s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/927926991e4b4b2a8d1702bd9da180c8 : FAILURE in 44m 59s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/83cee824557042edbd0ca91ac60d2d5c : FAILURE in 44m 18s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/ef321a08e14f425da75a5339baf74c2f : FAILURE in 44m 49s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/e9bfea7c802e4dcfb4b6a57b279a3ad0 : SUCCESS in 35m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/f3e151304c4047b4b93ff97c10463e2f : SUCCESS in 1h 14m 14s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fde278e7df9a434eb4ec63397162d053 : SUCCESS in 1h 05m 44s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ab842beeb6214741bf9d74da65c19d9c : SUCCESS in 1h 10m 50s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/14c67872d51f47a7b9bbf23025c4555e : SUCCESS in 1h 19m 05s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/e67841947be7422fbb25430cda3bbbac : FAILURE in 43m 21s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/440634c21c704c07a917200465a4e621 : FAILURE in 40m 05s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/460e795b95ea484089c9d24b1df7b5f3 : FAILURE in 44m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1e08875cf4c044bbb6093280cba0e4a9 : SUCCESS in 7m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0852019f58ff4c95826843387c193226 : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/854847978d594668bec7ebb48a82468e : SUCCESS in 4m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/db39e2422b284fe78daef62d54c2811f : SUCCESS in 9m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01bead5b3de2452d92bd16cc0955aac5 : SUCCESS in 9m 01s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7d33dadf3a694a338f05488ed3632330 : SUCCESS in 35m 40s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/2fefdc3642df4bc5b2aa050fd91c7c8e : SUCCESS in 5m 19s","accounts_in_message":[],"_revision_number":13},{"id":"162cb3166990838938819929a2c05ee2d5996486","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-16 09:52:17.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"2f7f4dd22635273a12687bd8b8a465b6f81308ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-16 11:02:23.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2f6c354e8aef4296a674ea64c56a6110\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/056cf53d578c4062bbd4626bc35ec176 : SUCCESS in 2m 22s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/a48cbe113d1a4b2b9a5ce22c378b610f : SUCCESS in 45m 48s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/0b202746248640ca9225dc109b6c9ba6 : SUCCESS in 41m 51s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/b58a065765a64d44b978ba46ec5d1112 : SUCCESS in 40m 41s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/5e93d60b7aa74498b7094d64de7b5257 : SUCCESS in 30m 22s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/82d9659e57634926ad15a43c6fa36ec8 : SUCCESS in 1h 04m 33s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/78a7c6a9a1fd4a0d9452f8ab0256e889 : FAILURE in 1h 02m 02s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e26d016971a54beb80e6e32615edf993 : SUCCESS in 1h 04m 27s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b9d90685b45e433c97430cb7badce1b6 : SUCCESS in 1h 02m 47s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/7b0efc9f85ff48618817129cb3cc4268 : SUCCESS in 48m 45s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/e0982b4e151540c4ad7fa300b38668dc : SUCCESS in 46m 31s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/94791f176ee948e19fb92307b0ad10d1 : SUCCESS in 50m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3be2ac04c56a475982ad5451807f6c46 : SUCCESS in 7m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bc9f7e0ff884314bd5b4aafd7b0a29f : SUCCESS in 3m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/da540ba4f30843f8940f68368e7cfc8b : SUCCESS in 4m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/cbdf587bfc684891829e346184f5be09 : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bebdb8dec85c477584167e7001cddeeb : SUCCESS in 9m 21s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/00cc7e9924d9424695f64314cbb5e1a3 : SUCCESS in 25m 24s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/85ac149b36ca4a40974a9a38bfcbebf6 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":14},{"id":"73dbddc35a2212e9b7036bf5f9cf3a06d4894b12","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-17 12:13:32.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"5a401f868ca28bc595500f70977865a28358b814","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 13:32:23.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a0c25e62e3cf43abb8d83194f13db08d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2cebc41205064e1099cbeefbe73ebbb8 : SUCCESS in 2m 26s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/4ee095604077475a8d9ed8f2d49fe281 : SUCCESS in 48m 36s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f8be30fb8acc4e389a5648f5f98ba511 : SUCCESS in 35m 27s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/9875cdd42ab040bcaf5a5b534128e466 : SUCCESS in 50m 08s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/d7f754a6217d41f2bdb64ab5ba7c0b72 : SUCCESS in 33m 53s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/6ec836b294c848a09af4b98849b0adae : SUCCESS in 50m 00s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a4d6604d45e141d5af968bab46209d52 : SUCCESS in 1h 00m 48s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/8800c858b75d4cd7813aa1a09b3c1fe6 : SUCCESS in 1h 13m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bd4225e7c6134fe08f998a9327a9354c : SUCCESS in 52m 30s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/81993d66099d4586bb03bd83190eb8ed : FAILURE in 48m 05s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/6381c6fba1ab4b92962a60c21c163e9f : SUCCESS in 42m 11s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/153d61de1a054c79a35090135e5f6cd5 : SUCCESS in 47m 52s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/580ff1e9a53148aaa84287f710ca2108 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9bec2e420b341769dac375620ff3329 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6a2fd69a77fb4896bc1277937db15bae : SUCCESS in 4m 47s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c03af3c9cab4f1eb36a6fabe00288ba : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/238dd70f53bb46c1b5395e82c73a74ce : SUCCESS in 7m 41s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/50dd2f992b374dffa2410faeaad58165 : SUCCESS in 31m 59s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/4499105d53b644aa86e65bbbe9fa6bc6 : SUCCESS in 5m 27s","accounts_in_message":[],"_revision_number":14},{"id":"0fae3513e87843d33c8956a929b6f9c8571a65f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-11-05 14:34:02.000000000","message":"Uploaded patch set 15: Patch Set 14 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"c953465659563bb5747a8a57b46d34e590d79825","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-05 15:58:48.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e41ae92892342e3be2399d7e9d99bff\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/761b2ab6787548d1b2d4832bb338c8ae : SUCCESS in 2m 41s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/b1582e8d094645ca82589688d7816a87 : SUCCESS in 50m 50s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/4e2d5064276c44368622fea240a0c9e4 : SUCCESS in 48m 45s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/fc8827a2ce85453eb13c1b93cd0bb483 : SUCCESS in 42m 18s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/7b61a62e5a1a4fefb3361680a3f0a377 : SUCCESS in 31m 42s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/69e317dceaf849439839eec927152786 : SUCCESS in 1h 09m 34s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/35d2b6651ff64412845f88bb5d25cbe5 : SUCCESS in 1h 07m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/160038f3156940f2a736121c601d2c63 : SUCCESS in 1h 19m 33s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/0ffd737997c345f490074a44664f2bd1 : SUCCESS in 1h 13m 12s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/b5d30bcf510f42678259f517a58d5e09 : SUCCESS in 49m 28s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/730270c9525a4a97a6219ab21b4ea852 : SUCCESS in 57m 19s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/28f92b8c934a498385ade7014979a9f9 : SUCCESS in 46m 31s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/74704659b9834a198798619572270c68 : SUCCESS in 7m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed5b6befe183487883ffad01cf2e3438 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/963d30b806484b27b09a02819161d706 : SUCCESS in 5m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/907705f0ca8a407489f1942d6abc4a5e : SUCCESS in 8m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aea2127059e7493baec524095c46520b : SUCCESS in 10m 00s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/8f6ea3b87dbf45b1bfc9549b92b2488f : SUCCESS in 39m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/d9af728fb58641a1b0d34ee4ba9251e5 : SUCCESS in 5m 33s","accounts_in_message":[],"_revision_number":15},{"id":"95f4168620247b0bf8c9ebb0b9f090fe4ac0e13b","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-11-07 00:52:57.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"c6a3c1e7c70873204bff015cdd8da7e5755b24df","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-11-07 10:09:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ed2747992499207455acdec29cceb94f48c6df06","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-11-14 16:07:03.000000000","message":"Patch Set 15: Code-Review+2","accounts_in_message":[],"_revision_number":15},{"id":"c9da564fc9e4fbd588e688bdabe155add85a6c6a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-09 15:14:39.000000000","message":"Patch Set 16: Patch Set 15 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"e45d2d1a8012c23e350777e633c7a8acddcb0e1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 17:10:01.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f0d4a01038f04712bb78b3631b0e10dd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fb5f055a6a8a4eba8d75c3612681201e : SUCCESS in 2m 29s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/d7674c95d54d44ae98fef1bce30fd0d9 : SUCCESS in 48m 54s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a5b16422b1c043f1bd39fa7d59b003f0 : SUCCESS in 47m 52s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/368e2852e0394b03b08e62fc2bb256ad : TIMED_OUT in 1h 14m 36s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1ca406f791374bca9cf376988af9bbf9 : SUCCESS in 35m 15s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/fb6d36272e1f426b93a6353bfe92fb19 : FAILURE in 1h 15m 29s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/120dd10f91f14cc09f1f12c70effe949 : SUCCESS in 1h 11m 12s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/88e07d2dffb44b98ad46f36b0ebfe116 : SUCCESS in 1h 06m 43s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/51332d8e53024a66b9cd7a2fadb3ea2a : SUCCESS in 1h 09m 45s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d7fd1bd573bf4c799472b8970c8768ac : SUCCESS in 39m 29s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/485d1b8c99ca4750ae0894801f0334ee : SUCCESS in 47m 52s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/1cb72d1986b34f839df1de40198032f9 : SUCCESS in 1h 36m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa4264fdc8654edd925c7523d71a49e7 : SUCCESS in 7m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2da667b129384e2dac45edf02d0de6cc : SUCCESS in 3m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ee88c2b437ff415f979587f35e1cf263 : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0964c4d084ef42049bcc03d736025d29 : SUCCESS in 8m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbfe736baf004747b87e1934fc23c9be : SUCCESS in 8m 48s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/b4cf2d8b7f8043da91cad502746382f0 : SUCCESS in 32m 55s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e384872bbcbe4c79ad34436e817cb829 : SUCCESS in 5m 31s","accounts_in_message":[],"_revision_number":16},{"id":"f97b0b71c3b3dfe0f1316c27a26cf4fbe301b130","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-09 17:23:02.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"60f853eaedacdea4bb83133cadefa2aeac9eba9c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-09 18:46:49.000000000","message":"Patch Set 16: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf9cc5890d414242a5d80ec7a274a6b5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7f6574f17f7f4bd48f090cce13621280 : SUCCESS in 2m 31s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/bd89cf05b38d4633a3871987f58bd534 : SUCCESS in 54m 43s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/cbe09d5b7c534bebaf82937bf99c5d7f : SUCCESS in 47m 47s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/45bc4c1bfdf54a8086b785cb1e75c93d : SUCCESS in 44m 46s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/0ed5134ac4ea48e59699ff3114697a66 : SUCCESS in 15m 27s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/162f2fe5a4554eb6902a0dd1ffdc4abe : SUCCESS in 56m 28s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/d5768d9df31b44d1b42edfc06a98b482 : SUCCESS in 1h 04m 41s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/b5afd0300af84d76b872df10190a6ad8 : SUCCESS in 51m 46s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/ae9d543b07214435b1f0fca74991f237 : SUCCESS in 1h 07m 45s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/69a2134ee1cf45b193c106c0a3d5463c : SUCCESS in 52m 31s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/93ab381f827b40a48560ec040e40d7bd : SUCCESS in 46m 47s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/f74e5896d97049af997f882f1558f6f0 : SUCCESS in 47m 45s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/464aa822470d4f6facfff1a4472a6c5d : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/12c60f7743b94833a4a401551359ad7a : SUCCESS in 4m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad337476f2654acbb0870ca71ab70d19 : SUCCESS in 6m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d1c026fbd8a04029a873993628369a8b : SUCCESS in 9m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b721c1fa5f74877aac7b9dd6e63cb56 : SUCCESS in 8m 42s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/4a7760ee64c54829a0133090f2f3b8c9 : SUCCESS in 33m 31s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/50333cab163e4501973df1def74a8b36 : SUCCESS in 6m 54s","accounts_in_message":[],"_revision_number":16},{"id":"cbb28289e24a2b5b64cd1ceff1874f86c49aa075","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-12 21:06:49.000000000","message":"Patch Set 17: Patch Set 16 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":17},{"id":"d8322502173a7729de5dfcc0716ec0ba5b572b19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-12 22:27:39.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be12e0a408d44a6985ddd693a7681730\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a8a39e6a177d41c98ec65cbf4cc147e4 : SUCCESS in 1m 45s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/f6640bbc093046a19c2717dc15cac021 : SUCCESS in 52m 49s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/a37678be94be461eaf93a39393b18123 : SUCCESS in 46m 13s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/a3a2c8fc76144f9ca381d768a323d4ed : SUCCESS in 47m 46s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/f217c9a4fe584c5493a3e645f0e71abb : SUCCESS in 21m 29s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/1915836dd55f4119a4d58298379c747c : SUCCESS in 1h 07m 46s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/19d31f62c47148bc926181ec2bb1c81c : SUCCESS in 1h 08m 41s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/4472650ea6464b7d861bb4d95b55b83c : SUCCESS in 48m 53s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/1d7d7ad8718142e3b3607eba37312dd1 : SUCCESS in 1h 11m 35s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/4732f46c81b5440f8ccbce3412a6624b : SUCCESS in 48m 04s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/9a9a49abfc764423a84d1e9791ec1509 : SUCCESS in 42m 37s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/6ba6ef3f895d40a09f327a6de8d782e8 : SUCCESS in 50m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1263ef0f46244d33bedc706a07182e09 : SUCCESS in 4m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c473a28c998143478d5e11e10c882273 : SUCCESS in 2m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cacf65666565409b95c56c3a43c9aa91 : SUCCESS in 4m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a100b43937c842f599a7e575e1fe438c : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca064d92b9c94a4792da4c5c54754238 : SUCCESS in 7m 18s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/c9106d654a5447f29928e6bf18869013 : SUCCESS in 34m 12s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/5caf5fb9d38f46d88a0ecc95c2f37ef2 : SUCCESS in 4m 45s","accounts_in_message":[],"_revision_number":17},{"id":"56f41b9808e4c917b32910e430a8dcfcbcc795a3","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-01-15 22:25:40.000000000","message":"Patch Set 17: Code-Review+2 Workflow+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":17},{"id":"d13c710f7ec73a32f17721086a738b54bdd7da12","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-20 19:05:50.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"db89b654ff9316b8503e5d946d2e06f02d5a8ddc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-20 19:06:24.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":18},{"id":"8f636aa1cfa0b2f8576bf045eed823d98f8a1957","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-20 20:26:06.000000000","message":"Patch Set 18: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f1b06a4a5df14af3a855d30cf14dbdfb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/cbdac4fbbd844b21ad05e10263f506e9 : SUCCESS in 3m 45s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/49992563cf854b07b9f6f5c68de10f72 : SUCCESS in 49m 56s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/ee0e54f684304b74bde408785d595bb6 : SUCCESS in 49m 35s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/0e1a78d7ae724a198742897324a8a1ae : SUCCESS in 51m 26s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/fd9a9ae76503408485f528a10875ea7f : SUCCESS in 40m 57s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/9d6f9f7989b448e9b8aa2c1f42fd98db : SUCCESS in 1h 08m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/023099ffaeae48e5b60dc37ef1449322 : SUCCESS in 1h 09m 36s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/7bfca06dd10e448983016db9b1c5f159 : SUCCESS in 1h 06m 39s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/f2305cfd2abb4b0983e40a989bb8e799 : FAILURE in 27m 55s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/b6f23ae5bf04437d8dae2693c784812a : SUCCESS in 50m 57s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/5ac34e7c60d742bc8626249999fad450 : SUCCESS in 55m 05s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/2a41d2d8f3a441cea5fa9da8c258deb1 : SUCCESS in 50m 22s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/94f66522d0c2456a85b7f7158254b974 : SUCCESS in 7m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab61cd736f98439aa9211a599a995657 : SUCCESS in 2m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3d98ce42dbff4e99bd447346cf1ce49d : SUCCESS in 5m 13s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/491741b9962548ffb2f852ccdeced328 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce0741fb6adf430f99a39f1896c56295 : SUCCESS in 8m 58s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/cf8c5f7c73324be8b0d6e33e717fd944 : SUCCESS in 32m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/21bf69665d364259a5f5a9c35a25f3f4 : SUCCESS in 5m 07s","accounts_in_message":[],"_revision_number":18},{"id":"6e0f0423ec053668c386ffccc6160c32e92faaf9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-21 00:36:45.000000000","message":"Patch Set 18:\n\n(1 comment)","accounts_in_message":[],"_revision_number":18},{"id":"98627000f8d2805a71e3e69bba6baf292b37e60f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-21 01:49:52.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e3ec6ba3b976463187221f34ccb1c17e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4feb8f92a07d45b9bad121ef120fc99f : SUCCESS in 2m 25s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/ad1b41a1e7bc4756a6510b47b2899f81 : SUCCESS in 45m 17s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/f1db7a517da444fc89fa77f0323daad5 : SUCCESS in 53m 03s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/46ade91cf9474bb484bf8466b66a6c6f : SUCCESS in 47m 00s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/ea6144b44bf249938d560fcd29bb8729 : SUCCESS in 30m 58s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/5e5bdb0a1c2749a7a4e6b232b85af975 : SUCCESS in 1h 02m 10s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/4f2db08705014643ad390e0ae8be64eb : SUCCESS in 52m 33s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/782f06a895284341a118e853ab919753 : SUCCESS in 1h 07m 04s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/908a266debea448cb3dbc823f2ab3834 : SUCCESS in 1h 06m 30s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/d5d93461892042fcb0321d8a693fbd75 : SUCCESS in 47m 00s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/0626bba74f2841c98dac361cba2d497e : SUCCESS in 42m 31s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/eecb08cfbe374549bed5e6e12e060096 : SUCCESS in 52m 36s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f82d9cf78e6c4fe4b15fb49edc7903ed : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a26d207c1804485ae57e2d142fb16b6 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ee0665ceb7f46a1852a70cdd4148fba : SUCCESS in 4m 25s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/48f35c95ae1c48868584261b6e32731d : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d15083549334342a8ae0c4ce53591d0 : SUCCESS in 5m 23s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/41f45ce5b835463589c24c7272b0f9d9 : SUCCESS in 33m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/414752d859c4404c962ad35d8b525b81 : SUCCESS in 5m 30s","accounts_in_message":[],"_revision_number":18},{"id":"93a64d11d31d3c37d72d69870a78779050da8390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-01-23 12:16:07.000000000","message":"Patch Set 19: Patch Set 18 was rebased\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":19},{"id":"e372b0139f568c5f20deddba5684eaeb2e3dcf7c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-23 13:42:56.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/675d43db864a4328acdf0ca7bea9a974\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/43947503fc80433493b7710dab6a6a2f : SUCCESS in 1m 57s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/75e628561fb24f97a9a9f8c849903e5d : SUCCESS in 47m 47s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/5b1af3eae3f8402c9144b75e92384860 : SUCCESS in 51m 36s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/26ba5116fa74469b858802206469ff54 : SUCCESS in 42m 51s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/a5cfa313e99f4ce1ab1691d9418b539e : SUCCESS in 36m 40s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/30ee6bbba6234f2d99606fee345f9102 : SUCCESS in 1h 08m 59s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/412c75e410364224b02d58beb3e61fe7 : SUCCESS in 1h 14m 11s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e9dc127e59c242b58fe627f9de36d21a : SUCCESS in 1h 05m 32s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/39a736c6941a494dafaf8a8436fdf777 : SUCCESS in 1h 11m 46s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/a195d99c1dba4faabf97f0059cc14f5f : SUCCESS in 39m 28s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/22472e52eeb943558e6fdfd615dc8a70 : SUCCESS in 37m 30s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/ee7f5dac84244fd1b62eea67975fcb10 : SUCCESS in 35m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d0066d888d074092bc3127877cfef1da : SUCCESS in 6m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2be5c01910be40b8bf9ff70a03e72d37 : SUCCESS in 3m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2747081f987a4133b9a5f461735f97c5 : SUCCESS in 4m 38s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/d760d63eedf44916a41557f49669e499 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c728944097404696b13ab2e0ec56b9c3 : SUCCESS in 8m 28s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/f24585d56c5b4f33a3580021252a047f : SUCCESS in 33m 41s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/23e2a003e94d438d9af77deeb625198e : SUCCESS in 5m 20s","accounts_in_message":[],"_revision_number":19},{"id":"8fa5672f946dfdd845220fae19dac8a019e50994","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2026-02-05 21:22:23.000000000","message":"Patch Set 19: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"57f54f08adf5cd5e4a943e05545dd400370b37c4","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-02-13 09:41:53.000000000","message":"Patch Set 19: Review-Priority+2","accounts_in_message":[],"_revision_number":19},{"id":"ddae2e6ad2cb1ceb71ff82e220419925660c3e0a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 09:17:45.000000000","message":"Patch Set 19: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":19},{"id":"1866a7a9982aa7c75ab7d88654ecf7e7cb33a316","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 10:37:11.000000000","message":"Patch Set 19: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e9272477e45e4268a31008872b5a0594\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/56871bb873fb449b897377d0bebe13b6 : SUCCESS in 2m 58s\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/7975b1aedda9436ba555cbfbd7d08a64 : SUCCESS in 46m 01s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/3f304acd1b3c495786ee4ca721e5c31e : SUCCESS in 49m 37s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/522bbd6f168245079b9944d104d24dc4 : SUCCESS in 52m 45s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/56b518f906fa4d5589098c2c6a924d9a : SUCCESS in 22m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6a87c6d3450f495c8ec1f357e281a495 : SUCCESS in 53m 43s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/ca61829a954d4a4e8b09d4fdb79871b7 : SUCCESS in 59m 07s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/fe0dfeceb7ca43f8818569a73bbdc8ed : SUCCESS in 1h 08m 00s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/79e5ddadb48a4006bd5cf1784cc5e872 : SUCCESS in 1h 06m 11s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b897b36efa314c09bdb352a48f920f92 : SUCCESS in 43m 13s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/5abf9494da4c446b90e6a29871461a11 : SUCCESS in 51m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2a3715bcd5244d8841308af329c25a3 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/938fe57a045447659ff8c320ec176c82 : SUCCESS in 5m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bcd8954b25694593ad4d7ad3a93edf3c : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06d34c88a99c42c4aec37d5a968c1dcd : SUCCESS in 8m 10s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/6f6b263ea00d4999a728efdaec9769c5 : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":19},{"id":"e19ee87a393f7b759d6da19cb7552e816f87d2d0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 10:37:11.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":19},{"id":"6052f0e8546c34a0de0194148045d9de8f76a817","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-16 10:38:08.000000000","message":"Patch Set 19:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb6a208291c345a185b86fcc97a69d84\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/40903077b6ec4e6c94f5c7c0212e92e7 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":19}],"current_revision_number":19,"current_revision":"f25a4b6f6fa91332fd4ecc55d02f31621cb7cd64","revisions":{"02b75713a92bf44d02492815ab7c273232ec5920":{"kind":"REWORK","_number":1,"created":"2025-08-06 15:17:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/1"}}},"commit":{"parents":[{"commit":"85c0388cf439b84bd6bf54d3d75f7d56e349f4cc","subject":"Migrate sink service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/85c0388cf439b84bd6bf54d3d75f7d56e349f4cc"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:16:50.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02b75713a92bf44d02492815ab7c273232ec5920"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/02b75713a92bf44d02492815ab7c273232ec5920"}]},"branch":"refs/heads/master"},"a55a8d6ba9bee556516c1dbbb95b90908f7e18a5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-08-06 15:36:11.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/2"}}},"commit":{"parents":[{"commit":"5f51f1fd11e0d1fb3ac6082001a2b1876a1a218b","subject":"Migrate sink service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5f51f1fd11e0d1fb3ac6082001a2b1876a1a218b"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:36:03.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a55a8d6ba9bee556516c1dbbb95b90908f7e18a5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a55a8d6ba9bee556516c1dbbb95b90908f7e18a5"}]},"branch":"refs/heads/master"},"79fec7b42dbda1a009a36e434ce7545a76348a61":{"kind":"REWORK","_number":3,"created":"2025-08-07 09:10:21.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/3"}}},"commit":{"parents":[{"commit":"df2f6e133bbc9c5bbb6f90dd8efc844e2ab1b2fb","subject":"Migrate sink service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/df2f6e133bbc9c5bbb6f90dd8efc844e2ab1b2fb"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-07 09:10:06.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79fec7b42dbda1a009a36e434ce7545a76348a61"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/79fec7b42dbda1a009a36e434ce7545a76348a61"}]},"branch":"refs/heads/master"},"175a1a916f9364d81c5b45cecc582662d852c8e0":{"kind":"REWORK","_number":4,"created":"2025-10-07 16:36:39.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/4"}}},"commit":{"parents":[{"commit":"32b85398bd51b69b7bb35de6a47d6e5d31904da0","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/32b85398bd51b69b7bb35de6a47d6e5d31904da0"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-07 16:36:02.000000000","tz":120},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/175a1a916f9364d81c5b45cecc582662d852c8e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/175a1a916f9364d81c5b45cecc582662d852c8e0"}]},"branch":"refs/heads/master"},"a33cf70d6092ca149e3855e87bd666fef6ca32a2":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2025-10-08 14:46:18.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/5"}}},"commit":{"parents":[{"commit":"6044597564207dd06ce7458f3fe8e9871940dfa4","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/6044597564207dd06ce7458f3fe8e9871940dfa4"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-08 11:42:35.000000000","tz":120},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a33cf70d6092ca149e3855e87bd666fef6ca32a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a33cf70d6092ca149e3855e87bd666fef6ca32a2"}]},"branch":"refs/heads/master"},"178068cef62c4e1fcbec278a9f8a0ab91a2f8ae0":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-10-08 15:09:42.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/6"}}},"commit":{"parents":[{"commit":"e8b3616f116e7300a83a6db6a821a40494e88f84","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e8b3616f116e7300a83a6db6a821a40494e88f84"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-08 15:09:05.000000000","tz":120},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/178068cef62c4e1fcbec278a9f8a0ab91a2f8ae0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/178068cef62c4e1fcbec278a9f8a0ab91a2f8ae0"}]},"branch":"refs/heads/master"},"c28065a2a89bbf9805a66d2162efa7932d656b44":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-10-09 08:54:56.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/7"}}},"commit":{"parents":[{"commit":"3820b763c6e40f86ca1387c9da3197900f1bc38a","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3820b763c6e40f86ca1387c9da3197900f1bc38a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-09 08:51:05.000000000","tz":120},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c28065a2a89bbf9805a66d2162efa7932d656b44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c28065a2a89bbf9805a66d2162efa7932d656b44"}]},"branch":"refs/heads/master"},"101606a344b85004f26629d95e4b5b56a18d7e1e":{"kind":"REWORK","_number":8,"created":"2025-10-14 12:28:12.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/8"}}},"commit":{"parents":[{"commit":"d6b4b6f96c361b108da54aaf07e1f77bfb961b43","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/d6b4b6f96c361b108da54aaf07e1f77bfb961b43"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 12:22:09.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/101606a344b85004f26629d95e4b5b56a18d7e1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/101606a344b85004f26629d95e4b5b56a18d7e1e"}]},"branch":"refs/heads/master"},"4264802de39ca4563f547c37bfcd18265fd8d485":{"kind":"REWORK","_number":9,"created":"2025-10-14 13:00:16.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/9"}}},"commit":{"parents":[{"commit":"f5a25a08fe3ccdfc6ba65b4a1c474dffe6df9426","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f5a25a08fe3ccdfc6ba65b4a1c474dffe6df9426"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 12:59:59.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4264802de39ca4563f547c37bfcd18265fd8d485"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4264802de39ca4563f547c37bfcd18265fd8d485"}]},"branch":"refs/heads/master"},"1278b7fd422aad850e9e19c4ccc41ddc0b1d23da":{"kind":"REWORK","_number":10,"created":"2025-10-14 17:55:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/10"}}},"commit":{"parents":[{"commit":"9f88624b37d1a9d7a1e500fc5c0d31a12b734333","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/9f88624b37d1a9d7a1e500fc5c0d31a12b734333"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 17:48:51.000000000","tz":120},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1278b7fd422aad850e9e19c4ccc41ddc0b1d23da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1278b7fd422aad850e9e19c4ccc41ddc0b1d23da"}]},"branch":"refs/heads/master"},"c0c5332b67d8b5e36995f46510640a2ed627ba3a":{"kind":"REWORK","_number":11,"created":"2025-10-14 20:54:03.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/11"}}},"commit":{"parents":[{"commit":"bcbacc05d5bfcd7828512564af48570d14579eec","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bcbacc05d5bfcd7828512564af48570d14579eec"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 20:49:39.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0c5332b67d8b5e36995f46510640a2ed627ba3a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c0c5332b67d8b5e36995f46510640a2ed627ba3a"}]},"branch":"refs/heads/master"},"a6d7fd98e9be6c4b47b0a7601aa78d10bfe43ac6":{"kind":"REWORK","_number":12,"created":"2025-10-14 23:27:54.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/12"}}},"commit":{"parents":[{"commit":"36f510f1e95676c02048a241a8e86efb5d3610f3","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/36f510f1e95676c02048a241a8e86efb5d3610f3"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-14 23:27:34.000000000","tz":120},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6d7fd98e9be6c4b47b0a7601aa78d10bfe43ac6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a6d7fd98e9be6c4b47b0a7601aa78d10bfe43ac6"}]},"branch":"refs/heads/master"},"2360e0b1cce28ebe7ea28bdf9b29eae04dc8c35d":{"kind":"REWORK","_number":13,"created":"2025-10-15 13:11:08.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/13"}}},"commit":{"parents":[{"commit":"0b205d526bc9199e780f452ec50f6d3f9814eae1","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0b205d526bc9199e780f452ec50f6d3f9814eae1"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-15 13:08:56.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2360e0b1cce28ebe7ea28bdf9b29eae04dc8c35d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2360e0b1cce28ebe7ea28bdf9b29eae04dc8c35d"}]},"branch":"refs/heads/master"},"e4939a564a882d8b4c3c445495a3def3c3365f03":{"kind":"REWORK","_number":14,"created":"2025-10-16 09:52:17.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/14"}}},"commit":{"parents":[{"commit":"ec5d0e70a80c0c0eac7a371b18b06472a9e5e37c","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/ec5d0e70a80c0c0eac7a371b18b06472a9e5e37c"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-10-16 09:30:14.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4939a564a882d8b4c3c445495a3def3c3365f03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e4939a564a882d8b4c3c445495a3def3c3365f03"}]},"branch":"refs/heads/master"},"f2b6398e9a5dcd5e5473fee882ad4bd65354e00b":{"kind":"TRIVIAL_REBASE","_number":15,"created":"2025-11-05 14:34:02.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/15"}}},"commit":{"parents":[{"commit":"be0df9fa86954056db3aca358de156bbdbbcac6f","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/be0df9fa86954056db3aca358de156bbdbbcac6f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-11-05 14:31:25.000000000","tz":60},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2b6398e9a5dcd5e5473fee882ad4bd65354e00b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f2b6398e9a5dcd5e5473fee882ad4bd65354e00b"}]},"branch":"refs/heads/master"},"1b59dc24d8766aa372fc2d6490638aee59489cb4":{"kind":"TRIVIAL_REBASE","_number":16,"created":"2026-01-09 15:14:39.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/16"}}},"commit":{"parents":[{"commit":"4599190fdeaec2593fd5184a9f432be3554d9b2f","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/4599190fdeaec2593fd5184a9f432be3554d9b2f"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2026-01-09 15:14:39.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b59dc24d8766aa372fc2d6490638aee59489cb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/1b59dc24d8766aa372fc2d6490638aee59489cb4"}]},"branch":"refs/heads/master","description":"Rebase"},"49fa6ff0db71c539328b5c8954071aae1bc4e95b":{"kind":"TRIVIAL_REBASE","_number":17,"created":"2026-01-12 21:06:49.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/17"}}},"commit":{"parents":[{"commit":"003240100e026683a323a6148a4e8863bedaead7","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/003240100e026683a323a6148a4e8863bedaead7"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2026-01-12 21:06:49.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/49fa6ff0db71c539328b5c8954071aae1bc4e95b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/49fa6ff0db71c539328b5c8954071aae1bc4e95b"}]},"branch":"refs/heads/master","description":"Rebase"},"67c4ec3f3589254ffe3388e88e96c3cef218b461":{"kind":"REWORK","_number":18,"created":"2026-01-20 19:06:24.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/18"}}},"commit":{"parents":[{"commit":"5c45d30a30cb5f56bb04906667a4dd7f59e96b04","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5c45d30a30cb5f56bb04906667a4dd7f59e96b04"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer","email":"oschwart@redhat.com","date":"2026-01-20 19:06:00.000000000","tz":60},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/67c4ec3f3589254ffe3388e88e96c3cef218b461"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/67c4ec3f3589254ffe3388e88e96c3cef218b461"}]},"branch":"refs/heads/master"},"f25a4b6f6fa91332fd4ecc55d02f31621cb7cd64":{"kind":"TRIVIAL_REBASE","_number":19,"created":"2026-01-23 12:16:07.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/97/956697/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/97/956697/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/97/956697/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/97/956697/19"}}},"commit":{"parents":[{"commit":"7bce7f1f41d8ca00344f2985c6baddeef8d8fbac","subject":"Migrate producer service to use native threading","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/7bce7f1f41d8ca00344f2985c6baddeef8d8fbac"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2025-08-06 15:09:24.000000000","tz":0},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2026-01-23 12:16:07.000000000","tz":0},"subject":"Migrate worker service to use native threading","message":"Migrate worker service to use native threading\n\nThis commit migrates the worker service from eventlet to standard\npython threading.\n\nKey changes include:\n* Worker service main entrypoint (designate/cmd/eventlet/worker.py)\n    now explicitly initializes oslo_service threading backend under\n    designate/cmd/threading/worker.py.\n    It will be removed when the migration is over.\n* Setup.cfg path change.\n* Removing eventlet usage from do_axfr function in\n    designate/dnsutils.py.\n    The DNS query now uses the native timeout parameter.\n\nChange-Id: I3fd123451dea9cbd1615211b41ae1c085537129b\nSigned-off-by: Omer \u003coschwart@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f25a4b6f6fa91332fd4ecc55d02f31621cb7cd64"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f25a4b6f6fa91332fd4ecc55d02f31621cb7cd64"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
