)]}'
{"id":"openstack%2Fdesignate~956879","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd","project":"openstack/designate","branch":"stable/2025.1","topic":"add-service-clean-cmd-stable/2025.1","attention_set":{},"removed_from_attention_set":{"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2026-06-22 14:01:15.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd","subject":"Add `designate-manage service clean` cmd","status":"ABANDONED","created":"2025-08-08 12:19:51.000000000","updated":"2026-06-22 14:01:15.000000000","total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":949126,"cherry_pick_of_patch_set":12,"meta_rev_id":"b9a47934cda9745e77505e77f98e0e711f19302c","_number":956879,"virtual_id_number":956879,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2025-08-08 13:17:53.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":-2,"date":"2026-03-16 16:44:23.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-08 12:19:51.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"CC"},{"updated":"2025-08-08 13:17:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-08-15 20:45:41.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"CC"},{"updated":"2026-03-16 16:44:23.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"4289f68c45d6a61a12dd642b27a6a5df95290cb2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2025-08-08 12:19:51.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c826c35bc319e52b75c176e06a59fab2dd748150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 13:17:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/15850fe55adf4f3098c3ddac0bf2744a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/07045b53187c4505b3eb5a2b98c32fdf : SUCCESS in 2m 04s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/bd11f74343674a56b09f514a54063cc7 : SUCCESS in 40m 12s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/4595431631ca4ac1be6b143442b9b005 : FAILURE in 41m 18s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/fd0126cc04da4fb1ad2ee02b73f0b4cc : FAILURE in 33m 31s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/1bc2aa0ab1974bd8a5ab8afedff64ac8 : SUCCESS in 43m 09s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/ff2e51163e5a43f786b16e6d820f5438 : SUCCESS in 42m 14s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e7f8c6f646f344ac8e07698e15df06de : FAILURE in 43m 49s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/6c379d02960f408d9aaa38f15789bc25 : FAILURE in 43m 10s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/bbb1924dea9a4a09a343459c1220d6f5 : FAILURE in 41m 47s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/b3d658926c5042e4881787acf3a8b9f7 : FAILURE in 43m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/478bfc5a91244f35a4c45cf7c4493d10 : SUCCESS in 45m 20s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/d70773dd477c4757b0b27019ba31edf9 : SUCCESS in 55m 09s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8413911eb0b24270aa85429bcc9346a1 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ce97c15e4174504b69ed0b18f82dda7 : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ea20251cabb405d9434bdb5ec60fd7c : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d0e152e1055742388c6eebdff2861190 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/415687f912c54b20bfeae7f846aa21b9 : SUCCESS in 8m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/559709a2d8a24e2c8181bc45bd6dd999 : SUCCESS in 5m 40s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/ab9240c4099d47c48d108768be94683b : SUCCESS in 31m 32s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/c1bf26bc66a14dd0952a843bbb8e9d23 : SUCCESS in 6m 31s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/331188b6fa984449925f9a39dc8b9ae8 : FAILURE in 48m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"405f9b94f0a9aa7fb30f57b9fe2cb44f3e867403","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2025-08-15 20:45:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cd7aac757b5a484a3d79c5448a0faea68bf0673","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-15 21:35:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/437b3f595ec744da87f26cf137b2cc49\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/a75db0bb30f24531a30fb8c5d027a223 : SUCCESS in 1m 29s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/7f74a987ab92451fa800f109347b5f2e : SUCCESS in 44m 47s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/850766732a6f4208a82350e1bd4f8003 : FAILURE in 45m 10s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/1efe0c5447e1460fb6a0ec1a8a3369bd : FAILURE in 17m 00s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/a642fba3bbd344b3ad4feab393c31533 : SUCCESS in 37m 07s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/842912ed14cd4aec860adceddc722182 : SUCCESS in 38m 34s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/a4e607666f27416796cbe8feea4f1f51 : FAILURE in 37m 06s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/846ca1cfafaa49ebad8281d193286121 : FAILURE in 39m 08s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/e04c69004b3645cda41e748b44e40a20 : FAILURE in 23m 51s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/8010f49d672d4630bb707d935ffa91a3 : FAILURE in 25m 33s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/c515b1f90b5e45f8b0f06feea7b84e75 : SUCCESS in 45m 59s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/9ed01810fc8e42598bcce32d03dce1e3 : SUCCESS in 41m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c2c0bce36f348bc928dc6d68373822d : SUCCESS in 4m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97dea16a8c764e3f8bf1439198ef0f7c : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4130ff49791d4ec8b1691fe0d88a2328 : SUCCESS in 5m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/98f63f62b1044e6c8486deabb617c5ad : SUCCESS in 3m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba27afd97a5e47159513df860a382687 : SUCCESS in 10m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dd628ec8ac58435f87da1c81f2c17c44 : SUCCESS in 3m 45s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/84f7d3497fe449958cdfaf6967b13eaf : SUCCESS in 32m 07s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/ddf5009d95774d39ac22a8b773f27b41 : SUCCESS in 4m 48s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/20987ba1cde24293af24d6b0dbbbf8b1 : SUCCESS in 46m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ea9e5e8eceb7923c5a4dce67de86481ed2ddc1a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-03-16 16:44:23.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b9a47934cda9745e77505e77f98e0e711f19302c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-06-22 14:01:15.000000000","message":"Abandoned\n\nWe don\u0027t backport features","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2b086ed1b677808094788b9ef516d8d993999448","revisions":{"2b086ed1b677808094788b9ef516d8d993999448":{"kind":"REWORK","_number":1,"created":"2025-08-08 12:19:51.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/79/956879/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/79/956879/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/956879/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/956879/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/79/956879/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/79/956879/1"}}},"commit":{"parents":[{"commit":"a891d81974a4c3aa56a143252b26df9dcbd5fba5","subject":"Merge \"Fix neutron job for stable/2025.1\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a891d81974a4c3aa56a143252b26df9dcbd5fba5"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2025-08-08 12:19:51.000000000","tz":0},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n(cherry picked from commit 6c610e23ce3937087a86e2599c9ca18b4bf988c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b086ed1b677808094788b9ef516d8d993999448"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/2b086ed1b677808094788b9ef516d8d993999448"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
