)]}'
{"id":"openstack%2Fdesignate~956880","triplet_id":"openstack%2Fdesignate~stable%2F2024.2~I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd","project":"openstack/designate","branch":"stable/2024.2","topic":"add-service-clean-cmd-stable/2025.1-stable/2024.2","attention_set":{},"removed_from_attention_set":{"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2026-04-30 09:16:01.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd","subject":"Add `designate-manage service clean` cmd","status":"ABANDONED","created":"2025-08-08 12:20:49.000000000","updated":"2026-04-30 09:16:01.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"cherry_pick_of_change":956879,"cherry_pick_of_patch_set":1,"meta_rev_id":"990d120428d93f09b5f8570f26b92ece48c04622","_number":956880,"virtual_id_number":956880,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-08 13:40:10.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":-2,"date":"2026-03-16 16:44:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-08 12:20:49.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":12404,"name":"Rico Lin","email":"ricolin@ricolky.com","username":"rico.lin"},"state":"CC"},{"updated":"2025-08-08 13:40:10.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-16 16:44:11.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2026-04-30 09:16:01.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"a411923a20480359219c8d3b040f1e1c3baa7b61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2025-08-08 12:20:49.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2025.1.","accounts_in_message":[],"_revision_number":1},{"id":"19d9f0e352afc5edbf64c5ea241a48563d691df2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-08 13:40:10.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/524949e41ca242ac8edd88d4af860961\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4deebf45db23425d8ce71a83c50ab5d7 : SUCCESS in 2m 10s\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/76ef2a29a5ef471aa755ec6c07544d5d : SUCCESS in 41m 19s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/db631267812b41dcb06b13e412fd51ce : SUCCESS in 55m 43s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/4f24f7bc5e9847159388d1e1f708fb58 : SUCCESS in 56m 20s (non-voting)\n- designate-bind9-keystone-default-roles https://zuul.opendev.org/t/openstack/build/409c00351df443ea9118b73f39f4ea33 : SUCCESS in 44m 17s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/d1ba6f916cd644dd9a59e84709718a8d : SUCCESS in 43m 59s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/27d41a1f526a4f3487198b0397e3aa1b : SUCCESS in 1h 12m 52s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/fc54e4ecb66f45ad99bee8d82f0fd2ed : SUCCESS in 57m 43s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/dc8e2d0a580e48d189fc7becc95d1b7b : SUCCESS in 44m 51s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/b4ca9857a2d84975817a6acf659c5d7a : SUCCESS in 38m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/20c510d461114de9b1b71394b04c8daa : SUCCESS in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a652904ed96499580ea5124a4a2149f : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b5deda416ceb4b64802583eea42353e2 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2149dfb8c9fe45a59c1dafbcf334ad9b : SUCCESS in 8m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eed7762563fc4b1f84e2c90645d4105d : SUCCESS in 5m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb36e7a3dab84edfb875b12f1e63bf48 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b143098a5d2645ce823985b6c25bb287 : SUCCESS in 5m 23s\n- neutron-tempest-plugin-designate-scenario-2024-2 https://zuul.opendev.org/t/openstack/build/178d7a4d44c74d0ba827317f22617f46 : SUCCESS in 26m 26s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/e25945ae5dec4eebb0a4aeac66efe6ff : SUCCESS in 5m 37s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/80e8b5f7076b4ba1a2b0e8bc8f67aab0 : SUCCESS in 39m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"48dbde51ad36031a7567e61f0d8314fdd05da7f9","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2026-03-16 16:44:11.000000000","message":"Patch Set 1: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"990d120428d93f09b5f8570f26b92ece48c04622","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-04-30 09:16:01.000000000","message":"Abandoned\n\nstable/2024.2 branch of openstack/designate transitioned to End of Life and is about to be deleted. To be able to do that, all open patches need to be abandoned.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3dab87a634aa5e8cc4ae4824d022a90768710ef8","revisions":{"3dab87a634aa5e8cc4ae4824d022a90768710ef8":{"kind":"REWORK","_number":1,"created":"2025-08-08 12:20:49.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/80/956880/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/80/956880/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/956880/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/956880/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/80/956880/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/80/956880/1"}}},"commit":{"parents":[{"commit":"5bfc35864c28a97ded6104dea05fa9691a7f14d8","subject":"Add an designate-api-wsgi compatibility script","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/5bfc35864c28a97ded6104dea05fa9691a7f14d8"}]}],"author":{"name":"ricolin","email":"rlin@vexxhost.com","date":"2025-05-12 14:50:20.000000000","tz":480},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2025-08-08 12:20:49.000000000","tz":0},"subject":"Add `designate-manage service clean` cmd","message":"Add `designate-manage service clean` cmd\n\nAdd `designate-manage service clean` cmd to allow clean stale services.\nAny services not provide heartbeat within double heartbeat_interval time\nneeds to be consider as dead services. And should be removed ASAP.\n\nAnd in any case, even service is actually alive (maybe in a crazy rpc\ntimeout case), it can alway update and create it\u0027s service status back.\nSo we\u0027re fine to delete the service status in this case too.\n\nCloses-Bug: #2110262\n\nChange-Id: I211d0a60aa19ab4c0ae0fbb808e2d9080ccbbedd\nSigned-off-by: ricolin \u003crlin@vexxhost.com\u003e\n(cherry picked from commit 6c610e23ce3937087a86e2599c9ca18b4bf988c9)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3dab87a634aa5e8cc4ae4824d022a90768710ef8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3dab87a634aa5e8cc4ae4824d022a90768710ef8"}]},"branch":"refs/heads/stable/2024.2"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
