)]}'
{"id":"openstack%2Fdesignate~958394","triplet_id":"openstack%2Fdesignate~master~If23b7fa0e72ee086d8134776df3460df29c2dcf9","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-09-01 14:23:51.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"If23b7fa0e72ee086d8134776df3460df29c2dcf9","subject":"Add synchronized lock for recordset operations","status":"ABANDONED","created":"2025-08-25 08:56:34.000000000","updated":"2025-09-01 14:24:35.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"b37702c9d6f78c0b6dcbb7726208219ec6abf550","_number":958394,"virtual_id_number":958394,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-08-25 10:15:15.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-25 10:15:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-01 09:38:39.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"CC"}],"messages":[{"id":"56c08f027e56fe0c0bffe7d29f99d2ba531c1906","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-08-25 08:56:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5dfb10f85c86fdf8946fae2f4489df74abb6b9a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-25 10:15:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d17377ceeb834d968e5f31bc724e38b6\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/85aeac6239a84d53a6c63eb3336d476a : SUCCESS in 43m 28s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/a59fc614450c43c7baba480838b26c6a : SUCCESS in 45m 40s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/68bcbebf3fd44b6a9e5126a833998805 : SUCCESS in 38m 17s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/577928f6879844e996bb79718f2a6f62 : SUCCESS in 1h 05m 05s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/abf075679edd4a1aac79dc62bcf1c0e2 : SUCCESS in 1h 15m 55s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d4e0e8cc476642548e243c2add6c251f : SUCCESS in 1h 16m 38s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/de7ed46b099b4a1b9d4e38d431ef84f1 : SUCCESS in 1h 01m 45s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/f608048df3b4421bbbc0f8bb61f132ef : SUCCESS in 39m 37s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4efde6cad48c466aabe4baf5f271353d : SUCCESS in 51m 55s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8aadc8871fab471d94557d071a18c16b : SUCCESS in 6m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4124245845de4f168b11db6551f8e671 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/482d1e97bf6540f1b6faea4ace59085e : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/d6ac70b59fa9478f8347d84ea046cefb : SUCCESS in 5m 12s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/676e2d8596884ac89096ffcdf67a0442 : SUCCESS in 6m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbacb909aeea4a24b670aea55d29cd62 : SUCCESS in 5m 20s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/86b8fabd0f254c028d53cc541709c495 : SUCCESS in 30m 08s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/72320c54bbe14b478560d5a12e66adc5 : SUCCESS in 5m 32s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/ce42b88323d348f8bfb7ad87de9988f4 : FAILURE in 49m 19s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42f5305da530ed8a487e7ac7ef64210d6ae3f71a","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-09-01 09:38:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0d96e97a4b3436c235234da076dc582771cfb818","tag":"autogenerated:gerrit:abandon","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-01 14:23:51.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1},{"id":"b37702c9d6f78c0b6dcbb7726208219ec6abf550","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-01 14:24:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8b192f5498e113fce89dff8c983db55e35ccdb92","revisions":{"8b192f5498e113fce89dff8c983db55e35ccdb92":{"kind":"REWORK","_number":1,"created":"2025-08-25 08:56:34.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/94/958394/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/94/958394/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/958394/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/958394/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/94/958394/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/94/958394/1"}}},"commit":{"parents":[{"commit":"f307d62faf300222bfe62a1d52f850a8a2868121","subject":"Merge \"Add pyproject.toml to support pip 23.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f307d62faf300222bfe62a1d52f850a8a2868121"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-08-25 08:55:00.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-08-25 08:56:23.000000000","tz":120},"subject":"Add synchronized lock for recordset operations","message":"Add synchronized lock for recordset operations\n\n... in the central service because when we create,\nupdate or delete recordset it will touch the zones\nserial that can cause race conditions.\n\nChange-Id: If23b7fa0e72ee086d8134776df3460df29c2dcf9\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b192f5498e113fce89dff8c983db55e35ccdb92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/8b192f5498e113fce89dff8c983db55e35ccdb92"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
