)]}'
{"id":"openstack%2Fdesignate~958986","triplet_id":"openstack%2Fdesignate~master~I93c1c38de5ddf188523601d646b48a67c132a4ac","project":"openstack/designate","branch":"master","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-09-04 13:08:04.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-09-04 14:19:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I93c1c38de5ddf188523601d646b48a67c132a4ac","subject":"Migrate designate-bind9-keystone-old-default-roles to Noble","status":"MERGED","created":"2025-08-31 07:28:54.000000000","updated":"2025-09-04 14:22:22.000000000","submitted":"2025-09-04 14:19:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"958986","meta_rev_id":"ec1adfa60c8ba1917699dc61471cf2aab51e6cea","_number":958986,"virtual_id_number":958986,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-09-04 14:19:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-09-04 13:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-09-03 09:05:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-09-04 13:08:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-09-04 13:08:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-08-31 08:54:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-03 09:05:14.000000000","updated_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"},{"updated":"2025-09-04 10:43:42.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"062055fce013d9f4bc79a2cc906dcae8f569a47b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-08-31 07:28:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"92f6a96181e33a699962f03f601217a5160fe59f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-08-31 08:54:15.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd5b93be0b584dcabc0d1825de25a12a\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9a2e5256df344713b674f95644cac51d : SUCCESS in 43m 23s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/e68482d9ae5a4f8fa26f65fc9bad4eb8 : SUCCESS in 45m 05s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/1031d30529734d12ba6ca82b542f16e9 : SUCCESS in 45m 08s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/8ea6c1700e37491ca7c17e7beb684c3e : SUCCESS in 1h 19m 17s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/727dbd2b5cd1413d8c5c0960db22e8d3 : SUCCESS in 1h 11m 03s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/ff4699607b4d4fd89b45f14ed25385a4 : SUCCESS in 56m 23s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/05f7734786fa4c23b96532396e918ff8 : SUCCESS in 1h 03m 16s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/00ecc28e835c432ea5ef0eb48427496d : SUCCESS in 38m 05s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/a4200ec9b8304c2c8f44001916b76e7e : SUCCESS in 37m 57s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a45efa91d0ae442298d811e216e8f4af : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2a255ebb8354fffbe8811c45ad128be : SUCCESS in 3m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/810121cb482d4473927e4b947a3803c9 : SUCCESS in 4m 32s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6dac805da3e24ab3853520e28ba29bdb : SUCCESS in 6m 17s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c125830e0cf04e76b2d182c618c75b0c : SUCCESS in 7m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/79252946ffc446078637aacb1ba11990 : SUCCESS in 9m 14s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/0100e407459c4abdb5cdc9c7b5ae1d4c : SUCCESS in 32m 04s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/b18a8ecd5d0b4249b94eece8fdf80f9a : SUCCESS in 4m 59s\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/10e59ede7f0c4ecab8510ae245c91021 : SUCCESS in 43m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff89cf3993715f5a37415e643e48de1ac2eefcfc","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-09-03 09:05:14.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7892a7ae8fd20cf324b1603f9cad3a3e2069cd8b","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-04 10:43:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"21ba55e5b26e78d70e41ed38e6dbfa2faccf68e4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2025-09-04 12:44:10.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"863346c4639603e03d3e50a79eed657757e75964","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-04 13:08:04.000000000","message":"Patch Set 1: Code-Review+2 Review-Priority+1 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7507f9d7d895cf526c2ebfbf5adb848f92f8c0b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 13:08:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a93771233aba0783cdc50d04b5217441fdab285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 14:19:19.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/877a155a7d444cbc9e1753a70885336f\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/37a32be3165c4ba7a335a2f2f5971f2d : SUCCESS in 44m 23s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/c5fbec1a23b542e3bf89a809351a6a8d : SUCCESS in 44m 18s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/cf0c2512afaf4f5cadff926c69a6eff4 : SUCCESS in 48m 56s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/7f2d1f41417446bcb93cbe61aef9656e : SUCCESS in 1h 05m 27s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/bac5965c6c664fc4b73fdfc67f54d734 : SUCCESS in 50m 35s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/7feb0451360948c6bbefceb8acffbb6e : SUCCESS in 49m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/99320cf918df4b7d978f8cfdad4a674b : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dc27d1517ea47cfaab3166731396de3 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f721dfa4684348b9b921e9bc29d803ba : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8717e93eb61432e8cefa3c467a4df35 : SUCCESS in 5m 27s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/396199c0549b4479b10e1cb5f168f48c : SUCCESS in 35m 54s","accounts_in_message":[],"_revision_number":1},{"id":"bd15fd33a47003051669de81a12750106fc4676d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 14:19:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ec1adfa60c8ba1917699dc61471cf2aab51e6cea","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-04 14:22:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4af4d05e1ce413aae869688bf57a87f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/413d8b7b7df9463495907d589e4f6f3e : SUCCESS in 1m 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"15879d6d1a636cb7d43ec4987727e1b51a1522ec","revisions":{"15879d6d1a636cb7d43ec4987727e1b51a1522ec":{"kind":"REWORK","_number":1,"created":"2025-08-31 07:28:54.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/86/958986/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/86/958986/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/958986/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/958986/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/86/958986/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/86/958986/1"}}},"commit":{"parents":[{"commit":"f307d62faf300222bfe62a1d52f850a8a2868121","subject":"Merge \"Add pyproject.toml to support pip 23.1\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/f307d62faf300222bfe62a1d52f850a8a2868121"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-31 07:28:20.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2025-08-31 07:28:44.000000000","tz":540},"subject":"Migrate designate-bind9-keystone-old-default-roles to Noble","message":"Migrate designate-bind9-keystone-old-default-roles to Noble\n\n... because Jammy is no longer supported since 2025.2 .\n\nChange-Id: I93c1c38de5ddf188523601d646b48a67c132a4ac\nSigned-off-by: Takashi Kajinami \u003ckajinamit@oss.nttdata.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15879d6d1a636cb7d43ec4987727e1b51a1522ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/15879d6d1a636cb7d43ec4987727e1b51a1522ec"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
