)]}'
{"id":"openstack%2Fdesignate~959305","triplet_id":"openstack%2Fdesignate~stable%2F2025.1~I60ed37719fd0411159b8ec4f68ca7b001eb8a7be","project":"openstack/designate","branch":"stable/2025.1","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-09-29 15:12:17.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"16137":{"account":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"last_update":"2025-10-14 10:23:24.000000000","reason":"Change was submitted"},"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-10-14 10:23:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I60ed37719fd0411159b8ec4f68ca7b001eb8a7be","subject":"Remove next pagination link for floating IP list","status":"MERGED","created":"2025-09-03 11:48:59.000000000","updated":"2025-10-14 10:24:26.000000000","submitted":"2025-10-14 10:23:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"959305","cherry_pick_of_change":946228,"cherry_pick_of_patch_set":6,"meta_rev_id":"78ad2effa1185cd14c46dfa70878d12fa557420a","_number":959305,"virtual_id_number":959305,"owner":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-14 10:23:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-10-14 09:17:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-10-14 06:45:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2025-10-14 09:17:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-09-03 12:42:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-09-29 09:22:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2025-09-29 09:22:16.000000000","updated_by":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"reviewer":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"state":"REVIEWER"}],"messages":[{"id":"66206b2656176ea123bf08c24fbc3bc386bc3689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-03 11:48:59.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"1286b84d0a5d6a3b57ebb12bb9c610ac40bb68d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:42:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f819f97d72ff4ecbae685a335b7ec570\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/9c74b4c2d8ce45bda927e1080739e07f : SUCCESS in 39m 22s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/e3e41326bb0744bfa74425f29c0c155f : FAILURE in 26m 46s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/053c869964e54a638ce2404c4f00afbd : FAILURE in 15m 50s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/f6911cbfc3f74730b49db312b62cf8b2 : SUCCESS in 47m 15s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/4c797847761844ee96ddcb52778ca7e2 : SUCCESS in 43m 01s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/78e7cc880eba41eabb80ae6c2872bac6 : FAILURE in 34m 24s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/87fb6f244b754c3a97b20d25e9be1b60 : FAILURE in 37m 28s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/faa922477a0b4f369fb772533595c471 : FAILURE in 38m 57s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/a8851aea48ce46a487d453f614799436 : FAILURE in 31m 01s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/235d78cbd5834fe1999cc668ca810cb7 : SUCCESS in 42m 14s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/92d7f921fe7348eb8a58f68e4c2e6910 : SUCCESS in 40m 01s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7479cd92952a43f7b3438d47afe08c13 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58e4a6ff35634751893852699d4f3f29 : SUCCESS in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b0628b041afe48ae978094c6190373e1 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0c98a5ed87f940db8331e97e35164c9f : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d70e8755620b4585b6d9bf2ef050ffb9 : SUCCESS in 8m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/522f962bfb31445b84d233b56dc428eb : SUCCESS in 2m 42s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/8f35c8ef22fb4f0e8fbc21323ec364bd : SUCCESS in 30m 44s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/50fc47948f4c423a8a749224047b3f08 : SUCCESS in 6m 07s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/176cedee2cc142f8944271e03d439cf1 : FAILURE in 41m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6815bc24261c32139d61f6334f02d2cad4f0376b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-09-03 13:10:51.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a4309a2f94f3ccde0582e3859b124be85dce9643","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 14:19:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b604b34fb20c4b65994def77c65a3733\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/fa8adf596bb54f85a1911caf40f2e862 : SUCCESS in 50m 15s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/22e818c6f70c4ff684b271313c8a1bb9 : FAILURE in 16m 11s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/ef842b9f4681484f94d5b7231c48e890 : FAILURE in 23m 14s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/26992d02788543928627e87cc3f54f8f : SUCCESS in 43m 53s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/72ab7164795c4a899de0844f5764ddec : SUCCESS in 45m 11s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/85a71724d28c4fbb84def9c6f93be7b1 : SUCCESS in 1h 00m 47s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/172e60db614f4d60a17838a266a3219c : SUCCESS in 58m 03s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/86e829443c794e4db9da52e1b0ecef64 : SUCCESS in 51m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/9f91e3bd74bb4c608e5d64466be34d30 : SUCCESS in 1h 03m 15s\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/7f3642fbf4b54a1fb3d5c4cd41e6551b : SUCCESS in 42m 19s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/ee053407a1034f0eb30701a7a572d76d : SUCCESS in 44m 47s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4d7797815e4049bfb3754551962acfb7 : SUCCESS in 4m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7de363501bcb4535b6f379e94c551b18 : SUCCESS in 2m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13a9967cb7d74762b5170afb94ec0381 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/021efb7e68ec41089184074d5e2b8e7c : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce74fe7a2fbc42938966ec324d0efea3 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/16622960273c4c4aa0e10c3410ed1f6d : SUCCESS in 4m 57s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/0512ac3ce5f04c96bd6adec235be5624 : SUCCESS in 31m 05s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/1fb6ff1e755c4d6897287546bde1cc15 : SUCCESS in 7m 01s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/c03ebc7e6dea40048e8daab54a92e9d0 : SUCCESS in 44m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"defb9daef27cedd95b9a1289f068ea7073646dba","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-09-29 15:12:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e2c3527b4d503b0210c88732f582377fe10652be","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-09-29 15:32:47.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"b756acf8470d9d01edffc31ab009edf1c31671af","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"date":"2025-10-14 06:45:01.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"8208f481ad2f59f5943c996f2a2b4c15f6841560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 07:47:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d73970e989a04c2880121343d49e14f1\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/335b70a46ae944f2a2039b2e961d45b9 : SUCCESS in 37m 06s\n- designate-bind9-centos9stream-fips https://zuul.opendev.org/t/openstack/build/a7b8936c812a4bf4ad749631c6bae833 : FAILURE in 16m 08s (non-voting)\n- designate-bind9-centos-9-stream https://zuul.opendev.org/t/openstack/build/dec812fa15a343c2bf3bea69f061b7fd : FAILURE in 22m 31s (non-voting)\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/eacafaf1e9e94507b2cca4a0d0918a56 : SUCCESS in 42m 59s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/aab604d667a64fe2a35f42ea04124a39 : SUCCESS in 45m 07s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/be42a9849bd241e09f8defb1d7646f33 : FAILURE in 33m 31s (non-voting)\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/b56dfbd36893415996c57433872ba94c : FAILURE in 47m 00s (non-voting)\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d900511a362f4a978e13dd1dfaa1f3ff : FAILURE in 42m 25s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/bb7245ba66f7436c8704fe0053a54ea7 : FAILURE in 44m 09s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/ea2e27e39f4f4723aa9ac991676a16e2 : SUCCESS in 45m 40s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/4b8466b1d68c4b48ae78662d2f88fb91 : SUCCESS in 41m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/15d1261609284f8c9132de0f2f1b5de0 : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e94816dd0aa448b2bdec943424ae4c8e : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4c2940c5184f44a3b5753a89954e2d17 : SUCCESS in 8m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4bf0a2ddc1a8486fa71f6d1e2dfa3b25 : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c6863d62cc4ceb84eec00b97f16d63 : SUCCESS in 9m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50971e4c18c64422bc08641998e0bc7e : SUCCESS in 4m 55s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/bfd3b68203484232b1995504c8081f9e : SUCCESS in 34m 22s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/eeed86c61ca84df2b4a25f373a2bedb5 : SUCCESS in 3m 16s (non-voting)\n- designate-bind9-catalog-zones https://zuul.opendev.org/t/openstack/build/339003fcc8be492a80fad848c23564e4 : SUCCESS in 53m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"64c6ad8a4afa48bbd1218864ba345f04bdeec23f","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-14 09:17:28.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c340570920557192797c31da7d6631101c9f43f1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 09:17:51.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"692bdfec7e01e56904731a0d15dba41ea1dfcfb3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 10:23:24.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe977e22b6cb46c1841d9232361dc6a8\n\n- designate-bind9 https://zuul.opendev.org/t/openstack/build/a79a9f92ad3f4718bb73e3e85c957e93 : SUCCESS in 53m 19s\n- designate-bind9-keystone-old-default-roles https://zuul.opendev.org/t/openstack/build/7d27aa2c96a740febf8cf0296e4c6dfc : SUCCESS in 42m 14s\n- designate-pdns4 https://zuul.opendev.org/t/openstack/build/7c173ac0599b4efe87123488fdb284a5 : SUCCESS in 41m 20s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/a186c6b440194ce6a48615a604863a5e : FAILURE in 41m 56s (non-voting)\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/25fb75eeab8745f4ad917ac398202655 : FAILURE in 31m 18s (non-voting)\n- designate-ipv6-only-pdns4 https://zuul.opendev.org/t/openstack/build/b0d87ef3f92c4e18aa3a64eca12ddf59 : SUCCESS in 50m 12s\n- designate-ipv6-only-bind9 https://zuul.opendev.org/t/openstack/build/259ba075cc24472c8f113a7212e48adc : SUCCESS in 50m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ec21a7a37274250b7e335f0fa1f2e96 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9812ef1bd1347b1973ec6c91e4f0b79 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/cb41192a47a24212ada629eb049a6585 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bba17d8e1c7342f2bbbe6a3d7e8dccf7 : SUCCESS in 5m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d772201299884f5da40ffb4dfac4869a : SUCCESS in 5m 45s\n- neutron-tempest-plugin-designate-scenario-2025-1 https://zuul.opendev.org/t/openstack/build/58cb9b5182f142c18d8be562e1c806fa : SUCCESS in 32m 42s","accounts_in_message":[],"_revision_number":3},{"id":"f9d61218fcfb6df309fa9a38787b9efe67140bad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 10:23:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"78ad2effa1185cd14c46dfa70878d12fa557420a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-14 10:24:26.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8cc944b4b58644d7a75d6668ac29d019\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c0660603c4a41299b20d0bb514869b9 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e793dc24fa34fb3be35817fb6a411c6 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"bbb241f89620faa6d99ec98f78d77ca4dd32417a","revisions":{"febee3f7907f77fca72cd4b40e64d27f870a26b8":{"kind":"REWORK","_number":1,"created":"2025-09-03 11:48:59.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/05/959305/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/959305/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/959305/1"}}},"commit":{"parents":[{"commit":"a891d81974a4c3aa56a143252b26df9dcbd5fba5","subject":"Merge \"Fix neutron job for stable/2025.1\" into stable/2025.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/a891d81974a4c3aa56a143252b26df9dcbd5fba5"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-03 08:06:37.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-09-03 11:48:59.000000000","tz":0},"subject":"Remove next pagination link for floating IP list","message":"Remove next pagination link for floating IP list\n\nFirst of all the floating IP list API does not\nsupport pagination as it doesn\u0027t pass any kind\nof limit or marker down to the central API so\nwe simply get all the floating IPs returned.\n\nBut the floating IP API still accepts the limit\nand marker query parameters, if limit is not given\nwe default to the default_limit_v2 config option\nthat has default value of 20.\n\nIf a project has exactly 20 floating IPs we hit\nthe bug documented in [1] where the returned list\nis exactly the size of the limit used.\n\nIn the openstacksdk [2] this raises an exception\nwhen the last item is the marker in the next url.\n\nIn the designateclient OSC plugin [3] this will\nloop forever and generate HTTP requests until\nstopped.\n\nWe resolve this by making the next link configurable\nin the options for the object adapters but default\n`next` to True to keep backward compatibility and\nthen set this to false for the floating IP list.\n\nWe add that since we cannot set the `links` option\nto False as that would change the returned API\nresponse body by removing links from the body\nentirely.\n\n[1] https://github.com/openstack/designate/blob/595894be2202bfd98805a0ed87281e69c59b5c1d/designate/objects/adapters/api_v2/base.py#L120\n[2] https://github.com/openstack/openstacksdk/blob/79fbbbabff69e8e40e725ebbb8c605d2cc70ccae/openstack/resource.py#L2080\n[3] https://github.com/openstack/python-designateclient/blob/490b0d6e7623859704746fdfd6d67e0d3a737ac8/designateclient/v2/base.py#L36\n\nCloses-Bug: #2106112\nChange-Id: I60ed37719fd0411159b8ec4f68ca7b001eb8a7be\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n(cherry picked from commit e80666a1ddb6a43c2760505d0020d392cde87d21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/febee3f7907f77fca72cd4b40e64d27f870a26b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/febee3f7907f77fca72cd4b40e64d27f870a26b8"}]},"branch":"refs/heads/stable/2025.1"},"3355117473376d948e263ec02b66e9f8d1400386":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-09-03 13:10:51.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/05/959305/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/959305/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/959305/2"}}},"commit":{"parents":[{"commit":"3837c294cae734b9d2cab47871a6643dc6f5ab8a","subject":"[Stable-only] Deploy using module path","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3837c294cae734b9d2cab47871a6643dc6f5ab8a"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-03 08:06:37.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-09-03 13:10:51.000000000","tz":0},"subject":"Remove next pagination link for floating IP list","message":"Remove next pagination link for floating IP list\n\nFirst of all the floating IP list API does not\nsupport pagination as it doesn\u0027t pass any kind\nof limit or marker down to the central API so\nwe simply get all the floating IPs returned.\n\nBut the floating IP API still accepts the limit\nand marker query parameters, if limit is not given\nwe default to the default_limit_v2 config option\nthat has default value of 20.\n\nIf a project has exactly 20 floating IPs we hit\nthe bug documented in [1] where the returned list\nis exactly the size of the limit used.\n\nIn the openstacksdk [2] this raises an exception\nwhen the last item is the marker in the next url.\n\nIn the designateclient OSC plugin [3] this will\nloop forever and generate HTTP requests until\nstopped.\n\nWe resolve this by making the next link configurable\nin the options for the object adapters but default\n`next` to True to keep backward compatibility and\nthen set this to false for the floating IP list.\n\nWe add that since we cannot set the `links` option\nto False as that would change the returned API\nresponse body by removing links from the body\nentirely.\n\n[1] https://github.com/openstack/designate/blob/595894be2202bfd98805a0ed87281e69c59b5c1d/designate/objects/adapters/api_v2/base.py#L120\n[2] https://github.com/openstack/openstacksdk/blob/79fbbbabff69e8e40e725ebbb8c605d2cc70ccae/openstack/resource.py#L2080\n[3] https://github.com/openstack/python-designateclient/blob/490b0d6e7623859704746fdfd6d67e0d3a737ac8/designateclient/v2/base.py#L36\n\nCloses-Bug: #2106112\nChange-Id: I60ed37719fd0411159b8ec4f68ca7b001eb8a7be\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n(cherry picked from commit e80666a1ddb6a43c2760505d0020d392cde87d21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3355117473376d948e263ec02b66e9f8d1400386"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3355117473376d948e263ec02b66e9f8d1400386"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"},"bbb241f89620faa6d99ec98f78d77ca4dd32417a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-10-14 06:45:01.000000000","uploader":{"_account_id":16137,"name":"Tobias Urdin","email":"tobias.urdin@binero.com","username":"tobasco"},"ref":"refs/changes/05/959305/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/05/959305/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/05/959305/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/05/959305/3"}}},"commit":{"parents":[{"commit":"239af3482a009fe04ac4c8e6686c812780a3ad4f","subject":"zuul: change designate-grenade-common to be non voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/239af3482a009fe04ac4c8e6686c812780a3ad4f"}]}],"author":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-04-03 08:06:37.000000000","tz":120},"committer":{"name":"Tobias Urdin","email":"tobias.urdin@binero.com","date":"2025-10-14 06:45:01.000000000","tz":0},"subject":"Remove next pagination link for floating IP list","message":"Remove next pagination link for floating IP list\n\nFirst of all the floating IP list API does not\nsupport pagination as it doesn\u0027t pass any kind\nof limit or marker down to the central API so\nwe simply get all the floating IPs returned.\n\nBut the floating IP API still accepts the limit\nand marker query parameters, if limit is not given\nwe default to the default_limit_v2 config option\nthat has default value of 20.\n\nIf a project has exactly 20 floating IPs we hit\nthe bug documented in [1] where the returned list\nis exactly the size of the limit used.\n\nIn the openstacksdk [2] this raises an exception\nwhen the last item is the marker in the next url.\n\nIn the designateclient OSC plugin [3] this will\nloop forever and generate HTTP requests until\nstopped.\n\nWe resolve this by making the next link configurable\nin the options for the object adapters but default\n`next` to True to keep backward compatibility and\nthen set this to false for the floating IP list.\n\nWe add that since we cannot set the `links` option\nto False as that would change the returned API\nresponse body by removing links from the body\nentirely.\n\n[1] https://github.com/openstack/designate/blob/595894be2202bfd98805a0ed87281e69c59b5c1d/designate/objects/adapters/api_v2/base.py#L120\n[2] https://github.com/openstack/openstacksdk/blob/79fbbbabff69e8e40e725ebbb8c605d2cc70ccae/openstack/resource.py#L2080\n[3] https://github.com/openstack/python-designateclient/blob/490b0d6e7623859704746fdfd6d67e0d3a737ac8/designateclient/v2/base.py#L36\n\nCloses-Bug: #2106112\nChange-Id: I60ed37719fd0411159b8ec4f68ca7b001eb8a7be\nSigned-off-by: Tobias Urdin \u003ctobias.urdin@binero.com\u003e\n(cherry picked from commit e80666a1ddb6a43c2760505d0020d392cde87d21)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbb241f89620faa6d99ec98f78d77ca4dd32417a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/bbb241f89620faa6d99ec98f78d77ca4dd32417a"}]},"branch":"refs/heads/stable/2025.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
