)]}'
{"id":"openstack%2Fdesignate~963623","triplet_id":"openstack%2Fdesignate~master~Ib21b6f052a33dc5819750f852502c7a49480c758","project":"openstack/designate","branch":"master","topic":"fix_designate_backend_coverage","attention_set":{},"removed_from_attention_set":{"22623":{"account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"last_update":"2025-10-17 15:17:27.000000000","reason":"\u003cGERRIT_ACCOUNT_22623\u003e replied on the change","reason_account":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}}},"hashtags":[],"change_id":"Ib21b6f052a33dc5819750f852502c7a49480c758","subject":"Fix designate backend coverage","status":"MERGED","created":"2025-10-09 19:35:58.000000000","updated":"2025-10-17 16:33:03.000000000","submitted":"2025-10-17 16:32:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963623-fix_designate_backend_coverage","meta_rev_id":"3b843b0f7218c6fe452af6449d1930390465a778","_number":963623,"virtual_id_number":963623,"owner":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-17 16:32:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":2,"date":"2025-10-17 12:30:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-10-17 15:17:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-09 21:28:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-17 12:30:28.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"2a0d67bf032526167a1fe7a24cb058d59718005f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-10-09 19:35:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23ac40d8e8946fdc52adba60a8186678426adf35","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-10-09 19:38:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ad9e77404aef2b6484fe07b74e2e0a83683375f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-10-09 20:13:25.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"142762735ab0f6c7e698cd121a5faad34f55f99e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-09 21:28:21.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/57fc7638f00b4027845d1d1941bbc594\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/6f3ed642f942423eaa1542e973c24b41 : SUCCESS in 55m 37s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/6bd530ca8f644688a0026b476da8255e : SUCCESS in 42m 28s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/7ccb5129e9dd43928a9cc06325c031db : SUCCESS in 50m 29s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/81887ebea7d94a9688c5bd998f1f06fe : SUCCESS in 36m 06s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/44909abd73f44dc58a00ab2f380d5e22 : SUCCESS in 1h 06m 15s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/e042df8b9a624c14b099eb1bcaeab74a : SUCCESS in 53m 42s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/1b0337b7c9ea47f884c673721b4fb0a8 : SUCCESS in 1h 08m 46s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/c378be7a8b95456690b8c38ca572ca2c : SUCCESS in 1h 08m 19s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/891c0a4a55e44f3c8e26a7dddc91f7e7 : SUCCESS in 49m 12s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/a5430f1a43d146f0b5e4e57445d0601c : SUCCESS in 50m 30s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/034add558fdc41c2b22914e7791872c3 : SUCCESS in 48m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8661d52e39c84dc9ab82efcce24c5258 : FAILURE in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/66e520114a3e4335be6bcdda2500b8f3 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5019c80b65824e4bb8ac4b0dad4a8475 : FAILURE in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/39992e7de99949069171422907d25251 : FAILURE in 5m 30s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7abd8e7132ce451bb60219ddc18e5a5c : FAILURE in 8m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b058b530ebc430baf04fa40aa3f1571 : SUCCESS in 4m 33s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/bee6602e4c5e4840a25dfcb586091b5e : SUCCESS in 36m 15s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/dd6f867eb35c4c0eb2921a63faa032c3 : FAILURE in 7m 07s","accounts_in_message":[],"_revision_number":3},{"id":"14e350791828d3dd1f8fba7c199db25f58e30f9c","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-10-10 08:05:23.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"60dd7487d9d15aec570bde3dc1e66a96d1081675","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-10-11 08:35:20.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"733237653c82110c4232b17dc2cf3b81ad0b87d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-11 09:33:07.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fec7cd1a6cf48dd8a7682f2768dc4db\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/27e88c8e8fc94ee3adee86bd679762fa : SUCCESS in 44m 10s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/9212a9a0489c4b7cb7c9904cc9fb480e : SUCCESS in 49m 06s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/11de07c4e388442692100874688afb35 : SUCCESS in 45m 27s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/05b440b558584743acfa5987c2ef082c : SUCCESS in 29m 39s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/7bfb76d676474eadaf2a64d4393c58b1 : SUCCESS in 52m 04s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/9bfcafc2a6964fb3bc2c4f4ceefeef21 : SUCCESS in 48m 49s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/fd23886c3e8f4231b2c95ba4dfe1b8db : SUCCESS in 53m 10s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/aa0a387fd11845d5ad268d3fb9ca2d36 : SUCCESS in 56m 15s\n- designate-bind9-core-catalog-zones https://zuul.opendev.org/t/openstack/build/c9f0896bdd0947788eb3cb8ea3d91817 : SUCCESS in 47m 17s (non-voting)\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/b88104c273ef4fe3a2a270bdda511807 : SUCCESS in 50m 48s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/07439345dda642aba5702f1af872b737 : SUCCESS in 52m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0760fe55ac004b088adadec2470d825e : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4cbd3fb8f340cfa37f218c2e4b0222 : SUCCESS in 2m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c5fec801b5649dbac4baeddb68740e7 : SUCCESS in 4m 38s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7df6ef174ada4682b708bf678dcf50b5 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/33c91819435e4602b0c80bda7041d285 : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b72f25032454dbcb62af55db5a3cd18 : SUCCESS in 4m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/7aa8a6ec14df4406b298773d939149cf : SUCCESS in 35m 18s\n- designate-tox-dnspython-latest https://zuul.opendev.org/t/openstack/build/569de0ed4ae249cbab52358c91058a5d : SUCCESS in 3m 22s","accounts_in_message":[],"_revision_number":4},{"id":"910e943b54af7fe91f3cc2529492f90dc9e4cad6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2025-10-17 12:30:28.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"11ec795761869f36990dcde9d6e7aaf1eab92855","author":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"date":"2025-10-17 15:17:27.000000000","message":"Patch Set 4: Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"0bd9051a6c1957343e5869618241fb5614ca96e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 15:17:53.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f78546610e7a1840d47f108307171aa8ace5f433","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 16:32:05.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ffa73d31bddb4a8592f6b580d6186eb0\n\n- designate-bind9-core https://zuul.opendev.org/t/openstack/build/12fa9b8177d04f1c8beb873bab275770 : SUCCESS in 38m 40s\n- designate-bind9-with-keystone-default-roles https://zuul.opendev.org/t/openstack/build/16dcc0780067477b8402080f331b731f : SUCCESS in 44m 02s\n- designate-pdns4-core https://zuul.opendev.org/t/openstack/build/6c00a6a1318341ce944844b9e87936d4 : SUCCESS in 46m 22s\n- designate-bind9-multipool https://zuul.opendev.org/t/openstack/build/1c19b9d3e59443d19a3bf6701ffb2d8f : SUCCESS in 33m 35s\n- designate-grenade-pdns4 https://zuul.opendev.org/t/openstack/build/3cc51f904b8740a680daa598229f068b : SUCCESS in 1h 09m 28s\n- designate-grenade-bind9 https://zuul.opendev.org/t/openstack/build/e57a48184a4e4fe7a94e95bb924c4629 : SUCCESS in 1h 07m 13s\n- designate-grenade-pdns4-skip-level https://zuul.opendev.org/t/openstack/build/43f9d72966bf4f17ae64dfd543a2f7b7 : SUCCESS in 1h 06m 22s\n- designate-grenade-bind9-skip-level https://zuul.opendev.org/t/openstack/build/d9b177de932841f8948b30c8c13c9101 : SUCCESS in 50m 54s\n- designate-only-ipv6-pdns4 https://zuul.opendev.org/t/openstack/build/cd70bb993e0545509eeba158fd0cda9c : SUCCESS in 43m 58s\n- designate-only-ipv6-bind9 https://zuul.opendev.org/t/openstack/build/5f6e61478d6945ba9f922a869f404a8d : SUCCESS in 42m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2686e29c081740e9b937ac6a950139e3 : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/34e4c83073f6461fa0eed55da72b1def : SUCCESS in 4m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c0e9e9cc6be6439cad6e804c2df3c8d9 : SUCCESS in 8m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb77efeb5e7d4ef789d3d797ee2bc40f : SUCCESS in 8m 59s\n- neutron-tempest-plugin-designate-scenario https://zuul.opendev.org/t/openstack/build/a6bbe50643d24b4c9094bb42efa3a495 : SUCCESS in 32m 48s","accounts_in_message":[],"_revision_number":4},{"id":"a35494ff85a8e0cf9002488487d65e0e074d3d19","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 16:32:05.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"3b843b0f7218c6fe452af6449d1930390465a778","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-17 16:33:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4eb8ffc516804a65bf49467e696954b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f078dfd210fe4d4ea31406fc0cce8c08 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3bc81788322f10a8ada3e685a4417f4cc88f3942","revisions":{"c3cd87cc14453fee2c71079db7d9c55c6d857a66":{"kind":"REWORK","_number":1,"created":"2025-10-09 19:35:58.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/963623/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/963623/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/963623/1"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"me@eandersson.net","date":"2025-10-09 19:31:50.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"me@eandersson.net","date":"2025-10-09 19:35:42.000000000","tz":120},"subject":"Fix designate backend coverage","message":"Fix designate backend coverage\n\nThe driver has been broken since its introduction and this was\nrecently fixed. Fixing the unit test that was meant to cover\nthis.\n\nChange-Id: Ib21b6f052a33dc5819750f852502c7a49480c758\nSigned-off-by: Erik Olof Gunnar Andersson \u003cme@eandersson.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3cd87cc14453fee2c71079db7d9c55c6d857a66"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/c3cd87cc14453fee2c71079db7d9c55c6d857a66"}]},"branch":"refs/heads/master"},"35c3eabafe298e38c7194e0b140c93a2d0570721":{"kind":"REWORK","_number":2,"created":"2025-10-09 19:38:37.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/963623/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/963623/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/963623/2"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"me@eandersson.net","date":"2025-10-09 19:31:50.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"me@eandersson.net","date":"2025-10-09 19:37:30.000000000","tz":120},"subject":"Fix designate backend coverage","message":"Fix designate backend coverage\n\nThe driver has been broken since its introduction, but this was recently fixed.\nThis change updates the unit test that was originally meant to cover it.\n\nChange-Id: Ib21b6f052a33dc5819750f852502c7a49480c758\nSigned-off-by: Erik Olof Gunnar Andersson \u003cme@eandersson.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/35c3eabafe298e38c7194e0b140c93a2d0570721"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/35c3eabafe298e38c7194e0b140c93a2d0570721"}]},"branch":"refs/heads/master"},"0ee80ba59a6eb4ac2397932c3bfdf8bd49afe383":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2025-10-09 20:13:25.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/963623/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/963623/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/963623/3"}}},"commit":{"parents":[{"commit":"652a8180f03bc35f6769bbd2e43aecd5bd71602e","subject":"Fix zuul job names in projects.yaml","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/652a8180f03bc35f6769bbd2e43aecd5bd71602e"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"me@eandersson.net","date":"2025-10-09 19:31:50.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"openstack@eandersson.net","date":"2025-10-09 20:13:03.000000000","tz":120},"subject":"Fix designate backend coverage","message":"Fix designate backend coverage\n\nThe driver has been broken since its introduction, but this was recently fixed.\nThis change updates the unit test that was originally meant to cover it.\n\nChange-Id: Ib21b6f052a33dc5819750f852502c7a49480c758\nSigned-off-by: Erik Olof Gunnar Andersson \u003copenstack@eandersson.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ee80ba59a6eb4ac2397932c3bfdf8bd49afe383"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/0ee80ba59a6eb4ac2397932c3bfdf8bd49afe383"}]},"branch":"refs/heads/master"},"3bc81788322f10a8ada3e685a4417f4cc88f3942":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-10-11 08:35:20.000000000","uploader":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"},"ref":"refs/changes/23/963623/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/designate","ref":"refs/changes/23/963623/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/designate refs/changes/23/963623/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/designate refs/changes/23/963623/4"}}},"commit":{"parents":[{"commit":"e773de35e72ee46a3eced71e2678342271fc692f","subject":"Merge \"Initialize the client attribute to prevent AttributeError\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/e773de35e72ee46a3eced71e2678342271fc692f"}]}],"author":{"name":"Erik Olof Gunnar Andersson","email":"me@eandersson.net","date":"2025-10-09 19:31:50.000000000","tz":120},"committer":{"name":"Erik Olof Gunnar Andersson","email":"openstack@eandersson.net","date":"2025-10-11 08:35:20.000000000","tz":0},"subject":"Fix designate backend coverage","message":"Fix designate backend coverage\n\nThe driver has been broken since its introduction, but this was recently fixed.\nThis change updates the unit test that was originally meant to cover it.\n\nChange-Id: Ib21b6f052a33dc5819750f852502c7a49480c758\nSigned-off-by: Erik Olof Gunnar Andersson \u003copenstack@eandersson.net\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/designate/commit/3bc81788322f10a8ada3e685a4417f4cc88f3942"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":22623,"name":"Erik Olof Gunnar Andersson","email":"eandersson@blizzard.com","username":"eoandersson"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
